)]}'
{"id":"openstack%2Fnetworking-odl~708133","triplet_id":"openstack%2Fnetworking-odl~master~Iaf163f48b5f89d7585525e241670761177cb2d14","project":"openstack/networking-odl","branch":"master","topic":"drop_py27","hashtags":[],"change_id":"Iaf163f48b5f89d7585525e241670761177cb2d14","subject":"Change the remaining non python3 compatible code","status":"MERGED","created":"2020-02-17 14:46:58.000000000","updated":"2021-05-06 13:26:10.000000000","submitted":"2021-05-06 13:24:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"708133-drop_py27","meta_rev_id":"dbe3d5d66fa0006b967259adef9dc51ffe7b66ae","_number":708133,"virtual_id_number":708133,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 13:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-04-21 13:50:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-05-06 12:39:59.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-05-06 12:39:59.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 07:04:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-17 15:39:23.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-09-21 10:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-20 11:17:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-04-20 11:18:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-05-06 11:15:03.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"6ba16971eab960d4b9911d2c3b64bfa67e6a427c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-17 14:46:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13361a88cf70a64d98954ae879e04669a11f4f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 16:31:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f216ff663f41898fac33c865c49779 : SUCCESS in 10m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1666a2dc57224233a785e656667fc614 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cd615b7b716405eae8d8284cc6aad10 : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43504376975643e38cf6c0699707cab2 : SUCCESS in 7m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/843fb87e1df2416490c1e72a5a613ec5 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32158adb09324bc1846e1a40d540c6ec : SUCCESS in 10m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45bb094004c64c9585e05b06ad818706 : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee4819bec2ae42f586089b37878f6a38 : SUCCESS in 10m 09s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/328ae06d2e714b65aae62e877955ea38 : FAILURE in 22m 32s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/6af6fc07aa414c2d8b4a55c550f042a4 : RETRY_LIMIT in 9m 02s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/86263618a3d248d7a4a2e62e90304e73 : RETRY_LIMIT in 7m 48s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/257e0306423042dba668e03c019425f4 : FAILURE in 16m 52s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/ccf25012fab74a1bacb7d52f7e596244 : FAILURE in 18m 16s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/aa75c4593380491b8bb777f391d77d7c : FAILURE in 22m 18s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/d39b50536cee44c6be151122d583b88e : SUCCESS in 1h 16m 57s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/80425ef5070f4868becf7bf596e20e71 : RETRY_LIMIT in 16m 38s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/48b5d58fc9134ae79c3f7fcb3db16a1f : SUCCESS in 21m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e2953a979ec437f762068286d111d76ec19f27b0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-19 08:16:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0d4d23abce7b40ab89210460693d3e1964d9584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 09:49:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e45c5e38a014ecb80e1a6e51a7cc09c : FAILURE in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ddb5547b3e0468daa2b56d8f587c33f : FAILURE in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8c5e2adfb4f488aab82128979ff875a : FAILURE in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c31603093f444b18eb118dc5a869232 : SUCCESS in 8m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45ee220d96504e0b8c7b52427ed2f96b : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/534029a96dcb4c60bc46f1f3e81200c3 : SUCCESS in 10m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70aa2da720a409da2467bf9418e3597 : FAILURE in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26ec951577be42129ab36b6dfa3ad2ab : SUCCESS in 8m 14s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/0825206d588f4a6c87624bb2efa48ea8 : FAILURE in 21m 28s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/9c0d763d1d454fd5b01947390a28b154 : RETRY_LIMIT in 16m 05s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/c443dd6d822c4aa1b2e142220ea522fd : RETRY_LIMIT in 14m 06s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/8ce6dc1577594c6a9d903b5be8add11d : FAILURE in 14m 52s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/a9999b7da3d340b981f6f9c6cb2813b7 : FAILURE in 20m 55s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/090f80b2c29846bbba467c2536379211 : FAILURE in 21m 42s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/1e45dd740ab3407788b1b15e173f95f3 : SUCCESS in 1h 25m 12s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/704e1b9ce69d433b9c371be32dfabe7a : RETRY_LIMIT in 13m 21s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/eb914a927b8c483dbf23986e7f551d68 : FAILURE in 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"214e22d463ab709d5ed8b4ecf8ec076109e8af35","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-19 13:42:10.000000000","message":"Patch Set 1: Workflow-1\n\nHmmm, I come back later to this, as it seems that I stepped into some deep dependency pitch","accounts_in_message":[],"_revision_number":1},{"id":"87a324537c725343e349876eac7601951158ef5a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-02 13:44:24.000000000","message":"Patch Set 1: -Workflow\n\nI think when the horizon \u0026 networking-bgpvpn issue will be solved this can move (see the related patch: https://review.opendev.org/708898)","accounts_in_message":[],"_revision_number":1},{"id":"08aa37473094356c2e941e4da4419c9a87628227","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-20 09:17:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0004e76f80c3b28c7f07331167054264ef7ddb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 12:34:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791c94d6a22f4bc2bb22dd55fe5e72e9 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b37b28d012941c3b83d63d754ad52e0 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf7dbbea02e24a2e91f0872571d5f066 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc93105dec44f97b347656b350164d2 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c508cb9028c425a8702bd11284ea601 : SUCCESS in 11m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1da884a9a174856a8846e541afb41a8 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e9bc8e282c0494c99a950e491ec23f4 : SUCCESS in 9m 42s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/bc54dfd74119422dbe67644c04be9341 : FAILURE in 22m 45s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/e58ef73739b844f78ff0041b9ee00515 : TIMED_OUT in 3h 07m 48s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/23cd68273e5843aeb4efb5420c7fb7dd : TIMED_OUT in 3h 10m 55s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/69538796c31c4b35966fe7e77b7cb357 : POST_FAILURE in 47m 30s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/1a3bcb14377a45ae9f0ba6c312943cbf : POST_FAILURE in 46m 32s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/bdf855005df94fa5a4f703fe8683b241 : SUCCESS in 1h 24m 01s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/6ddd187b3e634c8bab0b298f293e249d : SUCCESS in 1h 11m 50s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/6698dfbb97064466a7aff57cf7005cef : SUCCESS in 22m 20s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e32f10b214a44ec0905ff5d135088545 : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":2},{"id":"392c4037e9829eed5b1ee50506a1d005c4af078e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-27 05:32:11.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b07a32f985898c2cd826ed3558062b5acf4d4b67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 08:42:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e52a0f6aa3490f8c98dbdf272af47b : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c013cb2935248af94fc7176f675e522 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d197704c252841a08dda39bb9b9640ac : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72e2a87234124cfb82d139b0f5a3cbba : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e5ebd8b0744bd385e8dfb18377697d : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4db85afb1d994b5b9ba6b49837f2c2db : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3adbba8cfbb64a70a3863868010d29b1 : SUCCESS in 8m 03s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/16f20ee6e499411f95de61d47c10b60d : FAILURE in 18m 37s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/d4e12f0bc3414537a3608af69d8c002a : TIMED_OUT in 3h 08m 25s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/c853ff9273884031a6e67936bcdead47 : TIMED_OUT in 3h 09m 21s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/e118a9331707478c985012679a7d5564 : POST_FAILURE in 39m 48s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/757f237c754248cb9bad3ed00ef64a30 : POST_FAILURE in 42m 12s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/f69cac105baa4aaa8241e5cb65fadb05 : SUCCESS in 1h 28m 06s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/f35af9c688ea435ca7f984d4843732a9 : SUCCESS in 1h 24m 21s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/3891bbb45f694f0f902abf213142305d : SUCCESS in 19m 38s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/f476e364e5b34b6792b7831f366affd4 : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":3},{"id":"eb37330ee465324b9ca2cddd74c7e6df6635b349","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-14 07:04:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e93cf2c76ac8eafc80269f505cfe60545b15edf4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-24 11:54:33.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d156827082452493f9675d4065530897ccedbe54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:36:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9dcce0b8b684a8f88331ce00acb132c : SUCCESS in 11m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e035a7a63d4d4ccda8ed3c145459f706 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5949771c4f0f45698c71e9f7b2b6dd59 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98bb79e4775d4b53be16ee8f84b797ba : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e999dff19fd4c7480a352108019beb0 : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/479e458472cf4c6d9db6c1dd8060643d : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44f5630e83d74cfb944d2b2b78738452 : SUCCESS in 6m 10s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/829f27d8a45b474498b12cac8b33f731 : FAILURE in 14m 47s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/e2ae76699afc487c8fa092eac78dd287 : TIMED_OUT in 3h 07m 47s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/059e5edeff5544868a388acfc2d6cc36 : TIMED_OUT in 3h 08m 06s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/e1a0f429eb4a4ab8a4bbebb359ed295e : POST_FAILURE in 44m 02s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/2c7444d4affc4ec5a3288f70879ff17a : POST_FAILURE in 33m 47s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/098545dd58b24883823cd36dd1e89fd8 : FAILURE in 40m 54s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/633b4f3a8d65486782663eab0df2127f : FAILURE in 38m 05s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/4fdd9788ebb04f49bb2cf1ea8a4205b0 : SUCCESS in 22m 34s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/62ee06a3967843f2874f0d4e15619b5c : SUCCESS in 22m 47s","accounts_in_message":[],"_revision_number":4},{"id":"84d8da0b406fa061d13d8448ff12ed6d54c19a2f","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-17 05:58:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e913d8b4fb2a81c7838a302501237cc91b5c2061","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-17 06:00:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5b36ec4ed0b04affdc28aa8e198440f2a8c56326","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-17 06:00:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bd44cfa7a9d4406be0e0ece28b68a08d25838b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 07:51:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff1a4712edc43758da1db3b162b1a1e : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/863e32b135054bff9d3e2f8aa3747604 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96232e34930c41b2b3ca8551b8911f06 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54d8804a7f024106a022a93da95b48f3 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef54174d9c7e4624821428c05aed0c58 : SUCCESS in 22m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3510d17122f04ae6ba2894642d31d450 : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a275321bffc46429e8d99134d101533 : SUCCESS in 8m 37s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/2a90cf846ce94ec6803a4848149c58ec : FAILURE in 15m 41s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/b421a779e2cc474e84fe10ba895314fe : FAILURE in 46m 45s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/97ea0bf0f3da46c48b8a8c59961b8fcb : FAILURE in 50m 13s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/ff77865097574407a070522f3aaf3c6f : POST_FAILURE in 36m 10s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/24e4b2a5bed2496695e6acdd6b345cf2 : POST_FAILURE in 35m 38s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/9c27c635792e497982c8b04a60b721b4 : SUCCESS in 1h 10m 10s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/05ff8654dfc1499a9ce7a3d1ffd79636 : SUCCESS in 1h 05m 17s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/587bcedffae84400bb60ad2704afcd38 : SUCCESS in 21m 10s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/ad5cfeeefb244c4aa1c6ab00b0e8b6a4 : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":6},{"id":"ec91b1727429567786d5136ce529ad67f3b031f0","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-17 15:39:23.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1a6af717f39ca8625d04d161766c02e3cf205563","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-18 08:37:42.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"05af3c736618051aab7711508a8eb54894b39d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 10:15:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c61dd636924448497ff220b4be09b02 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a90bf4d9a67b45da9c89ae095a2893cf : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7878bd6465ad49b88e99036e6c396efe : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/886eba4999bd48ae89b6b27395ef6a96 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a5658a071143eca08c84bc3966ceb7 : SUCCESS in 17m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91ab74b505984fcc8f1863fc3217a000 : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52b1c44faa294626a68cc5cffc27be4b : SUCCESS in 8m 38s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/01e30ebf755b4a80bdd781c446875d0f : FAILURE in 22m 09s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/3afc8af30f3a465d8fb385455318b4c5 : FAILURE in 51m 14s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/b494d0e7a8a04536a8163fee593353e1 : FAILURE in 56m 37s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/55c4495128a2432ea1adefcdee61a4fc : POST_FAILURE in 34m 36s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/0cc2d032047a4e5a9cf5f84f8f732e7a : POST_FAILURE in 40m 26s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/6f951762c72f4cc99b178e9b4147523f : SUCCESS in 1h 33m 51s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/b3577da7f4cc4fa0beab553b9a28df1a : SUCCESS in 1h 17m 39s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/65feefcbfe404dff8ce792c51a1a177f : SUCCESS in 20m 58s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/3d3ee34040374302991d89af56cdb3b8 : FAILURE in 25m 25s","accounts_in_message":[],"_revision_number":6},{"id":"3cc05bddedd70a4359c70df6794b8357d24cd0ad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-21 08:32:45.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"75314ec88c24f84f45786b11a12cba1c7f8405d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:13:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0f6658c950441e9958ab59fe6e32dd : SUCCESS in 10m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdd91959faf8406c803acd373705ca62 : SUCCESS in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fee1f114bd7449f286901e36636c412b : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10df66128cd240ae988355b5cf93e233 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdfa2bd69f8b4fcaae20772cc7a9de23 : SUCCESS in 20m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f696bcdbb124b818a3e0d56d453a250 : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ff888ccbfc54438a2ec188ac34dcded : SUCCESS in 12m 06s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/030c6656561048c4adea8c9805f5bae4 : FAILURE in 27m 28s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/475f7807a59844a88459360541e738c9 : FAILURE in 54m 21s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/e92aa93512ae4d168680629e4d8735b8 : FAILURE in 46m 26s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/675150407d9441979b4a116ef5542c23 : POST_FAILURE in 35m 38s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/312243a639c84cfd8d8bc6deb4492192 : POST_FAILURE in 50m 21s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/0fdd292a05104be6b5d1ab61c610f163 : SUCCESS in 1h 25m 15s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/fe2bd3da8b5940b4b1e2a69fbd962ff7 : FAILURE in 1h 28m 21s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/ffa4fd79a1ae4e5687a6e10234dbbfb5 : SUCCESS in 20m 11s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5710789fd34d4532a02d5746897971c8 : SUCCESS in 19m 07s","accounts_in_message":[],"_revision_number":6},{"id":"1155818c834a2776378d890668cb66835ea487d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-20 11:00:13.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"15ff1ff11f7384fb6a3af0cb890fdd1a9b71ca58","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-20 11:17:46.000000000","message":"Patch Set 7:\n\nThis project is still using six. This library should be removed too.","accounts_in_message":[],"_revision_number":7},{"id":"b828b02f347c050ea3f2b2b3a73d436adfbcbe14","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-20 11:18:18.000000000","message":"Patch Set 7: Code-Review-1","accounts_in_message":[],"_revision_number":7},{"id":"7e9a3f48441cf05930356629968486af3e9b908c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 12:01:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6717d9342734745809b2847e2bdec83 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d771d381d0458088b0a6a898e273c2 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/124f4c1cda1845babfdf325c5fa65225 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c95df7b9a3664579b81f7d9ff5d52632 : SUCCESS in 6m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f33f67871eb49f1a20096fa7a428691 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a34902ef844472fac414709c78a3ca9 : SUCCESS in 7m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41b3b039146747479aee771698d74155 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1908bdd162724574ba434dd91eca57d9 : FAILURE in 4m 07s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/e7e05ba6adb14f39853d2b11fcdd1562 : FAILURE in 45m 17s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/4a494b4743fd4cc7a7d0a765ae30491e : FAILURE in 45m 31s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/a999ec3e754e4394b2800f95cc0f11d8 : FAILURE in 50m 12s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/61cccf888b274bbc9c18186868f50c0a : POST_FAILURE in 33m 39s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/c6d0796ba74441aeb029ec60c7450d03 : POST_FAILURE in 35m 58s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/e32e120b412549b2b377d05cffe2b444 : SUCCESS in 54m 08s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/e6b8564362014aee8f9ca6b233b6bd58 : SUCCESS in 59m 18s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/33cd2bf7e42c4efdad53e56f36b2dddc : SUCCESS in 18m 31s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/93ed2cdaf57847da8ab33ba5436ec31f : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":7},{"id":"4bdfa1e9a3d2a7a776a30fb74569a8a307d062bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-21 13:25:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"11e4d6e432c79567a2f494cd4fd74ec0d5920507","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-21 13:26:15.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 7:\n\u003e \n\u003e This project is still using six. This library should be removed too.\n\nYou are right, I have no idea why I skipped six.","accounts_in_message":[],"_revision_number":8},{"id":"4eb01030f378dd7825044e01b4c2af240ba82f96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-21 13:28:06.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"b25e086a5e92c0eafaddf182d3b17c02c9b5c81b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-21 13:39:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"99a7f9befd5654e881ba5f7137393202b9b2bf7c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-21 13:50:06.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nlooks good, waiting for zuul","accounts_in_message":[],"_revision_number":9},{"id":"e0a94d00dab9e159ef5c7c2b10526491f0b07c14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 14:41:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c9af5a44df740f891c3a2f19ae0545b : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d5325d3f8c348388894156b31fc5cd3 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cd01bae388b4863ab1f6eda26a48799 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/542a9a57d43f45e195fb7cafd7562b0c : SUCCESS in 7m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cea8dc10d41d43d789d7178984fe95a6 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab908bacde1f45df8c55efe3e169b87a : SUCCESS in 9m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7fe384cb818b4537a5536de6eebdbef9 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c48b96c970ef472ca55372bb728ced92 : SUCCESS in 9m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f7ebb322ec148639667e95e81e9bf62 : FAILURE in 4m 30s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/998b1f6d592244c6aaa9a953e29e1aa0 : FAILURE in 1h 04m 59s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/33b3600e30e047de8a9e27b249d9e9db : FAILURE in 49m 41s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/7f2bfdaed842445da1cfd24b853d7785 : FAILURE in 49m 08s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/ff465385ff444ad9a40a1df8abdfbff5 : POST_FAILURE in 39m 43s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/b5401c89863743c68a66d3507ad23ebe : POST_FAILURE in 37m 12s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/1ccefa79d74a42fa8bed5bbac9ffb42a : SUCCESS in 1h 00m 12s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/cf68bb45eb164bcea2107a91306502cb : SUCCESS in 1h 01m 26s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/8b1b0fb773304672a1720650079d789f : SUCCESS in 19m 06s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/96013bb5a924488aaeb107deb380d70b : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":9},{"id":"cedb2f11d6b912223be24d40b8c60af9811c62fe","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-06 12:39:59.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"d5f5349df63eebaeb001b72bcc2de9aed9b9ec3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 12:40:20.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"226fda11b08ed335661d93c36408135cfca60932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:24:00.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7492eded74b4521a07ba940fdbd3cae : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29aa9241adeb4ed19a5885ab12aec251 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/349ec55675ce4af4830c90d53763b1a1 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b852f5971ac14132bee2d534ccb8a3db : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e8316df0cd4eb8830d07087eba3469 : SUCCESS in 9m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e0ea7f826b504de98d9d5acc955644f7 : SUCCESS in 4m 42s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/337c81ccf0304526b21cf2c0d349ebcd : SUCCESS in 19m 38s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/52dc7a4f10ed40ff885a3d897eba25bd : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":9},{"id":"7ad9d142924fb3d4deb1557f4cdcd768309193eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:24:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"dbe3d5d66fa0006b967259adef9dc51ffe7b66ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:26:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d687caa8e694502be74bac677da4ffa : SUCCESS in 1m 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f50b18561e427e884b9642a8638933 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"732a90818df5ff33e2a237ae166ea2d530040fee","revisions":{"85fd9b786d3372b47a0123b5f38438cec593f267":{"kind":"REWORK","_number":1,"created":"2020-02-17 14:46:58.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/33/708133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/1"}}},"commit":{"parents":[{"commit":"d3323d05d7dfb4dff5498030f300314c1219e689","subject":"Remove references for unittest2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d3323d05d7dfb4dff5498030f300314c1219e689"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:43:56.000000000","tz":60},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nThe change is based on 2to3 -p networking_l2gw\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/85fd9b786d3372b47a0123b5f38438cec593f267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/85fd9b786d3372b47a0123b5f38438cec593f267"}]},"branch":"refs/heads/master"},"36a0f50746f82c356e31e6753992181eb510199a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-20 09:17:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/33/708133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/2"}}},"commit":{"parents":[{"commit":"6628ed21767899ea7eec4927d5843407bf518cb4","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6628ed21767899ea7eec4927d5843407bf518cb4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-20 09:17:39.000000000","tz":0},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nThe change is based on 2to3 -p networking_l2gw\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/36a0f50746f82c356e31e6753992181eb510199a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/36a0f50746f82c356e31e6753992181eb510199a"}]},"branch":"refs/heads/master"},"e7777798cc57f006ca40d13343adc388fb2bc8b8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-27 05:32:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/33/708133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/3"}}},"commit":{"parents":[{"commit":"6628ed21767899ea7eec4927d5843407bf518cb4","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6628ed21767899ea7eec4927d5843407bf518cb4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-27 05:32:09.000000000","tz":0},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nThe change is based on 2to3 -p networking_odl\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e7777798cc57f006ca40d13343adc388fb2bc8b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e7777798cc57f006ca40d13343adc388fb2bc8b8"}]},"branch":"refs/heads/master"},"fa4c0a7c8e66f78a4cd9781c4797aad4759cb085":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-24 11:54:33.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/33/708133/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/4"}}},"commit":{"parents":[{"commit":"b3fc2f302f44e986c38c137ea1128eed81e141d3","subject":"Merge \"variable type of port-id in ceilometer driver is changed from integer to string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b3fc2f302f44e986c38c137ea1128eed81e141d3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-08-24 11:54:33.000000000","tz":0},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nThe change is based on 2to3 -p networking_odl\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/fa4c0a7c8e66f78a4cd9781c4797aad4759cb085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/fa4c0a7c8e66f78a4cd9781c4797aad4759cb085"}]},"branch":"refs/heads/master"},"eae76520d898367d3082f2d07a6cb29f31b0136e":{"kind":"REWORK","_number":5,"created":"2020-09-17 06:00:10.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/33/708133/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/5"}}},"commit":{"parents":[{"commit":"30ee733387efb4323175c1df372e9a344c520fcc","subject":"Merge \"Use importlib to take place of imp module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/30ee733387efb4323175c1df372e9a344c520fcc"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-17 05:59:45.000000000","tz":480},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nThe change is based on 2to3 -p networking_odl\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\nCo-Authored-By: zhoulinhui \u003cdf.some@foxmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/eae76520d898367d3082f2d07a6cb29f31b0136e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/eae76520d898367d3082f2d07a6cb29f31b0136e"}]},"branch":"refs/heads/master"},"26ffa24a6f62a93d34ab313e50619dbce6069a94":{"kind":"REWORK","_number":6,"created":"2020-09-17 06:00:58.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/33/708133/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/6"}}},"commit":{"parents":[{"commit":"30ee733387efb4323175c1df372e9a344c520fcc","subject":"Merge \"Use importlib to take place of imp module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/30ee733387efb4323175c1df372e9a344c520fcc"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-17 06:00:47.000000000","tz":480},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nThe change is based on 2to3 -p networking_odl\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\nCo-Authored-By: zhoulinhui \u003cdf.some@foxmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/26ffa24a6f62a93d34ab313e50619dbce6069a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/26ffa24a6f62a93d34ab313e50619dbce6069a94"}]},"branch":"refs/heads/master"},"87ae7c6daa74145dfa6a2841b7d92c20508c3555":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-04-20 11:00:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/33/708133/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/7"}}},"commit":{"parents":[{"commit":"50356ed5c8753b4424815e3082d7fd147c0c0212","subject":"Merge \"Update master for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/50356ed5c8753b4424815e3082d7fd147c0c0212"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-04-20 11:00:13.000000000","tz":0},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nThe change is based on 2to3 -p networking_odl\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\nCo-Authored-By: zhoulinhui \u003cdf.some@foxmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/87ae7c6daa74145dfa6a2841b7d92c20508c3555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/87ae7c6daa74145dfa6a2841b7d92c20508c3555"}]},"branch":"refs/heads/master","description":"Rebase"},"5d85d822f0322b622bc59db49b3054b6a02b5efa":{"kind":"REWORK","_number":8,"created":"2021-04-21 13:25:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/33/708133/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/8"}}},"commit":{"parents":[{"commit":"50356ed5c8753b4424815e3082d7fd147c0c0212","subject":"Merge \"Update master for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/50356ed5c8753b4424815e3082d7fd147c0c0212"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-04-21 13:24:29.000000000","tz":120},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nRemove all non py3 compatible code and six.\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\nCo-Authored-By: zhoulinhui \u003cdf.some@foxmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5d85d822f0322b622bc59db49b3054b6a02b5efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5d85d822f0322b622bc59db49b3054b6a02b5efa"}]},"branch":"refs/heads/master"},"732a90818df5ff33e2a237ae166ea2d530040fee":{"kind":"REWORK","_number":9,"created":"2021-04-21 13:28:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/33/708133/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/33/708133/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/33/708133/9"}}},"commit":{"parents":[{"commit":"50356ed5c8753b4424815e3082d7fd147c0c0212","subject":"Merge \"Update master for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/50356ed5c8753b4424815e3082d7fd147c0c0212"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-17 14:23:39.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-04-21 13:28:00.000000000","tz":0},"subject":"Change the remaining non python3 compatible code","message":"Change the remaining non python3 compatible code\n\nRemove all non py3 compatible code and six.\n\nChange-Id: Iaf163f48b5f89d7585525e241670761177cb2d14\nCo-Authored-By: zhoulinhui \u003cdf.some@foxmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/732a90818df5ff33e2a237ae166ea2d530040fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/732a90818df5ff33e2a237ae166ea2d530040fee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
