)]}'
{"id":"openstack%2Fnetworking-odl~712900","triplet_id":"openstack%2Fnetworking-odl~master~I5f9e0c417336c8ff9e417a668268784485d8d2fa","project":"openstack/networking-odl","branch":"master","topic":"hacking3","hashtags":[],"change_id":"I5f9e0c417336c8ff9e417a668268784485d8d2fa","subject":"Switch to hacking 3.0","status":"MERGED","created":"2020-03-13 10:24:37.000000000","updated":"2020-09-02 20:11:48.000000000","submitted":"2020-09-02 20:09:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712900-1599077349098-b29784c5","meta_rev_id":"027dd9964a339275dcef841b0157c19b1b081332","_number":712900,"virtual_id_number":712900,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-08-24 16:11:47.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 20:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-08-14 06:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2020-09-02 19:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2020-09-02 19:34:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-08-28 07:37:07.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"date":"2020-08-28 07:38:08.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 12:47:47.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2020-05-06 09:55:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-08 03:08:31.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-08-14 06:59:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-08-28 07:37:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-08-28 07:38:08.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-09-02 19:34:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-02 20:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58bef604ee2f81d74cd88b19013902a53d2ff18c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-13 10:24:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"108135b7fd8cdd8c91ecb336fa2ea32bd84d786e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 11:51:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62488d77498b400491995e246cfadaca : FAILURE in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/166b56b474504ccd99901d8a87a33611 : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f8a7ce083554692adeb6a74bc4f4e69 : FAILURE in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22eefd4ade8e47e09064dcfff8fd3889 : FAILURE in 4m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b30f08eb3d4d4ee882b0c609ad9475cd : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20fb206ba4e14e4a890aa83fa51993a2 : SUCCESS in 7m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5697d4d353d5432eb760c1200e9331bf : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3b7c43866814555b487ab6cd2b4f28a : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6f8c43381e94c9493ab8d81a17ca651 : SUCCESS in 8m 21s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/b4d3d8cdc87f4a23949386e0ea50bea9 : FAILURE in 17m 59s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/af1d84181676435ea5c506cd94c76314 : RETRY_LIMIT in 9m 35s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/39acf551e4fd48b0a4ac08212357aa57 : RETRY_LIMIT in 7m 06s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/fd13b7e26fe5468b937aea3ee50897f6 : FAILURE in 11m 45s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/e6d664410b124f58b45f7ccbefd3fc38 : FAILURE in 21m 28s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/9762432f31484048acb7164b10c47803 : SUCCESS in 1h 24m 05s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/6536954ae2d64e49a697ad50c1c48565 : SUCCESS in 1h 14m 31s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/eeb6b49ecd284d05bec14e0f59514fe3 : SUCCESS in 20m 45s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e2c579767c6e4ae58fb9fb7d1c66362c : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ef5803b7626032cd4bf542a15f0735327f755f33","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-13 12:47:35.000000000","message":"Patch Set 1:\n\nAs I see some setuptools version brakes py37 job....","accounts_in_message":[],"_revision_number":1},{"id":"9236df742475c5425b60780f42bb65713583de8d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-16 14:17:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c1231e15e169f1c4ec9e9ce9ff3910c6e8a092fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 16:11:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114b764df64a4273af31fc11c9349fb6 : FAILURE in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14005784eefc4eb9ab70303c0ff49a02 : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5faa1004502c4cb0afb1676107ffa94d : FAILURE in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94bc9ad0ceb2481984bece90cd1c6554 : FAILURE in 6m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df0f218331c14cceab3b27a870ef7e51 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a66e93c163a40e9a1e28750fffc225a : SUCCESS in 6m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/01152063ab0a48fb8b49785b98a6b459 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bafb0c29be04517ac66315d3f9917f9 : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/008ab6d922ea457d9d6bebda4e0007dc : SUCCESS in 11m 33s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/150405dcdda14ccc86ee14515f06d656 : FAILURE in 19m 21s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/17baf545ef7d407b9d02eea1bbc3121e : RETRY_LIMIT in 8m 07s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/7352ad25d78c4690a4e55b7842c97905 : RETRY_LIMIT in 6m 56s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/79003e7d653b41c2ba1e21c287d4b4fb : FAILURE in 15m 50s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/0987260df2d442a18684f7f11e59f277 : FAILURE in 18m 47s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/90350fd2910341adbadc9204e744b1f1 : SUCCESS in 1h 45m 25s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/e0c87d5a35624e04b01635424148f3de : SUCCESS in 1h 24m 30s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/f4940fff9bbd474b9001a44b5526e92c : SUCCESS in 21m 45s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/977bcae762ae4348a9ba520e8d0da27f : SUCCESS in 21m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c79e22268499d151e0540ac3c7773725cf432609","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-29 14:16:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3125e9e2724611ba37fd454da2c6b64c6c578334","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-30 10:06:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"781085c3ce5c1133c125d5b04f1f77642630045f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-30 10:07:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c87876f1d7920bd509f38104de2841207d7e33d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 10:10:58.000000000","message":"Patch Set 3:\n\nNote: 3.0 is out, I suggest to update. This should not cause any problems.","accounts_in_message":[],"_revision_number":3},{"id":"c89ca68d3b1443c944a156ab815f1973bbe49c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 11:54:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50ab3459ea2b47858255acb6e080f941 : FAILURE in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c521526422d4b9f842116f819a144c6 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db2df4b34ca045258e6f538d2afb1bb3 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94f476c80fa24b6cb2dc6d7d78afb7eb : SUCCESS in 7m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b53e455664b14496ba53007743cc96e1 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf056ebf0b7144a9aab674decb691cf6 : SUCCESS in 7m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54d34821e98a48a3a2f1626011618c00 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbe89781f43a42c88486213be9cfac18 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c518f22f16e434286e170c1dda31558 : SUCCESS in 8m 46s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/bc3dea531d884d0abd24d7e3129f3522 : FAILURE in 22m 49s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/1fa11f42469343d58e29fbe784414397 : RETRY_LIMIT in 10m 26s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/e73c0744e8ea447eb3771f0377ce2446 : RETRY_LIMIT in 12m 24s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/e5337750059943308eb06b58a1b7bb8d : FAILURE in 16m 02s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/890ef569c2df48818e099f35e35b4aec : FAILURE in 14m 37s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/2bb87de597b349b7ad7c188887f3aedc : SUCCESS in 1h 32m 14s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/b89b948529f24326ae196ab54d902fd5 : SUCCESS in 1h 25m 09s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/fe05dfc6a6034234898bb2cc160c3f33 : FAILURE in 19m 20s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/288e0f94da1c4d47bd38c1438d4ccb12 : FAILURE in 21m 55s","accounts_in_message":[],"_revision_number":3},{"id":"7eba0fa184707c04c64e3dafce5416765e4654c0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-31 09:56:07.000000000","message":"Patch Set 3:\n\n\u003e Note: 3.0 is out, I suggest to update. This should not cause any\n \u003e problems.\n\nGood to know, I ask around neutron to have consensus about it.","accounts_in_message":[],"_revision_number":3},{"id":"f9a7e948e3ce81319a60af91319e3bfcdf7d470b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 09:54:51.000000000","message":"Patch Set 3:\n\nrecheck\n\nDependency has merged, you can continue here. Let\u0027s use neutron 3.0.","accounts_in_message":[],"_revision_number":3},{"id":"fc393ea51fcb123d37799d753e723dd5b9763cc0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 09:55:47.000000000","message":"Patch Set 3:\n\n\u003e Dependency has merged, you can continue here. Let\u0027s use neutron\n \u003e 3.0.\n\nhacking 3.0","accounts_in_message":[],"_revision_number":3},{"id":"2323cdec50f4a99c112fa3119cec4d2bf2760574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:05:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/589a9323fad24c97b35618e956ea8fd9 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39c50d5a5083479097b9d86a9385416b : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40c7fec5a2de45f9bd703e24efc6137e : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d06e0b7f8852433a9a9a1788a388c579 : SUCCESS in 5m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f09c4cc2d1d4013aae65caa16373778 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a34ae6973d4d68b16f8df4262a59b9 : SUCCESS in 8m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5255342cec1347ccb310fd11b63c0e07 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/369e0b4bba2c4072b1a3241c459033af : SUCCESS in 11m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac34a9c576f6481ba88aff706e18685e : SUCCESS in 12m 47s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/75498ff687fc44168d64e0f4eae33eee : FAILURE in 17m 16s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/d0da87f0fa0642c8a96ac58593e4f269 : TIMED_OUT in 3h 07m 51s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/2f7215f8373f47e69f0f69bb3ff266a1 : TIMED_OUT in 3h 08m 51s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/ee99acffcfc94cf6bffbdb3d09ca1de1 : POST_FAILURE in 37m 16s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/46f75c6b92fe45ef9ffacf0e33cb9452 : POST_FAILURE in 40m 31s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/dc996ae1fcf34992ad6c24fe8ddd70bb : SUCCESS in 1h 11m 50s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/414662e0d94d4200ac4e75da2b6b667a : SUCCESS in 1h 09m 34s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/54bf70941a0b45a693961fc557ab15c4 : FAILURE in 17m 52s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/62ce5647b46e475bbfa8763279a09a73 : FAILURE in 18m 45s","accounts_in_message":[],"_revision_number":3},{"id":"7e33a889f7155a077e540235b558633c3261f730","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-07 09:32:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1272d6ac28212063f537c7c0e6c641eab43e4241","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-07 09:52:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0f770ede0423af912844da06a36c9464ab10fcb7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-07 09:52:22.000000000","message":"Topic changed from bug/1876033 to hacking2","accounts_in_message":[],"_revision_number":5},{"id":"38f85f6819a74bdfd161be4239be0bab64b865ea","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-07 10:00:35.000000000","message":"Patch Set 5:\n\n\u003e \u003e Dependency has merged, you can continue here. Let\u0027s use neutron\n \u003e \u003e 3.0.\n \u003e \n \u003e hacking 3.0\n\nI changed, thanks","accounts_in_message":[],"_revision_number":5},{"id":"322bff119e07eed437d1e37ed077e812ff1beda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:05:18.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c4b93d35e024bf4a669ce1cd1fb86b8 : FAILURE in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abbaa07299b6423f9c16155aa3dc884d : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/668e3d3014fe47b09f5aec8a9c7a5fc9 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4da1d6343273447188e18c118440eb1e : SUCCESS in 7m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/727bef5987044a0c9e6a0210e0509d03 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa3b099ded544d4a68518fcfd37f130 : SUCCESS in 13m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb737b956aac4829991eafd1735cd995 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e47623e3fc35490389024461f0d749d2 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b16f7371619f42ed814abf6f4c517ed4 : SUCCESS in 10m 43s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/03a49252cda14e83b6cea97b0cdc323f : FAILURE in 22m 33s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/e07b006317584c50a5816cf7df179cbb : TIMED_OUT in 3h 09m 57s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/7882dbc6349940329038b3d3b843516f : TIMED_OUT in 3h 11m 13s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/0c0b4d41fa364950a60c80027fc0832f : POST_FAILURE in 43m 37s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/b739fe70ef6e4169bfd4f5f1bb66e3c5 : POST_FAILURE in 50m 05s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/d613d6ce30a04ca1978a1bb76e8ce057 : SUCCESS in 1h 33m 35s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/1d97b5c50eed4663851bd35f2d7826d7 : SUCCESS in 1h 32m 41s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/a2994555eef94c688972f64c0d1f2915 : SUCCESS in 19m 09s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/8fd74826b59747f1a6739c1ae75315f5 : SUCCESS in 19m 31s","accounts_in_message":[],"_revision_number":5},{"id":"d749c23dd816c925d3bae76ca8eafe6ed7f92d64","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-08 03:08:31.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ead8c71e08d91353fcfe5dd3a4ebd819cf5cf60c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-15 07:06:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d2e35ffb0664c69fd271b15736f18f9ed43b6736","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-15 07:11:28.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nThanks Akihiro for the detailed review","accounts_in_message":[],"_revision_number":6},{"id":"f999d46333348c108ea06e4059fef14b2d2b64db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:24:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf462d325a4144abae72b3bfefe49098 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09303e34f18f4d3a951e96ad377a6b6d : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/577abaeefdde4b46afbd9d31a3b0a338 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/747950a817d9411aa18b378e797fe130 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9e3aa48e694b5fa62c5da8aa51645f : SUCCESS in 9m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a798a831df2f4396befa6d03fb81322d : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/069a6f04ddcd4633a7cf1cfa7b026ec1 : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e15ac54e5234ba3adb1f42c028ad1b5 : SUCCESS in 8m 38s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/4ad6befe5bd84aca8cb00b9813cf47f3 : FAILURE in 17m 38s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/c8f8e0136ce4483cb03fef46a87174cc : TIMED_OUT in 3h 09m 47s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/5fefe0a7425e4a759fb488e67f7e3823 : TIMED_OUT in 3h 16m 09s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/c1d8b745695b4b8585a72a096e024ad9 : POST_FAILURE in 41m 38s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/bbedbf1b39004a4a933388c5914179b4 : POST_FAILURE in 51m 10s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/2d39bb29154f453cb138e4d9cba634a2 : SUCCESS in 1h 20m 30s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/f62447ee95e741b9aa79df87bbabc200 : SUCCESS in 1h 16m 31s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/06064384960f4e52bfbd56854df83e3a : SUCCESS in 21m 21s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/50e7398a1080448780f5b9e40476daae : SUCCESS in 18m 45s","accounts_in_message":[],"_revision_number":6},{"id":"96752b10dc2e86eaf7990a530b74d00559564ffe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 14:59:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"43faccae910ca8d6a35b20441ffc85cd703734ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:14:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03bdf7121a5b4284ab8cef83ca60e159 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fead634258345c988491403b65ff2f3 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbdb324441b1456da5f72e8ea08fcec1 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b854f18e36744bb19b13d4d117e560a3 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc168af3c043467eb800ee924bfbf491 : SUCCESS in 10m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8872ff636a3841fca03caa4ea04cd147 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4427766fa84451a87d0f10f2410ee17 : SUCCESS in 8m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc5a60d8746240f3a0a451718b9a1e57 : SUCCESS in 11m 37s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/1ee9172840ef461f8dc3cf802dbaf06a : FAILURE in 21m 31s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/6612de0739b64ac8b0e031fc75d42b06 : TIMED_OUT in 3h 09m 56s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/81c3368f4d85445283d4c4ac3874b6a5 : TIMED_OUT in 3h 10m 20s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/ed06d5b75df14e4295d3a82d27dcd16a : POST_FAILURE in 50m 43s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/4af9123fc34e4a64b10a9f969219e2fd : POST_FAILURE in 45m 59s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/89b73f3ccc30435fad06647561afb151 : FAILURE in 33m 45s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/f5d463ccf3a44d3289b564c790565861 : FAILURE in 33m 03s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/4cd1ffd151934f338e9ecc0e2e2b6abb : SUCCESS in 22m 43s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/b38bba390ac74679aff336e2a07eb11f : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":7},{"id":"f2f8026bb58b7b2403c2f2c73f891b97c7cd449b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-14 06:59:38.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"bd7fafe63c353becb4c38c79c910eac94a9e4ae2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-24 16:11:47.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"f5955b6eff71e1e5fa3e43e851c6f058832e8db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 19:27:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0229a52acd734b968b8174110f540141 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43684eea24444e53840933a9b1fdb521 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db504fe68aba4d478011fc499a1341dc : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18fe70aa2f924a018d894eff3594ebe2 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bb6a91b7bf4f4992895a976fe20da3 : SUCCESS in 7m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c356e97b96e44071816aa71187774c26 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4b386965c61441ebb34ab3262a34376 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39c940acefac42909bab7b89cd1a3576 : SUCCESS in 11m 20s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/c2342f0d808c48c0889ae4351d96653a : FAILURE in 18m 45s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/fa888d2ad3e54f67814be1d8ad9061d0 : TIMED_OUT in 3h 08m 41s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/5b422534377b4fe39c830a9e0635989f : TIMED_OUT in 3h 09m 52s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/60025a5f203941ff979aa527f41440ee : POST_FAILURE in 34m 44s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/9c78e6b86b9a43cfb11660754aec0715 : POST_FAILURE in 46m 12s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/43f9ee3d51a443c1908707be5ec10828 : FAILURE in 36m 06s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/4e886a93ba8c41fa8a52fbd3f44aecd1 : FAILURE in 36m 13s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/0ac1988a81e4435abc5daab6229623b0 : SUCCESS in 20m 36s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/aa6f7e021b15439aa888daeec95a5543 : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":7},{"id":"bc9fc6027fb634d46ffe94a9d22cdf1d492178fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-02 19:34:57.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e7e640045ecab2bedfe935340329ee466e3c7f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 19:35:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"dbce7b611b3787eb1f553d67b7f37a55d1514d15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:09:08.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67d38291a354e378652c81f94ea8037 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/313b0cb30545473bb70b9946541ab430 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9abe1f35e0af4c259d2b3d4d4d8985fa : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97f56393a5c94f70a9010b75f497ed46 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6effa73c54410cbf57e82a2ecc10c7 : SUCCESS in 8m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aaed3f84744440fcaead989a7540dc23 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0638ead7dd14b86852929814d2d85bb : SUCCESS in 11m 17s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/3cf0a454dd1c46b5a8ab1fb27cef6e7e : SUCCESS in 20m 20s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/deded3bf72d447148ea791d04475491a : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":7},{"id":"5476a1a276f194ecb389c1f779ae20805a26c8a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:09:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"2fc221fa298dc214cde6e8bc6bf59f539f5a39ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:11:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7091e71fad74bec94052a8a5bae9a6b : SUCCESS in 1m 53s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e62c825dcb4d26b45af05f1ad7e40a : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"84ebb940a6d5fa05cb69e3bdbf0cea517b1e22d3","revisions":{"4979ee78b48a0073cfe0efeca7f00bfbb4a0c2ba":{"kind":"REWORK","_number":1,"created":"2020-03-13 10:24:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/712900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/00/712900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/1"}}},"commit":{"parents":[{"commit":"3de47a82996c66a3349d880c7a0c34a0b0e0ae6f","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3de47a82996c66a3349d880c7a0c34a0b0e0ae6f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-12 14:00:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-13 10:23:28.000000000","tz":60},"subject":"Switch to hacking 2.0","message":"Switch to hacking 2.0\n\nMake the code compatible with hacking rules:\n* line break before/after binary operator (for this added ignore for\nW503, line break before binary operator, as if I understand well the\nsuggested is to have the line break before the operator)\n\nChange-Id: I5f9e0c417336c8ff9e417a668268784485d8d2fa\nDepends-On: https://review.opendev.org/#/c/708965/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4979ee78b48a0073cfe0efeca7f00bfbb4a0c2ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4979ee78b48a0073cfe0efeca7f00bfbb4a0c2ba"}]},"branch":"refs/heads/master"},"b3be7511b0ea2a2de417ca9db00430f98451321d":{"kind":"REWORK","_number":2,"created":"2020-03-16 14:17:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/712900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/00/712900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/2"}}},"commit":{"parents":[{"commit":"3de47a82996c66a3349d880c7a0c34a0b0e0ae6f","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3de47a82996c66a3349d880c7a0c34a0b0e0ae6f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-12 14:00:03.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-03-16 14:17:04.000000000","tz":0},"subject":"Switch to hacking 2.0","message":"Switch to hacking 2.0\n\nMake the code compatible with hacking rules:\n* line break before/after binary operator (for this added ignore for\nW503, line break before binary operator, as if I understand well the\nsuggested is to have the line break before the operator)\n\nChange-Id: I5f9e0c417336c8ff9e417a668268784485d8d2fa\nDepends-On: https://review.opendev.org/#/c/708965/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b3be7511b0ea2a2de417ca9db00430f98451321d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b3be7511b0ea2a2de417ca9db00430f98451321d"}]},"branch":"refs/heads/master"},"75cdc055f42ac5392c634ada5b00110b68e797fc":{"kind":"REWORK","_number":3,"created":"2020-03-30 10:06:35.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/712900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/00/712900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/3"}}},"commit":{"parents":[{"commit":"3de47a82996c66a3349d880c7a0c34a0b0e0ae6f","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3de47a82996c66a3349d880c7a0c34a0b0e0ae6f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-12 14:00:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-30 10:01:52.000000000","tz":120},"subject":"Switch to hacking 2.0","message":"Switch to hacking 2.0\n\nMake the code compatible with hacking rules:\n* line break before/after binary operator (for this added ignore for\nW503, line break before binary operator, as if I understand well the\nsuggested is to have the line break before the operator)\n\nChange-Id: I5f9e0c417336c8ff9e417a668268784485d8d2fa\nDepends-On: https://review.opendev.org/#/c/708965/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/75cdc055f42ac5392c634ada5b00110b68e797fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/75cdc055f42ac5392c634ada5b00110b68e797fc"}]},"branch":"refs/heads/master"},"eb381c2a8a7da7cd687d6fbb5ced487e63218bae":{"kind":"REWORK","_number":4,"created":"2020-05-07 09:32:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/712900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/00/712900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/4"}}},"commit":{"parents":[{"commit":"3de47a82996c66a3349d880c7a0c34a0b0e0ae6f","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3de47a82996c66a3349d880c7a0c34a0b0e0ae6f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-12 14:00:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-07 09:32:08.000000000","tz":120},"subject":"Switch to hacking 3.0","message":"Switch to hacking 3.0\n\nMake the code compatible with hacking rules:\n* line break before/after binary operator (for this added ignore for\nW503, line break before binary operator, as if I understand well the\nsuggested is to have the line break before the operator)\n\nChange-Id: I5f9e0c417336c8ff9e417a668268784485d8d2fa\nDepends-On: https://review.opendev.org/#/c/708965/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/eb381c2a8a7da7cd687d6fbb5ced487e63218bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/eb381c2a8a7da7cd687d6fbb5ced487e63218bae"}]},"branch":"refs/heads/master"},"07cc2de515e6c90dc1fda375572bc436e21c2de5":{"kind":"REWORK","_number":5,"created":"2020-05-07 09:52:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/712900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/00/712900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/5"}}},"commit":{"parents":[{"commit":"12661e9d451143ecc4fa64d1c99da623de306bc5","subject":"Add subnet deleted handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/12661e9d451143ecc4fa64d1c99da623de306bc5"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-12 14:00:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-07 09:50:58.000000000","tz":120},"subject":"Switch to hacking 3.0","message":"Switch to hacking 3.0\n\nMake the code compatible with hacking rules:\n* line break before/after binary operator (for this added ignore for\nW503, line break before binary operator, as if I understand well the\nsuggested is to have the line break before the operator)\n\nChange-Id: I5f9e0c417336c8ff9e417a668268784485d8d2fa\nDepends-On: https://review.opendev.org/#/c/708965/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/07cc2de515e6c90dc1fda375572bc436e21c2de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/07cc2de515e6c90dc1fda375572bc436e21c2de5"}]},"branch":"refs/heads/master"},"829d1db15741dca42f112ea77bc0357667dfe400":{"kind":"REWORK","_number":6,"created":"2020-05-15 07:06:53.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/712900/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/00/712900/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/6"}}},"commit":{"parents":[{"commit":"ef420a935d042e47b21d5fa7a2e3cf048b2935ad","subject":"Add subnet deleted handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/ef420a935d042e47b21d5fa7a2e3cf048b2935ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-12 14:00:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-15 07:02:13.000000000","tz":120},"subject":"Switch to hacking 3.0","message":"Switch to hacking 3.0\n\nMake the code compatible with hacking rules:\n* line break before/after binary operator (for this added ignore for\nW504, line break after binary operator)\n\nChange-Id: I5f9e0c417336c8ff9e417a668268784485d8d2fa\nDepends-On: https://review.opendev.org/#/c/708965/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/829d1db15741dca42f112ea77bc0357667dfe400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/829d1db15741dca42f112ea77bc0357667dfe400"}]},"branch":"refs/heads/master"},"84ebb940a6d5fa05cb69e3bdbf0cea517b1e22d3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-04 14:59:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/712900/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/00/712900/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/00/712900/7"}}},"commit":{"parents":[{"commit":"52a6f383f437400f9cee1c587850af961f761e69","subject":"Merge \"Remove neutron-fwaas from the required projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/52a6f383f437400f9cee1c587850af961f761e69"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-12 14:00:03.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-08-04 14:59:46.000000000","tz":0},"subject":"Switch to hacking 3.0","message":"Switch to hacking 3.0\n\nMake the code compatible with hacking rules:\n* line break before/after binary operator (for this added ignore for\nW504, line break after binary operator)\n\nChange-Id: I5f9e0c417336c8ff9e417a668268784485d8d2fa\nDepends-On: https://review.opendev.org/#/c/708965/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/84ebb940a6d5fa05cb69e3bdbf0cea517b1e22d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/84ebb940a6d5fa05cb69e3bdbf0cea517b1e22d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
