)]}'
{"id":"openstack%2Fnetworking-odl~722263","triplet_id":"openstack%2Fnetworking-odl~stable%2Fussuri~Ie33b1bcc9aed0c9ad1da911adbbaf0254a83eafa","project":"openstack/networking-odl","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ie33b1bcc9aed0c9ad1da911adbbaf0254a83eafa","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-23 09:49:47.000000000","updated":"2020-07-31 10:33:09.000000000","submitted":"2020-07-31 10:31:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722263-1596191480294-7dc4a3d4","meta_rev_id":"010839ec078b6190995359cb9f497c4cac4c6206","_number":722263,"virtual_id_number":722263,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 10:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-07-30 15:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-07-30 15:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-07-30 15:09:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2020-07-30 20:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 15:17:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-04-28 10:02:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-28 22:27:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-30 20:43:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-31 10:31:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb2a7adffbc632822b7101d9ee2122fff6d9abe8","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 09:49:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4df5bbf35aaa88c148ad5e3d2b4dac053925230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 10:20:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b5cae0e8aa4d74a944a3f8f7d0e360 : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c19388f4dd84de0954e07d0611f6601 : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd62768bf11145679af41ac15924a099 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1b35c769b94ac69634552564c5a298 : SUCCESS in 12m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2d49fb0ca64d15b2f4b15925180042 : SUCCESS in 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f59bdc364a0e4c56acf4e31544ced005 : SUCCESS in 13m 56s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/b351f394e9ae42e8bf9c754ece16535e : FAILURE in 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"797e306525244438af0bf5d5001fd43a0166c82a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-23 15:17:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e5927e7ff1f9aa21f7d85217607c4410654315a4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-28 10:02:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c943b134129f8cfa27add9eec5022b4ac3603e82","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-28 16:00:18.000000000","message":"Patch Set 1:\n\nrecheck zuul restart lost events","accounts_in_message":[],"_revision_number":1},{"id":"509494b41372ed66dae055af8d31adce4b50026f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:22:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9f1e218b904f83b1dcb2741bc2c836 : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/127c152bc5e64344aac70f1c64df139f : SUCCESS in 9m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a07f68538057497495e01c0944afcd95 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f384e08964f496cbff11545d16aa888 : SUCCESS in 11m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ec8833ccaf749389481cc901ba83f6c : SUCCESS in 11m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7a8e578a1d345e2a2ee74c147073eb0 : SUCCESS in 12m 08s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/e50a697e359b423cbd99a89c8b12ce80 : FAILURE in 22m 36s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/b62f8f49de1a4f0885920864dbb6454d : TIMED_OUT in 3h 08m 58s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/e06ddd25e0cf446ba98fea8d9d6b06a6 : TIMED_OUT in 3h 11m 45s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/1df9e696f93c4c3f9ccb994a417fcf72 : POST_FAILURE in 51m 40s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/91d2d814130b4524a225622c891896fd : POST_FAILURE in 45m 52s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/d0886819c0824c709ade71826326bde8 : SUCCESS in 1h 13m 30s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/1a4ca44a82554accbcc5f86271c0ef5d : SUCCESS in 1h 32m 37s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/684314b78ada4ba7b107481d7fcfb614 : FAILURE in 23m 34s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/b0df735fde96442ea2a7ab4e7b86d1dc : FAILURE in 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a57853566a1129d5521c353264f0fbd3d47b9b96","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-28 22:27:54.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"90e34dc27e9a20a3fdc9e7962eb0c723ebdb7476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:43:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b85b5f369594d00b753cf3fc4a5d67a : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31df16b7b0854574ae68174472dd01ff : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/653b39c46d6447e2812e6ac19d8163c7 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962363faebe64304a6af37be00a55749 : SUCCESS in 7m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0d99e7f103d42dbbf0aa0fd0195d725 : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/685a336c91c146789fb8c45caaa4909a : SUCCESS in 8m 43s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/611fbd7a065b41df9e496bfa4a0ac6b8 : FAILURE in 20m 42s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/cbc945636b544e54ae33500fa2b167b6 : TIMED_OUT in 3h 07m 32s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/cfb39f4a8e854134bf82c7b7a3d64a29 : TIMED_OUT in 3h 08m 05s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/6eb0fde847ce499db83960ef9c73968c : POST_FAILURE in 48m 21s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/efa5becc9eaf4b62a4cf3ce1e3e92aad : POST_FAILURE in 1h 00m 30s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/16183c88892a4c0e9574169b5c118f86 : SUCCESS in 1h 23m 58s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/a98232a63d12467bae0c0a7456ba3abf : SUCCESS in 1h 27m 25s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/18467b32a26b45e5857e011f62116fe1 : FAILURE in 19m 00s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/210058c6dfc840fe811e310059685460 : FAILURE in 23m 34s","accounts_in_message":[],"_revision_number":1},{"id":"79cf7d1d64c1616835c3b3fd3657e82f12c4ab5a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-29 08:31:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6329e192acd0f6b54367cd74bab010ca0acab5aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:49:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748e4f15cc5b4fe9b73202ec40b45332 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e417e2ba63c2460c8585f9824f91ce44 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6bcca772d8747aa81f4e9a923db3d94 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f49325290bba43949a9f4709d2babdb0 : SUCCESS in 7m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f3468495c804e1992b98bb319153afb : SUCCESS in 12m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/feaa2e04ec3645c3827b1bc2e3fe24e3 : SUCCESS in 9m 22s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/950167e3fa2347749e97a1a77a4e1428 : FAILURE in 23m 04s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/3c9eda22802748808bf6258595aec072 : TIMED_OUT in 3h 11m 13s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/8d7e865186c5499782c75f6b8459748c : TIMED_OUT in 3h 09m 40s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/b4ae09d643a84b6ca96a3b2b2bbc3679 : POST_FAILURE in 41m 22s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/aebdf54f1dfb49dba5283d0fe0739160 : POST_FAILURE in 42m 37s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/85cc1c71078642bb896a55ad816ebb54 : FAILURE in 1h 26m 32s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/0b08de71b1f14ad2a738c03973738022 : SUCCESS in 1h 23m 25s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/0b952195f4994bd297efe0e4b226d8bf : FAILURE in 22m 15s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/c13b3bcb9e5b4aa89526d0fc9810a2db : FAILURE in 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"21767a03d2b1faa8f2f3d339e58790f38676a4d0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-29 15:54:21.000000000","message":"Patch Set 1:\n\nrecheck\nThe error for me looks like some terrible timing issue, let\u0027s see","accounts_in_message":[],"_revision_number":1},{"id":"2c8792a15372ef4aa07b05261ab8017b16653741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:05:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9929b84704054942a30babfc238d6a83 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e5ed6ed948c43a296d2005098e5b580 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a254ef7c57b4a28b534d4ab91ee47e9 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30fcae565c52476f8a3ddaff5d081ba8 : SUCCESS in 9m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23e0a3187dc1450d9237c7becd3d968f : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78ff8d7b9fc04860b72dbb1a61d9e0ca : SUCCESS in 9m 31s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/dd9f5955935542eeafbb5e3b188cfa26 : FAILURE in 24m 04s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/8488407c81304309b0df50abfad741ed : TIMED_OUT in 3h 08m 49s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/c53bf8688ea44304b8bb8b327057a97f : TIMED_OUT in 3h 08m 22s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/a84bb2f64c7d4f9592d6e7e2e55681dc : POST_FAILURE in 41m 30s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/83a19049467e4f088e33cdf82203e83c : POST_FAILURE in 46m 19s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/4ba33c8c343644d590d3d0bbf3ae0664 : SUCCESS in 1h 32m 51s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/57ff52762bed4074a7442b19813597d1 : SUCCESS in 1h 23m 17s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/f1372e7691cb48c0b55c1fbb0c540b67 : FAILURE in 20m 24s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e8c7fa93970549dd9a941f3d9639a13b : FAILURE in 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d027f001baa5b5374c7fb47d6f097ac1d0982777","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-30 14:51:49.000000000","message":"Patch Set 1:\n\nThe fix is on its way:\nhttps://review.opendev.org/724656\n\nAs soon as that is merged to master I backport it to stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"ea587f65374e8e366ac79acf14d2ce4b069a036a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-05 11:09:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"21f59bad97efdecacd0b4c9fcf8b819e473b87a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:23:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35176f97c18e4244abdeef369a4d4f1f : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd43f0678c2f4ab79a2d77d6c9e73c4a : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5193f947f1934aa5a73b8ad22c21725a : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d354bd51d5204a3aa0975239e9e065de : SUCCESS in 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01f606a821164fa3b273f169ec76675c : SUCCESS in 12m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80e70ec1d42f486b929ee16cd0c12b4b : SUCCESS in 9m 26s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/b3a4e47b136b4860b72e26651a932520 : FAILURE in 19m 31s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/b810c14774ba4b908b1c831bdf5a8b42 : TIMED_OUT in 3h 08m 19s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/b32df46bde0542d99825e3f2f6b3b4f7 : TIMED_OUT in 3h 09m 59s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/5113e38b6bda4061bfb16d44e64bff89 : POST_FAILURE in 42m 02s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/3a45dd5dd18c41d78cbbc67f6b201a1a : POST_FAILURE in 45m 50s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/839f71e5e4834472906ee8cacd97e979 : SUCCESS in 1h 37m 37s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/30f018c4d51c4932a9d92fc0fdad5e7d : SUCCESS in 1h 32m 55s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/c94187b5d8be4d8db2a30af434d7fc91 : SUCCESS in 17m 03s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e4a182ab6b9a472ab2131ca79845a066 : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f136d30d9680f0b94b0eaed56e618eaba5605de0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-06 15:21:24.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ef37c821fc678a18163312ba759123b648f9d1fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:42:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4deb667a92e415384365599d4839d43 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/927fc75055334296a35902bb3342b7bb : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/459b4d3b5c5c460e8844f3dc2564ae97 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56805709a02e43218b2807fedcafafea : SUCCESS in 10m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2293f9f430d4a718da82d50b38253b4 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42091e1b7d4f4ef1927f507b8b661907 : SUCCESS in 9m 46s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/130c7466b8694e968bc094f41e22c507 : FAILURE in 20m 23s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/c2a99eda9a3e44549d10164e54989abc : TIMED_OUT in 3h 11m 42s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/7ee7733ef3ab4be4a67ac5ce60ef14b7 : TIMED_OUT in 3h 10m 53s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/1df8756128c1484fba97b784c0fa77a3 : POST_FAILURE in 44m 11s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/a3f801b02bba4ba9bea85b874c8026d0 : POST_FAILURE in 47m 17s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/9d5afa29be9a4c8f83586bb6a24c879d : SUCCESS in 1h 28m 07s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/7a16a8b522914e1a9386cf2accfbedd9 : SUCCESS in 1h 23m 33s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/c7ce1f75922c442cac4958653c69f0e1 : SUCCESS in 25m 07s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/2a05380e89354cfbb67bed1047bb4992 : SUCCESS in 23m 04s","accounts_in_message":[],"_revision_number":3},{"id":"7a2741b29d3e077cb0a24f6616301ec8d770c98d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-11 08:58:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f728fa54fc23a7e0ef797a044abf0da04fffad00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:23:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9d8fa8c1b44cc38ff3b0e48cd977f3 : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd933cf60f614a74b514cfbf5dd1d252 : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5437c88a218a49348c68688c7716fd1c : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee7030f651a4ab1a4ebc0c8c8f4b789 : SUCCESS in 8m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8f5d31710994905a33f685664e334fa : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af7ab2d3f26745bcace92db09c6cdff4 : SUCCESS in 9m 04s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/fe119980ec21425d9b1c293429b6a130 : FAILURE in 28m 58s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/806ff352eb004e8cb70ac1a93040b790 : TIMED_OUT in 3h 11m 38s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/b30651ad1ae14265ae8f4ab2711285fb : TIMED_OUT in 3h 08m 32s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/370a8d8ab89441bbbe28d43a63086433 : POST_FAILURE in 39m 38s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/bd339c1ffe7247b49ba365fd07b2cdc4 : POST_FAILURE in 41m 40s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/812dab6c88d44490a4f0afe08c7ac220 : SUCCESS in 1h 20m 48s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/223746d56e884a80a2feeae16ad32850 : SUCCESS in 1h 16m 34s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/6db9060b0063448981e44db75ee124d4 : SUCCESS in 21m 38s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/c08995cd96a0406dac86cb33d215ecf8 : SUCCESS in 22m 45s","accounts_in_message":[],"_revision_number":4},{"id":"3388f5ec7c1c61f55a61bfd04ed2adfe42eae692","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-07 09:48:50.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2acf964a3071c4908ad55b735a9351597d2b5505","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 10:36:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c31c52b2caa21d73a0c3507d3a316375c59425f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 11:40:15.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"1c5b95f6d49642b9759d1df887bfbd84445afdf9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-30 15:09:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"be74925e86152ec4986ef7d69b4cd5ea8e55d011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 18:30:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e100d17b90504f91b7945d0d1b8d04d3 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4f1222d0273477685bd05e4ab3111cd : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8944665dac6140268f367334e0040483 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/369c3dbbf3de495795777b80859c9587 : SUCCESS in 11m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/537a5f4b609d4c6ca451c3b03f73b453 : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c5b0b9ce2794ad5afebba19d62ad523 : SUCCESS in 8m 46s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/410c78633e2e4d04bbbf8bc01f9c0a93 : FAILURE in 26m 03s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/159fb043c03e482283c3bfe9738f044d : TIMED_OUT in 3h 07m 14s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/85c1713ba51b4b4583483f7bba89af16 : TIMED_OUT in 3h 08m 01s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/862b8512bd3d44a3b657631f84e55cbd : POST_FAILURE in 44m 53s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/c55ad8ccb03a407a8d5217dd779f793b : POST_FAILURE in 40m 59s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/9d206dc4724448b4a5ab8aa9c157c840 : FAILURE in 35m 42s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/46ceab1479074304824e3bcfdb49f1f3 : FAILURE in 29m 11s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/7ce1a20c9ebd441bbfab1b06f84b197e : SUCCESS in 21m 18s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/6f431f5000064cd7a66a0207091a0433 : SUCCESS in 16m 58s","accounts_in_message":[],"_revision_number":5},{"id":"f8c07201d4503b8bffcd6c732cecffceb2430c73","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-30 20:43:45.000000000","message":"Patch Set 5: Workflow+1\n\nadding again +W","accounts_in_message":[],"_revision_number":5},{"id":"a41e90963f98b9d1f60d1021c934064d634c4ab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 20:44:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fe34578a9d5e2dafad4e409aa34ee8d0b1e39d98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 21:10:14.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0410851da26c4699a283cd7d4379e4de : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eec706d72844b06b370e1f2f9770afb : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89267e9c5d3546ce8fdd37846dad348b : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b3bfc1c1214682b94ad7bb65f3467c : SUCCESS in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27231b0fb7cf4bfe937c059f71da008b : SUCCESS in 6m 27s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/ca7e512162ba4b3bb0523f709e68c15a : FAILURE in 22m 09s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/fc1edd37cea94d8b87503199fc765177 : SUCCESS in 17m 35s","accounts_in_message":[],"_revision_number":5},{"id":"9f8caf6c24a8815dcdefd35110e66172b740cbbd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-31 06:40:00.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c84510321b2cde0c513dcd14fee3998b52a3b540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:02:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74e1074a1bc49e8a99ef607c4911ea5 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10917e4d30644fdfa948fc8c8fc67dd4 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59a0eb2a52674b66a9f0e4b90e517cb9 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2048722a2bd4b33ba6a2ded5199c149 : SUCCESS in 8m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e650a21e3a5a4e91a463e79906df2ad4 : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d17187a1ab945b59ca93a762d51bedb : SUCCESS in 12m 42s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/f71d84ff1b104a3081a1cb5ed193c095 : FAILURE in 19m 31s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/bc45d5e432ec40b8a41e33ed28981b2c : TIMED_OUT in 3h 12m 14s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/c3eda4382eef4b829800e7c830daa674 : TIMED_OUT in 3h 06m 46s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/65249a6638b24d0ead96fbfed50d549a : POST_FAILURE in 33m 35s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/02fb276c6b024aa7ae1cd58c13d36da3 : POST_FAILURE in 43m 04s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/584d03c987ce457f8a91649934cc94fe : FAILURE in 30m 22s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/7694c99815d843fc86ba5a07ca815265 : FAILURE in 36m 56s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/cbdbca350887420c822a7e35668c0c0a : SUCCESS in 20m 12s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/d3a543d6be4642a59ebdbe82bd2a279c : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":5},{"id":"011ae62c80271a08dff0f81986f84ff450990969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:02:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e9934a3399fa0913595f08c1c1b6bee8afe731c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:31:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2ab004dba240a5b50f87ce5de57a2a : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/315236b376ac49169769233483200f10 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7e7f58be82949c49a1123d0bf562d42 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7720f0c7fb44b32a8a1c27d121ac039 : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3097541df61b4649953cf7f4ad8dafcf : SUCCESS in 8m 47s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/1d9b110f6a364331bdde6e964b70b7f6 : SUCCESS in 21m 08s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/eea9da89aa5644b891e5841563c6a9bf : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":5},{"id":"04cd4e40ad3f6197badde77474210c97f98bdb4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:31:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d09a1a1475fa9dd5d66b67f58bdc8e8d0ba4deaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:33:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fdd44eca39429d8ff7c2059e64eeb0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"18021af42857ba3137fc273e2377ac7ab2a15426","revisions":{"7088fa0c212c193cfb6dd1ac7e303867fb9baaba":{"kind":"REWORK","_number":1,"created":"2020-04-23 09:49:47.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/63/722263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/722263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/1"}}},"commit":{"parents":[{"commit":"56d5d9594f1254b313670ba73e1b4ac13f021f1a","subject":"Make tempest jobs use the latest ODL and zuulv3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/56d5d9594f1254b313670ba73e1b4ac13f021f1a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:45.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:45.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Ie33b1bcc9aed0c9ad1da911adbbaf0254a83eafa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7088fa0c212c193cfb6dd1ac7e303867fb9baaba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7088fa0c212c193cfb6dd1ac7e303867fb9baaba"}]},"branch":"refs/heads/stable/ussuri"},"d80982f479456411ff21e4d4df87464f3f56b763":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-05 11:09:40.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/63/722263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/722263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/2"}}},"commit":{"parents":[{"commit":"f417fe2871154e2ba7eacabbd6b08b9017e05ab1","subject":"Add subnet deleted handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/f417fe2871154e2ba7eacabbd6b08b9017e05ab1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:45.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-05 11:09:40.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Ie33b1bcc9aed0c9ad1da911adbbaf0254a83eafa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d80982f479456411ff21e4d4df87464f3f56b763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d80982f479456411ff21e4d4df87464f3f56b763"}]},"branch":"refs/heads/stable/ussuri"},"830863f4c24c580e698d9e5e9a43d6063af80f00":{"kind":"NO_CHANGE","_number":3,"created":"2020-05-06 15:21:24.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/63/722263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/722263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/3"}}},"commit":{"parents":[{"commit":"2f1cda3ff9de426cdebb2d1360ae7ab9010d3de0","subject":"Add subnet deleted handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2f1cda3ff9de426cdebb2d1360ae7ab9010d3de0"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:45.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-06 15:21:24.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Ie33b1bcc9aed0c9ad1da911adbbaf0254a83eafa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/830863f4c24c580e698d9e5e9a43d6063af80f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/830863f4c24c580e698d9e5e9a43d6063af80f00"}]},"branch":"refs/heads/stable/ussuri"},"94fe2017dadd0aef4c8d024065e52b1c117cbfd6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-11 08:58:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/63/722263/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/722263/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/4"}}},"commit":{"parents":[{"commit":"90f5bb0e10ba16d83cfd15e9b51893434fc7a1ea","subject":"Add subnet deleted handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/90f5bb0e10ba16d83cfd15e9b51893434fc7a1ea"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:45.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-11 08:58:13.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Ie33b1bcc9aed0c9ad1da911adbbaf0254a83eafa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/94fe2017dadd0aef4c8d024065e52b1c117cbfd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/94fe2017dadd0aef4c8d024065e52b1c117cbfd6"}]},"branch":"refs/heads/stable/ussuri"},"18021af42857ba3137fc273e2377ac7ab2a15426":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-30 15:09:03.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/63/722263/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/722263/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/722263/5"}}},"commit":{"parents":[{"commit":"d31b0ef9319dacc7cd4cc89e7963f2a1e82df1fa","subject":"This change is a squash of 2 different patches to pass the gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d31b0ef9319dacc7cd4cc89e7963f2a1e82df1fa"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:45.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-07-30 15:09:03.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Ie33b1bcc9aed0c9ad1da911adbbaf0254a83eafa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/18021af42857ba3137fc273e2377ac7ab2a15426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/18021af42857ba3137fc273e2377ac7ab2a15426"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
