)]}'
{"id":"openstack%2Fnetworking-odl~722264","triplet_id":"openstack%2Fnetworking-odl~stable%2Fussuri~I28ccba1b07097dccabf664dc928f0893b410ad8a","project":"openstack/networking-odl","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I28ccba1b07097dccabf664dc928f0893b410ad8a","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-23 09:49:49.000000000","updated":"2020-07-31 14:19:13.000000000","submitted":"2020-07-31 14:17:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722264-1596205073972-3ec634e1","meta_rev_id":"8c61ca2f6ee73c7721a0d6ddb94db40957f90f13","_number":722264,"virtual_id_number":722264,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 14:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-07-30 15:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-07-30 15:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-07-30 15:09:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2020-07-30 20:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 15:18:07.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-04-28 10:01:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-28 22:27:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-30 20:46:59.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-31 14:17:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"356b8fdb43803be9bc50c6d14beaf510a2949baf","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 09:49:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fa5da5d69c19a5ffb723b868375e0db0b2ce692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 10:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876884d5308b4f2c911f2fe95cc6de0d : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18ca78b6ec564858acdf439f17691656 : SUCCESS in 10m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/935356550ba84b6ba013e6226d63a9bb : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae9e4fab54140358a8a9863765e4b1c : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2b04a35b8f499399977681959d51b4 : SUCCESS in 12m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ac128b4eb854cecb1355d4c7acace7d : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b8b965d6c04a08b41593ff54be5d3b : SUCCESS in 11m 01s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/9300c604e3134bee82d0ed4311df1b4c : FAILURE in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba430c60abdba5c9c4d7ccf1a57ecb192f706c0c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-23 15:18:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7c747abc8231ee020621c374e8dd9584319fabd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-28 10:01:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d92f90ad510cb7c1301bf12692dcf23d09a18d5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-28 16:00:21.000000000","message":"Patch Set 1:\n\nrecheck zuul restart lost events","accounts_in_message":[],"_revision_number":1},{"id":"0ade57edefe69418b592d0367355ef6833b3c78e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:50:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d5b7bf448b425bb6614f3a2d229a92 : SUCCESS in 11m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ed775f6395945dea8b48d7e0680bf83 : SUCCESS in 11m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4863ce979fb84776a736d0449643d931 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab3e7cf165eb4103be92151ce5e451ff : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/157ebd48323245be80a68654c2e66765 : SUCCESS in 9m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/586d4cb0c6004c939ceb58d100a4ec2d : SUCCESS in 11m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d7f325cf0324867a9d2a156780357b3 : SUCCESS in 12m 05s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/9b5828f9714b4132afbf5dc041622f2c : FAILURE in 22m 12s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/2646331ae512492abe223783585c57b9 : TIMED_OUT in 3h 07m 26s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/1391ac67b81546a7be6ca077db5aaa9e : TIMED_OUT in 3h 07m 50s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/a510f509be0e481492c7492ec70ad0d8 : POST_FAILURE in 51m 51s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/e6491f762a61460ebb68843bbf6f546a : POST_FAILURE in 52m 15s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/445b89c2588e4db79b65ea6148034760 : SUCCESS in 1h 28m 37s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/a2e898b45caf4cd6be30dc773e1b1e14 : SUCCESS in 1h 17m 17s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/b4fccaf6dc8d465eae94bd0069e79abc : FAILURE in 21m 18s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/9799e9b4d6fc4829ac15f6bdc3b0460e : FAILURE in 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"09e772015b437dd5a2f1b51644933f4f69bbbaec","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-28 22:27:24.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49f4a54abfd9d90a1d93158ec7451a921626e077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:40:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da1e59c771ce41f6bdb989700b512ff0 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/167e9acb0be448289136fbdddd3334ef : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddd7fc8759724a40a5f2dd59f6524a63 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61f21bd327d44da28448af6175066916 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2927691e6ad54edead4e9f90a6166b7c : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/417d55c9216c41ec9062083e6d8e12cd : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96489b07027249a89296dbf343c2a197 : SUCCESS in 8m 54s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/30f1be23623d434ca2b1fb36a341bcf1 : FAILURE in 23m 49s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/ee126f1e8d89462da97b5d04caf2d784 : TIMED_OUT in 3h 07m 45s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/51230f1cc3c647b3a2b791de442124fa : TIMED_OUT in 3h 07m 35s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/1d9e4333b7d7474fb73e998f1eee3226 : POST_FAILURE in 46m 43s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/e1ee7edc29e545339348ab4fd11f8cde : POST_FAILURE in 45m 28s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/73ed5a27eabe4adca47241d98ebb5bed : SUCCESS in 1h 08m 24s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/cb95267bf1ce45eca97b761dd4855097 : SUCCESS in 1h 07m 40s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/74e07656aacc4991aba561c820fc1a30 : FAILURE in 20m 30s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5661939581c042199fbf6dd221019d5e : FAILURE in 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"00979472a1a1c44c2999c6464392dcca4eb06cd8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-29 06:54:44.000000000","message":"Patch Set 1:\n\nrecheck TIMED_OUT","accounts_in_message":[],"_revision_number":1},{"id":"917433520090a93e8ebf1e77344338d28979ed5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 10:11:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79c53bfae524465abb40f33c932e0b4 : SUCCESS in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2fe4b3a72964cc4b47d7e2a84de4044 : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddfea447878d47bd9908c85f00f77bc3 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/216fdb2d91394a8bb4017c1666864e02 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f0cd1d70f44e71b75a0914cfd4dd27 : SUCCESS in 11m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25ad7b7348fa425bb48cc8f203824a60 : SUCCESS in 13m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4322a85f4df4d8f9f4fdc04fd869f5f : SUCCESS in 11m 27s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/dfc186e69add456ca4f6b9a2091aa3cc : FAILURE in 30m 02s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/77982d8d49f8474a935023b479bfda1c : TIMED_OUT in 3h 09m 40s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/814942edbe594b9fa98ea98afa738764 : TIMED_OUT in 3h 08m 41s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/86babc9f48384b6fbb59b2e682e699a1 : POST_FAILURE in 55m 36s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/cb5dee1b6b9f4a329e0f8edf154195f4 : POST_FAILURE in 54m 42s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/9d4ad6abc9004aea9155163dc5835ee3 : SUCCESS in 1h 21m 12s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/5a8905c671b94a9495170d5a4b489a72 : SUCCESS in 1h 23m 22s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/a6bb7cfda12e40aa996cbadecb26344d : FAILURE in 19m 24s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/7289485550a04aa9a476abd7d4a4d568 : FAILURE in 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1ba93e20b52990ce1419310dc2b89dce81a4ae7e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-29 11:54:38.000000000","message":"Patch Set 1:\n\nhmmm, test_subnet_delete seems to be failing constantly","accounts_in_message":[],"_revision_number":1},{"id":"9b33168f19185d7acb5ceb26b2a5f27df655206d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-30 14:51:58.000000000","message":"Patch Set 1:\n\nThe fix is on its way:\nhttps://review.opendev.org/724656\n\nAs soon as that is merged to master I backport it to stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"98faeff4cefbae4e5fc8a28ba9ebb815aa194e1b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-05 11:09:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b7d1f4853dd55f0bda065fbd732ebceeba8ed9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:22:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa3996166d44acda1f5d7097361712e : SUCCESS in 10m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e077c5825c4e40a2ad9f54fd63f5c1cf : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e180a7addcf6460e897e9426d384c2da : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b14b2fd28e154ef78a2595cad506eb54 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8ae2b13f684e5fa8ca49d01484a8f0 : SUCCESS in 14m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8a59c715b6e4df3a65d8f1c854ecd29 : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c87235b13e75494e89aa51b3724b4406 : SUCCESS in 7m 53s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/44d6a77c1a41486ea11e78035659c33e : FAILURE in 20m 18s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/54335af41a324a6b9b98afc9a414c5ac : TIMED_OUT in 3h 09m 34s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/78908794beda4976bcc3a658a568c863 : TIMED_OUT in 3h 08m 13s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/f8485cb296844af093a7380209937a2a : POST_FAILURE in 43m 12s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/056b190f692445bd90d7336e66b784c3 : POST_FAILURE in 45m 00s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/f6e34513e0fd4a8abcab163e9ca2ee33 : SUCCESS in 1h 15m 23s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/609be46d472b43f89541e04de91316ee : SUCCESS in 1h 18m 11s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/8f4405ee73e44e3cbc987343335d2588 : SUCCESS in 22m 37s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/f77ba4e2ce2e4ba28c6195fcc39a1871 : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":2},{"id":"30fc2e1abd665b1288aed6ae9c67d8384950242a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-06 15:21:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"80d2461eec6fcdff125c7737bcc219249d7e1f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:46:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ce2f68db1b4bee826a31fc40ba7743 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d91754c4a294fc5acbd6ac1db0a7ea5 : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28a876a3520644338af0e3da2dccc97b : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26c47c7fa2a24b17a14ac6ef74d01360 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64daf96a6f0479e8dd8da8f4c3be144 : SUCCESS in 11m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1f10b8c112f4e31990571d53b393553 : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d294e3ec2d794a47a9c25a768de1b4e0 : SUCCESS in 8m 07s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/227ff1c10fb840ab97d0baa0b11cd587 : FAILURE in 21m 36s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/5fdf45378128407e8ead1bd2c79b3396 : TIMED_OUT in 3h 13m 41s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/9ca64416c7c943608de8fb6190ec340b : TIMED_OUT in 3h 11m 00s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/b120cb167aef407dacb07ebb901fed7a : POST_FAILURE in 40m 30s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/ac5fc336a6d5405c89b7bdd4dcc176f8 : POST_FAILURE in 47m 03s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/7673d494c3c347b886d07049787308a9 : SUCCESS in 1h 22m 58s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/e3f238ff69be48e9a8c2da943275db0c : SUCCESS in 1h 23m 43s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/8f102fd4b8764f9fad56a4b1faba81a1 : SUCCESS in 26m 19s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/d85671fd2f3f4121a19d00e4acd048e3 : POST_FAILURE in 26m 04s","accounts_in_message":[],"_revision_number":3},{"id":"20f95f6caa21e1b3b6203d1ba0779fa33ae9e78a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-11 08:58:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"be108368d5499bc3b29c222e3cc88aa266e84594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:25:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d34d3f02d41d411fad5e6431a6aba109 : SUCCESS in 10m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31b44c72039240b2bb6093bef503e340 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0488052a34b64eee842d7aa8e3216745 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c01bb60f3b842b5835a08b5953d6ea2 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c0a588df8e401eba1adc6514869b71 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10a8885b8aca4fad976664874c75ed13 : SUCCESS in 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/963c6a9bcf6b4698885b29e408bcf672 : SUCCESS in 12m 12s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/5c4acca21a084e82b5b5fd4f135e7dcd : FAILURE in 26m 18s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/3eff0f7d0e764ad28e919fbcf9369502 : TIMED_OUT in 3h 09m 50s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/0cdb6154a53e44f0ab9eab3bc2217d0a : TIMED_OUT in 3h 10m 23s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/f6e521d157914c5d9aff8d0953ae389c : POST_FAILURE in 49m 36s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/32fd8f18878e4de58faeb43fabeb971b : POST_FAILURE in 49m 13s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/c6128b2b71a5467698cfc593bc1dab47 : SUCCESS in 1h 23m 04s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/eadd062b319740b1a7989c69ae7aefbd : SUCCESS in 1h 17m 23s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/55f1aa171707471091a71c9ffac67148 : SUCCESS in 19m 01s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/528d747d2c024024b6db2909b32e8c70 : SUCCESS in 21m 26s","accounts_in_message":[],"_revision_number":4},{"id":"4d1e835168f6ce98fcfe5ad1ab52ea5457cc73fc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-07 09:48:53.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"29427c1b93bff70c36d7d03318bb180150f2c2cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 10:36:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7109a291008a6c089be73261b19a3678d4a8a63e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 11:40:16.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"290862875c592700ce047eca1172004cfb309746","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-30 15:09:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4c7a4723ef421897151c3dbaa546fdb52cf8887e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 18:46:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/611c20ed0e1f4880b95258aaf394788d : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb0e2b9827e4e38855b4b10269f15ce : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a890dc3975c4faabfb40384968ba671 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca0bca0f49ee4b42a4d9c1f0f4958572 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7a213e59f14e0bbaea94b2bdbed2fe : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2104d9b3ef3492db352d42c4c628938 : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47acb65024c640ad9e8f7dd83a13bc96 : SUCCESS in 9m 00s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/08decf2551f04ce58dd56926af59f549 : FAILURE in 28m 52s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/32f0a3a627ac4e389bd28f65a0462c1d : TIMED_OUT in 3h 08m 31s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/b1feceeb9c714bd2a834cc6a456f8492 : TIMED_OUT in 3h 09m 01s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/bdcd7f1819d243348a3334164c96509e : POST_FAILURE in 43m 01s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/81bd5973a2e84f30bd927d603de04ff9 : POST_FAILURE in 41m 51s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/d74c2894bc19410880e0b72f7bb50375 : FAILURE in 34m 40s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/f245247ec0bb4ab889c77b62d53f0c5c : FAILURE in 32m 36s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/1dae92d239f94141a07ab8855e88021f : SUCCESS in 19m 09s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/047fa9b684424a4f9d3b7734842b254d : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":5},{"id":"1baa0715f6483be382d7f68e8880c0a9312aa55a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-30 20:46:59.000000000","message":"Patch Set 5: Workflow+1\n\n+Ws were lost during rebases. Adding it again.","accounts_in_message":[],"_revision_number":5},{"id":"f5925620e93e3d3aa0916c91254a654a981a5f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 20:47:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c6c58c5d4e5a0779d23688271b3cede59ae4cc2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 21:10:14.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"7b7d9b248e505a2aed8d3e0c3345721e7183cd06","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-31 10:31:59.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5e63c5619d4623560ec86e74b914211aebbbf632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:49:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d9aa3e8afa42f4a1834c27937df646 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5f295d2212d4b45b4585ce09440f40d : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79af790b2d3043de8023cecc8c07e53c : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f321dd6c7eb5412dbfca1757f2886a10 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13707e2ad6b4573945d68b71cba4f70 : SUCCESS in 8m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16cec47f46b940419782f40c99c8d301 : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56e27d56a57c4c3282ca4f39f136840f : SUCCESS in 6m 36s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/e4802ce198d94cc2a2d6c7b398d8b746 : FAILURE in 19m 44s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/4f392fb8340e4caaaf94ba3ad72b0a30 : TIMED_OUT in 3h 07m 39s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/08fa2342ce0b4dc4a3e571ca430f3fde : TIMED_OUT in 3h 07m 33s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/db6f43261a9740978e4338869781fcef : POST_FAILURE in 51m 41s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/e5e4f2b3005f43ea9b0d420321ad28b3 : POST_FAILURE in 41m 14s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/54c64447dead45fe8fc6405fd88338c2 : FAILURE in 38m 26s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/ab4bd7cc74c847c19554c03423b2dfc3 : FAILURE in 32m 35s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/913c4dfd0a574df9bfbff7fadf92469c : SUCCESS in 18m 37s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5da67a6f9a7748c79e8d79eb34b7acbc : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":5},{"id":"7a546a2f5fdc2233dd76600af06f32ee59945696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:49:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7e0acbc9b718b60bd2539a58bb8993391f5e9022","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 14:17:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e863b03d6a54f2cbfd5e1be402ce050 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/792be6eeb6d24b47b810e9a563009f39 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c54ec13e6b64dcc8abbb4b147667ced : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2510d54bfc0457e8396fec900fe33fb : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93d7767768d4f69a491ac319fca8b1d : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89d8d40107e748fca11b34f17fcf9dc2 : SUCCESS in 10m 27s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/c079eb3e37294695a072745d12a03f55 : SUCCESS in 18m 26s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/740111c491f44a9d80fbfc5cadd9cc7f : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":5},{"id":"d116778d71af6c514886b5d349aabedec6798c9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 14:17:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3f5508936dc0139f147bf7672deb17cde25b446d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 14:19:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54f0c98125c04ce386d4eb52a397cf39 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04a96d3f4f247cdada838eb7d20be65 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"02a05f4f6a24ae891701f41c8a9394cdc152d3e0","revisions":{"e099fb2dfa3f584b580f5ded4e73b42b6dc71a7c":{"kind":"REWORK","_number":1,"created":"2020-04-23 09:49:49.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/64/722264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/64/722264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/1"}}},"commit":{"parents":[{"commit":"7088fa0c212c193cfb6dd1ac7e303867fb9baaba","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7088fa0c212c193cfb6dd1ac7e303867fb9baaba"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:48.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:48.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I28ccba1b07097dccabf664dc928f0893b410ad8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e099fb2dfa3f584b580f5ded4e73b42b6dc71a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e099fb2dfa3f584b580f5ded4e73b42b6dc71a7c"}]},"branch":"refs/heads/stable/ussuri"},"4ebb46893eadbc65ec0199f6ff9a3c810f23cc2f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-05 11:09:48.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/722264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/64/722264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/2"}}},"commit":{"parents":[{"commit":"d80982f479456411ff21e4d4df87464f3f56b763","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d80982f479456411ff21e4d4df87464f3f56b763"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:48.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-05 11:09:48.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I28ccba1b07097dccabf664dc928f0893b410ad8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4ebb46893eadbc65ec0199f6ff9a3c810f23cc2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4ebb46893eadbc65ec0199f6ff9a3c810f23cc2f"}]},"branch":"refs/heads/stable/ussuri"},"e07fe0d1600ba286d39282828bf90a624912cf63":{"kind":"NO_CHANGE","_number":3,"created":"2020-05-06 15:21:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/722264/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/64/722264/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/3"}}},"commit":{"parents":[{"commit":"830863f4c24c580e698d9e5e9a43d6063af80f00","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/830863f4c24c580e698d9e5e9a43d6063af80f00"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:48.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-06 15:21:29.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I28ccba1b07097dccabf664dc928f0893b410ad8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e07fe0d1600ba286d39282828bf90a624912cf63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e07fe0d1600ba286d39282828bf90a624912cf63"}]},"branch":"refs/heads/stable/ussuri"},"38a86cd0d97069c0c0e86ebb94422b992198a574":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-11 08:58:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/722264/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/64/722264/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/4"}}},"commit":{"parents":[{"commit":"94fe2017dadd0aef4c8d024065e52b1c117cbfd6","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/94fe2017dadd0aef4c8d024065e52b1c117cbfd6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:48.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-11 08:58:19.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I28ccba1b07097dccabf664dc928f0893b410ad8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/38a86cd0d97069c0c0e86ebb94422b992198a574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/38a86cd0d97069c0c0e86ebb94422b992198a574"}]},"branch":"refs/heads/stable/ussuri"},"02a05f4f6a24ae891701f41c8a9394cdc152d3e0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-30 15:09:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/722264/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/64/722264/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/64/722264/5"}}},"commit":{"parents":[{"commit":"18021af42857ba3137fc273e2377ac7ab2a15426","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/18021af42857ba3137fc273e2377ac7ab2a15426"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 09:49:48.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-07-30 15:09:10.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I28ccba1b07097dccabf664dc928f0893b410ad8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/02a05f4f6a24ae891701f41c8a9394cdc152d3e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/02a05f4f6a24ae891701f41c8a9394cdc152d3e0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
