)]}'
{"id":"openstack%2Fnetworking-odl~738074","triplet_id":"openstack%2Fnetworking-odl~master~I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5","project":"openstack/networking-odl","branch":"master","topic":"complete-retirement-networking-l2gw","hashtags":[],"change_id":"I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5","subject":"Use the networking-l2gw project new location","status":"MERGED","created":"2020-06-25 17:41:49.000000000","updated":"2020-09-11 08:25:50.000000000","submitted":"2020-08-31 13:17:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738074-1598879855249-9d651cf1","meta_rev_id":"2e69d0683f5d021847631f277b7fcf1d6b1aa1bc","_number":738074,"virtual_id_number":738074,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 13:17:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-08-31 11:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-08-28 06:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-08-28 18:49:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-08-31 11:32:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-08-28 06:19:04.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 18:43:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-28 06:18:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-08-28 06:19:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-08-28 18:49:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-31 11:32:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-08-31 13:17:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4755bfcead35d26e7b95237f5948e5bca9b0612d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 17:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daaa8edd3114cb5c8518d47644357d681a5ceba8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 17:45:45.000000000","message":"Patch Set 1: Workflow-1\n\nwe have code dependency also, discussing with neutron team.","accounts_in_message":[],"_revision_number":1},{"id":"1a85dc12b88fbaac1fb6edfc523470a375aad40d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 18:01:01.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, needs backporting to all stable branches","accounts_in_message":[],"_revision_number":1},{"id":"273711e77e845984a4ab50b1922a1ae95cdfc0c1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 18:43:14.000000000","message":"Patch Set 1: -Code-Review\n\nNeeds removing here as well, doesn\u0027t it?\n\nlower-constraints.txt:networking-l2gw\u003d\u003d12.0.0\nrequirements.txt:networking-l2gw\u003e\u003d12.0.0  # Apache-2.0\ntox.ini:    pip install -q -e \"git+https://opendev.org/openstack/networking-l2gw#egg\u003dnetworking_l2gw\"","accounts_in_message":[],"_revision_number":1},{"id":"59f9cebb517239d7807a6abc55e6042418e247ad","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 19:24:45.000000000","message":"Patch Set 1:\n\n\u003e Needs removing here as well, doesn\u0027t it?\n \u003e \n \u003e lower-constraints.txt:networking-l2gw\u003d\u003d12.0.0\n \u003e requirements.txt:networking-l2gw\u003e\u003d12.0.0  # Apache-2.0\n \u003e tox.ini:    pip install -q -e \"git+https://opendev.org/openstack/networking-l2gw#egg\u003dnetworking_l2gw\"\n\nyeah but before that we need to remove the code dependencies otherwise networking-old become uninstallable.","accounts_in_message":[],"_revision_number":1},{"id":"ae4091b1af68eee8b0071600eca7f28b8220867f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 21:00:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5577b0a2a442426ea5e72ab7b6876289 : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bac82454d2e74373933019e54d736855 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34ecfc9a81eb4cc194f85c30a4a40b04 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1922d664b94d25854ace076ea67a2c : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/996d384b4cb44e5e8d8ed45f89787f3c : SUCCESS in 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4895a3f90e1d4ebfaf697d23d0e53686 : SUCCESS in 9m 12s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/6e238db8607b4d8da2d45a8a848f2cb7 : FAILURE in 22m 26s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/8b134165584b40b8b24a5fdedd48a26e : TIMED_OUT in 3h 15m 24s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/06a1047b84e4450aaa08cbf4214087c8 : TIMED_OUT in 3h 07m 31s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/d6b8beef90864227ac7b867bb6a920fb : POST_FAILURE in 43m 11s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/78dcf090a67c45b5a3444c2e5699eba8 : POST_FAILURE in 44m 51s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/b463cdea3dc44f44a87a4e0e4afd8570 : SUCCESS in 1h 24m 46s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/8ba27b9d946140f4998d9b2df4562003 : SUCCESS in 1h 20m 45s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/1bcd8c990c3d4110abea3b57133be066 : SUCCESS in 16m 59s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/927fbc185047459c8d06ca14e7a974d0 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a729e73c2f33d193baf948fd62208e40905a262e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-27 22:35:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c5b56134d70923557d889ce6e061457556d30cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:35:25.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"94a9524e0b6e2877cc3b52468b4636f690ad6b8a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-27 22:47:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c55e08bb8eb06b7498bd72254f650c88621cde4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 02:01:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea6b3b844ed491788062f551debb639 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93c26c04eabf4fdcae2e9773dcb028ba : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebb00445b3024e4091fc7e1ebcf1893e : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30ced29fc7ab40c7a0454beb68025175 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3198e99b1d471bad6e3f449ce8000d : SUCCESS in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/656e94ed6a84430a9e55e247789aa267 : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fd02a422ac7493b9e54b58e52c6e168 : SUCCESS in 9m 02s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/fa14677ec97c4dce9fd715af8c89f062 : FAILURE in 23m 28s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/ac7ca8ee86f241e1a9b1897d842d74f8 : TIMED_OUT in 3h 07m 18s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/3dee406c4c9f431eb34ec7fefadd4513 : TIMED_OUT in 3h 06m 35s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/49e3d6022103438f87c3a91af52d6448 : POST_FAILURE in 44m 20s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/7731d9ddb8e84a45b0c6c7145a3c0540 : POST_FAILURE in 41m 40s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/59387ac1b45542468780728e14ac1846 : FAILURE in 34m 54s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/dddf5b40882747c7993af27fc5517d64 : FAILURE in 35m 16s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/6edce6e64d264c8dafa8e45e224f6657 : SUCCESS in 20m 37s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/d8213c7cc7d448f49dcd21012a931667 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":3},{"id":"01d179837eda2d3822c0e6101d156f1081c058eb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-28 06:18:51.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks gmann","accounts_in_message":[],"_revision_number":3},{"id":"14cfcc85367333725c19f15d8204d63860eb4a23","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-28 18:49:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2e35c7559f7f8dab6933ced8eff2b0cf2f71a05a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-31 11:32:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"867aedc5d8dd4699bcbefb0d4fd3ba0a873932ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 11:32:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7a35a9a31b29f4085cfec8d51caad6d4a499a2e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:02:29.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f43cd73a1e45ff8c235d98b558d410 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d6f04dd91654c04ad1c7d1a0bfd70fb : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecb72b626cf745cf96b55ecadef48f73 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70dffb2a11fd4e139051b0c1e4b2bfff : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb63cd9cff34a73a3356d246219a87a : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a70071a3a99b4c9fa28680a0afd28bd2 : SUCCESS in 8m 38s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/8b1f6ee1dcef4f219fbe2a88895ac78a : FAILURE in 27m 29s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/c80fe8f64dec4213bb851e4a72e5879d : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":3},{"id":"a5f1a92c65e9eb43df41918efc598122f1aad5e9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-31 12:22:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7ecc514509d1b61233ff98768618953ae89f07bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:52:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f99a91808864d488a21d078bf6ec1cf : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142fd4fead774c4480248fdedc466ba6 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60acfe5b917145d6a1cf69e9a8a481d6 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f04d35aab1b4c8eacdd6cae65439d8d : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79d16acacc8448480f71a2ae49a97bd : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a1785693f2d48a18e850ef167996d93 : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c8cfa9df4864fcc90ce3721b4377862 : SUCCESS in 9m 05s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/70099254a217498087a38665f57b666e : FAILURE in 17m 42s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/d80544dc3e0b43a7a14c37d4f028d6e2 : POST_FAILURE in 22m 19s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/df34354508f04c139ce108a2be2efbd0 : POST_FAILURE in 20m 00s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/fe51cffbcb4b4d7abe768e3be94b9d91 : POST_FAILURE in 23m 30s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/152b6ff89a7f447fafe844bae4885384 : POST_FAILURE in 18m 53s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/72c8e684ec0e4cb48a5971cf534aa046 : FAILURE in 14m 50s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/a168395e9b00469ab88add82f7c01db0 : FAILURE in 19m 13s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/957fc9f45a7e4cb5a9d8a4f24884ca5f : SUCCESS in 15m 51s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5e8835928f24473bbba9d17fa43ea80d : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b0e44c2a2c8a870cd712ab0cd575b245c3bc32d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:52:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c7b2c3790a3b46e50b229df5c40ea5a13c97e704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:17:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902c8f7d7dde40a8ab65da4a46ed4c90 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d6c36eacc84f498b5fdcbb64190949 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea5381ed562b4724a931dc5985d0b2e3 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4741c21a0fa54ca7ac8a4778ac77010f : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae16b77c68b34e038f15155331729066 : SUCCESS in 15m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf64960a3ed44348b268960d0d2b07ee : SUCCESS in 8m 04s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/b510bf49fe0247fc96c371ef95dd41bf : SUCCESS in 23m 00s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/cd41f5184a5b48de83d6be1f3fc1f2b9 : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d75be1de272ed920cc9d5aee3a9612ab29fcb0db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:17:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"32f22456445e1998830c2130fd21cb07c6b6e63a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:22:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b143521a5fe5460bbcf9981d288246ec : SUCCESS in 1m 30s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1ff9ba688c40e195aecdb8ada56036 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f48102932a294a712be32cc18a5d3547edb0d7db","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-11 08:25:24.000000000","message":"Patch Set 3:\n\nthis needs backporting to all stable branches!","accounts_in_message":[],"_revision_number":3},{"id":"584825c2b49115d97b59ab2bb001952b46ba4ff6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-11 08:25:50.000000000","message":"Patch Set 3:\n\nCould somebody do the backports, please?","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bcac237d81a56f9ad61df0256e0b706ee95e0a0e","revisions":{"5666b3070341da0e06776bf3ecbc104816846ed0":{"kind":"REWORK","_number":1,"created":"2020-06-25 17:41:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/738074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/74/738074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/1"}}},"commit":{"parents":[{"commit":"6253d8baa44c3293ee070f59c184dd7322d59947","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6253d8baa44c3293ee070f59c184dd7322d59947"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 17:38:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 17:39:42.000000000","tz":-300},"subject":"Remove the networking-l2gw from zuul jobs dependency","message":"Remove the networking-l2gw from zuul jobs dependency\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nSee also:\nhttps://review.opendev.org/#/c/738035/\nhttps://review.opendev.org/#/c/738038/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5666b3070341da0e06776bf3ecbc104816846ed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5666b3070341da0e06776bf3ecbc104816846ed0"}]},"branch":"refs/heads/master"},"883017a5d5b1bd31789f1554b19e688d62262d31":{"kind":"REWORK","_number":2,"created":"2020-08-27 22:35:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/738074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/74/738074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/2"}}},"commit":{"parents":[{"commit":"6253d8baa44c3293ee070f59c184dd7322d59947","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6253d8baa44c3293ee070f59c184dd7322d59947"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 17:38:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-27 22:35:10.000000000","tz":0},"subject":"Use the networking-l2gw project new location","message":"Use the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/883017a5d5b1bd31789f1554b19e688d62262d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/883017a5d5b1bd31789f1554b19e688d62262d31"}]},"branch":"refs/heads/master"},"bcac237d81a56f9ad61df0256e0b706ee95e0a0e":{"kind":"REWORK","_number":3,"created":"2020-08-27 22:47:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/74/738074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/74/738074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/74/738074/3"}}},"commit":{"parents":[{"commit":"0cc249bae09b721a7fa5a50a86069cd307cdcd01","subject":"Workaround mock._is_started","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/0cc249bae09b721a7fa5a50a86069cd307cdcd01"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-27 22:47:21.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-27 22:47:48.000000000","tz":-300},"subject":"Use the networking-l2gw project new location","message":"Use the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/bcac237d81a56f9ad61df0256e0b706ee95e0a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/bcac237d81a56f9ad61df0256e0b706ee95e0a0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
