)]}'
{"id":"openstack%2Fnetworking-odl~763210","triplet_id":"openstack%2Fnetworking-odl~master~I79729d1cf9e911d7b8ec063520ddcedb442b725b","project":"openstack/networking-odl","branch":"master","topic":"sodium_latest","hashtags":[],"change_id":"I79729d1cf9e911d7b8ec063520ddcedb442b725b","subject":"CI: change sodium from snapshot to latest","status":"MERGED","created":"2020-11-18 16:01:57.000000000","updated":"2020-11-24 16:06:02.000000000","submitted":"2020-11-24 15:53:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763210-sodium_latest","meta_rev_id":"99d92be6932e81961dad86502951813b7c43eaa0","_number":763210,"virtual_id_number":763210,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":25617,"name":"Slawek K","email":"s.kmiecik@samsung.com"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-24 15:53:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-11-24 15:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":25617,"name":"Slawek K","email":"s.kmiecik@samsung.com"},{"value":1,"date":"2020-11-19 02:03:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2020-11-24 15:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-11-23 14:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-11-24 15:15:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":25617,"name":"Slawek K","email":"s.kmiecik@samsung.com"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-11-19 05:08:16.000000000","_account_id":25617,"name":"Slawek K","email":"s.kmiecik@samsung.com"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25617,"name":"Slawek K","email":"s.kmiecik@samsung.com"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 17:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 02:03:29.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-11-19 05:08:16.000000000","updated_by":{"_account_id":25617,"name":"Slawek K","email":"s.kmiecik@samsung.com"},"reviewer":{"_account_id":25617,"name":"Slawek K","email":"s.kmiecik@samsung.com"},"state":"REVIEWER"},{"updated":"2020-11-23 13:27:56.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-23 14:21:32.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-24 15:09:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"8914b7a08f597e36ffcf902710e898e720244095","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-18 16:01:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d42e47ccbb311169268003723bee3868ba60641e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 17:40:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87c192400f54ebe904fa7cbe0c0d04e : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a86b0a6c02184b529cd77c82a16edac9 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70b4650b661749b7845afbe9a0257c46 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0fa4d718cc04bd3b54bcd18863f2404 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b506f3511df34b4981efb77cb4e8c260 : SUCCESS in 8m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fd9e8db28e2477aba69c3afcf2aaaf6 : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f6c4684917b4070ba6d687522a8ef13 : SUCCESS in 8m 31s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/9e1c9f330a764d759bb26500c8c17bb4 : FAILURE in 58m 48s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/19ab47d3a39b4c69b46b04897c438c5b : FAILURE in 1h 09m 26s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/dc47330f9d6e4e9b931bed929289674b : FAILURE in 57m 36s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/967bce9fe73c44218cad2149416e5107 : POST_FAILURE in 48m 11s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/9c6ad98902a1470f92b41fbeccb8aa03 : POST_FAILURE in 46m 32s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/5862fdd0c9c9427f8a1808a018cab8fc : SUCCESS in 1h 28m 49s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/044ff18352a84fd9985268409bb6c47c : SUCCESS in 1h 28m 44s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/5ea158c263dc4e9ab523e088f6368cad : SUCCESS in 21m 32s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/efe12718d0e94c18aa7a2c28e462df38 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"57a62ca2e5a374f03b0501d5571de864b9ace470","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-19 02:03:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6938a0e8355f82ba47435c05d464c27038db5519","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-23 14:21:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1491b4767bcce43848431e80ee3588f99d3483b6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-24 15:09:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e0274288b5936f1c5d3cb7feb48ae25c42d4b78f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-24 15:15:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38b86c80b3a86aadef25c2576ea8450ba22cb45f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:16:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd9f60d9c832a0586594f6b27a487bfef8c94e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:53:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd2e29746b64bbfa1ee69fa8a7cb5b0 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9dc370b96824704b0b47138b5d7d971 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b16e94d7ea26458db4b2a944301df574 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01376d0f783e4f5e91e24e099e7719cd : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a6a71d2fa3d47afa91d4710349c7119 : SUCCESS in 8m 41s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/0b1ffba9d3a04e81a462ea4ff42e8c2d : SUCCESS in 23m 37s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e78b54a37f4b4983b92dc6eebdde59fc : SUCCESS in 27m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9e24b2835cab685da4efabedf202803797a2374a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:53:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"99d92be6932e81961dad86502951813b7c43eaa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:06:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c658638ff334d2c8fa7df32b1f2d031 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bea2c6b55b9b2b9e50e7e9fe0dcf7f587b2c66f7","revisions":{"bea2c6b55b9b2b9e50e7e9fe0dcf7f587b2c66f7":{"kind":"REWORK","_number":1,"created":"2020-11-18 16:01:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/763210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/10/763210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/10/763210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/10/763210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/10/763210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/10/763210/1"}}},"commit":{"parents":[{"commit":"4d7848c277c97d44a064dcd9f114f874a80e542e","subject":"Merge \"Migrate legacy grenade job to be native Zuul v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4d7848c277c97d44a064dcd9f114f874a80e542e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-11-18 15:58:15.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-11-18 15:58:15.000000000","tz":60},"subject":"CI: change sodium from snapshot to latest","message":"CI: change sodium from snapshot to latest\n\nSodium release of ODL is not available anymore as snapshot, so change\nthe reference to latest in job definitions.\n\nChange-Id: I79729d1cf9e911d7b8ec063520ddcedb442b725b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/bea2c6b55b9b2b9e50e7e9fe0dcf7f587b2c66f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/bea2c6b55b9b2b9e50e7e9fe0dcf7f587b2c66f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
