)]}'
{"id":"openstack%2Fnetworking-odl~770191","triplet_id":"openstack%2Fnetworking-odl~stable%2Fussuri~I6911923fa68c8ea44cc3436901c9be8344ae21e3","project":"openstack/networking-odl","branch":"stable/ussuri","topic":"oslo_lc_drop","hashtags":[],"change_id":"I6911923fa68c8ea44cc3436901c9be8344ae21e3","subject":"Dropping lower constraints testing (stable Ussuri)","status":"MERGED","created":"2021-01-12 15:24:14.000000000","updated":"2021-02-19 15:28:05.000000000","submitted":"2021-02-19 15:22:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"770191-oslo_lc_drop","cherry_pick_of_change":770364,"cherry_pick_of_patch_set":1,"meta_rev_id":"d16fe1a83d1b69c4ecffdc0795aac6f4d372dc09","_number":770191,"virtual_id_number":770191,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 15:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-02-19 14:45:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-02-03 09:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-02-19 14:45:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 15:27:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-12 15:27:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-12 15:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 15:52:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-19 14:45:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"cb82ed591902a0234865ad09c09418f139a0a6bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:24:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* .zuul.d/project.yaml\n* lower-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"08b5989ac21040e2e79d4175c63201269b7b2ed1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:26:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ca13cd2307400c66997110846868437180738af5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:27:13.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"0f87e017a92b6340b45aa9dd9ee489a90f56610f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:28:04.000000000","message":"Topic oslo_lc_drop-stable/ussuri removed","accounts_in_message":[],"_revision_number":2},{"id":"c63ba5f9ba7d3fab85ce0e43e49f8108434f0bf2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:28:35.000000000","message":"Topic set to oslo_lc_drop","accounts_in_message":[],"_revision_number":2},{"id":"94f7f2eb79562bd2a3f955136ed1ee093b93cae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:31:31.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-odl on branch stable/ussuri.  The error was:\n\n  while scanning a simple key\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/ussuri\", line 32, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/ussuri\", line 33, column 42","accounts_in_message":[],"_revision_number":2},{"id":"57fe4f26db776002bd541c73fd9ece67a3f47b5c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-20 11:23:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f60558bf8c228c71f694ae21ae020ccc39368ba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-27 16:03:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed9cab9e22815b16198b56560e9ee5201fb6a0d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 16:04:53.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-odl on branch stable/ussuri.  The error was:\n\n  while scanning a simple key\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/ussuri\", line 32, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/ussuri\", line 33, column 42","accounts_in_message":[],"_revision_number":3},{"id":"22f85ffd24b91b240867ca8d8713cee2ebeb56cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-28 08:33:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"87dc40c33da134dd3e66168bb1320fb68430139c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-28 08:33:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"90799218ad778501792c9042fbe4500f70a05c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:42:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770b00e5979e462aa9e33227f40422c5 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f19415ac6cb6479996e1da09cdfee689 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b56d12620c5941a3b4af846aa726fb0b : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac96de936b5f47e48825db9d841a89fa : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493fdfe8fb534ac9bc814cbfd5366291 : SUCCESS in 9m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a6103827f7c4f108fbaa320dd3e1262 : FAILURE in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e7218a5dd21405ab31a449d72bc4d21 : SUCCESS in 7m 51s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/b024424001424d77a520e0a852d4f9fe : FAILURE in 20m 00s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/6ff967c71b7349d6a19656fd0555e57e : TIMED_OUT in 3h 07m 16s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/52e0d1af2d854dc2a711f322adea48bc : TIMED_OUT in 3h 07m 19s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/475bd1fc9c7f4f41a33923f995c7306c : POST_FAILURE in 47m 45s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/35e4ecf0457b483e972e36db79b11828 : POST_FAILURE in 38m 33s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/c7a6ed6bfcff4eb0a9a92ca5d88dd87c : FAILURE in 43m 11s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/23e1394485834ebcbcc07d8e68aae59b : FAILURE in 31m 25s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/5bba6af8f0334690adc2b3b676dca78b : SUCCESS in 21m 44s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/48ac0089a8214e89938b0617dce6811c : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":4},{"id":"5721d4fde1e5aa7d25d39607c540a538c776b385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-28 13:41:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"79370b631c269058921ade4486c1095c644fef17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:58:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3827235b80554ec59b3185d23f6b5fdf : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf0d522dc1e049e582ff56db34061c84 : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5331a38b73748f588ca26d0e5005b6b : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da76a9e97c0248e0917af1a83236373e : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78301f4d79464e52a6f518729975ff7a : SUCCESS in 10m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8efa7a7e98f4a65b3b596aa6e33d9e2 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81221c6d4d8a455dbbb7678d2ac91f5d : SUCCESS in 9m 35s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/51257c4d7bb14c50820790b3373c6dc1 : FAILURE in 16m 17s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/7ae502f9bd684c99b0082a343eec78f6 : TIMED_OUT in 3h 08m 08s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/02130ebcaa9445d5935920a12b9eddd7 : TIMED_OUT in 3h 08m 19s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/2074a602d4a946c089c03666dc7a0908 : POST_FAILURE in 42m 22s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/b726fa22036244d5b0376aaafed6337a : POST_FAILURE in 39m 25s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/036c7efbb5434ec0878fa918cc9a75da : FAILURE in 44m 24s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/fa78aa25ba4a4c129998bbac5c79be9e : FAILURE in 29m 36s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/42324802b73a4f869f0bdf854f35c96b : SUCCESS in 20m 44s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/92bcea2862dc486b9d54a8334a87bf9f : SUCCESS in 20m 06s","accounts_in_message":[],"_revision_number":5},{"id":"aca3dc9faa199b7caf5123a52661ad70444abea2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-03 09:33:20.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a8989ae5a3d906e147c3020a9e8e0ddbf55696a0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-19 14:45:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nSimilarly, like in the victoria version: looks OK + Zuul is happy with it. LGTM.","accounts_in_message":[],"_revision_number":5},{"id":"0b1258e125ff68c8e175ce3993e4b01efc553a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 14:46:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"241bbe91f0f6efe7020e8f646b1db3ed23a00e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:22:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d7cbe1abb14dfca34c470f22cb491e : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88f19fb26fb046f2b10f331ecfa5fdc8 : SUCCESS in 7m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70c60292eafc4c0489d16c7ef26f0e79 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86ee1cbc31c547aa9aa5214968b1f1d7 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6617eb70c0694e8b9dbf7cb75a2f77f8 : SUCCESS in 12m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2aa0e79115a5436ab3236d993cf90330 : SUCCESS in 3m 17s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/12366ef1369b454b8a2906989e0947fb : SUCCESS in 22m 03s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/308064c0b8304877a153b3faad308143 : SUCCESS in 22m 45s","accounts_in_message":[],"_revision_number":5},{"id":"39740e4255d1e686b0b060e21d176ed6c14098bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:22:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d16fe1a83d1b69c4ecffdc0795aac6f4d372dc09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:28:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5166ae90330a42db8e700fb37d8afc03 : SUCCESS in 1m 01s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b344e2f260fb4252860ab5e0bd8a771b : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"07d98314db6d6b513900fbb1b0536f221145244c","revisions":{"6c1f0ba338cbc64426111df70bf4f047f603a7a5":{"kind":"REWORK","_number":1,"created":"2021-01-12 15:24:14.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/1"}}},"commit":{"parents":[{"commit":"4d10d1b3f6932ac2d34edc7240180a6976a54112","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4d10d1b3f6932ac2d34edc7240180a6976a54112"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:16:18.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-01-12 15:24:14.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Victoria)","message":"Dropping lower constraints testing (stable Victoria)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I6911923fa68c8ea44cc3436901c9be8344ae21e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6c1f0ba338cbc64426111df70bf4f047f603a7a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6c1f0ba338cbc64426111df70bf4f047f603a7a5"}]},"branch":"refs/heads/stable/ussuri"},"28a08f424850867c90559b14eb65af453627d0ab":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-12 15:26:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/2"}}},"commit":{"parents":[{"commit":"4d10d1b3f6932ac2d34edc7240180a6976a54112","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4d10d1b3f6932ac2d34edc7240180a6976a54112"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:16:18.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-01-12 15:25:22.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Ussuri)","message":"Dropping lower constraints testing (stable Ussuri)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I6911923fa68c8ea44cc3436901c9be8344ae21e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/28a08f424850867c90559b14eb65af453627d0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/28a08f424850867c90559b14eb65af453627d0ab"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"},"8555b31d41e564314c0f73b50730820700ce8c88":{"kind":"REWORK","_number":3,"created":"2021-01-27 16:03:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/3"}}},"commit":{"parents":[{"commit":"4d10d1b3f6932ac2d34edc7240180a6976a54112","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4d10d1b3f6932ac2d34edc7240180a6976a54112"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:16:18.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-27 16:02:39.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Ussuri)","message":"Dropping lower constraints testing (stable Ussuri)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nChange some test requirements up-to-date.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I6911923fa68c8ea44cc3436901c9be8344ae21e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8555b31d41e564314c0f73b50730820700ce8c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8555b31d41e564314c0f73b50730820700ce8c88"}]},"branch":"refs/heads/stable/ussuri"},"3240a36f48022a40ae84ccd98f4fb5a1583ad16d":{"kind":"REWORK","_number":4,"created":"2021-01-28 08:33:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/4"}}},"commit":{"parents":[{"commit":"4d10d1b3f6932ac2d34edc7240180a6976a54112","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4d10d1b3f6932ac2d34edc7240180a6976a54112"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:16:18.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-28 08:31:41.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Ussuri)","message":"Dropping lower constraints testing (stable Ussuri)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nChange some test requirements up-to-date.\nUse sodium-latest instead of snapshot, as snapshot is no more available\non maven.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I6911923fa68c8ea44cc3436901c9be8344ae21e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3240a36f48022a40ae84ccd98f4fb5a1583ad16d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3240a36f48022a40ae84ccd98f4fb5a1583ad16d"}]},"branch":"refs/heads/stable/ussuri"},"07d98314db6d6b513900fbb1b0536f221145244c":{"kind":"REWORK","_number":5,"created":"2021-01-28 13:41:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770191/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770191/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770191/5"}}},"commit":{"parents":[{"commit":"4d10d1b3f6932ac2d34edc7240180a6976a54112","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/4d10d1b3f6932ac2d34edc7240180a6976a54112"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:16:18.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-28 13:40:44.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Ussuri)","message":"Dropping lower constraints testing (stable Ussuri)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nChange some test requirements up-to-date.\nUse sodium-latest instead of snapshot, as snapshot is no more available\non maven.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I6911923fa68c8ea44cc3436901c9be8344ae21e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/07d98314db6d6b513900fbb1b0536f221145244c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/07d98314db6d6b513900fbb1b0536f221145244c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
