)]}'
{"id":"openstack%2Fnetworking-odl~770255","triplet_id":"openstack%2Fnetworking-odl~master~I3c797d6d246d0fa6105c4a2b01d2b1eab5c33e4d","project":"openstack/networking-odl","branch":"master","hashtags":[],"change_id":"I3c797d6d246d0fa6105c4a2b01d2b1eab5c33e4d","subject":"remove unicode from code","status":"ABANDONED","created":"2021-01-12 09:43:59.000000000","updated":"2021-02-19 05:48:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56ac9368127eb2800ea2517754167ba331e0a322","_number":770255,"virtual_id_number":770255,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-12 11:01:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 11:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 13:11:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"2e0afa8ddf8f3b6444da0f30d5e877a7505f6033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-01-12 09:43:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5881389440d4c03aaf43a67c0d632c3e2fc81574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:01:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70877a9b0bb45c1b909cadba5dad70d : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c2525877d84471dbd4a0079630298e1 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/051008a4db7348fb936e861161236c97 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5faabbca13e460aac33132afdad19c3 : SUCCESS in 6m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a773fc527e9543c384a5fa2409b94b3d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ac322d935b44e0ae8dcd065ca53290 : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6af7b1ac93844d1986adf7e4d127946f : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18429feb205e4ea7ada941e673293190 : FAILURE in 4m 20s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/a88a10cc54654851a736729630268694 : FAILURE in 49m 26s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/17c55c6fb46645c3ac8bb6721d131afb : FAILURE in 1h 05m 02s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/e51fd857de264537bb0611646dabea8d : FAILURE in 55m 47s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/73081f44b0a2429baf2f886d6dd37a9e : POST_FAILURE in 50m 28s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/1b0ad23c2eb84c3594fc9bb8793c3e84 : POST_FAILURE in 46m 31s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/3dc798d14cb843c4b9ef021f5abd69c1 : SUCCESS in 1h 14m 13s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/2dd62aea08ae4217a7d0d2e345313475 : SUCCESS in 1h 12m 11s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/d0cedbf18a4f4a0fb830220870b9e732 : FAILURE in 20m 54s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/ad3cac15559845e6aad5aded8f3f3605 : FAILURE in 16m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5114400eed7c1755c74bd430b31dafdc7045da33","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-01-12 12:02:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f2b863151a9cd69f05e6f14d4ef6500c06f5899","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 13:11:49.000000000","message":"Patch Set 1:\n\nThere\u0027s a patch to fix the gate: https://review.opendev.org/c/openstack/networking-odl/+/769877","accounts_in_message":[],"_revision_number":1},{"id":"5ab59637e2f00fb0c54885a96bbecca873869247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 13:35:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae44b035d0464279ab294d14d1a04c69 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f1430a5836145f9a0e7859d5c59fca9 : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a244fc613e54c76a14163ef3e049232 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0030f1af8a354cda891d212b9f71700e : SUCCESS in 5m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c5220d05cbc4142a9bbf092fe2bfc2a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b66b1c39b3f4f73a4e410128a2fbcb0 : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/653c689267234640b46bff2a1ac7956c : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a84ccbb52634fc1951655f894b78558 : FAILURE in 2m 56s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/f74b365178344d9f915221b6be44eae1 : FAILURE in 41m 30s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/da98b70b80784059be7572be04f02a47 : FAILURE in 57m 27s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/cc4aac5865694e48b7b98587fdc066ec : FAILURE in 56m 20s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/c3a4a5babeb842689fda61056bd281ed : POST_FAILURE in 37m 20s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/9c6fe920c0714bbe9e568fb080e6aad6 : POST_FAILURE in 38m 16s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/80b1f92507574db9ab865c8552aacfd2 : SUCCESS in 1h 06m 41s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/9712e107b55944089ac8edaf5359abbd : SUCCESS in 1h 07m 39s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/f6c931b0dbed430c9f850979764fd7d5 : FAILURE in 12m 08s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/4186ff9b62594f2e81175b072c855334 : FAILURE in 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"56ac9368127eb2800ea2517754167ba331e0a322","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-19 05:48:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2951b646e64d44b03d79fa1c2385d4e6eaca9db6","revisions":{"2951b646e64d44b03d79fa1c2385d4e6eaca9db6":{"kind":"REWORK","_number":1,"created":"2021-01-12 09:43:59.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/55/770255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/55/770255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/55/770255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/55/770255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/55/770255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/55/770255/1"}}},"commit":{"parents":[{"commit":"7e65409f774102a372fc08f7c4b197eab867dfd3","subject":"Update TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7e65409f774102a372fc08f7c4b197eab867dfd3"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-12 09:43:37.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-12 09:43:37.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I3c797d6d246d0fa6105c4a2b01d2b1eab5c33e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2951b646e64d44b03d79fa1c2385d4e6eaca9db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2951b646e64d44b03d79fa1c2385d4e6eaca9db6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
