)]}'
{"id":"openstack%2Fnetworking-odl~770385","triplet_id":"openstack%2Fnetworking-odl~stable%2Ftrain~Idcd1e39b4d60d33c2396f9495b66b089341748ff","project":"openstack/networking-odl","branch":"stable/train","topic":"oslo_lc_drop","hashtags":[],"change_id":"Idcd1e39b4d60d33c2396f9495b66b089341748ff","subject":"Dropping lower constraints testing (stable Train)","status":"MERGED","created":"2021-01-12 15:32:59.000000000","updated":"2021-03-05 15:27:23.000000000","submitted":"2021-03-05 15:16:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"770385-oslo_lc_drop","meta_rev_id":"b696a3ba5ca70960c3289d9a6c91943209a6d79b","_number":770385,"virtual_id_number":770385,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 15:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-03-05 14:24:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-05 14:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-03-05 14:24:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 15:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 15:53:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-20 11:24:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-19 15:30:42.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-03-05 07:35:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"fc142960783ef0dc23aeddd3ead16e325c3f5563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:32:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"934b59ca61f6f9a2bd113b16c6fe52a3cdbc4a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:34:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-odl on branch stable/train.  The error was:\n\n  Unknown projects: openstack/networking-l2gw\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python3-train-jobs-neutron\n        - release-notes-jobs-python3\n        - periodic-stable-jobs-neutron\n        - publish-openstack-docs-pti\n        - check-requirements\n      check:\n        jobs:\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - networking-odl-grenade:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-oxygen\n          - networking-odl-tempest-fluorine\n          - networking-odl-tempest-oxygen-multinode:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-fluorine-multinode:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-rally-neon:\n              voting: false\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-rally-sodium:\n              voting: false\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-functional-neon\n          - networking-odl-functional-sodium\n      gate:\n        jobs:\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - networking-odl-functional-neon\n          - networking-odl-functional-sodium\n\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/train\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"6afa21f12f0660062731ecc0b814a0cf35471d48","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-20 11:24:22.000000000","message":"Patch Set 1: Code-Review-1\n\nYou will need https://review.opendev.org/q/I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5 to make that working. Maybe You can rebase it on top of that other patch","accounts_in_message":[],"_revision_number":1},{"id":"2e794d351e7521bb961f3515b1406db1176fc679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 12:14:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a3624a35ef322453e4497f88d8448cc7d57ea25","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 12:14:52.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e You will need https://review.opendev.org/q/I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5 to make that working. Maybe You can rebase it on top of that other patch\n\nFinally I came back again to these old things....","accounts_in_message":[],"_revision_number":2},{"id":"1a7d91bf84e4a900e76c74118ecfa9fcc00292ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 13:16:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c982f588441b4159abe8e057c568ac01 : FAILURE in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c528276218074c32a620524b1198fd45 : FAILURE in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c62f6670e7504e948d9229e91ce0a169 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47a2ae3b4037446c9874c343fcc35eef : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65a4a1cd49044e59ae10ab27470c8942 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb6b8ea3463b456090713168f2e5a415 : SUCCESS in 9m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e0c66886f321491bb62a7041591da5f9 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32e98dd3a7df4cefb3967eab139ea53d : SUCCESS in 7m 57s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/f427e845543c4e6ab6e52cbe6c32fd78 : FAILURE in 23m 31s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/cc0ba53b082a47fb988737a9f4db3d99 : RETRY_LIMIT in 18m 55s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/49ed4d731bd045be8e9d6984da177f60 : RETRY_LIMIT in 10m 27s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/1e35b42039e54409a02a35a75f006e8c : FAILURE in 23m 03s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/e77af915b0ff4e44b57e49e2aff4d90a : FAILURE in 24m 25s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/6e80d36f7e024682956e953a80336ebc : FAILURE in 14m 13s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/da956dccaa504dc0b84dbce6495d225e : FAILURE in 21m 12s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/db42f246bec54df2abe7770050066639 : RETRY_LIMIT in 13m 13s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/76ad9fcbb0e54a1599264979899a1ed8 : RETRY_LIMIT in 13m 28s","accounts_in_message":[],"_revision_number":2},{"id":"183298b397d29f1e9c55e3a4bea171d29e59746f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-19 15:30:42.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7e26e05adcbd5c429166edbca5894e94aef97965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 15:52:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fefed8d0256364b5ae9fb7797fa7437973fcf506","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 15:53:33.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThanks","accounts_in_message":[],"_revision_number":3},{"id":"b896865b824d57b1a9dbc7467956b04238a6aeb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:05:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30103b32576d4221a833daa9a3ead667 : SUCCESS in 8m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0bb572744cd4bbe93cfd8324b3e4d6c : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13e95ba0c6fc46e1a5b439a379e7aea0 : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f434750d968f40a68d1d654cf3e31537 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/448545f6632d4376be32e3f6815569a5 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ae870d4a6546a8a38974c224df5c5b : SUCCESS in 9m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93fd63c7a8d347e48209567c3a7b107a : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a068cc6584c744c295886c77c85525fb : FAILURE in 8m 05s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/de8630a8c77a456d959df3704d6531a1 : FAILURE in 20m 47s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/5e02c2983e694cccad1e22c1402f9e69 : RETRY_LIMIT in 11m 11s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/3e91c866f8474a92af43c49dfc2faef8 : RETRY_LIMIT in 12m 06s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/52e079ca1d994d19a5051745a4577818 : FAILURE in 27m 07s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/016b915f722c464da03f75f4cd6013cc : FAILURE in 24m 32s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/99974d0e70f046a69573e39c369bc50d : FAILURE in 21m 37s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/fd805524914447718bfd11f87f116c4f : FAILURE in 15m 49s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/ef81926aaa304bf182421bd9163394bd : RETRY_LIMIT in 15m 49s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/c2665c47bd844ab485a40d45dc8049b4 : RETRY_LIMIT in 17m 56s","accounts_in_message":[],"_revision_number":3},{"id":"10adca198f1ec31fc566b22f542bd4d3b9a3e1ee","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-22 08:23:08.000000000","message":"Patch Set 3:\n\nrecheck\n\nNo logs for functional","accounts_in_message":[],"_revision_number":3},{"id":"6d7de14cc602b5c447a6cc64afcb7f97ef3c2b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:21:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0fe361c1ef24518b37e63109356d7f5 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f73b0eb97e534f35b356c0b4cd59cc25 : SUCCESS in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8194202411e84c939f113ff97153a400 : SUCCESS in 9m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60ca593951dd4d3f9ca2de2d31d0890f : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce8d382a915846ddb6c9753ba4405681 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4868e589e94da6958aaec0bd176abd : SUCCESS in 15m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/44591cf427a0436d9d0530138531fc0b : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f8908ed46f145179ad03f0aeaa24597 : SUCCESS in 9m 52s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/afc73a11e7e34c8c992f89bbe60cc17f : FAILURE in 18m 02s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/23f24e205c6f48ffabd780b88d0462aa : RETRY_LIMIT in 8m 44s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/610348a76a8b426a85601c6a1547d5af : RETRY_LIMIT in 8m 54s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/36862c71fd124832b7b60187e04a3fe3 : FAILURE in 25m 34s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/22d83fe274304bb2bb388fa3a13b08f2 : FAILURE in 22m 45s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/becf33881f2e432f839a54afd009ad6e : FAILURE in 17m 25s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/1e24cfcdb8ed4634997126caa5c301a6 : FAILURE in 18m 53s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/e0c44f484e31488ab10c531b593b470a : RETRY_LIMIT in 11m 44s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/1fa6e25230284d2eaa4bf9c28e779cd7 : RETRY_LIMIT in 19m 48s","accounts_in_message":[],"_revision_number":3},{"id":"d510208f6eecd6fcbecaf5cccb34640c70176f4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-22 09:59:05.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8594877c79a0c068427299d0a69af4fa9077f7f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 11:13:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2400cf018da24700ae261ff00455eecd : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ddbb6a283038430e8f7cda925289fb1b : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/907ffebccc7740f086b2f91787571db6 : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b9dde179fa94fe0bf57bd1c65cb3f5c : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d96ea64100c4a5f8d2ed27f43eb2a77 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eade9755c51c4ebe9cc48251945f41cf : SUCCESS in 10m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2946a8d4380b49bfa1aa58e218455476 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/984694994a0f4284a0674a7e01f38a0f : SUCCESS in 8m 59s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/92d8926ae0104b44865c1d8dfcec81c8 : FAILURE in 23m 43s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/4ce0ae0cbd3e4256a0f968463e5f133b : RETRY_LIMIT in 6m 20s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/7ff420d294f14cedb3b74da70662d734 : RETRY_LIMIT in 10m 12s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/00cc58a448b343c4928183cd0e3ffa6d : FAILURE in 21m 07s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/16fad4ebfe104c7f97810585daf8d0aa : FAILURE in 18m 00s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/ca74752b876249f58495ac14fa379daf : FAILURE in 14m 01s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/614fb32714574a5c98f73dd8ab36aea1 : FAILURE in 17m 56s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/363dd979731e4d978f77d1d3526abede : RETRY_LIMIT in 13m 17s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5ef9a64b1a8a4bcfabb18f39b00c9ac9 : RETRY_LIMIT in 13m 29s","accounts_in_message":[],"_revision_number":4},{"id":"950203051ada3c7b20c0286128184626811c349d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-22 14:30:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"860969bc8ce9d44224acd6d4215abe3f1a741719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:30:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac7aa9fc43d6439da63c58337fc3bc79 : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/153482ba1a634ad780421e8cf8eb5dbb : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fd05b48fb284f558ab2295e7ced0357 : SUCCESS in 10m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a51a682f25d641d59327c4349884c7cf : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bb4506e48a840efa2001d5a4d7d419d : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4da86a968b94e889317c9745b35da9d : SUCCESS in 10m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/282ff11a5cad4e3785201e79ae32bdec : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9779fdf761c4a08b1a001e7e55533ee : SUCCESS in 12m 55s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/0e3769f8b4a345538a6ec5faea1970f2 : FAILURE in 20m 42s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/f6a02de0040c4aa2966d06f04076786f : RETRY_LIMIT in 6m 11s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/80ed52d3f9c642d3b8944a88f77ec71e : RETRY_LIMIT in 10m 50s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/284186765ae54c25bd22520ec764b38f : FAILURE in 22m 08s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/e7afe4ba6b554a429b4b6c4c4137f1c7 : FAILURE in 24m 45s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/a27a27bd606f4511b29a3dbf64d761db : FAILURE in 18m 56s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/259893fba6094c88a845d8484b5ddf57 : FAILURE in 19m 29s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/fe8b7debade34a63ab9940b7111c71cc : RETRY_LIMIT in 8m 18s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/336259905fed4fb8ae4ff228b5aaa316 : RETRY_LIMIT in 12m 11s","accounts_in_message":[],"_revision_number":5},{"id":"5b67d3e7867d868fb9b67977d6e9b3ac37d595af","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-23 14:10:23.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nSo if I understand it correctly, we have 2 ways to solve the functional job issue:\n- either use python3 in functional job (in this case the best is to see what change was merged in ussuri and apply the necessary changes here)\n- or enforce py2 pip install (see my comment in pre.yaml)\n\nboth ways should work I think. maybe I\u0027d start with the \u0027extra step in pre.yaml\u0027 option first. :)","accounts_in_message":[],"_revision_number":5},{"id":"53b039e7bfc220d61c7195e80716885d64c4f6ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-23 17:20:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ee702c454319bfa7d6659d57fea352f722c21464","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-23 17:24:36.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nThanks, let\u0027s see","accounts_in_message":[],"_revision_number":6},{"id":"24a9ccc83b773b5cf16abf2a3581a8958b368870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:21:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baca3d9e34a744f7a3848368b1814e36 : SUCCESS in 9m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b536f0bdcc342a2a7186f6d9c0bc7c7 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7400ca2d28284aeeb09263c75482c6ea : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e2338fd77f34ca894a42ed34e9be2f1 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d312c1558c64614a5cf575e55ecfda9 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8962f8c553d4bef9b0e349544a3ea12 : SUCCESS in 10m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f1d01e4ec7048ae87422a4ec4e633d3 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb7c0d6277fb453ea56ceb7a1e6a387d : SUCCESS in 8m 02s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/7b58370e9b4141e5bdb7c18e4110868d : FAILURE in 14m 40s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/bca7dfb6516342eaadbb1219d2f4a9d2 : RETRY_LIMIT in 10m 26s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/27cc9adc7cdf4ca0a5cfd3a199faca22 : RETRY_LIMIT in 10m 20s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/d1341ae236a84ead88b5793aa5e1853e : FAILURE in 19m 03s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/dc5193787164477fad0f411d1791426b : FAILURE in 17m 51s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/56eaa91bd99d49bd93cab2e027289d8b : FAILURE in 14m 04s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/fc43ec63c5c943b28fc52b0ffb7b8ff1 : FAILURE in 17m 20s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/435b1a18a5bf4b08a4e2e5e01b9d7374 : RETRY_LIMIT in 13m 05s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/f56c738d6439486d8180a774fca9276a : RETRY_LIMIT in 12m 15s","accounts_in_message":[],"_revision_number":6},{"id":"1f9545f381b56efb78dc2fac82d13d6b5c9f4a3e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-23 20:03:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"36442c974595fe720c9bcf000767a0d65478b479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-24 09:16:57.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"651c0c8d1742b2d991ba4d4c258748c40090f4ce","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-24 09:18:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cff2bf327e474eb42845ce0ecc68677a54a52798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 10:16:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99b1296dac6408985053b1d081f7426 : SUCCESS in 11m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/854cd907dcba44019b091fb7dadc7215 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95f7a3524b554cb0895d5c2a40597cff : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/821cac53db514097834e5359969324e7 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c95835a524de4c748838d0afecce7c6d : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85b7e2ae02847bb882fc363bd497224 : SUCCESS in 10m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90515949bad441c5ae2a4973b5e66db7 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa53269c63bc4301bb0d3890b7c9c5f9 : SUCCESS in 11m 13s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/4ae27c28515d4aabb153044539bec3dc : FAILURE in 20m 58s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/dd0d9ef9b9744732920b4edc13f4be2d : RETRY_LIMIT in 10m 30s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/ea9e0e59cba148409132e6b4cc37462f : RETRY_LIMIT in 7m 09s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/bd61ea1c4a034ff98e0d7c9705d0be23 : FAILURE in 26m 44s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/8de7116fc3664817b64fc1b70fd42b09 : FAILURE in 28m 14s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/105b771ce12d432e9f5d0c91badc5ebc : FAILURE in 18m 59s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/34278ea55fb34fe9b80c7bd90924261e : FAILURE in 16m 56s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/f5ee46a16dd5419fb5b52afbf3bb4d73 : RETRY_LIMIT in 19m 25s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/c2fdd57962854027b08dbb6f620c5f3e : RETRY_LIMIT in 15m 23s","accounts_in_message":[],"_revision_number":7},{"id":"b2bc0ef7863dedeec96200f1db262e2eaed7fa22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-01 10:54:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ab8b10bafcb6257a1862d9e81be96abce11e1340","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:56:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e889f46c35f94c6180fc5732439ed6b8 : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b33f99f18694d8ba5ef98620f118133 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8d4e88b310145d89bf86ab633f34553 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afd9d985fb85435190c479061ceeacf8 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e844cf7939574b088ca7d27669ac55ef : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aafea196d0364ad49aab0cd94331166c : SUCCESS in 10m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51468f2f4ea14349bddb030cdfe39b71 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f8e4785c56f4757b94f4d8972dc51a3 : SUCCESS in 8m 03s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/f5fe2b568cc544e1ac53c9b7a4fc25a0 : FAILURE in 20m 59s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/159037785eca48178fc46c0f8d0ef202 : RETRY_LIMIT in 12m 46s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/77bea8cecdc942cbadffdbdaf0c90809 : RETRY_LIMIT in 11m 10s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/a1710de209c04f09a40a0001d4d66e6f : FAILURE in 20m 52s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/d7c509a6c6754b9083911aee87db90c3 : FAILURE in 25m 09s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/5bec508e18234d3cb93329aa4a5b94b0 : FAILURE in 20m 25s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/d4f861dc53ae4e068867ef857348176f : FAILURE in 20m 14s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/302200a5bf3c40389c0a51a07fce877f : RETRY_LIMIT in 13m 40s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/1fcc2e18a44941939374020f37039015 : RETRY_LIMIT in 16m 38s","accounts_in_message":[],"_revision_number":8},{"id":"4df2eb77eb311273c030dd03139b92780e893873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-01 12:05:39.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"05344252e1a4b251bb8f419df1dc0bee16472bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:54:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd941982905a44b3b18b5a475a58b31e : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d634876b01d4ae1b1f9996689e38981 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6516f673cb9b49468461e1c07e1bb630 : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94199f7b9c52406fb066adadb77b17ca : FAILURE in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f0c50d756404538be54893c3d134e4d : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec84dfe8cba941a99e0c935568361fc9 : SUCCESS in 10m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93165841f2e445e9b67bf67da2427ee1 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53da4c2009314b50948f9db571b1602e : SUCCESS in 8m 17s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/5f298ae752f1489ea12a5abfcd634750 : FAILURE in 27m 37s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/f83cf61f797b44b3abcb5eca45fb0941 : RETRY_LIMIT in 13m 10s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/bdbfbe2bc36c4da69e4f42d10c9aa999 : RETRY_LIMIT in 7m 08s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/ecaaad36cc93431d9cb2ff5f2d0c2378 : FAILURE in 22m 36s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/9417041a359d4775a25eb196e7f2988f : FAILURE in 27m 48s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/2a875f14a3434ec2a07d1dd59543f19f : FAILURE in 17m 08s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/aed0fb11314342d6bc42a6b0fe02ea86 : FAILURE in 17m 03s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/f8fd60d5ef044164b10f579f1c2c6f29 : RETRY_LIMIT in 12m 10s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5845b21359fa472496d015a4759157c2 : RETRY_LIMIT in 10m 50s","accounts_in_message":[],"_revision_number":9},{"id":"140bc1183acb40a67d6cda2affaedb26502bb12e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-04 11:04:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"02d87168a98f780fd24d9a3ffb2f1b137f23e84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 11:04:27.000000000","message":"Patch Set 10: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-odl on branch stable/train.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/train\", line 88, column 19","accounts_in_message":[],"_revision_number":10},{"id":"dddb5a20d6f3519a4954d6c95534f34decd6e45b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-04 11:10:19.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"74bf2ac449350aa99cf86de9b9d3035fa733c344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 12:12:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cbb9734b044232a2a9ba760b902627 : SUCCESS in 7m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe7daf575b934b2ebb5c5f0a692470d2 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345573f6e4ed40bd9355d4de9663ac76 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/120b56d3c19f4b0da2deb512bb86f92c : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92401904806048d5ac25b71b563a1310 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffdd5897d3c452ba82dc97d4c0d381c : SUCCESS in 16m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd6c8d0fd3ed4909bc99d1f94aeb5cd7 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2a77be567ef445b8a130c9d03568653 : SUCCESS in 9m 55s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/b9cb63e42d0344539fdaa240c4bbc7ed : FAILURE in 16m 45s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/ea0946ea1de646749fef91482e1dd611 : RETRY_LIMIT in 7m 11s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/31dd4abefaaa45b991be58ae5c8d27a9 : RETRY_LIMIT in 8m 19s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/2d78c4c92f094b70a8f87288bc36d8ba : FAILURE in 17m 19s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/a37e856ca06f41bba72b37037b1ec6a3 : FAILURE in 24m 10s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/34556bd79e1b43879be6f06dc566b122 : FAILURE in 16m 37s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/2b97a75a799c463f979af694a2109f0c : FAILURE in 26m 18s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/f013f6bf754845cf89871ec561bebf62 : RETRY_LIMIT in 8m 56s (non-voting)\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e47c88304efc4ca0ab7b8f76cb9018b6 : RETRY_LIMIT in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"343480dc86bdf9a562764a16bf8dabb0a711ea7e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 14:15:43.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"45501f75db8d5604804ed2e19f0512e51bd803aa","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-05 14:24:16.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nSo is backport in https://review.opendev.org/c/openstack/networking-odl/+/745806 still useful after that? The isort part is fixed here, but there was also \"Backport of subnet deleted handler\"\n\nAnyway this one is good to go for gate fixing definitely!","accounts_in_message":[],"_revision_number":11},{"id":"fb9e3ae66ada9a4565532004298c2bbd83629d1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:24:44.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"084c11e4aef7105b6923003929b59357c688454f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 15:16:29.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/477642053f3b463e97861daef2b5d2b8 : SUCCESS in 12m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c44c1b5f43a42269365685994565df4 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0444a420f7bf49edb63093ee20021455 : SUCCESS in 12m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f69bbc72482a4090a2ab24abffb90c3d : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/103d5c73db26433a8a3643bb6568aef8 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a904c9db1e14f71b7833783a29649d9 : SUCCESS in 13m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b87516d85aec44ffbcd799c7267352f9 : SUCCESS in 4m 13s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/86770f9392b045bc81198b6968a032cd : RETRY_LIMIT in 11m 06s (non-voting)\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/796824d369d44ae0be4b94ddaba2e7e8 : RETRY_LIMIT in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8d80d59f24680d04f65b0a610ab848186d84dad0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 15:16:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"b696a3ba5ca70960c3289d9a6c91943209a6d79b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 15:27:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e421eb0a557f4d78a0965afb7320c2f3 : SUCCESS in 1m 23s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ead089e25ad45d8877ca434f81e2e9c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"20f118035802ec38aa407132707819c15e9c300a","revisions":{"b24771fb15a61cfb1e7bb5cfac6c4ff25194a5f2":{"kind":"REWORK","_number":1,"created":"2021-01-12 15:32:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/1"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b24771fb15a61cfb1e7bb5cfac6c4ff25194a5f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b24771fb15a61cfb1e7bb5cfac6c4ff25194a5f2"}]},"branch":"refs/heads/stable/train"},"91fb906821db736732862254361add1da79268f6":{"kind":"REWORK","_number":2,"created":"2021-02-19 12:14:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/2"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-19 12:08:30.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/91fb906821db736732862254361add1da79268f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/91fb906821db736732862254361add1da79268f6"}]},"branch":"refs/heads/stable/train"},"9dd7628ee318ed187ff7c603458e2feafe82466c":{"kind":"REWORK","_number":3,"created":"2021-02-19 15:52:16.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/3"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-19 15:50:37.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/9dd7628ee318ed187ff7c603458e2feafe82466c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/9dd7628ee318ed187ff7c603458e2feafe82466c"}]},"branch":"refs/heads/stable/train"},"17e39fc2a258669bf20eaf842f305914b81e0372":{"kind":"REWORK","_number":4,"created":"2021-02-22 09:59:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/4"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-02-22 09:58:09.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/17e39fc2a258669bf20eaf842f305914b81e0372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/17e39fc2a258669bf20eaf842f305914b81e0372"}]},"branch":"refs/heads/stable/train"},"d613743b10756373258e670ac5cd58d86708f380":{"kind":"REWORK","_number":5,"created":"2021-02-22 14:30:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/5"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-22 14:29:48.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d613743b10756373258e670ac5cd58d86708f380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d613743b10756373258e670ac5cd58d86708f380"}]},"branch":"refs/heads/stable/train"},"7b8ee4a8ad6c3e9d8027172ca8bc34e08a27b46f":{"kind":"REWORK","_number":6,"created":"2021-02-23 17:20:26.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/6"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-23 17:19:46.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7b8ee4a8ad6c3e9d8027172ca8bc34e08a27b46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7b8ee4a8ad6c3e9d8027172ca8bc34e08a27b46f"}]},"branch":"refs/heads/stable/train"},"2732bbf59e438f5c93629d2b41f4987af00fc7be":{"kind":"REWORK","_number":7,"created":"2021-02-24 09:16:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/7"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-02-24 09:16:49.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2732bbf59e438f5c93629d2b41f4987af00fc7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2732bbf59e438f5c93629d2b41f4987af00fc7be"}]},"branch":"refs/heads/stable/train"},"e13b53c0f25e5d1a586a1365ad861f7a97dc9eb6":{"kind":"REWORK","_number":8,"created":"2021-03-01 10:54:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/8"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-03-01 10:53:29.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e13b53c0f25e5d1a586a1365ad861f7a97dc9eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e13b53c0f25e5d1a586a1365ad861f7a97dc9eb6"}]},"branch":"refs/heads/stable/train"},"0d13cffb15ade88bce7e40d19c71868148f2f351":{"kind":"REWORK","_number":9,"created":"2021-03-01 12:05:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/9"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-03-01 12:05:34.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/0d13cffb15ade88bce7e40d19c71868148f2f351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/0d13cffb15ade88bce7e40d19c71868148f2f351"}]},"branch":"refs/heads/stable/train"},"3a9f0e0a02de064300bb5a3c55cf75894f2edaea":{"kind":"REWORK","_number":10,"created":"2021-03-04 11:04:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/10"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-03-04 11:00:00.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3a9f0e0a02de064300bb5a3c55cf75894f2edaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3a9f0e0a02de064300bb5a3c55cf75894f2edaea"}]},"branch":"refs/heads/stable/train"},"20f118035802ec38aa407132707819c15e9c300a":{"kind":"REWORK","_number":11,"created":"2021-03-04 11:10:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/770385/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/85/770385/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/85/770385/11"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:32:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-03-04 11:10:15.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Idcd1e39b4d60d33c2396f9495b66b089341748ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/20f118035802ec38aa407132707819c15e9c300a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/20f118035802ec38aa407132707819c15e9c300a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
