)]}'
{"id":"openstack%2Fnetworking-odl~770391","triplet_id":"openstack%2Fnetworking-odl~stable%2Fstein~Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b","project":"openstack/networking-odl","branch":"stable/stein","topic":"oslo_lc_drop","hashtags":[],"change_id":"Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b","subject":"Dropping lower constraints testing (stable Stein)","status":"MERGED","created":"2021-01-12 15:34:59.000000000","updated":"2021-03-10 15:19:37.000000000","submitted":"2021-03-10 15:11:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"770391-oslo_lc_drop","meta_rev_id":"8c5e09af85228e731bc4320fefb52f750d37ff55","_number":770391,"virtual_id_number":770391,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 15:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-09 13:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-10 11:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-03-10 11:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 15:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 15:53:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-13 09:22:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-03-09 11:57:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-03-09 12:35:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"c701d9c3e99ed1904298ee4d694841703836ec1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 15:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ec4fc4fdc25adcbdfe4d605365681a80fefafb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:35:21.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-odl on branch stable/stein.  The error was:\n\n  Unknown projects: openstack/networking-l2gw\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs-neutron\n        - openstack-python35-jobs-neutron\n        - openstack-python36-jobs-neutron\n        - release-notes-jobs-python3\n        - periodic-stable-jobs-neutron\n        - publish-openstack-docs-pti\n        - check-requirements\n      check:\n        jobs:\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - networking-odl-grenade:\n              voting: false\n              branches: ^(?!(driverfixes|stable/(mitaka|newton))).*$\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-oxygen\n          - networking-odl-tempest-fluorine\n          - networking-odl-tempest-oxygen-multinode:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-tempest-fluorine-multinode:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n          - networking-odl-rally-oxygen:\n              voting: false\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-rally-fluorine:\n              voting: false\n              branches: ^(?!stable/(newton|ocata|pike)).*$\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n          - networking-odl-functional-oxygen\n          - networking-odl-functional-fluorine\n      gate:\n        jobs:\n          - openstack-tox-pep8:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-docs:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - openstack-tox-py36:\n              required-projects:\n                - openstack/ceilometer\n                - openstack/neutron-fwaas\n                - openstack/neutron-lbaas\n                - openstack/networking-l2gw\n                - openstack/networking-sfc\n                - openstack/networking-bgpvpn\n          - networking-odl-functional-oxygen\n          - networking-odl-functional-fluorine\n\n  in \"openstack/networking-odl/.zuul.d/project.yaml@stable/stein\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"0b2d1e0a3bb6758d6ff1e08eafd3a150c62b631f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-20 11:24:36.000000000","message":"Patch Set 1: Code-Review-1\n\nYou will need https://review.opendev.org/q/I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5 to make that working. Maybe You can rebase it on top of that other patch","accounts_in_message":[],"_revision_number":1},{"id":"73bdc0207705590ecbdc696b903ab18a11c4a8e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 13:00:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e0157aff22b996d3b31970a671d35d0f142fcdb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 13:02:28.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e You will need https://review.opendev.org/q/I38509a2fafc07e5196ba9f8f1dc11e617cbc74f5 to make that working. Maybe You can rebase it on top of that other patch\n\nFinally I spared time to these old things","accounts_in_message":[],"_revision_number":2},{"id":"36d8d0d4ce2ce9b7eaa86e67cae062134757f570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 14:07:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7965301126ea4d928a2334c3e45c6101 : FAILURE in 8m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5e54e394b174c3aa4ba4c1c7242db1b : FAILURE in 3m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/309e58409509459cb749d940ee684ba5 : FAILURE in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c966226254fc477b841702db958a0b6b : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3db3a619df7f4f9b8d8ea91668a1cdbb : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a04d36bc4c846a7a21a1d73de742569 : SUCCESS in 12m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/564ecfb653364944abc1bf2f3a202a46 : FAILURE in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e357a29724864839820099573a207414 : SUCCESS in 8m 19s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/e2499ae00ecf4fc8b184973c9cd033af : FAILURE in 24m 33s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/caf0c86caa99429782ad323432a8ce4f : RETRY_LIMIT in 6m 53s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/d8d0b475e7e24fb5aae2135c66727950 : RETRY_LIMIT in 10m 59s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/7b4c595de3aa413b8ad80061ebfd6499 : FAILURE in 22m 59s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/ff1db6ddca484d8dbc1b532207015e44 : FAILURE in 21m 13s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/0eff97764da04b2c94f55d8a51dc575b : FAILURE in 22m 00s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/cabc2e60ec0b4ecebcb3cc98bcf1ccc3 : FAILURE in 31m 42s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/0e5e7bb9aeb64dfb9756ad92f0887aa7 : RETRY_LIMIT in 9m 50s\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/a7b11cdfe75c4d718d176197ddbd7d92 : RETRY_LIMIT in 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"83cdfe1740e07acad932d21cee9b0a2e1126138a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-22 10:15:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f6062da5436d57a816456ee816525125e8000677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 11:17:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157c428d12784f2fb50565fb1a3e69f2 : FAILURE in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3ba4841fe6042cc83ced688cf817ee3 : SUCCESS in 8m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9510f619e7734aa88bb5f2416bd8343f : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62467ece9b8747f69aa05c6326c4fbe9 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15d2adc8c6f14a9f9ef7c15ed5a59d7a : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caba1a6fc0af46a79089fcb74ffb2e3d : FAILURE in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/024ee4ad40eb48fe8bb426ee798f32ec : FAILURE in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26007423e5db45d1aacbc8169677bf6e : SUCCESS in 8m 59s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/35b3d0644b65454ca74de724771adec4 : FAILURE in 23m 28s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/4fa87ef946db483698bbcde12bbaccac : RETRY_LIMIT in 12m 31s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/628bbb55c1f34b3daf6d1977aea50bf7 : RETRY_LIMIT in 5m 30s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/3c69f4e89c444a16b9dc88e338aee30f : FAILURE in 18m 45s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/8612021a52fe4f6992f34a8a2e2276fd : FAILURE in 17m 56s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/56034ba332724860a19c6e62afc3464b : FAILURE in 36m 10s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/349cb948afab43c8b0e43384b082ed34 : FAILURE in 20m 25s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/edfd670f795344f28016d34409f1d293 : RETRY_LIMIT in 13m 18s\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/972e2b5b12a34949a81416fb4d068b25 : RETRY_LIMIT in 14m 26s","accounts_in_message":[],"_revision_number":3},{"id":"69ceb9668de1939838f1ab8d4ea0265256e49630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-08 10:11:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"be2f2cd381816201f33219b07ca7be1f54bb0ef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:01:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58efdf222e2a4abd9d2f8d42ea5b9f18 : FAILURE in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/786b824038a347c2afcf3b32616ec923 : SUCCESS in 8m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/299459847e6f4e0fa5218e0cc12ec667 : SUCCESS in 9m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dee3b9d474f740c8aa96f9f814c1575b : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f6bf0c6373941608b66c180bc505edd : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1c4648215a46e2bdbaf91f422633f2 : FAILURE in 4m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99aaa74c4cfc44618663c5c427692e2a : FAILURE in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de6241c33aba4204a85490c79504c72d : SUCCESS in 10m 00s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/b1bf38d55337440e949f3679b1a3e82a : FAILURE in 13m 57s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/a2e0a195d4434d6e856d16457542a5d7 : RETRY_LIMIT in 6m 39s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/0edac3203c4041b9905ac74de85d761f : RETRY_LIMIT in 6m 56s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/a1946426c0d042b2a996311aedb5e459 : FAILURE in 20m 28s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/b0df05f906934d84a1fe85d5f1ab9bd8 : FAILURE in 17m 52s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/93f704c21bb04153ac4a43e4e6bd307d : FAILURE in 12m 11s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/0ef16ccade17465f8a4523ef96600f26 : FAILURE in 13m 29s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/4f758ffd7dab4505b605633b1c82992b : RETRY_LIMIT in 12m 53s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/09f8591a2afe4dc280fb88f9d4340899 : RETRY_LIMIT in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6c76e0c18b50b7d85194637d4f0ed8d217c2574e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-08 12:11:56.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f905cfe55540e508295b02be3660e4bbc4649648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 13:23:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6464421d9a48bfb62d609c58e0c9dc : SUCCESS in 8m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a522c7a56a9441285054226182a1d6f : SUCCESS in 8m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6513af27ea3748a7a5d25b0453da5f0e : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2b5519e26c943f3b5479826c9b85c94 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a32f5522fe4f3799d1c2675191aa8b : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e08df9e22d4bdaad823a54b086890d : SUCCESS in 8m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/314ab1ef24a74191af8ff3892b6c82e1 : FAILURE in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f87f0fc0564d42c69710d42a534eded9 : SUCCESS in 7m 44s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/80fc35dca71e45d3b469cc651c7586a3 : FAILURE in 13m 04s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/1963fb90453e4f3eb0df518011446587 : RETRY_LIMIT in 7m 43s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/0f90e46d71cd45ebb1885ac225253af3 : RETRY_LIMIT in 9m 04s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/140575faf62b4264bfcc7d94d77790ab : FAILURE in 12m 15s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/7520f3a2d9ce4536a9921fc7ccd72670 : FAILURE in 13m 10s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/e2e2f9f4f941402e9e0d89ed04d3d2bc : FAILURE in 14m 43s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/b64b74c47286456fb3bbec3f027f6eb6 : FAILURE in 14m 25s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/e61260d437854d8f8e22b7752e723813 : RETRY_LIMIT in 14m 46s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/0abe1a829a84480dbe3f1c7087e6e2a8 : RETRY_LIMIT in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6fa547f9ca9a7b59d021949f204395ba2f8f0b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-09 07:48:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aa9853f67680a5e7f50ee7247aeef2d8b5ee351c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 08:42:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241fb5f9eb2c4bc593c19211e6023eed : SUCCESS in 9m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f601ab05facd4b168355315cdd51e4ba : SUCCESS in 6m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5947aca43ef947e592abf3613e2dd483 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39f6dc3f008347e1a6fa30cc82e42d13 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aee774230fa4216aaeef4b39e225aab : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b05fe3251a4fff817581f34828b6de : SUCCESS in 7m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/26de98a2bb3b4c779acab819d29d9266 : FAILURE in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5ecb3c6ef654579bae82304b3cd9a9d : SUCCESS in 7m 13s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/6f67952f589a49289c7ebb9150ef5dec : FAILURE in 22m 23s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/b114774a92464bc5ac89514737c442e0 : RETRY_LIMIT in 9m 26s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/8e10d7ddc1f24115922ba1ab615f2dc4 : RETRY_LIMIT in 7m 40s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/56eaa8c4111348fb9d346c7ef89bfe16 : FAILURE in 16m 19s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/57343f5e03e144ec8b41bde8f1312cb2 : FAILURE in 23m 39s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/0a1fcca7ca8f481f98784301235d6696 : FAILURE in 19m 59s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/c832a3088c444129926da100f73948fe : FAILURE in 21m 38s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/b758fb31ba844f19a894e36489424883 : RETRY_LIMIT in 11m 44s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/911de0ddede24217a3ac179e19053a04 : RETRY_LIMIT in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cc581c44e3bfc129cf6016877fb96e6ccfcec845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-09 09:57:27.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"4947f9e1d5d3d8ecabe555f64bbaa6c439e24ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:53:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bfd3e53349461da2f19f9729e626dc : SUCCESS in 8m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ea31d3b62924b089d46a7ee234f87c1 : SUCCESS in 9m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ddb7a05315a34a1db914ccb8f3c91319 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34d7a1b11a28442086731826d67bc840 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebf2b4cf891e49b1afd276385b59a4e9 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9cc8d1ab9f4e8aa37031e553083fc2 : SUCCESS in 10m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/26b99343f61a4671bcd96426e919df75 : FAILURE in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7719e7417f7147159ca60d8fe35de5fc : SUCCESS in 8m 57s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/4ba883a0f85e4126b15c9661877b7ccc : FAILURE in 17m 32s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/52ce48b28eb242a2bd5200d28d6f40c4 : RETRY_LIMIT in 6m 07s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/db3c776be83d44acac4b6b2bd34abb26 : RETRY_LIMIT in 10m 32s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/4fd922596b7b4c18b9420625ed4ce280 : FAILURE in 19m 52s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/f2d20ac6d1f24c6689eb88896eec0d06 : FAILURE in 25m 43s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/7a9e520cbcab4c6a9e8b6c03d6d6891e : FAILURE in 29m 07s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/5f6879fabb5a47d09a1e51567464df90 : FAILURE in 21m 58s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/09e526eee05b492e92ffa3fc8e86322f : RETRY_LIMIT in 11m 27s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/c38c004f9a8542dfba2dc4ed3e6917e6 : RETRY_LIMIT in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c12e3ed74e005e113803aebdd2e722e5fe40f48c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-09 10:56:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"72941c99bf97ea49b3d7005eb9abdc25646777a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 11:50:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4138bb80057443b9934ff8be16b8fb15 : SUCCESS in 8m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bbfbf0da7554fac835756c368b7ed34 : SUCCESS in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/960193b3d1464fba8babc89f22ce8df6 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c4461a8df434a30bef72dbdf6b963ca : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd32682f75ab48f192bf63bcfa015292 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82d16e423ff45a59dd02531d5df8c71 : SUCCESS in 10m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f2dc28fb45446248fc6a6dcf6c5ad76 : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b47d5ed704e4398857eb027388c131b : SUCCESS in 12m 04s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/738949ff2bc44f39884ff890ec1218d1 : FAILURE in 15m 46s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/3f06c3b04fa24bbeba244572b3049bf4 : RETRY_LIMIT in 11m 21s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/bee0c739fee54a1fbf4a21e13a3259a5 : RETRY_LIMIT in 17m 39s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/84cd2668462f4ed6802070ce5427abc2 : FAILURE in 22m 51s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/68d58eafdf2b419cab8c284b0080afff : FAILURE in 34m 50s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/75566f141131497eb406cad8ecfda221 : FAILURE in 26m 26s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/a93482e87b8f435080483885e5f5ada4 : FAILURE in 17m 27s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/9de4f0f8372d4ea8ad714c81a75ff2ad : RETRY_LIMIT in 10m 25s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/c95b3332bd044e7cbb951edf9274fd34 : RETRY_LIMIT in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"63f61379266193866cec00c144e4bb0e835c181b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-09 13:33:46.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)\n\nOne minor comment for the bandit constraints, nevertheless the patch looks OK to me.","accounts_in_message":[],"_revision_number":8},{"id":"50ff88b750941b93d1620c4304eaa32226f8b9a6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 11:46:17.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLooks good and similar to train one","accounts_in_message":[],"_revision_number":8},{"id":"5ac91bc6b4b4d184d2a1416dfce2e6a8ea2532ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:46:37.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"224d11162d4fa0d24bf272155c795beb4b38f036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:39:48.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ecae103209e457aae421c7ff9c57ff4 : SUCCESS in 9m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64a085fac74f40e1962fa209dbfd4d67 : SUCCESS in 7m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0ebcf375fc4741a69d69408987ec9db8 : FAILURE in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22aacaa18a97482c9f6e040c72992e2d : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9600d7debf76423799b644179d9d54e9 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8590cdfe5b234e27a732b39eb9846c1b : SUCCESS in 9m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/755b5ae0c7074b3d9bc0a8ea3f35b539 : SUCCESS in 3m 17s\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/3178614d845a4d0e8fb143e61dd493d5 : RETRY_LIMIT in 11m 57s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/32179184d33c4045b2aaee1bb518fbbf : RETRY_LIMIT in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4cc35dee2dff2577ac2827e16d09f8e66fbcf989","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-10 13:09:41.000000000","message":"Patch Set 8:\n\nrecheck - test_feature_configs_does_not_mutate_default_features - testtools.matchers._impl.MismatchError: {} !\u003d {\u0027bgpvpn-vni\u0027: \u0027\u0027, \u0027operational-port-status\u0027: \u0027\u0027}\n\n(I\u0027ve seen this intermittent failure in the past if I\u0027m not mistaken)","accounts_in_message":[],"_revision_number":8},{"id":"6bec722b037157d2b0826f771a90932ced6a73a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:12:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/775e344d4ac942f8aeac0ae4388a9731 : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7017f8aeeff84807a57d22f70f81c6ff : SUCCESS in 7m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d1705b12395424c99c7067e08aab05b : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5592a044089844dd842eb90cf279e58d : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dea6acc6b164d7ca04a8af94a869b67 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdbaca4a01c145f28c21854d50c4be6e : SUCCESS in 9m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/480787a3e04d4514afe77e4121d4c37f : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84584486290c40659b13b6ed2e2a680a : SUCCESS in 9m 35s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/7ec71fd7054c4039a3ba640d3198b196 : FAILURE in 13m 58s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/ea7ec2fd75554477a8e3d864941a3743 : RETRY_LIMIT in 7m 29s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/93ae6416a4ea4afb9ae44f3ef265d4ba : RETRY_LIMIT in 14m 31s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/d19818341f4b456fb9a112500528d1a6 : FAILURE in 24m 05s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/4827d91a52bc4f40bfe88a47033b85b0 : FAILURE in 28m 32s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/40e4b1c727d04d9cb7015cea9b45d613 : FAILURE in 31m 59s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/4240a9f43162486f961c488bc9b5a503 : FAILURE in 23m 55s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/f9fdf4d012ea434a9d47cd3ac102a039 : RETRY_LIMIT in 17m 20s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/9021e8fd213d44919caa69e7a833f293 : RETRY_LIMIT in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1d1e58defdbf685200f80aa179f16d0631f9e3ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:12:16.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b7947e900b66a236f1cb9bae77da85b1b0cc981b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:11:53.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398e9ab33bc34c1da1a9973a3978af70 : SUCCESS in 11m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc4633441ffa4c1fa8f798055b6c552f : SUCCESS in 7m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d513c58027204b03a3f0f3258fd88f20 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ddd959864bb41018550278ad84d22ff : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8d5ece6a0184c59822d136efb5a4c3d : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea36b47c6d04d88b7ff2c76061f3f4c : SUCCESS in 9m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f31c5ac28f6413cbed2a3ad3f0cf5e9 : SUCCESS in 3m 23s\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/9a3d5f6c322942658440cc213566c482 : RETRY_LIMIT in 14m 58s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/d782993665c44e5fb4aec99a65b8d937 : RETRY_LIMIT in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f02cb07d09ed6ae21cdad4f852e1c3b5c0ae52e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:11:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"8c5e09af85228e731bc4320fefb52f750d37ff55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:19:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2989e71c6364cde9caeab4a0c31466e : SUCCESS in 1m 13s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affc4b5d57264f3391519253402018b2 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1e3a1230fd9270280faa16336690d3c9f00a9cb3","revisions":{"e8bbb9539d24a35dfd27b5a3f7d5fc4c18a65e41":{"kind":"REWORK","_number":1,"created":"2021-01-12 15:34:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/1"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e8bbb9539d24a35dfd27b5a3f7d5fc4c18a65e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e8bbb9539d24a35dfd27b5a3f7d5fc4c18a65e41"}]},"branch":"refs/heads/stable/stein"},"0f88bc70a126abccdf1f1a4ca8ee3a3cc4ce8671":{"kind":"REWORK","_number":2,"created":"2021-02-19 13:00:58.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/2"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-19 12:43:48.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/0f88bc70a126abccdf1f1a4ca8ee3a3cc4ce8671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/0f88bc70a126abccdf1f1a4ca8ee3a3cc4ce8671"}]},"branch":"refs/heads/stable/stein"},"6fde2beba51dc8254fc3573cfc5d084575cc024d":{"kind":"REWORK","_number":3,"created":"2021-02-22 10:15:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/3"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-22 10:15:13.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6fde2beba51dc8254fc3573cfc5d084575cc024d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6fde2beba51dc8254fc3573cfc5d084575cc024d"}]},"branch":"refs/heads/stable/stein"},"a2c7bac4cad8ada4541f39a40d5a86e34d0aea42":{"kind":"REWORK","_number":4,"created":"2021-03-08 10:11:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/4"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-03-08 10:10:44.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a2c7bac4cad8ada4541f39a40d5a86e34d0aea42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a2c7bac4cad8ada4541f39a40d5a86e34d0aea42"}]},"branch":"refs/heads/stable/stein"},"743ceaeba3af0dc0b349b8db108ec5123e708e0d":{"kind":"REWORK","_number":5,"created":"2021-03-08 12:11:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/5"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-03-08 12:11:51.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/743ceaeba3af0dc0b349b8db108ec5123e708e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/743ceaeba3af0dc0b349b8db108ec5123e708e0d"}]},"branch":"refs/heads/stable/stein"},"3c96d608f9293340fc076b8fae6237e8d9ac1ba0":{"kind":"REWORK","_number":6,"created":"2021-03-09 07:48:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/6"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-03-09 07:47:55.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3c96d608f9293340fc076b8fae6237e8d9ac1ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/3c96d608f9293340fc076b8fae6237e8d9ac1ba0"}]},"branch":"refs/heads/stable/stein"},"dd1331de1e4303866769a95e30a41440db052562":{"kind":"REWORK","_number":7,"created":"2021-03-09 09:57:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/7"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-03-09 09:57:20.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/dd1331de1e4303866769a95e30a41440db052562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/dd1331de1e4303866769a95e30a41440db052562"}]},"branch":"refs/heads/stable/stein"},"1e3a1230fd9270280faa16336690d3c9f00a9cb3":{"kind":"REWORK","_number":8,"created":"2021-03-09 10:56:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/770391/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/770391/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/770391/8"}}},"commit":{"parents":[{"commit":"5b062f03e795809760b706d7700b53c8ef038a7f","subject":"[stein-only] Constrain l2gw to Stein version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5b062f03e795809760b706d7700b53c8ef038a7f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-03-09 10:55:34.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/1e3a1230fd9270280faa16336690d3c9f00a9cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/1e3a1230fd9270280faa16336690d3c9f00a9cb3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
