)]}'
{"id":"openstack%2Fnetworking-odl~789074","triplet_id":"openstack%2Fnetworking-odl~master~I11e2508dced505dbba13f95650f1ba9d696be6e9","project":"openstack/networking-odl","branch":"master","hashtags":[],"change_id":"I11e2508dced505dbba13f95650f1ba9d696be6e9","subject":"Dropping lower constraints testing","status":"ABANDONED","created":"2021-05-02 02:20:32.000000000","updated":"2021-05-07 12:48:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dd88df8fba6252a7fc11fb19b96132c6b3993521","_number":789074,"virtual_id_number":789074,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-02 03:26:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":-2,"date":"2021-05-04 16:35:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-02 03:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 16:35:28.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"9ced49b34bc1a8a6dd5202ddb3c5f76036bb4937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-05-02 02:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"226df8625d5a42190c24ededa1850b913b5c8e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 03:26:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2619eff273e54b568e57a3c0da8dc225 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc037f449fb8495bbc0c9e06b1972738 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/500bd5255f364372b9d385d7d5cd9705 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9e6e774054b49d98b9e04260c644c7f : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48536e9b920845339f46828d71b57399 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e600c47ab0e74349b53a6e5b18f3818f : SUCCESS in 7m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4fa0fe26542a49d98b93c7c3b0688ace : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b57b841ff9ea42a4817c819e8bfd5e62 : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de4d8092580048d98451a5edfe4bd97a : FAILURE in 3m 32s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/73d44b3f879743f4a774b872a24ef45c : FAILURE in 49m 08s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/11ea12452ee742e19fed4bdb071bd99c : FAILURE in 54m 22s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/4531d7ef87fa4aca9ca3b38fe392bed7 : FAILURE in 39m 45s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/b6fe11f199d4453c85855fdf875b2d1c : POST_FAILURE in 39m 47s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/ba4907fbbd544afba98aee7cc69fbb8f : POST_FAILURE in 32m 35s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/4d9e7074cef048519054e18339f320d4 : SUCCESS in 1h 00m 54s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/74dda44144c543789bf34c44b1c6641f : SUCCESS in 1h 01m 02s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/91b4d6f04281444192a5c5501fa3b271 : SUCCESS in 22m 35s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/c5044bab703b4459a813fe148ed10cf4 : SUCCESS in 21m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b6dd15b5a270d2fb3a5a688f77559c25ce75fc41","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-05-04 16:35:28.000000000","message":"Patch Set 1: Code-Review-2\n\nThe neutron team discussed this topic in our team meeting and the current agreement is to drop l-c job from stable branches and keep it in the master branch. If you propose to change this agreement, orpose it to the mailing list or the neutron team meeting first.","accounts_in_message":[],"_revision_number":1},{"id":"dd88df8fba6252a7fc11fb19b96132c6b3993521","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-05-07 12:48:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc4974def4351c8067a41005d7d066c9e2b4acb6","revisions":{"fc4974def4351c8067a41005d7d066c9e2b4acb6":{"kind":"REWORK","_number":1,"created":"2021-05-02 02:20:32.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/74/789074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/74/789074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/789074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/789074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/74/789074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/74/789074/1"}}},"commit":{"parents":[{"commit":"50356ed5c8753b4424815e3082d7fd147c0c0212","subject":"Merge \"Update master for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/50356ed5c8753b4424815e3082d7fd147c0c0212"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-05-02 02:19:43.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-05-02 02:20:26.000000000","tz":480},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: I11e2508dced505dbba13f95650f1ba9d696be6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/fc4974def4351c8067a41005d7d066c9e2b4acb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/fc4974def4351c8067a41005d7d066c9e2b4acb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
