)]}'
{"id":"openstack%2Fnetworking-odl~792014","triplet_id":"openstack%2Fnetworking-odl~master~Iaadc40382efa8562bc706ba3dc95e87718d4b152","project":"openstack/networking-odl","branch":"master","topic":"get_agents_objects","hashtags":[],"change_id":"Iaadc40382efa8562bc706ba3dc95e87718d4b152","subject":"Retrieve the agent using \"get_agent_objects\"","status":"MERGED","created":"2021-05-18 16:51:50.000000000","updated":"2021-05-20 08:45:17.000000000","submitted":"2021-05-20 08:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792014-get_agents_objects","meta_rev_id":"cbf784074178f02a5a94529a24f161f2a2519fe9","_number":792014,"virtual_id_number":792014,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-20 08:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2021-05-20 08:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-20 08:20:32.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-20 08:20:32.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 17:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 08:00:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-05-20 08:05:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-20 08:05:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-05-20 08:16:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a3212818ec3aa7ea71e847a45e5c8b9c30b133a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-18 16:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bca8a3c6896f8b8f64af858b06e993697191290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 17:48:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/789fd4971c7b413d88465ce7eef7a12c : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/620be273aeb24d5d908776b993329dd8 : FAILURE in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da26c8d517484816829edce9dd2179e0 : FAILURE in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e416079fccc40daade97049641ff923 : FAILURE in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4aca54ab224366ade3afe7c5f99674 : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae3c76c18dcd451d8cd04e47d06ca041 : FAILURE in 7m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7a9568514d34983bb3a0ce092027cce : FAILURE in 3m 48s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/774e44d489fa438493a146ad6eb221a5 : FAILURE in 43m 34s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/6bd7329b86184001ac2b7e533a21a9ec : FAILURE in 36m 50s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/e2e3d0e49e304e0ebff9d41bc681f920 : FAILURE in 40m 05s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/9665394e36204953badef8410c95e916 : POST_FAILURE in 36m 13s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/b1f9ebc8a433408a92b385a253b3d24c : POST_FAILURE in 32m 16s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/2c7279d641ef4b20a1ea0d5e0460667d : SUCCESS in 55m 37s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/705622e43d6a45ed8b4c1ef06b05b80e : SUCCESS in 43m 49s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/3b17f31d53ee4a6eb28b18b52622ab70 : FAILURE in 18m 31s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/096fa988f25e482e9e231c8ce5822741 : FAILURE in 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fd27b7c8e8d6c8d4a3a7b2ac103d19e52b0e443f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-19 08:00:14.000000000","message":"Patch Set 1:\n\nI think this should be merged first: https://review.opendev.org/c/openstack/networking-odl/+/635179","accounts_in_message":[],"_revision_number":1},{"id":"ab583408ddd0e1a871250c87b78e6795caef5156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-19 13:45:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fd68f2d83868f3faba8eadc526ae0770707600fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 15:01:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d006e2f5c24c3aa391612768a7f9c2 : SUCCESS in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19ff8ce5629340ee9850f405e1b69a94 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29b9951b20e64ac88271329cbbd98534 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cf2773dd6ac41449f2f5d1d3e3301ac : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980ecbd26506418dbc9b150a44b2c508 : SUCCESS in 9m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2533ad5ad064f648707b891d0fedb51 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/749e99933b044312a37419d53f6d0521 : FAILURE in 4m 30s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/ecdadeb2bfcc4d10ad25afffbdbb3c67 : FAILURE in 54m 07s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/a591b527555c4a7bb779131d7ae7bf76 : FAILURE in 1h 02m 38s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/e7a1604608f64a9a8782d3ce1730e09a : FAILURE in 51m 50s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/60c7aa3f118d40a9ba323fea2f8ba74f : POST_FAILURE in 53m 32s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/f63383e3084f4fd6a2f9a9a537e46131 : POST_FAILURE in 31m 50s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/907706a442df41f3be85b5ad2dca5c3f : SUCCESS in 56m 20s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/3382d26156094118987226f0b085f029 : SUCCESS in 1h 09m 18s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/d3380e8c80174f53962a7020954a7423 : SUCCESS in 23m 05s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e0271a9a33ad4f98be9c0dece4625e66 : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":2},{"id":"5d3d7dd0357f419d28629ae900956b84876882d6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-20 08:16:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e4f5095bd4236e74a9843d0661ff65922a74bea","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-20 08:20:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae6d99c23a4be5536314b21fd17eeda4f28af0eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 08:20:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4bfe1e050a9c8c0b4ea4ade3023f0de9cac9ad81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 08:43:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f34b78d6206d4a4d986b4677fd7226ad : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f7984dd7e854aaab5e999b1a3304c0d : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ca4c80d85df4378a33541f2d409205b : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8705a4da3d6d43e39a35e259d57a3b83 : SUCCESS in 7m 55s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/b734ab4bd6ae43b3b39b8ecdb82c4905 : SUCCESS in 18m 19s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5d052ac6b4f046d0a76314888cdf8114 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e933183b7e025e98f36545423efccb8a0e9c9001","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 08:43:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cbf784074178f02a5a94529a24f161f2a2519fe9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 08:45:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c66478334a4244b89609c93f2a092055 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2470f7f02d0959078abf3021118bb30cfcfc4a7c","revisions":{"5492a67071697d21a91eb87239f5e225783530d7":{"kind":"REWORK","_number":1,"created":"2021-05-18 16:51:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/14/792014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/14/792014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/1"}}},"commit":{"parents":[{"commit":"29ef940cb517bc984efa8ddd9d92459d929d2b85","subject":"Merge \"Change the remaining non python3 compatible code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/29ef940cb517bc984efa8ddd9d92459d929d2b85"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 16:49:57.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 16:51:47.000000000","tz":0},"subject":"Retrieve the agent using \"get_agent_objects\"","message":"Retrieve the agent using \"get_agent_objects\"\n\n\"AgentDbMixin.get_agents_db\" is going to be deprecated.\n\nTrivial-Fix\n\nChange-Id: Iaadc40382efa8562bc706ba3dc95e87718d4b152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5492a67071697d21a91eb87239f5e225783530d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5492a67071697d21a91eb87239f5e225783530d7"}]},"branch":"refs/heads/master"},"2470f7f02d0959078abf3021118bb30cfcfc4a7c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-19 13:45:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/14/792014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/14/792014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/14/792014/2"}}},"commit":{"parents":[{"commit":"12d78c690f72ad51e2bda7a1ebbfdbd76032897f","subject":"Merge \"use payloads for TRUNK and SUBPORTS callbacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/12d78c690f72ad51e2bda7a1ebbfdbd76032897f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-05-18 16:49:57.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-05-19 13:45:08.000000000","tz":0},"subject":"Retrieve the agent using \"get_agent_objects\"","message":"Retrieve the agent using \"get_agent_objects\"\n\n\"AgentDbMixin.get_agents_db\" is going to be deprecated.\n\nTrivial-Fix\n\nChange-Id: Iaadc40382efa8562bc706ba3dc95e87718d4b152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2470f7f02d0959078abf3021118bb30cfcfc4a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2470f7f02d0959078abf3021118bb30cfcfc4a7c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
