)]}'
{"id":"openstack%2Fnetworking-odl~796789","triplet_id":"openstack%2Fnetworking-odl~master~Ice43e4a8a43ae3a56fb20e41b026fbc7f0162a44","project":"openstack/networking-odl","branch":"master","topic":"exame-tls-proxy","hashtags":[],"change_id":"Ice43e4a8a43ae3a56fb20e41b026fbc7f0162a44","subject":"Enable tls-proxy support for test job","status":"MERGED","created":"2021-06-17 08:10:36.000000000","updated":"2021-09-07 19:45:11.000000000","submitted":"2021-09-07 19:40:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796789-exame-tls-proxy","meta_rev_id":"2802f775042cdee5efbccc4a5036b5cb54cc1211","_number":796789,"virtual_id_number":796789,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 19:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-06-18 07:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-10 13:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-08-10 13:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-17 09:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-18 07:16:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-10 13:56:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-07 18:01:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"bf8af553f01721a7c7f46884f96f490c1e7fe155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-06-17 08:10:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d68ce5271c1228baa5c142654b2d0e883261f8d","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-06-17 08:11:14.000000000","message":"Patch Set 1:\n\nHi, this is effort for early goal survey, which you can provide/find more information on https://etherpad.opendev.org/p/support-tls-default \nFeel free to update this patch and work on this. Please help to provide information if you think this makes no sense to this project.","accounts_in_message":[],"_revision_number":1},{"id":"3c9ca64a8ab76df2986f0be2dc22073507c1b8ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 09:11:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1073d2e781ec47d293c34be7ca987106 : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e589d06dd02b43c28a927ee1bf833c09 : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/226715290e3647b8884ec145e38b1b7c : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fb335bd9162480593cdba9d2f053840 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f13fc7eb2d6c4b3cbcc3d4171a9a3d87 : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05d7440858f54ceb826daed63e78dd8d : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dadb2f78fa2646f1bfa6bc81bc42c1f3 : FAILURE in 4m 28s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/aa16c643f7b74ceeb13e50104577b525 : FAILURE in 52m 17s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/821bf16d8740462e80b7c906fe8075e0 : POST_FAILURE in 16m 06s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/5ae4e72d3a4b4159ad832a5149dc9b8b : POST_FAILURE in 22m 06s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/7ccfe71fb78447d183e5a0502a2c5eae : FAILURE in 26m 25s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/851ced703bf34af29f1d330bb3b1da3e : FAILURE in 20m 08s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/174d6b238eac4ae2b69904b2f66e35c1 : FAILURE in 17m 11s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/4e04b05fa13f4ecbad491fadbabf0662 : FAILURE in 15m 25s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/f849180634374409977055223a2e4204 : SUCCESS in 24m 22s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/c50f4c012c944d529276dc46a89975e7 : SUCCESS in 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0f864d98a5c2982bd118b53167efdcc479dcd5b7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-18 07:16:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa5c2ef33a9f32a910d6da186eccce91deefc6e1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-10 13:56:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c35714356d3123adb514a0ad3571b602e7b6afd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:56:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"945fab10916c822631087d607c87196fc55ce4bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:20:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af92a811ac44b73b2d85f48024b7ccd : SUCCESS in 9m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d66a7388d387402fb7b7b0104445b7d4 : FAILURE in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/369f446979414692b0f2be8f3623901e : FAILURE in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4073f2b6ca74848be605ac9be5d5fa8 : SUCCESS in 8m 48s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/5a4c78795c10409d9aa2a267506f2afc : FAILURE in 18m 15s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/34b6499c06ae4e63989be675543ee2f2 : FAILURE in 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"63a3cf80a92645214069c76c4ed7ba8b3433c382","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-07 18:01:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b54d64c60970408a79d39eecd32fef681dd4e5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:10:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66fe21b90d614ca19830abc9c67b9d53 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/150822135f88476da8c2278e9a1fb054 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59b14b4e8806405a8c7e529fe1a3d05e : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3478cfe202a542e1bbf81de50be9296e : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5145d19ee74e82921f4e517e344e0e : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37e4375c5b484c2a9831df86cd3dafc3 : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/929116a75f3f4170923bd7ac8960b4fa : FAILURE in 3m 57s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/3dbd7b797f654dd788c99b2b62e12dc2 : FAILURE in 58m 45s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/1043638cf3d84858b7cf9401b71becdf : POST_FAILURE in 19m 43s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/74b71b7c19a245b1997eb60de5f51e8a : POST_FAILURE in 21m 27s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/61abbf7960f543cfaf42ff7ec586dc9e : FAILURE in 23m 30s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/96e97f189e71438caa25c4bad28170e2 : FAILURE in 24m 55s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/05139cf11cf7479596cd36b3c56c0af9 : FAILURE in 18m 36s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/4d888eecc4de473f82fca457895e2c7c : FAILURE in 18m 54s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/9538ef6720954e1a9f0cda7705346e7f : SUCCESS in 28m 12s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/a585485b588749818ab68e8d77e5fde2 : SUCCESS in 28m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bff4efbaf09f02cf57ce2bae062a2656ae114579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:10:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd95853f909c030d1a2253ee346b12937c950b88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:40:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db48809e9d4c47b68252d21724528a32 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8290fac4d2486091c71bdb9ecd9cca : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8534d7bc4bf846d98c51c7fe31959b4e : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/370ddf60fe184b8280e4bf9cb3bb85fd : SUCCESS in 9m 58s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/85b5faffe80d489b85a45a96b3bb4d40 : SUCCESS in 26m 41s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/e7220971ec704f7f9b2a7f6baa5dc9bb : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"927d6864ad59bc50e63b51a85ea0da6531823325","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:40:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2802f775042cdee5efbccc4a5036b5cb54cc1211","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:45:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e59b2dfd51749688c88f76281525f19 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b07846c9a581be5b65b55a6b6a9287081e3879d","revisions":{"6b07846c9a581be5b65b55a6b6a9287081e3879d":{"kind":"REWORK","_number":1,"created":"2021-06-17 08:10:36.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/89/796789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/89/796789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/796789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/796789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/796789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/89/796789/1"}}},"commit":{"parents":[{"commit":"a413efcbf389f2e9c71538365cad522ada219806","subject":"Merge \"Explicitly set jobs to ML2/OVS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a413efcbf389f2e9c71538365cad522ada219806"}]}],"author":{"name":"ricolin","email":"ricolin@ricolky.com","date":"2021-06-17 08:09:46.000000000","tz":480},"committer":{"name":"ricolin","email":"ricolin@ricolky.com","date":"2021-06-17 08:10:33.000000000","tz":480},"subject":"Enable tls-proxy support for test job","message":"Enable tls-proxy support for test job\n\nChange-Id: Ice43e4a8a43ae3a56fb20e41b026fbc7f0162a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6b07846c9a581be5b65b55a6b6a9287081e3879d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6b07846c9a581be5b65b55a6b6a9287081e3879d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
