)]}'
{"id":"openstack%2Fnetworking-odl~798298","triplet_id":"openstack%2Fnetworking-odl~stable%2Frocky~Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b","project":"openstack/networking-odl","branch":"stable/rocky","topic":"zuul-fix","hashtags":[],"change_id":"Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b","subject":"Dropping lower constraints testing (stable Stein)","status":"MERGED","created":"2021-06-28 13:12:48.000000000","updated":"2021-06-30 13:38:23.000000000","submitted":"2021-06-30 13:36:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"798298-zuul-fix","meta_rev_id":"dff7275eda116a5d65e0e5710114f6ccb1a089a5","_number":798298,"virtual_id_number":798298,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 13:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-06-30 09:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-06-30 13:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-06-30 13:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-28 13:12:48.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-06-28 14:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-28 15:34:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-30 08:05:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-06-30 08:05:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-30 08:05:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"83456e74d9bc264bd4d2b0c1ecb28ede9356a71b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-28 13:12:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9e07091ec13b4df6a31349da29c13cbf4dfde42","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-28 13:16:11.000000000","message":"Patch Set 1:\n\nThis fix is backported to address a zuul error mentioned in https://review.opendev.org/c/openstack/networking-odl/+/798298\nThe cherry-picked commit in stable/stein contains changes other than zuul configuration but they sounds reasonable for stable/rocky too, so I included them too.","accounts_in_message":[],"_revision_number":1},{"id":"191f5cfcd876c5ffb4bfea45740a43b639d0efa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:11:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b53625503634676b0b7ffc226940a53 : SUCCESS in 11m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aac255683d4a4d2a92c380cfecf987a9 : SUCCESS in 9m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9ee4ec9863fb4bce930625e6ea6c612d : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cee9828eb9c477db2326443124c3339 : SUCCESS in 6m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2dfbc221fcb84c28a94f7482fe289cde : SUCCESS in 8m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cdfbf62fec054efd89f31ffe0ce4c4e7 : FAILURE in 5m 01s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/bd41c2976d734f738c02f16c9598eaa2 : FAILURE in 24m 42s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/b54a10f565564e77aaee20e462c4da05 : RETRY_LIMIT in 14m 41s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/67ab5e58477647d1923efd7f7f986e38 : RETRY_LIMIT in 15m 10s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/91ab0178df3b4e5781b50844512ff887 : FAILURE in 45m 13s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/e8198b6dc95943f98cf4d5f0db2d79d7 : FAILURE in 27m 41s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/c78b58840f0d4e5b880d675b03298da0 : FAILURE in 33m 06s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/1b1082fe277149a99bb016fe2b40c389 : FAILURE in 24m 36s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/b738a3efad6b407fbac54940aa6af567 : RETRY_LIMIT in 10m 56s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/7119b00e42a243098526263c9af0fb8a : RETRY_LIMIT in 12m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08ac612cb0044d7796b54e7028b14094 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"10a14eff676a0ddb39b31f2d4bd4175d978fb0aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-28 14:54:34.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"48159a73139c5c030ad061ad28fab8099570ceb3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-28 14:55:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c7da0377b3527c8b689615802598abe450c11a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 15:53:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/570ba14c169846cc8cb990d3508c75a2 : SUCCESS in 12m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53a7b123cb44418caae2166ee2a43f3c : SUCCESS in 9m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/05864249ab8e4e9eb3a62af61f686046 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/838ccb12b2174c6a8e82d275497e51c8 : SUCCESS in 5m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b09b850a28ee4437a5fd4c8d4939ef0a : SUCCESS in 8m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c82645486e0249b391f72e6a819f5e6c : FAILURE in 5m 01s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/0682145ec8ca4165981ea4f46ba4730a : FAILURE in 27m 14s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/873eba690cc848a5b1a005d5164b8af8 : RETRY_LIMIT in 10m 19s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/6be8ad99128b4df3b6a203edf2fa7779 : RETRY_LIMIT in 14m 37s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/aea4a0d580e64173aa690befc02c726a : FAILURE in 38m 57s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/cf463776fdba4a29adb837fbe0ca00f6 : FAILURE in 25m 30s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/8dd689b165464615b0b8ce1895608c3b : FAILURE in 31m 58s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/7f3d548a28074bbb8fd3d420e4739ff9 : FAILURE in 21m 54s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/0a1659f49e7940eeac993b3b05b7e07d : RETRY_LIMIT in 14m 56s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/0e9f0566c14c48d2a5d63f1c4ee99fdd : RETRY_LIMIT in 16m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea9ab7a7b3840968ec1c9638b83a0c0 : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b198faa4aa9c6c5a24d000c8814ef1210f7ba117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-28 16:33:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"74cfb44a36bbb3687b86ad04418070c299cc9580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 17:33:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632996c1ddb34ca181ccbc479d8ec1f6 : SUCCESS in 11m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fbb9ed451404c70906fa51e0fdc9e7d : SUCCESS in 11m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f2b931167fca4507b4d37d4752091e56 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68138f17e3594e2c9c998b59a1319e1f : SUCCESS in 6m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1b24df602f1a4eecaf4f70730abfad59 : SUCCESS in 8m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92b581fe2c054dc9ad7abf7e42fe4691 : FAILURE in 5m 12s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/94afd76c6cb84e319837a44ad4370630 : FAILURE in 26m 08s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/f3b0189e346c4ee79cdaec02e12904ce : RETRY_LIMIT in 8m 38s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/b1bb9e3fa17141c3ba8bd28da50058f7 : RETRY_LIMIT in 8m 49s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/cfe2e9e9d8224f03a11f32e7c765b6bd : FAILURE in 50m 04s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/abc426e08f6444acabd27b557caf0b52 : FAILURE in 28m 27s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/04ad7985de2348b7a9394dfb5e3f0b98 : FAILURE in 38m 04s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/ced8fca21b6d45b2be3b0758308aabc5 : FAILURE in 26m 13s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/e8b246eefa474bef86b88103a55fa3fa : RETRY_LIMIT in 11m 20s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/8fad04dd1fe84bf5bddfb54c6129c10f : RETRY_LIMIT in 11m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ba21639c4504ece8f2cae99ec0f6d2b : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":3},{"id":"4da125b5869e8d323f821429e24f20d155d5369d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-06-29 15:33:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"50243f7c745efc879f2fd780da20f659ab5ec8cb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-30 06:39:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0eaf13e3b0ce4038ec132a06ce73a53ed332ec23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-30 06:56:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9791a041cfb975f6dc98a3c9fd0e112a6eabd390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 07:56:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b83d1c3077c6485e98cf2061a94bb77f : SUCCESS in 10m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b256f4a6bc64027ac4217e77079b42a : SUCCESS in 10m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/454ff598f5044540bd6019fc65c9b40c : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1c2716c58f54fe09f134ce0ff8051b1 : SUCCESS in 5m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0a32e3a6d1314d89a437b2797ad9c79c : SUCCESS in 8m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7edab9fcac04f24a621c79935a83d8a : FAILURE in 4m 59s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/2da400db2bb84fe7ae90b4b9b9dc23a2 : FAILURE in 22m 22s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/13f5b5b617954c588f41347d018f5d16 : RETRY_LIMIT in 14m 54s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/70d3657583084a24ac8a4dca00f75143 : RETRY_LIMIT in 9m 54s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/ba2cd3844a2f43879fb2a7c5e79541d1 : FAILURE in 43m 11s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/3406e000c2c0415da435f93ee8dc5051 : FAILURE in 25m 17s (non-voting)\n- networking-odl-rally-oxygen https://zuul.opendev.org/t/openstack/build/db8cf75a3e61499dbb34a270ae84a4c1 : FAILURE in 31m 21s (non-voting)\n- networking-odl-rally-fluorine https://zuul.opendev.org/t/openstack/build/aa6589c165ea4e95bd758f484bfb487f : FAILURE in 22m 39s (non-voting)\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/8d0e2997dd89461babbe16384dd2b07a : RETRY_LIMIT in 11m 27s (non-voting)\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/49b999b4caa24da6bb8bee1e9ee4bb22 : RETRY_LIMIT in 15m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2057b8d01284aada32d2fd53e6fc690 : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":4},{"id":"efb2530ac7884f3d86d1b6fc6ce7f7a850b68583","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-30 09:01:55.000000000","message":"Patch Set 4:\n\n(1 comment)\n\n1 minor comment for the test-requirements.txt, otherwise looks OK.\n\nRequirements-check job fails due to I158bcc8c249ad7ec981707c0d96ad8db87eb23d9 not present in rocky, so it\u0027s not a valid failure. Since stable/rocky in networking-odl is not really active I guess it is OK to set req-check non-voting for now.","accounts_in_message":[],"_revision_number":4},{"id":"415f42a9ab7511889f81e98fabc5dd4638377583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-30 09:35:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7041d0020adeed38cbcede67c47b2e539e277b2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 09:54:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f90d4f4574ea4c30b205e7d62d8f2074 : SUCCESS in 13m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6057f225f2d4b25bd6cd468fbfa7ebc : SUCCESS in 11m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/68221535deda4035a3985838c70b7850 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/762879b0d81b4ff682d794f4ee722a64 : SUCCESS in 6m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dd34cc73d86044eb942acebbea103ff4 : SUCCESS in 9m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f312eded6544a01b7b02ab1a543ca65 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e40d48c88819474cb42c1b03a8668e6c : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":5},{"id":"14860739e85194cca73a27862d9eab7be7e1622e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-30 09:58:05.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"84363bcc98b5905e502bd84917dcf70287f1b612","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-30 13:19:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nSounds good, for rocky making requirements non-voting is OK to me (and branch is in much better state with this change)","accounts_in_message":[],"_revision_number":5},{"id":"84c15ff819aa7d23a69efeaba39797b60809bc69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:20:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"24504e5da5239b041b20031ff02ed9e2322591c3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-30 13:22:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1eb130eefa582f106e11be00d3944864b6d39bd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:36:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d9d33219dab40599b011e4a23fe047e : SUCCESS in 11m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3af76ed8e5c64ed4bb62295922efcb4d : SUCCESS in 10m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9b4d83f8a26b4fbc9cd43d8f595ca56a : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8ec75462634f8fbe2cde77e1cb5c60 : SUCCESS in 6m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9998c0b226be4bcfa2c6bcc065053f97 : SUCCESS in 8m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/458f1363c56e49de82a5abdf77e8c51e : FAILURE in 5m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a326e5f7f05f4815b0b0fc4518e6406a : SUCCESS in 11m 12s","accounts_in_message":[],"_revision_number":5},{"id":"4c831adf074427ee27b295f091bc0840b0aa6c5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:36:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"dff7275eda116a5d65e0e5710114f6ccb1a089a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:38:23.000000000","message":"Patch Set 5:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84fad7b83af14d3082ff9ab1e9c4cec3 : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"872b9ed083a3fa5052d013148960c33afd3ec2cc","revisions":{"f07939c3f96b31b51a617ef2d764c85e45c89243":{"kind":"REWORK","_number":1,"created":"2021-06-28 13:12:48.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/98/798298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/98/798298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/1"}}},"commit":{"parents":[{"commit":"9ef63ea47bf64150878920444fb059349af5fef0","subject":"[rocky-only] Constrain l2gw to Rocky version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/9ef63ea47bf64150878920444fb059349af5fef0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-06-28 13:14:05.000000000","tz":540},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n(cherry picked from commit 1e3a1230fd9270280faa16336690d3c9f00a9cb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/f07939c3f96b31b51a617ef2d764c85e45c89243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/f07939c3f96b31b51a617ef2d764c85e45c89243"}]},"branch":"refs/heads/stable/rocky"},"901e5375be69d174c1fff187e0bfef34bb89da20":{"kind":"REWORK","_number":2,"created":"2021-06-28 14:54:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/798298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/98/798298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/2"}}},"commit":{"parents":[{"commit":"9ef63ea47bf64150878920444fb059349af5fef0","subject":"[rocky-only] Constrain l2gw to Rocky version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/9ef63ea47bf64150878920444fb059349af5fef0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-06-28 14:54:29.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n(cherry picked from commit 1e3a1230fd9270280faa16336690d3c9f00a9cb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/901e5375be69d174c1fff187e0bfef34bb89da20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/901e5375be69d174c1fff187e0bfef34bb89da20"}]},"branch":"refs/heads/stable/rocky"},"c919a067f0ce215e13af561108cfa4958d6e84f8":{"kind":"REWORK","_number":3,"created":"2021-06-28 16:33:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/798298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/98/798298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/3"}}},"commit":{"parents":[{"commit":"9ef63ea47bf64150878920444fb059349af5fef0","subject":"[rocky-only] Constrain l2gw to Rocky version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/9ef63ea47bf64150878920444fb059349af5fef0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-06-28 16:33:56.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n(cherry picked from commit 1e3a1230fd9270280faa16336690d3c9f00a9cb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c919a067f0ce215e13af561108cfa4958d6e84f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c919a067f0ce215e13af561108cfa4958d6e84f8"}]},"branch":"refs/heads/stable/rocky"},"8244586d56da1afeded40556efa9643b1aca7496":{"kind":"REWORK","_number":4,"created":"2021-06-30 06:56:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/798298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/98/798298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/4"}}},"commit":{"parents":[{"commit":"9ef63ea47bf64150878920444fb059349af5fef0","subject":"[rocky-only] Constrain l2gw to Rocky version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/9ef63ea47bf64150878920444fb059349af5fef0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-06-30 06:55:51.000000000","tz":120},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nMake functional tests non-voting now to keep the gating work for older\nbranches.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n(cherry picked from commit 1e3a1230fd9270280faa16336690d3c9f00a9cb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8244586d56da1afeded40556efa9643b1aca7496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8244586d56da1afeded40556efa9643b1aca7496"}]},"branch":"refs/heads/stable/rocky"},"872b9ed083a3fa5052d013148960c33afd3ec2cc":{"kind":"REWORK","_number":5,"created":"2021-06-30 09:35:00.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/798298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/98/798298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/98/798298/5"}}},"commit":{"parents":[{"commit":"9ef63ea47bf64150878920444fb059349af5fef0","subject":"[rocky-only] Constrain l2gw to Rocky version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/9ef63ea47bf64150878920444fb059349af5fef0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 15:34:11.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-06-30 09:28:34.000000000","tz":120},"subject":"Dropping lower constraints testing (stable Stein)","message":"Dropping lower constraints testing (stable Stein)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\nFor stable/rocky and older in Neutron meeting ([3]) the decision was to\nmake non-voting and faulty jobs (see [4]) dropped.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n[3]https://meetings.opendev.org/irclogs/%23openstack-neutron/%23openstack-neutron.2021-06-29.log.html#t2021-06-29T15:13:37\n[4]http://lists.openstack.org/pipermail/openstack-discuss/2021-June/023321.html\n\nUse the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nnetworking-l2gw location change original-Change-Id:\nI38509a2fafc07e5196ba9f8f1dc11e617cbc74f5\n(cherry picked from commit bcac237d81a56f9ad61df0256e0b706ee95e0a0e)\n\nChange-Id: Icccc14c49f5fcb6961adb0eb5fb99dff63ec102b\n(cherry picked from commit 1e3a1230fd9270280faa16336690d3c9f00a9cb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/872b9ed083a3fa5052d013148960c33afd3ec2cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/872b9ed083a3fa5052d013148960c33afd3ec2cc"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
