)]}'
{"id":"openstack%2Fnetworking-odl~866849","triplet_id":"openstack%2Fnetworking-odl~master~I3362c6aecba6e4a292f0b117ce0326bd08628bf9","project":"openstack/networking-odl","branch":"master","topic":"functional_timeout","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-12-15 16:23:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3362c6aecba6e4a292f0b117ce0326bd08628bf9","subject":"CI: Increase the timeout of functional jobs","status":"MERGED","created":"2022-12-07 12:32:49.000000000","updated":"2022-12-15 16:26:01.000000000","submitted":"2022-12-15 16:23:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866849-functional_timeout","meta_rev_id":"5c4b5e9bfb34a8693720538e8a56197148a45d61","_number":866849,"virtual_id_number":866849,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-15 16:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-12-15 15:42:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-12-15 15:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-12-15 15:43:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 13:22:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-15 15:42:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-12-15 15:43:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"27904c54397be3d900b855346f9ad800cee070d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-12-07 12:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88a656a0514ba7315de6a295e2775c7ef096c498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:22:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c3d0bba551540b0bcd9b8561a1c7e0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48f883e0d0f45a8a47d5d3a04ae246a : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac624ecee183454fb4d750347890cd02 : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15f7818ed6924ed5af7cafba726715ba : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03eaef9f649643b1a978465ed0a6da9b : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a78aab3fec89450d808533d63108aa97 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd9b6f11e0241db975957d18f3adedb : SUCCESS in 6m 39s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/566c9f6abe9249fe8e7e958b5f0669e0 : FAILURE in 39m 47s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/672d650acf8c4c61b27f49fda5f41da8 : FAILURE in 32m 01s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/f67c049c644f447bb4c36ccf3442eb9e : FAILURE in 32m 11s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/f0bcf21a58474c788791eff15cdf6465 : POST_FAILURE in 38m 16s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/4f7d7cb041a54d98ab9794c70fc5aaf5 : POST_FAILURE in 31m 30s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/d266d16099004df1956bd6757f782254 : SUCCESS in 44m 54s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/7d05763189614e70a21e5c1084e3065e : SUCCESS in 47m 27s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/6f49611110cf4d7a8f0f6c70935c8bcb : SUCCESS in 16m 06s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/5aeaf53429194f35b7a931b42582548c : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1f25419860b40d5d59ec57a79f0b2ec8d8056cc4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-15 15:42:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa85406114587f18361882ab4100a9e10cddeb74","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-12-15 15:43:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73fddd6bf95cd7e1bc1daf69b32e2c68d5af50ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:44:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a076c37c4e6b42200ecb02f5a6dd2462c0bed55b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:23:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb006ff86a094bbd83bee7168d30fdd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66fec1901c8447c786bcaec7c2d210e6 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/671d5b485fa64c7eb7c11fd3a6e31256 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15161afcc9bf4997b0ebccce144d6d00 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc15df9444f44468f502d04d3216aa2 : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2224644b1ec43f798e77fed479bcced : SUCCESS in 7m 36s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/a6d55911e126451584a51286315f641f : SUCCESS in 23m 05s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/fafc262379d64f47b19b3da2bbe6e649 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3d42d9293fcd8628e1e10b427a6beb1d796a3cfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:23:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c4b5e9bfb34a8693720538e8a56197148a45d61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:26:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16083433bf354bd5837d2ba8c8407bd0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad7752ddc3d4005810778388f452ad7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eecb7125a630c5790852739b8bad0306f1b3aa59","revisions":{"eecb7125a630c5790852739b8bad0306f1b3aa59":{"kind":"REWORK","_number":1,"created":"2022-12-07 12:32:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/49/866849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/49/866849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/49/866849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/49/866849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/49/866849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/49/866849/1"}}},"commit":{"parents":[{"commit":"e44ff511447743936f3dbf0c3b695935d9c8f4e0","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e44ff511447743936f3dbf0c3b695935d9c8f4e0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-12-07 12:31:03.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-12-07 12:31:03.000000000","tz":60},"subject":"CI: Increase the timeout of functional jobs","message":"CI: Increase the timeout of functional jobs\n\nBoth functional jobs have 30 minutes timeout, whichsometimes is\ntoo short, let\u0027s increase it to 45 minutes.\n\nChange-Id: I3362c6aecba6e4a292f0b117ce0326bd08628bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/eecb7125a630c5790852739b8bad0306f1b3aa59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/eecb7125a630c5790852739b8bad0306f1b3aa59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
