)]}'
{"id":"openstack%2Fnetworking-odl~879334","triplet_id":"openstack%2Fnetworking-odl~master~Ib88acb90258da0b7e96fabea2c40f4ab91232740","project":"openstack/networking-odl","branch":"master","topic":"bug/1989718","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-06-17 09:48:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib88acb90258da0b7e96fabea2c40f4ab91232740","subject":"Add context wrapper around periodic task methods","status":"ABANDONED","created":"2023-04-03 11:42:01.000000000","updated":"2023-06-17 09:48:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"37ce3dbe06949a4bf511669335f4e3e01febb9ed","_number":879334,"virtual_id_number":879334,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-13 09:09:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 12:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-17 09:48:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"0e49c7daf0b46a09faf1d60f08a914056619ac01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-03 11:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f8eb6aab4ded8a48478e31ac47d83a3be5f8d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:46:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a491e6005a5a446bb4ce9fe93202a501\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b6feeb01b4480fb1930d8bb982b95e : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2afc6be40f974813b05eedfca2adb1ef : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f184916023449d7b4cd148e362e053e : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dedc62e071b84409b59599f0c36fc4f8 : FAILURE in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa9d7579bc6d4a57b64695133bd641de : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ae152f4b240414c90e1d8c9646592d9 : SUCCESS in 6m 46s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/fc9952043dd2406d8ea3d77ae27d4396 : FAILURE in 44m 45s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/4cbd9f6ab2f6444c891a91101e521705 : FAILURE in 56m 28s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/90640de68a864c6f85ecff2028f04390 : FAILURE in 55m 32s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/3aa3c44dbcb845579c886743c9f5a814 : POST_FAILURE in 30m 23s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/51c4376a615348489ccfce11f0ab529c : POST_FAILURE in 35m 28s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/26c32552767d4ec4aa47e49b08b815f7 : SUCCESS in 39m 09s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/ea2e78d0ae244958aff481ed5a92b50a : SUCCESS in 59m 15s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/1da40038106740beb9198eb38748dd3c : SUCCESS in 17m 01s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/eaed7d9f97814cbd9bddbb0648c70905 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2621893356ed6ac0bedddc593da34c2aab96d9f2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-03 13:14:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ec011be873d7ab496731b3ea4ba83cd7b6767c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 14:14:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e02bf1cfaafe4df5a62d36be7e8f5fe1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a7d18e8e91642c5b6c865667c3b1acc : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25fae90840a343f88a404b583d0c83f1 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b3484b98a9f4713812a942ca9e0c8cc : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bace6b1c7894caeb63c4d51c0223672 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6fba569a9046a0a9d5a970274f62c9 : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7714996c5a3a4d5f9854ba8c88abd85d : SUCCESS in 7m 50s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/cc8cf323d94644eabc3133c35cb8839f : FAILURE in 42m 46s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/6cef400407894ae3b10e04d5cf610e4d : FAILURE in 46m 03s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/23a7513dbcab4451b8ec3dd116af6141 : FAILURE in 54m 11s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/807f0d458dc64bdeb5d859d1bddb1cb8 : POST_FAILURE in 33m 52s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/6b15d9a366814b5ba5a7e98e1266461e : POST_FAILURE in 41m 07s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/f5c82617b82448d19ddac4dd99dec955 : SUCCESS in 53m 33s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/44fb813a6dc94895a9c6c951d80952ec : SUCCESS in 53m 05s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/1975b05738c54c0399d4b6077a5980ef : SUCCESS in 25m 33s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/639823c6b22b4e06b75e736459cf8fc7 : SUCCESS in 27m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f89ce1e29e6bd2aee235dfcd288c161f91d65ed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-13 08:02:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"13124c7c3d6dca396915e742d4ded41c91890872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 09:09:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51ef39cc1f9746efa088f12e346c5575\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d56250adda641b2859287804295ff57 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12f280e2c65942508be665591f58d1de : FAILURE in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc9d751412824a8a945eb440a5f83aa0 : FAILURE in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5202ce81e364ceea846ca92eec5a65a : FAILURE in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a66917fa9b94e429ef85607dfd338a1 : SUCCESS in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2ea9bf575ad450491b123b2b8704483 : FAILURE in 6m 50s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/121ce6cee30a43eaa6bed420b6d270e7 : FAILURE in 43m 37s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/5a061be926704d788aff8603688ed160 : FAILURE in 48m 34s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/a1db879f9a944b5390dd6297e4be8424 : FAILURE in 58m 02s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/9a3014c6458a4d24b0d556cceb57081c : POST_FAILURE in 44m 04s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/855c171d7ba74386b42d32c033cd287f : POST_FAILURE in 42m 36s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/d4350e92188c4f1399693b491d58e484 : SUCCESS in 51m 43s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/113f2db9536b436394ca85eac69e7dd4 : SUCCESS in 59m 04s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/3f1eb95add8d48f090239ed30c81bebc : SUCCESS in 24m 04s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/eb8c88fb5a504d59a879574966c0848a : SUCCESS in 26m 03s","accounts_in_message":[],"_revision_number":2},{"id":"37ce3dbe06949a4bf511669335f4e3e01febb9ed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-17 09:48:18.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2cae4aceed98b0f27d75cc4dbdf8db275e6e935","revisions":{"ad1cad5d8441d9a764223999fce447bbe2cb16e8":{"kind":"REWORK","_number":1,"created":"2023-04-03 11:42:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/34/879334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/34/879334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/1"}}},"commit":{"parents":[{"commit":"db5c79b3ee5054feb8a17df130e4ce3a95ec64c2","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/db5c79b3ee5054feb8a17df130e4ce3a95ec64c2"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-29 09:15:46.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-03 11:38:17.000000000","tz":120},"subject":"Add context wrapper around periodic task methods","message":"Add context wrapper around periodic task methods\n\nupdate_periodic_task and _update_periodic_task_state lacked\ncontext wrapper and this resulted in ugly tracebacks in CI logs\nalso.\nAs an extra use exc_to_retry from neutron-lib db_api in journal\u0027s\nrecord method.\n\nRelated-Bug: #1989718\nChange-Id: Ib88acb90258da0b7e96fabea2c40f4ab91232740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/ad1cad5d8441d9a764223999fce447bbe2cb16e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/ad1cad5d8441d9a764223999fce447bbe2cb16e8"}]},"branch":"refs/heads/master"},"b2cae4aceed98b0f27d75cc4dbdf8db275e6e935":{"kind":"REWORK","_number":2,"created":"2023-04-13 08:02:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/34/879334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/34/879334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/34/879334/2"}}},"commit":{"parents":[{"commit":"db5c79b3ee5054feb8a17df130e4ce3a95ec64c2","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/db5c79b3ee5054feb8a17df130e4ce3a95ec64c2"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-03-29 09:15:46.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-13 08:01:41.000000000","tz":120},"subject":"Add context wrapper around periodic task methods","message":"Add context wrapper around periodic task methods\n\nupdate_periodic_task and _update_periodic_task_state lacked\ncontext wrapper and this resulted in ugly tracebacks in CI logs\nalso.\nAs an extra use exc_to_retry from neutron-lib db_api in journal\u0027s\nrecord method.\nTODO: finalize commit msg when things are finished and working\n\nRelated-Bug: #1989718\nChange-Id: Ib88acb90258da0b7e96fabea2c40f4ab91232740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b2cae4aceed98b0f27d75cc4dbdf8db275e6e935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b2cae4aceed98b0f27d75cc4dbdf8db275e6e935"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
