)]}'
{"id":"openstack%2Fnetworking-odl~884191","triplet_id":"openstack%2Fnetworking-odl~stable%2Fxena~Icff1b5737b4d2c52ead48e3700de668038f3c5c1","project":"openstack/networking-odl","branch":"stable/xena","topic":"remove-odl-periodic-stable","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-02 09:38:34.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-06-02 08:54:10.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"Icff1b5737b4d2c52ead48e3700de668038f3c5c1","subject":"Remove periodic-stable-jobs template","status":"MERGED","created":"2023-05-24 12:40:18.000000000","updated":"2023-06-02 09:40:21.000000000","submitted":"2023-06-02 09:38:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884191-remove-odl-periodic-stable","meta_rev_id":"a8e3bc60f026478def8d3a110398f9dc31cce3d9","_number":884191,"virtual_id_number":884191,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-02 09:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-06-02 08:53:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-06-02 08:54:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 14:08:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 13:41:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"04d56e28388b7469213f57cc84734433fdedc566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-05-24 12:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d1a764287e29f4399e91c8a31add795c9b4cc7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 14:08:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/585f1e5d33bf4393bd216e46f862ad06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19239ada66f4ffba4d201ec885a8eb3 : FAILURE in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7001d725ca064a608a228c9e2d5058d9 : FAILURE in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c36d4ca3abb4a86bf0b78f1d400a9a5 : FAILURE in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19f8468b6a0742f1b43cb6012d48bbb8 : FAILURE in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055cee42f91c488a97c54fa0e4340e67 : FAILURE in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5efc56b8364a4a87a787e7fccb4d865c : FAILURE in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dac1faf20b28463aa46ae47d883e42f9 : FAILURE in 4m 04s (non-voting)\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/cd953a69e7414b208ca8c2c3f15678f2 : FAILURE in 52m 34s (non-voting)\n- networking-odl-tempest-neon https://zuul.opendev.org/t/openstack/build/a73fecd7272b41d0a54a4293024705ce : POST_FAILURE in 21m 33s (non-voting)\n- networking-odl-tempest-sodium https://zuul.opendev.org/t/openstack/build/795ebcd981ec4db8ba97c7a00e04680e : POST_FAILURE in 25m 46s (non-voting)\n- networking-odl-tempest-neon-multinode https://zuul.opendev.org/t/openstack/build/2198975722dd48f2b2a546b73ce762dd : FAILURE in 22m 28s (non-voting)\n- networking-odl-tempest-sodium-multinode https://zuul.opendev.org/t/openstack/build/442b408330d24dc3989fd8a96909ae33 : FAILURE in 17m 06s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/c61d6c8def304391a6f70a077b6e625b : FAILURE in 22m 45s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/007b0c3544e1455c84b5328a31991549 : FAILURE in 20m 19s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/b1979d4d39f94b6cbcac6fd9bd02a977 : RETRY_LIMIT in 8m 26s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/fc3928c3c6194676b0ec7745e86c281e : RETRY_LIMIT in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"049f8ec0204279278dbf8891f7530beb14f2b364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-01 13:27:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d06c1516168d23d1236ff0eebb650dc242a7b08b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-01 13:41:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cff8e2a261244db3f5e3c69789cef287764ea5f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 14:19:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afc9ed842f024ae8b18a0ff217355deb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9893c9909443dfa22e89d168039703 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29b4f1a8a5d1406dad5dff6175beb4ac : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b15d62f5f2d4dbca54428226293778d : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73b31b01d4244b00beb9660c9f1526fe : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e026466061fd4d169ebda49cfcb512bb : SUCCESS in 10m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/410a56dfbcaf4600aa9f14702a1134f8 : SUCCESS in 8m 01s\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/6fb8d56ef8b049eab0f04e124b8d42d1 : FAILURE in 23m 05s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/c30a73741ce74fe5919990bb86c85ab3 : FAILURE in 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2173eb826c369b8880af4d1acd969173d628991f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-01 18:46:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08730e265cc48eaeb7778d7814d1a7e4d96fb375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-02 08:53:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"af2d2719b6e9d9be7697166b0aeec28252ebe109","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-02 08:54:10.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c1221a6936a7e4791f9de0d2d6a5f6107b58904d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 09:25:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/390ba84827a546cf819d7ab91cbb9172\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81397cf13d1c4e5595226dfb06026a99 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/748bf0e25ec948eeb767c794ef396df2 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50ebd3a08f34e01bffc5960f8e6b0c9 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b7f0da6e6284e1399dcf3b275dc2ad7 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6d6be9d09641e4a8d741530be691c2 : SUCCESS in 9m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad40d7fd30984be18b663adc9525260d : SUCCESS in 8m 01s\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/d4060e5c71734a8da1d96ee516f3e380 : FAILURE in 21m 15s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/8eb6a6457d1c43b5a492924b1b36be1b : FAILURE in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f2855bb2e8b28d0624fd9b5327698105e0ad3416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 09:26:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0feabdfae62b2db6ed06ce86f3c7187a8d2fb1f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 09:38:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5de9c9e592f7420e8e9133a258ee76b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3694adc88e474afdbf6ded6ef185d9be : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3b04ea8d20349d1b6c63cd1b3d81ca2 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7172bdc404104e3093e6500851d4b813 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fed5d33a324ce1aa8df0cd2cdb122c : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3b4490eeb10443a917252f527270506 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":3},{"id":"06aee6817222cb09fcadd880b03f31855fcd8042","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 09:38:34.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"a8e3bc60f026478def8d3a110398f9dc31cce3d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 09:40:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e686caf788d44d3bf4ad5ff11aed000\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09987f2953104d8ea33a2cca8947f117 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df16b1b4477463919587efcd83dd3bf58a268b01","revisions":{"7bbef0e1fecb9a988b6e59f10b41144ef7e7933c":{"kind":"REWORK","_number":1,"created":"2023-05-24 12:40:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/91/884191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/884191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/1"}}},"commit":{"parents":[{"commit":"d90a8f3aeac6d6a4477ffbf2ac86c167c2fb7304","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d90a8f3aeac6d6a4477ffbf2ac86c167c2fb7304"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-05-24 12:17:06.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-05-24 12:39:52.000000000","tz":120},"subject":"Remove periodic-stable-jobs template","message":"Remove periodic-stable-jobs template\n\nnetworking-odl is about to become deprecated [1] (furthermore, periodic\nstable jobs on yoga and older branches are failing for long time), so\nit is unnecessary to run periodic stable jobs against networking-odl\nanymore.\n\n[1] https://review.opendev.org/c/openstack/governance/+/884002\n\nChange-Id: Icff1b5737b4d2c52ead48e3700de668038f3c5c1\n(cherry picked from commit 0e6970ce0450a76bd0c115d65eff346f762cfc8e)\n(cherry picked from commit 439a2437385a9a948ccec234d79f63e7422b9fae)\n(cherry picked from commit 9b01ad6c9d391cd9da97d27f7f5e402eecd4a239)\n(cherry picked from commit 29a53d72c54ac29dd9bf765f0bb0681a8e93fa7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7bbef0e1fecb9a988b6e59f10b41144ef7e7933c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/7bbef0e1fecb9a988b6e59f10b41144ef7e7933c"}]},"branch":"refs/heads/stable/xena"},"d64be5f973098db4982e3308286477c7faecb309":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-01 13:27:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/884191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/884191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/2"}}},"commit":{"parents":[{"commit":"73df772f9318a82741513956bf4644977b70ad89","subject":"Fix tox4 issues and define project upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/73df772f9318a82741513956bf4644977b70ad89"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-05-24 12:17:06.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-06-01 13:27:35.000000000","tz":0},"subject":"Remove periodic-stable-jobs template","message":"Remove periodic-stable-jobs template\n\nnetworking-odl is about to become deprecated [1] (furthermore, periodic\nstable jobs on yoga and older branches are failing for long time), so\nit is unnecessary to run periodic stable jobs against networking-odl\nanymore.\n\n[1] https://review.opendev.org/c/openstack/governance/+/884002\n\nChange-Id: Icff1b5737b4d2c52ead48e3700de668038f3c5c1\n(cherry picked from commit 0e6970ce0450a76bd0c115d65eff346f762cfc8e)\n(cherry picked from commit 439a2437385a9a948ccec234d79f63e7422b9fae)\n(cherry picked from commit 9b01ad6c9d391cd9da97d27f7f5e402eecd4a239)\n(cherry picked from commit 29a53d72c54ac29dd9bf765f0bb0681a8e93fa7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d64be5f973098db4982e3308286477c7faecb309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d64be5f973098db4982e3308286477c7faecb309"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"df16b1b4477463919587efcd83dd3bf58a268b01":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-02 08:53:15.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/91/884191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/91/884191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/91/884191/3"}}},"commit":{"parents":[{"commit":"284099a77bd2f1c2f2f598d84f48b14e2364992b","subject":"Fix tox4 issues and define project upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/284099a77bd2f1c2f2f598d84f48b14e2364992b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-05-24 12:17:06.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-06-02 08:53:15.000000000","tz":0},"subject":"Remove periodic-stable-jobs template","message":"Remove periodic-stable-jobs template\n\nnetworking-odl is about to become deprecated [1] (furthermore, periodic\nstable jobs on yoga and older branches are failing for long time), so\nit is unnecessary to run periodic stable jobs against networking-odl\nanymore.\n\n[1] https://review.opendev.org/c/openstack/governance/+/884002\n\nChange-Id: Icff1b5737b4d2c52ead48e3700de668038f3c5c1\n(cherry picked from commit 0e6970ce0450a76bd0c115d65eff346f762cfc8e)\n(cherry picked from commit 439a2437385a9a948ccec234d79f63e7422b9fae)\n(cherry picked from commit 9b01ad6c9d391cd9da97d27f7f5e402eecd4a239)\n(cherry picked from commit 29a53d72c54ac29dd9bf765f0bb0681a8e93fa7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/df16b1b4477463919587efcd83dd3bf58a268b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/df16b1b4477463919587efcd83dd3bf58a268b01"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
