)]}'
{"id":"openstack%2Fnetworking-ovn~168021","triplet_id":"openstack%2Fnetworking-ovn~master~I081a6c4225f4046d299df3a26bffff5e132153c1","project":"openstack/networking-ovn","branch":"master","topic":"fira1","hashtags":[],"change_id":"I081a6c4225f4046d299df3a26bffff5e132153c1","subject":"Fix attribute access","status":"MERGED","created":"2015-03-26 14:48:05.000000000","updated":"2015-03-26 16:37:33.000000000","submitted":"2015-03-26 16:37:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"55f0ca9215a6d03cafcac2b3f477857c1682261b","_number":168021,"virtual_id_number":168021,"owner":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2015-03-26 16:37:31.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-03-26 14:52:22.000000000","_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-03-26 16:34:53.000000000","_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-03-26 16:34:53.000000000","updated_by":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"reviewer":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"state":"REVIEWER"},{"updated":"2015-03-26 16:37:31.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1ed2f1482c890488bd4efa297728ae2b3a04ba23","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2015-03-26 14:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62c0e0bacdb1ec7aecb337b7ba636fccaaee47a4","author":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"date":"2015-03-26 14:52:22.000000000","message":"Patch Set 1: Code-Review+2\n\nWaiting on Jenkins to approve.","accounts_in_message":[],"_revision_number":1},{"id":"12944fea9c8a0f8024f46ed5739988fd39b5a268","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-26 16:31:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-ovn-pep8 http://logs.openstack.org/21/168021/1/check/gate-networking-ovn-pep8/bbcbe3c/ : SUCCESS in 1m 32s\n- gate-networking-ovn-docs http://docs-draft.openstack.org/21/168021/1/check/gate-networking-ovn-docs/eb857eb//doc/build/html/ : SUCCESS in 1m 46s\n- gate-networking-ovn-python27 http://logs.openstack.org/21/168021/1/check/gate-networking-ovn-python27/1dc3999/ : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"36381bc2627e2ff69d040149bc519a7dac4a0f77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-26 16:34:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"a6c356f67715b2661c0ed6a027158ec554dc5650","author":{"_account_id":105,"name":"Kyle Mestery","email":"mestery@mestery.com","username":"mestery"},"date":"2015-03-26 16:34:53.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f094f32b5d4488b6e666805373326c4f8c27f7c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-26 16:37:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-networking-ovn-docs http://docs-draft.openstack.org/21/168021/1/gate/gate-networking-ovn-docs/9a29e58//doc/build/html/ : SUCCESS in 1m 42s\n- gate-networking-ovn-pep8 http://logs.openstack.org/21/168021/1/gate/gate-networking-ovn-pep8/1315dfc/ : SUCCESS in 1m 20s\n- gate-networking-ovn-python27 http://logs.openstack.org/21/168021/1/gate/gate-networking-ovn-python27/9c320e9/ : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6ad8b91783c2388d1dceef97534d103b0bda9c6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-03-26 16:37:33.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"836867e1463516ef22e487d5081f9e9fbbf8ae10","revisions":{"836867e1463516ef22e487d5081f9e9fbbf8ae10":{"kind":"REWORK","_number":1,"created":"2015-03-26 14:48:05.000000000","uploader":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"ref":"refs/changes/21/168021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/21/168021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/21/168021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/21/168021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/21/168021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/21/168021/1"}}},"commit":{"parents":[{"commit":"4e47753c9720fe6b6c4b652fb08a8b3038172961","subject":"Add CRUD for ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4e47753c9720fe6b6c4b652fb08a8b3038172961"}]}],"author":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2015-03-26 14:43:20.000000000","tz":-420},"committer":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2015-03-26 14:48:01.000000000","tz":-420},"subject":"Fix attribute access","message":"Fix attribute access\n\nEnsure that network and port access does not have an exception.\n\nChange-Id: I081a6c4225f4046d299df3a26bffff5e132153c1\nCloses-bug: #1436872\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/836867e1463516ef22e487d5081f9e9fbbf8ae10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/836867e1463516ef22e487d5081f9e9fbbf8ae10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4e47753c9720fe6b6c4b652fb08a8b3038172961","is_merged_in_target_branch":true,"change_id":"Idaccb7be516e2bd0d58fee0044c3a32ede1250a0","change_number":166851,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
