)]}'
{"id":"openstack%2Fnetworking-ovn~449621","triplet_id":"openstack%2Fnetworking-ovn~master~Ife416526c85ccf37138c4c596bbc649801a1e7ac","project":"openstack/networking-ovn","branch":"master","hashtags":[],"change_id":"Ife416526c85ccf37138c4c596bbc649801a1e7ac","subject":"Fix some reST field lists in docstrings","status":"MERGED","created":"2017-03-24 13:27:53.000000000","updated":"2017-03-27 16:23:45.000000000","submitted":"2017-03-27 16:23:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a192fe45172ea586251ddbba71b44a15c4932a97","_number":449621,"virtual_id_number":449621,"owner":{"_account_id":23823,"name":"Duan Jiong","email":"jduan@fiberhome.com","username":"duanjiong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-03-27 16:23:45.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-03-27 15:17:45.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":2,"date":"2017-03-27 08:55:41.000000000","_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-03-27 15:17:45.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-27 08:55:41.000000000","updated_by":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2017-03-27 15:17:45.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2017-03-27 16:23:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"d38f9b57500ddad7101d7d739f438087ba70e163","author":{"_account_id":23823,"name":"Duan Jiong","email":"jduan@fiberhome.com","username":"duanjiong"},"date":"2017-03-24 13:27:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a8455d6943c4438fee45ab743b3789a3ac67745","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-24 14:35:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-ovn-docs-ubuntu-xenial http://docs-draft.openstack.org/21/449621/1/check/gate-networking-ovn-docs-ubuntu-xenial/9b09cf0//doc/build/html/ : SUCCESS in 3m 39s\n- gate-networking-ovn-pep8-ubuntu-xenial http://logs.openstack.org/21/449621/1/check/gate-networking-ovn-pep8-ubuntu-xenial/cc8d721/ : SUCCESS in 4m 33s\n- gate-networking-ovn-python27-ubuntu-xenial http://logs.openstack.org/21/449621/1/check/gate-networking-ovn-python27-ubuntu-xenial/d126f11/ : SUCCESS in 9m 00s\n- gate-networking-ovn-python35 http://logs.openstack.org/21/449621/1/check/gate-networking-ovn-python35/d2df302/ : SUCCESS in 7m 33s\n- gate-tempest-dsvm-networking-ovn-ovs-master-nv http://logs.openstack.org/21/449621/1/check/gate-tempest-dsvm-networking-ovn-ovs-master-nv/98f0e98/ : FAILURE in 51m 25s (non-voting)\n- gate-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/21/449621/1/check/gate-tempest-dsvm-networking-ovn-ovs-release/0d6c90e/ : SUCCESS in 43m 50s\n- gate-rally-dsvm-networking-ovn http://logs.openstack.org/21/449621/1/check/gate-rally-dsvm-networking-ovn/5909f0e/ : SUCCESS in 1h 05m 12s\n- gate-networking-ovn-dsvm-functional http://logs.openstack.org/21/449621/1/check/gate-networking-ovn-dsvm-functional/14a71fe/ : SUCCESS in 8m 18s\n- gate-networking-ovn-dsvm-functional-py35 http://logs.openstack.org/21/449621/1/check/gate-networking-ovn-dsvm-functional-py35/6a513d4/ : SUCCESS in 16m 36s\n- gate-install-dsvm-networking-ovn-kuryr http://logs.openstack.org/21/449621/1/check/gate-install-dsvm-networking-ovn-kuryr/9e2b736/ : SUCCESS in 35m 28s\n- networking-ovn-coverage-ubuntu-xenial-nv http://logs.openstack.org/21/449621/1/check/networking-ovn-coverage-ubuntu-xenial-nv/e57475d/ : SUCCESS in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d34843e14be5a88d0c6506e2e936987fefeee30","author":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"date":"2017-03-27 08:55:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d48072035a2e522f23c2e4b84762cee25c07afb5","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2017-03-27 15:17:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f70dea978078d7f46690f5fb2ff6aa9941cdcb89","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-27 15:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"24b689d4ecafe9103fe6b6689dd78048bea93ad5","date":"2017-03-27 16:23:45.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"a192fe45172ea586251ddbba71b44a15c4932a97","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-27 16:23:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-networking-ovn-docs-ubuntu-xenial http://docs-draft.openstack.org/21/449621/1/gate/gate-networking-ovn-docs-ubuntu-xenial/3312ba0//doc/build/html/ : SUCCESS in 4m 49s\n- gate-networking-ovn-pep8-ubuntu-xenial http://logs.openstack.org/21/449621/1/gate/gate-networking-ovn-pep8-ubuntu-xenial/69b695f/ : SUCCESS in 3m 30s\n- gate-networking-ovn-python27-ubuntu-xenial http://logs.openstack.org/21/449621/1/gate/gate-networking-ovn-python27-ubuntu-xenial/c8156d7/ : SUCCESS in 8m 05s\n- gate-networking-ovn-python35 http://logs.openstack.org/21/449621/1/gate/gate-networking-ovn-python35/b2e86fa/ : SUCCESS in 7m 10s\n- gate-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/21/449621/1/gate/gate-tempest-dsvm-networking-ovn-ovs-release/41a93b8/ : SUCCESS in 54m 23s\n- gate-rally-dsvm-networking-ovn http://logs.openstack.org/21/449621/1/gate/gate-rally-dsvm-networking-ovn/d2eaa9e/ : SUCCESS in 1h 04m 39s\n- gate-networking-ovn-dsvm-functional http://logs.openstack.org/21/449621/1/gate/gate-networking-ovn-dsvm-functional/4415697/ : SUCCESS in 8m 19s\n- gate-networking-ovn-dsvm-functional-py35 http://logs.openstack.org/21/449621/1/gate/gate-networking-ovn-dsvm-functional-py35/df00852/ : SUCCESS in 8m 19s\n- gate-install-dsvm-networking-ovn-kuryr http://logs.openstack.org/21/449621/1/gate/gate-install-dsvm-networking-ovn-kuryr/d11627f/ : SUCCESS in 25m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"786e819f2c5bbf3689e809002affefcb12f1cb30","revisions":{"786e819f2c5bbf3689e809002affefcb12f1cb30":{"kind":"REWORK","_number":1,"created":"2017-03-24 13:27:53.000000000","uploader":{"_account_id":23823,"name":"Duan Jiong","email":"jduan@fiberhome.com","username":"duanjiong"},"ref":"refs/changes/21/449621/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/21/449621/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/21/449621/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/21/449621/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/21/449621/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/21/449621/1"}}},"commit":{"parents":[{"commit":"3bc4a5367b759486aed10dfb9258598fe77dd103","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3bc4a5367b759486aed10dfb9258598fe77dd103"}]}],"author":{"name":"Duan Jiong","email":"jduan@fiberhome.com","date":"2017-03-24 13:22:38.000000000","tz":480},"committer":{"name":"Duan Jiong","email":"jduan@fiberhome.com","date":"2017-03-24 13:22:50.000000000","tz":480},"subject":"Fix some reST field lists in docstrings","message":"Fix some reST field lists in docstrings\n\nProbably the most common format for documenting arguments is reST field\nlists [1]. This change updates some docstrings to comply with the field\nlists syntax.\n\n[1] http://sphinx-doc.org/domains.html#info-field-lists\n\nChange-Id: Ife416526c85ccf37138c4c596bbc649801a1e7ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/786e819f2c5bbf3689e809002affefcb12f1cb30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/786e819f2c5bbf3689e809002affefcb12f1cb30"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3bc4a5367b759486aed10dfb9258598fe77dd103","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
