)]}'
{"id":"openstack%2Fnetworking-ovn~656907","triplet_id":"openstack%2Fnetworking-ovn~master~I87cd76ab72168083520885dd1d71465a2bc7f714","project":"openstack/networking-ovn","branch":"master","hashtags":[],"change_id":"I87cd76ab72168083520885dd1d71465a2bc7f714","subject":"Update device_owner for subports","status":"MERGED","created":"2019-05-03 00:21:42.000000000","updated":"2019-05-09 18:21:05.000000000","submitted":"2019-05-09 18:21:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656907-1557426065864-84f15ffa","meta_rev_id":"eb4da669bbafcf95b446e52062de18da3334b2f2","_number":656907,"virtual_id_number":656907,"owner":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-05-09 18:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"all":[{"value":1,"date":"2019-05-09 13:00:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":1,"date":"2019-05-09 13:00:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":2,"date":"2019-05-09 13:00:38.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-05-09 13:32:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-05-09 13:00:38.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-09 13:00:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":1,"date":"2019-05-09 13:00:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-05-09 13:01:45.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 15:09:35.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-05-03 15:43:34.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-05-03 15:47:27.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-05-03 17:21:10.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2019-05-04 04:16:25.000000000","updated_by":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"reviewer":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"state":"REVIEWER"},{"updated":"2019-05-09 13:01:45.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-05-09 13:32:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-09 18:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3051cf6ab644a2a90b770e4acd348c0a9f5df2f5","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-05-03 00:21:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc92130263c2c1e368fe731ef0825fbe425901da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 02:53:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/check/openstack-tox-pep8/a3bdabd/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/check/openstack-tox-py27/527a894/ : SUCCESS in 9m 27s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/check/openstack-tox-docs/f7c9a8e/html/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/check/openstack-tox-py36/c63945f/ : SUCCESS in 10m 08s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/check/openstack-tox-py37/7eae3c1/ : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/check/openstack-tox-lower-constraints/dfee1f5/ : SUCCESS in 11m 04s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-master/e5b6a79/ : SUCCESS in 2h 12m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release/db83afc/ : SUCCESS in 2h 14m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release-python2/d48d422/ : SUCCESS in 2h 26m 46s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/check/networking-ovn-rally-task/bdbe41b/ : SUCCESS in 1h 41m 26s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional-py27/81c0dcc/ : SUCCESS in 20m 54s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional/cd8ec38/ : SUCCESS in 20m 18s\n- networking-ovn-tripleo-c7-s003-oooq-containers http://logs.openstack.org/07/656907/1/check/networking-ovn-tripleo-c7-s003-oooq-containers/7b8682b/ : FAILURE in 1h 36m 54s (non-voting)\n- networking-ovn-dsvm-grenade http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-grenade/562e934/ : SUCCESS in 1h 55m 25s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/07/656907/1/check/openstack-tox-cover/297e5d7/ : FAILURE in 10m 03s (non-voting)\n- networking-ovn-tempest-slow http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-slow/9b2a019/ : SUCCESS in 1h 58m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8832b4c1df2e477bdb48de7f647ad18f36e85d6f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-05-03 15:09:35.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"7e7b4a100a5477913863efd250e74514c66b3f78","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-05-03 15:43:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2d147104bd217cba0b5f06d9625ee6fe553eff8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-03 15:47:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"50dadfc2074a19b4eda7df6070c307312d172791","author":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"date":"2019-05-03 17:21:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7dace466537ff6a7850d077c74a9236b7d9ff526","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-05-04 04:16:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d3c80cdbe3a44c38065e813aef843ebd460d74f5","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-08 12:47:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e84419d03a555857135b9c72e0b1c1397dddce19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 12:47:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"767d06cab26745b7d7f714561c46e319ea6ddbb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 15:58:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/gate/openstack-tox-pep8/19ab540/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py27/7da4bb8/ : SUCCESS in 9m 46s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/gate/openstack-tox-docs/c2d52ba/html/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py36/fb38cf9/ : SUCCESS in 11m 29s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py37/921d8e5/ : SUCCESS in 12m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/gate/openstack-tox-lower-constraints/1977797/ : SUCCESS in 13m 55s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release/0adf21b/ : SUCCESS in 2h 28m 22s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release-python2/3dae66e/ : SUCCESS in 2h 34m 55s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/gate/networking-ovn-rally-task/ca490ab/ : SUCCESS in 1h 40m 38s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional-py27/b94c32d/ : SUCCESS in 17m 30s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional/4169a75/ : FAILURE in 17m 57s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/1/gate/networking-ovn-tripleo-ci-centos-7-containers-multinode/a209351/ : SUCCESS in 2h 07m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ce4c23e1741a7e1b30fcace961530f13aafe8909","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-05-08 18:56:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"edd7409315b2377d75bb92d42e09ac5a76d46b0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 21:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/check/openstack-tox-pep8/a3c43e3/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/check/openstack-tox-py27/651f8ca/ : SUCCESS in 11m 08s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/check/openstack-tox-docs/86aadef/html/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/check/openstack-tox-py36/948dbc4/ : SUCCESS in 11m 40s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/check/openstack-tox-py37/890e696/ : SUCCESS in 12m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/check/openstack-tox-lower-constraints/db8b454/ : SUCCESS in 12m 42s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-master/aeaa06c/ : SUCCESS in 2h 18m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release/cbfb210/ : SUCCESS in 2h 40m 01s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release-python2/12a3f61/ : SUCCESS in 2h 24m 30s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/check/networking-ovn-rally-task/9841766/ : SUCCESS in 1h 53m 07s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional-py27/ae287a0/ : SUCCESS in 20m 05s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional/70e0f27/ : SUCCESS in 17m 39s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/1/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/e3edf4a/ : SUCCESS in 2h 18m 49s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-grenade/67f87bd/ : SUCCESS in 1h 54m 13s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/07/656907/1/check/openstack-tox-cover/8dfb6f5/ : FAILURE in 9m 44s (non-voting)\n- networking-ovn-tempest-slow http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-slow/f1c313a/ : SUCCESS in 1h 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbf3e3126b51abf9d922e1137c45d404fe715ed4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 21:39:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b108b7ab445f52014805db9c8fd4577f32b5633","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 00:18:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/gate/openstack-tox-pep8/5fd0b24/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py27/1f3cc4d/ : SUCCESS in 9m 55s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/gate/openstack-tox-docs/0335367/html/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py36/a613623/ : SUCCESS in 9m 18s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py37/866ad10/ : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/gate/openstack-tox-lower-constraints/e4c2155/ : SUCCESS in 10m 38s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release/0e39f08/ : FAILURE in 2h 22m 25s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release-python2/fd5c3ad/ : SUCCESS in 2h 29m 20s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/gate/networking-ovn-rally-task/5c1dc9e/ : SUCCESS in 1h 31m 56s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional-py27/220fe6c/ : SUCCESS in 16m 36s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional/08b23b3/ : SUCCESS in 18m 20s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/1/gate/networking-ovn-tripleo-ci-centos-7-containers-multinode/24c0c6d/ : SUCCESS in 2h 02m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e7aa47e6419391d00c7ddd3214674c777d8554e1","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-05-09 01:08:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb2e03112de34221941b2175167575a1be47ed4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 03:45:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/check/openstack-tox-pep8/d16defd/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/check/openstack-tox-py27/833bb67/ : SUCCESS in 8m 50s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/check/openstack-tox-docs/f08b110/html/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/check/openstack-tox-py36/509cad3/ : SUCCESS in 10m 55s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/check/openstack-tox-py37/4b27b3a/ : SUCCESS in 14m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/check/openstack-tox-lower-constraints/b8ec1c8/ : SUCCESS in 12m 03s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-master/c4ffcd4/ : SUCCESS in 2h 33m 06s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release/8d9afad/ : SUCCESS in 2h 30m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release-python2/7030cec/ : SUCCESS in 2h 35m 15s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/check/networking-ovn-rally-task/601be4b/ : SUCCESS in 1h 48m 29s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional-py27/8c2161c/ : SUCCESS in 28m 17s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional/26dfa5a/ : SUCCESS in 14m 26s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/1/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/634b62f/ : SUCCESS in 2h 05m 52s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-grenade/2e311eb/ : SUCCESS in 2h 10m 31s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/07/656907/1/check/openstack-tox-cover/aaff4f4/ : FAILURE in 9m 59s (non-voting)\n- networking-ovn-tempest-slow http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-slow/3c0db7f/ : SUCCESS in 2h 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a821afe259444a10b9b870037e3765f4780bb798","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 03:45:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bc8806d832a63ba830d965ae277a84c4c9bbd31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 05:51:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/gate/openstack-tox-pep8/8131dd3/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py27/b0dd588/ : SUCCESS in 9m 49s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/gate/openstack-tox-docs/e76ee5f/html/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py36/42e8fa5/ : SUCCESS in 9m 30s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py37/cd24a4a/ : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/gate/openstack-tox-lower-constraints/499b996/ : SUCCESS in 10m 57s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release/3575d61/ : FAILURE in 2h 03m 44s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release-python2/00ba247/ : SUCCESS in 2h 04m 57s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/gate/networking-ovn-rally-task/2b55e92/ : SUCCESS in 1h 28m 36s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional-py27/258a35f/ : SUCCESS in 21m 20s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional/4590653/ : SUCCESS in 14m 54s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/1/gate/networking-ovn-tripleo-ci-centos-7-containers-multinode/1a8348d/ : SUCCESS in 1h 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"25d2f511c18eda2172e3bd72d2e3bcf7fdae22d2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-05-09 07:06:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e018f938085733df9ac5f5ead9a0c277f7422898","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:26:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/check/openstack-tox-pep8/4220593/ : SUCCESS in 6m 55s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/check/openstack-tox-py27/b52429a/ : SUCCESS in 11m 15s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/check/openstack-tox-docs/c64ee06/html/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/check/openstack-tox-py36/7a3ab31/ : SUCCESS in 9m 52s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/check/openstack-tox-py37/e17fe2b/ : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/check/openstack-tox-lower-constraints/02a582f/ : SUCCESS in 10m 39s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-master/d81db58/ : SUCCESS in 2h 11m 11s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release/80d667f/ : SUCCESS in 2h 18m 45s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-dsvm-ovs-release-python2/c512c54/ : SUCCESS in 2h 10m 50s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/check/networking-ovn-rally-task/8dd211f/ : SUCCESS in 1h 44m 04s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional-py27/c055134/ : SUCCESS in 20m 04s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-functional/314e7d7/ : SUCCESS in 20m 45s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/1/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/940f0aa/ : SUCCESS in 1h 54m 53s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/07/656907/1/check/networking-ovn-dsvm-grenade/aeb2a55/ : SUCCESS in 1h 53m 30s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/07/656907/1/check/openstack-tox-cover/f175232/ : FAILURE in 10m 41s (non-voting)\n- networking-ovn-tempest-slow http://logs.openstack.org/07/656907/1/check/networking-ovn-tempest-slow/f0c77a7/ : SUCCESS in 2h 03m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c387b25ab3504a7980ac83c2210eb5fe5b8c10c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:26:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b8d093e5ed1d80a96413792b1af585101491906","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:06:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/1/gate/openstack-tox-pep8/0d782ce/ : FAILURE in 8m 05s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py27/9cbc882/ : SUCCESS in 11m 36s\n- openstack-tox-docs http://logs.openstack.org/07/656907/1/gate/openstack-tox-docs/8e33283/html/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py36/e35aba0/ : SUCCESS in 12m 30s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/1/gate/openstack-tox-py37/39dae05/ : SUCCESS in 11m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/1/gate/openstack-tox-lower-constraints/f2666ff/ : SUCCESS in 14m 32s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release/ac95756/ : SUCCESS in 2h 36m 00s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/1/gate/networking-ovn-tempest-dsvm-ovs-release-python2/f0a7b94/ : SUCCESS in 2h 21m 49s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/1/gate/networking-ovn-rally-task/bd4e1bd/ : SUCCESS in 1h 44m 00s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional-py27/e898e79/ : SUCCESS in 22m 20s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/1/gate/networking-ovn-dsvm-functional/5ae1812/ : SUCCESS in 19m 13s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/1/gate/networking-ovn-tripleo-ci-centos-7-containers-multinode/8403476/ : SUCCESS in 2h 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"481a75de48198f5a3b056e28a15f672894beaa74","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-05-09 13:00:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ec40000f4eeb4d77accbe4ad38501ed1067f3e73","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-09 13:01:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63d70da78c3b6dab2812600b31f08ada6066a8a1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-09 13:32:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a4c5d99ccdad795caea6f3dc662092d429cda57b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:48:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/2/check/openstack-tox-pep8/3e6a3d3/ : SUCCESS in 6m 11s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/2/check/openstack-tox-py27/fc24396/ : SUCCESS in 9m 31s\n- openstack-tox-docs http://logs.openstack.org/07/656907/2/check/openstack-tox-docs/3a3d74a/html/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/2/check/openstack-tox-py36/dd34041/ : SUCCESS in 11m 15s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/2/check/openstack-tox-py37/bafa358/ : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/2/check/openstack-tox-lower-constraints/51b37e0/ : SUCCESS in 13m 47s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/07/656907/2/check/networking-ovn-tempest-dsvm-ovs-master/97e03b2/ : SUCCESS in 2h 15m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/2/check/networking-ovn-tempest-dsvm-ovs-release/dfabda5/ : SUCCESS in 2h 20m 32s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/2/check/networking-ovn-tempest-dsvm-ovs-release-python2/44b8b69/ : SUCCESS in 2h 39m 34s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/2/check/networking-ovn-rally-task/7e526ed/ : SUCCESS in 1h 38m 25s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/2/check/networking-ovn-dsvm-functional-py27/5dd5bf2/ : SUCCESS in 22m 08s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/2/check/networking-ovn-dsvm-functional/cda8461/ : SUCCESS in 24m 23s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/2/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/b5d2509/ : SUCCESS in 1h 56m 22s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/07/656907/2/check/networking-ovn-dsvm-grenade/da4aea3/ : SUCCESS in 1h 57m 05s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/07/656907/2/check/openstack-tox-cover/6d8cc7e/ : FAILURE in 11m 23s (non-voting)\n- networking-ovn-tempest-slow http://logs.openstack.org/07/656907/2/check/networking-ovn-tempest-slow/a983883/ : SUCCESS in 2h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09c7b404cb7a9a6161e22fa5587c46cf48d6e80b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:48:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b3fbf05ea2dd9d0320cdd0d3aae80c27b0abc27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:21:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"eb4da669bbafcf95b446e52062de18da3334b2f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:21:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/07/656907/2/gate/openstack-tox-pep8/ce98e3c/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/07/656907/2/gate/openstack-tox-py27/3e7ad54/ : SUCCESS in 10m 18s\n- openstack-tox-docs http://logs.openstack.org/07/656907/2/gate/openstack-tox-docs/078e05a/html/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/07/656907/2/gate/openstack-tox-py36/d0b3ff4/ : SUCCESS in 12m 14s\n- openstack-tox-py37 http://logs.openstack.org/07/656907/2/gate/openstack-tox-py37/f74ae7e/ : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/656907/2/gate/openstack-tox-lower-constraints/789b4b1/ : SUCCESS in 11m 52s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/07/656907/2/gate/networking-ovn-tempest-dsvm-ovs-release/b3f83a9/ : SUCCESS in 2h 18m 29s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/07/656907/2/gate/networking-ovn-tempest-dsvm-ovs-release-python2/6468128/ : SUCCESS in 2h 14m 37s\n- networking-ovn-rally-task http://logs.openstack.org/07/656907/2/gate/networking-ovn-rally-task/0b2f13c/ : SUCCESS in 1h 40m 00s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/07/656907/2/gate/networking-ovn-dsvm-functional-py27/294e75f/ : SUCCESS in 17m 13s\n- networking-ovn-dsvm-functional http://logs.openstack.org/07/656907/2/gate/networking-ovn-dsvm-functional/391ebd3/ : SUCCESS in 16m 48s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/07/656907/2/gate/networking-ovn-tripleo-ci-centos-7-containers-multinode/8d96f06/ : SUCCESS in 2h 12m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e384647b7cbac63680ef7b06d5d6002439296bd6","revisions":{"f37704bf85c41d04a2277697516a38b915912429":{"kind":"REWORK","_number":1,"created":"2019-05-03 00:21:42.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/07/656907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/07/656907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/1"}}},"commit":{"parents":[{"commit":"036030f54bb5d6db7995d54501a6999369c9283c","subject":"Merge \"docs: Fix migration commands so they can be copy\u0026pasted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/036030f54bb5d6db7995d54501a6999369c9283c"}]}],"author":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2019-05-03 00:20:23.000000000","tz":-360},"committer":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2019-05-03 00:20:23.000000000","tz":-360},"subject":"Update device_owner for subports","message":"Update device_owner for subports\n\nWhen a subport gets bound, we\u0027re now setting the device owner to\ntrunk:subport the same way as ML2/OVS is doing as other projects\nmay rely on this value such as Kuryr.\n\nChange-Id: I87cd76ab72168083520885dd1d71465a2bc7f714\nSigned-off-by: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f37704bf85c41d04a2277697516a38b915912429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f37704bf85c41d04a2277697516a38b915912429"}]},"branch":"refs/heads/master"},"e384647b7cbac63680ef7b06d5d6002439296bd6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-09 13:00:38.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/07/656907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/07/656907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/07/656907/2"}}},"commit":{"parents":[{"commit":"7a1de107d58ad72119945c4477c3551cc060324b","subject":"PEP8: Fix bandit exclude path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7a1de107d58ad72119945c4477c3551cc060324b"}]}],"author":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2019-05-03 00:20:23.000000000","tz":-360},"committer":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2019-05-09 13:00:38.000000000","tz":0},"subject":"Update device_owner for subports","message":"Update device_owner for subports\n\nWhen a subport gets bound, we\u0027re now setting the device owner to\ntrunk:subport the same way as ML2/OVS is doing as other projects\nmay rely on this value such as Kuryr.\n\nChange-Id: I87cd76ab72168083520885dd1d71465a2bc7f714\nSigned-off-by: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e384647b7cbac63680ef7b06d5d6002439296bd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e384647b7cbac63680ef7b06d5d6002439296bd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
