)]}'
{"id":"openstack%2Fnetworking-ovn~660369","triplet_id":"openstack%2Fnetworking-ovn~master~I6ade0cf6bed6224ecc0cad956bd8efa511811444","project":"openstack/networking-ovn","branch":"master","topic":"source_ip_port_octavia_ovn","hashtags":[],"change_id":"I6ade0cf6bed6224ecc0cad956bd8efa511811444","subject":"Update OVN LB Algorithm to SOURCE_IP_PORT","status":"MERGED","created":"2019-05-21 12:34:22.000000000","updated":"2019-09-20 22:16:30.000000000","submitted":"2019-09-20 22:15:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660369-1569017708962-17048323","meta_rev_id":"d5ca4ef086aee5bec02fd40025acdadf7b2c66ff","_number":660369,"virtual_id_number":660369,"owner":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2019-09-20 22:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":2,"date":"2019-08-22 08:56:52.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-16 07:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-08-22 08:57:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":2,"date":"2019-09-19 10:20:51.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":1,"date":"2019-09-19 10:20:51.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 12:35:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-23 08:24:50.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-05-23 14:15:26.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-05-27 09:07:43.000000000","updated_by":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2019-05-27 12:01:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-22 08:56:52.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-08-22 08:57:42.000000000","updated_by":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"reviewer":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"state":"REVIEWER"},{"updated":"2019-09-16 07:23:42.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-09-19 10:20:51.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-09-20 22:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34c5a97163f47992b17676f6e2884db9a399e8d0","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-05-21 12:34:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"620b87448fa8fe034d07aa0007c2a812110d9e67","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-05-21 15:04:23.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for looking into it.\n\nVoting ahead of the gate","accounts_in_message":[],"_revision_number":1},{"id":"cb4e114d5c554df95de0db2923716e37f18c736e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:19:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/69/660369/1/check/openstack-tox-pep8/3309053/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/69/660369/1/check/openstack-tox-py27/cadcf55/ : SUCCESS in 9m 52s\n- openstack-tox-docs http://logs.openstack.org/69/660369/1/check/openstack-tox-docs/89b35d6/html/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/69/660369/1/check/openstack-tox-py36/62a8ea0/ : SUCCESS in 9m 37s\n- openstack-tox-py37 http://logs.openstack.org/69/660369/1/check/openstack-tox-py37/f1e2598/ : SUCCESS in 11m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/660369/1/check/openstack-tox-lower-constraints/b7a91ee/ : SUCCESS in 12m 15s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/69/660369/1/check/networking-ovn-tempest-dsvm-ovs-master/57b3280/ : SUCCESS in 2h 24m 49s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/69/660369/1/check/networking-ovn-tempest-dsvm-ovs-release/2433039/ : SUCCESS in 2h 38m 36s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/69/660369/1/check/networking-ovn-tempest-dsvm-ovs-release-python2/a55f360/ : SUCCESS in 2h 27m 10s\n- networking-ovn-rally-task http://logs.openstack.org/69/660369/1/check/networking-ovn-rally-task/bdac023/ : SUCCESS in 1h 46m 40s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/69/660369/1/check/networking-ovn-dsvm-functional-py27/3ea7def/ : SUCCESS in 19m 51s\n- networking-ovn-dsvm-functional http://logs.openstack.org/69/660369/1/check/networking-ovn-dsvm-functional/4554199/ : SUCCESS in 19m 12s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/69/660369/1/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/3202fc0/ : SUCCESS in 2h 03m 35s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/69/660369/1/check/networking-ovn-dsvm-grenade/c1df51a/ : SUCCESS in 1h 47m 57s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/69/660369/1/check/openstack-tox-cover/c532187/ : FAILURE in 11m 42s (non-voting)\n- networking-ovn-tempest-slow http://logs.openstack.org/69/660369/1/check/networking-ovn-tempest-slow/113664b/ : SUCCESS in 2h 04m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5dd071b7acec51c0874ad3086b92704a3e1f60ef","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-05-21 17:34:18.000000000","message":"Patch Set 1:\n\n\u003e Thanks for looking into it.\n \u003e \n \u003e Voting ahead of the gate\n\nThanks, Gate agrees with you :D","accounts_in_message":[],"_revision_number":1},{"id":"ed1613219235456e6f12d05d181caa8fa7392edb","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-05-23 08:22:35.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, thanks!\nI\u0027ll wait for Numan before approving to confirm that for a given source IP address, all the connections to the same VIP will go always to the same member.","accounts_in_message":[],"_revision_number":1},{"id":"fc5bfb50cb5a4ffb303633b63d27e4c182837cbb","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-05-23 08:24:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"461156a2d8f1e4d58e6a8f352252d55c9498ca50","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-05-23 13:05:22.000000000","message":"Patch Set 1:\n\nSupporting patch here:  https://review.opendev.org/660202","accounts_in_message":[],"_revision_number":1},{"id":"e19aa0ea55ead350374c43674310ea3a204d1f2f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-23 14:15:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"629ece688ee303949d04a9ceb2fff0420bbf04f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:15:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c15cc3b456bcfe83c9e912f9e9e4d45de863839","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-23 14:15:47.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"fb4c866d832e70a938d8a11454e378a2fce0c71a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-23 14:16:33.000000000","message":"Patch Set 1:\n\nRemoved W+1 since I just read that we are waiting for Numan...","accounts_in_message":[],"_revision_number":1},{"id":"e64d281e184c7a60990659b9a3931b9ea92b46dc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-23 14:24:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"038a4642e03646d56175dddca03ed555de4b982f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:57:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/69/660369/2/check/openstack-tox-pep8/d51eff2/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/69/660369/2/check/openstack-tox-py27/da4c544/ : SUCCESS in 10m 25s\n- openstack-tox-docs http://logs.openstack.org/69/660369/2/check/openstack-tox-docs/0f6ba73/html/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/69/660369/2/check/openstack-tox-py36/1323ffb/ : SUCCESS in 10m 02s\n- openstack-tox-py37 http://logs.openstack.org/69/660369/2/check/openstack-tox-py37/c9cb938/ : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/660369/2/check/openstack-tox-lower-constraints/01ad7d1/ : SUCCESS in 13m 10s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/69/660369/2/check/networking-ovn-tempest-dsvm-ovs-master/b402dd2/ : SUCCESS in 2h 12m 29s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/69/660369/2/check/networking-ovn-tempest-dsvm-ovs-release/5653e32/ : SUCCESS in 2h 25m 59s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/69/660369/2/check/networking-ovn-tempest-dsvm-ovs-release-python2/4ee9e23/ : SUCCESS in 2h 12m 09s\n- networking-ovn-rally-task http://logs.openstack.org/69/660369/2/check/networking-ovn-rally-task/5a088e4/ : SUCCESS in 1h 33m 41s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/69/660369/2/check/networking-ovn-dsvm-functional-py27/213be72/ : SUCCESS in 18m 21s\n- networking-ovn-dsvm-functional http://logs.openstack.org/69/660369/2/check/networking-ovn-dsvm-functional/3ff04a7/ : SUCCESS in 18m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/69/660369/2/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/a908b7d/ : SUCCESS in 2h 13m 17s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/69/660369/2/check/networking-ovn-dsvm-grenade/a8c3bd0/ : SUCCESS in 2h 08m 23s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/69/660369/2/check/openstack-tox-cover/6b906d2/ : FAILURE in 12m 42s (non-voting)\n- networking-ovn-tempest-slow http://logs.openstack.org/69/660369/2/check/networking-ovn-tempest-slow/a01fcbd/ : FAILURE in 2h 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"efb7e2ea1a25d82101166429eac8dc95da95e725","author":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"date":"2019-05-27 09:07:43.000000000","message":"Patch Set 2:\n\n\u003e LGTM, thanks!\n \u003e I\u0027ll wait for Numan before approving to confirm that for a given\n \u003e source IP address, all the connections to the same VIP will go\n \u003e always to the same member.\n\nSince the src port is also part of the tuple, I am not sure for a given source ip, it will always chose the same member.\n\nFor new connections from the same source ip to the VIP, the src port can change right ?\n\n@Reedip - Can you confirm that from  a given src ip, it will always be sent to one member ?","accounts_in_message":[],"_revision_number":2},{"id":"44dba71038f03e335eea141b5a8ca8f77132a34c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-05-27 11:25:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3f4caf9c240b0d110a8a968866a7cb655379460f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-05-27 11:38:15.000000000","message":"Patch Set 2: Code-Review-1\n\nId rather wait for Carlos/Michael as i don’t think the SOURCE IP is totally accurate. If the source port changes, the hash will differ and a different member will be selected.","accounts_in_message":[],"_revision_number":2},{"id":"bf95a21ec6b990707ad79efec79c42cdc274025c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-05-27 11:58:11.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks @Numan for the reply.\n\nThe OVN algorithm seem to be pretty unique then. I wonder if we also should change it in OVN itself to make it more \"standard\" and not hash the source port for example.","accounts_in_message":[],"_revision_number":2},{"id":"cc634c958dfbac93ab775ffdc7c2753a87127248","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-27 12:01:30.000000000","message":"Patch Set 2:\n\nI agree with Lucas. What were the historical reasons for such hashing? What was the use case? IMO we should first answer to these two questions and only after consider changing the hash.","accounts_in_message":[],"_revision_number":2},{"id":"df1e96307910a34ed700acd1c2b4d3e75e29e7cb","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-05-27 17:58:30.000000000","message":"Patch Set 2: Workflow-1\n\nI think we can start a discussion on the Upstream ML instead of the comments here? Just asking. Meanwhile, putting this to WIP till a conclusion is found. For the time being, RoundRobin seems to be the supported Algo for OVN LB, though it makes no sense, till we do not conclude it in OVN.","accounts_in_message":[],"_revision_number":2},{"id":"a0e222e8b396c4b29797e46455d9d17e87991a82","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-06-24 10:01:13.000000000","message":"Abandoned\n\nThis needs to be looked into by the Core OVN and Octavia team in more detail; Can be taken up in a separate patch","accounts_in_message":[],"_revision_number":2},{"id":"1a9481bedf1921c04f18ad0315a4badb2bbcfd98","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-23 12:22:19.000000000","message":"Restored\n\nRe-opening this, maciej requested.","accounts_in_message":[],"_revision_number":2},{"id":"36440c26747586cc39d9019fb3aadde6be97804c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:22:32.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"e12972421d5fd408a2161b2a171a38214354d4d7","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-23 12:29:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"70ce267ed637c59db437652862a926c9efb2d47e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-23 12:34:15.000000000","message":"Patch Set 3:\n\nI\u0027ve updated this change. I propose to add new algorithm name: SOURCE_IP_PORT and add its support to Octavia API.\n\nThen It\u0027ll be clear how does it work. As follows, tempest could be run without to many workarounds.","accounts_in_message":[],"_revision_number":3},{"id":"c3b8c54009a71b41ac0e4d4d8fd635288dcc7730","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:39:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/69/660369/3/check/openstack-tox-pep8/bc0fa43/ : SUCCESS in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/69/660369/3/check/openstack-tox-py27/5521d10/ : FAILURE in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/69/660369/3/check/openstack-tox-docs/4bc3d16/ : POST_FAILURE in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/69/660369/3/check/openstack-tox-py36/67ee5f0/ : FAILURE in 5m 09s\n- openstack-tox-py37 http://logs.openstack.org/69/660369/3/check/openstack-tox-py37/2d1cdfc/ : FAILURE in 6m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/660369/3/check/openstack-tox-lower-constraints/432bea4/ : FAILURE in 7m 00s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-master/7e3069b/ : SUCCESS in 2h 22m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release/7b96e15/ : TIMED_OUT in 3h 08m 49s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release-python2/2391b64/ : SUCCESS in 2h 06m 01s\n- networking-ovn-rally-task http://logs.openstack.org/69/660369/3/check/networking-ovn-rally-task/fa8e1cd/ : SUCCESS in 1h 46m 58s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-functional-py27/d341730/ : FAILURE in 11m 23s\n- networking-ovn-dsvm-functional http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-functional/827a143/ : FAILURE in 13m 46s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/69/660369/3/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/76bb309/ : SUCCESS in 2h 37m 02s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-grenade/63cd58f/ : SUCCESS in 2h 09m 19s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/69/660369/3/check/openstack-tox-cover/53cc378/ : FAILURE in 5m 57s\n- networking-ovn-tempest-slow http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-slow/8fd749d/ : FAILURE in 25m 59s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-full-multinode-ovs-master/332ce34/ : FAILURE in 2h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b90b1194a75917020cdd72ae4add32f23e84024a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-24 08:25:33.000000000","message":"Topic changed from source_ip_update to source_ip_port_octavia_ovn","accounts_in_message":[],"_revision_number":3},{"id":"6808fbd65ee43217da4995eefff45043ee5a89f7","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-24 12:46:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"56085af8fa6e04aaf5386c4ff22f600380ef3a62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:28:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/69/660369/3/check/openstack-tox-pep8/4a5c3be/ : SUCCESS in 7m 18s\n- openstack-tox-py27 http://logs.openstack.org/69/660369/3/check/openstack-tox-py27/5c08b87/ : FAILURE in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/69/660369/3/check/openstack-tox-docs/48f819a/ : POST_FAILURE in 6m 19s\n- openstack-tox-py36 http://logs.openstack.org/69/660369/3/check/openstack-tox-py36/57f74a6/ : FAILURE in 7m 23s\n- openstack-tox-py37 http://logs.openstack.org/69/660369/3/check/openstack-tox-py37/1117ce9/ : FAILURE in 7m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/660369/3/check/openstack-tox-lower-constraints/fd4e3d5/ : FAILURE in 7m 06s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-master/cff91bc/ : SUCCESS in 2h 10m 23s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release/436615d/ : SUCCESS in 1h 58m 07s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release-python2/fcce38e/ : SUCCESS in 2h 17m 59s\n- networking-ovn-rally-task http://logs.openstack.org/69/660369/3/check/networking-ovn-rally-task/a074d2f/ : SUCCESS in 1h 43m 44s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-functional-py27/5239480/ : FAILURE in 10m 15s\n- networking-ovn-dsvm-functional http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-functional/2bc4cfa/ : FAILURE in 14m 15s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/69/660369/3/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/c26283e/ : SUCCESS in 2h 36m 47s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-grenade/f848a77/ : SUCCESS in 1h 55m 25s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/69/660369/3/check/openstack-tox-cover/3c6655b/ : FAILURE in 3m 33s\n- networking-ovn-tempest-slow http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-slow/f7a7577/ : SUCCESS in 2h 23m 20s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-full-multinode-ovs-master/ecd9de3/ : SUCCESS in 1h 58m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ad02a6135e664c999df7704072e409609e4b827","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-24 15:30:10.000000000","message":"Patch Set 3:\n\nThis needs to wait for octavia-lib:\n\nAttributeError: module \u0027octavia_lib.common.constants\u0027 has no attribute \u0027LB_ALGORITHM_SOURCE_IP_PORT\u0027","accounts_in_message":[],"_revision_number":3},{"id":"8cef7047a8fd9f34a8416fe7fc0a6c2e1720cc6e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-26 13:14:51.000000000","message":"Patch Set 3:\n\noctavia-lib patch has been merged so I\u0027m rechecking this one","accounts_in_message":[],"_revision_number":3},{"id":"f2fa58b917d3439aabec760487c0d69a0fec6ed7","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-29 12:13:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b72e3190ec96d081314501768f23cfefb5267661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 14:47:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/69/660369/3/check/openstack-tox-pep8/c611236/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/69/660369/3/check/openstack-tox-py27/eaaf5c8/ : FAILURE in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/69/660369/3/check/openstack-tox-docs/ab3e36c/ : POST_FAILURE in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/69/660369/3/check/openstack-tox-py36/f0adfe3/ : FAILURE in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/69/660369/3/check/openstack-tox-py37/7cdde2f/ : FAILURE in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/660369/3/check/openstack-tox-lower-constraints/fcb7cae/ : FAILURE in 5m 27s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-master/8feebe7/ : SUCCESS in 2h 03m 01s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release/f448650/ : SUCCESS in 1h 59m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release-python2/470f798/ : SUCCESS in 2h 04m 20s\n- networking-ovn-rally-task http://logs.openstack.org/69/660369/3/check/networking-ovn-rally-task/ad1be9c/ : SUCCESS in 1h 42m 12s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-functional-py27/521a9d9/ : FAILURE in 9m 28s\n- networking-ovn-dsvm-functional http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-functional/3fc0742/ : FAILURE in 10m 36s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/69/660369/3/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/a462b1a/ : SUCCESS in 2h 17m 46s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/69/660369/3/check/networking-ovn-dsvm-grenade/0c4828b/ : SUCCESS in 2h 08m 50s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/69/660369/3/check/openstack-tox-cover/6733726/ : FAILURE in 3m 50s\n- networking-ovn-tempest-slow http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-slow/7182049/ : SUCCESS in 2h 28m 05s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master http://logs.openstack.org/69/660369/3/check/networking-ovn-tempest-full-multinode-ovs-master/4814b42/ : SUCCESS in 2h 06m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"03e3c56f5449e83a952bc36ae4f926b7422a027d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-02 07:57:51.000000000","message":"Patch Set 3:\n\nThis change fails, even I\u0027ve added Depends-On: I605f44f0f50219aa003df477de9bae4062f3c308 that points to octavia-lib change.\n\nRelease is almost ready: https://review.opendev.org/#/c/673928/, so lets wait.","accounts_in_message":[],"_revision_number":3},{"id":"597f9184ebd6a34d434956f72d32cfe38d1aec9f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-05 08:27:17.000000000","message":"Patch Set 3:\n\nrecheck octavia-lib release merged so I\u0027m rechecking this","accounts_in_message":[],"_revision_number":3},{"id":"0b56a9daa81a1d77723ee1262351f6ffdbcf0180","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 11:02:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/69/660369/3/check/openstack-tox-pep8/bfefe70/ : SUCCESS in 6m 08s\n- openstack-tox-py27 https://logs.opendev.org/69/660369/3/check/openstack-tox-py27/cd07e95/ : TIMED_OUT in 41m 11s\n- openstack-tox-docs https://logs.opendev.org/69/660369/3/check/openstack-tox-docs/03f8962/ : POST_FAILURE in 3m 27s\n- openstack-tox-py36 https://logs.opendev.org/69/660369/3/check/openstack-tox-py36/cc8a924/ : TIMED_OUT in 41m 34s\n- openstack-tox-py37 https://logs.opendev.org/69/660369/3/check/openstack-tox-py37/0bc48dd/ : TIMED_OUT in 41m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/69/660369/3/check/openstack-tox-lower-constraints/3d7a374/ : FAILURE in 5m 38s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-master/263ea45/ : SUCCESS in 2h 05m 42s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release/3740274/ : SUCCESS in 2h 10m 09s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/69/660369/3/check/networking-ovn-tempest-dsvm-ovs-release-python2/71dfbfb/ : SUCCESS in 2h 04m 35s\n- networking-ovn-rally-task https://logs.opendev.org/69/660369/3/check/networking-ovn-rally-task/9e96e5a/ : SUCCESS in 1h 44m 37s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/69/660369/3/check/networking-ovn-dsvm-functional-py27/668ecd7/ : FAILURE in 28m 42s\n- networking-ovn-dsvm-functional https://logs.opendev.org/69/660369/3/check/networking-ovn-dsvm-functional/2282245/ : FAILURE in 32m 58s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/660369/3/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/ec7bd02/ : SUCCESS in 2h 33m 42s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/69/660369/3/check/networking-ovn-dsvm-grenade/9075608/ : SUCCESS in 1h 58m 47s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/69/660369/3/check/openstack-tox-cover/4850179/ : FAILURE in 47m 50s\n- networking-ovn-tempest-slow https://logs.opendev.org/69/660369/3/check/networking-ovn-tempest-slow/1d7f34c/ : SUCCESS in 2h 19m 01s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/69/660369/3/check/networking-ovn-tempest-full-multinode-ovs-master/dd6708a/ : SUCCESS in 2h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ac21dd744545d6913027d374df73214929bd6704","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-09 11:54:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"117b11c0d0c9aec8fa33f397b769a8ba599bcc85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 14:35:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/69/660369/4/check/openstack-tox-pep8/28f8674/ : SUCCESS in 6m 41s\n- openstack-tox-py27 https://logs.opendev.org/69/660369/4/check/openstack-tox-py27/daf4eca/ : SUCCESS in 12m 27s\n- openstack-tox-docs https://logs.opendev.org/69/660369/4/check/openstack-tox-docs/012a1bd/ : POST_FAILURE in 4m 00s\n- openstack-tox-py36 https://logs.opendev.org/69/660369/4/check/openstack-tox-py36/c4c3e1a/ : SUCCESS in 10m 19s\n- openstack-tox-py37 https://logs.opendev.org/69/660369/4/check/openstack-tox-py37/423d71f/ : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://logs.opendev.org/69/660369/4/check/openstack-tox-lower-constraints/4a9bdb3/ : FAILURE in 6m 08s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/69/660369/4/check/networking-ovn-tempest-dsvm-ovs-master/5905074/ : SUCCESS in 2h 23m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/69/660369/4/check/networking-ovn-tempest-dsvm-ovs-release/eedbeb1/ : SUCCESS in 2h 23m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/69/660369/4/check/networking-ovn-tempest-dsvm-ovs-release-python2/b53cc49/ : SUCCESS in 2h 03m 53s\n- networking-ovn-rally-task https://logs.opendev.org/69/660369/4/check/networking-ovn-rally-task/fa09eba/ : SUCCESS in 1h 47m 25s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/69/660369/4/check/networking-ovn-dsvm-functional-py27/dc83566/ : SUCCESS in 20m 34s\n- networking-ovn-dsvm-functional https://logs.opendev.org/69/660369/4/check/networking-ovn-dsvm-functional/a11a23a/ : SUCCESS in 19m 51s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/660369/4/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/84f54b1/ : SUCCESS in 2h 28m 43s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/69/660369/4/check/networking-ovn-dsvm-grenade/ae843f5/ : SUCCESS in 2h 28m 21s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/69/660369/4/check/openstack-tox-cover/e8f4abe/cover/ : SUCCESS in 11m 07s\n- networking-ovn-tempest-slow https://logs.opendev.org/69/660369/4/check/networking-ovn-tempest-slow/4f08203/ : SUCCESS in 2h 29m 22s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/69/660369/4/check/networking-ovn-tempest-full-multinode-ovs-master/87e3224/ : SUCCESS in 2h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e054afd3b759c86459974e14e0c1c43e0eff4227","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-09 14:45:13.000000000","message":"Patch Set 4:\n\nopenstack-tox-lower-constraints fails so we need to bump up lower constraints to 1.3.1","accounts_in_message":[],"_revision_number":4},{"id":"bf2e3c404a77ad80e2afadc98b827d818fbcc4f9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-09 14:46:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9c45b029e3ab21894b257f68a9ec1cf7dbe8e677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 18:12:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/69/660369/5/check/openstack-tox-pep8/d6c26c6/ : SUCCESS in 6m 16s\n- openstack-tox-py27 https://logs.opendev.org/69/660369/5/check/openstack-tox-py27/adb3ccc/ : SUCCESS in 11m 42s\n- openstack-tox-docs https://logs.opendev.org/69/660369/5/check/openstack-tox-docs/1402227/ : POST_FAILURE in 3m 57s\n- build-openstack-releasenotes https://logs.opendev.org/69/660369/5/check/build-openstack-releasenotes/7ba379b/html/ : SUCCESS in 5m 23s\n- requirements-check https://logs.opendev.org/69/660369/5/check/requirements-check/83dab98/ : SUCCESS in 3m 22s\n- openstack-tox-py36 https://logs.opendev.org/69/660369/5/check/openstack-tox-py36/635fadd/ : SUCCESS in 8m 13s\n- openstack-tox-py37 https://logs.opendev.org/69/660369/5/check/openstack-tox-py37/dfba504/ : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://logs.opendev.org/69/660369/5/check/openstack-tox-lower-constraints/1e3d9b8/ : SUCCESS in 11m 13s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-master/49a0edc/ : SUCCESS in 3h 04m 01s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-release/0735729/ : SUCCESS in 2h 17m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-release-python2/0efa52b/ : TIMED_OUT in 3h 09m 26s\n- networking-ovn-rally-task https://logs.opendev.org/69/660369/5/check/networking-ovn-rally-task/990a87b/ : SUCCESS in 1h 40m 13s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-functional-py27/628eb2e/ : SUCCESS in 26m 27s\n- networking-ovn-dsvm-functional https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-functional/9f4d8cd/ : SUCCESS in 18m 25s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/660369/5/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/23bd0ea/ : SUCCESS in 2h 35m 52s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-grenade/40106d3/ : SUCCESS in 2h 05m 13s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/69/660369/5/check/openstack-tox-cover/5685586/cover/ : SUCCESS in 11m 03s\n- networking-ovn-tempest-slow https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-slow/92085ba/ : SUCCESS in 2h 22m 58s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-full-multinode-ovs-master/749889d/ : SUCCESS in 2h 02m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ba471e1cbad705df514cc138c3c2789de880f483","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-09 20:52:56.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1243843c704bebf9e9bfc29d6b2299f52b2434db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 00:17:53.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/69/660369/5/check/openstack-tox-pep8/290248f/ : SUCCESS in 4m 54s\n- openstack-tox-py27 https://logs.opendev.org/69/660369/5/check/openstack-tox-py27/093b380/ : SUCCESS in 7m 57s\n- openstack-tox-docs https://logs.opendev.org/69/660369/5/check/openstack-tox-docs/98cafe2/ : POST_FAILURE in 3m 01s\n- build-openstack-releasenotes https://logs.opendev.org/69/660369/5/check/build-openstack-releasenotes/1c88fc7/html/ : SUCCESS in 4m 37s\n- requirements-check https://logs.opendev.org/69/660369/5/check/requirements-check/72fb2cb/ : SUCCESS in 2m 40s\n- openstack-tox-py36 https://logs.opendev.org/69/660369/5/check/openstack-tox-py36/2ef88e8/ : SUCCESS in 9m 34s\n- openstack-tox-py37 https://logs.opendev.org/69/660369/5/check/openstack-tox-py37/b76c31b/ : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://logs.opendev.org/69/660369/5/check/openstack-tox-lower-constraints/ead612c/ : SUCCESS in 10m 40s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-master/4112417/ : SUCCESS in 1h 54m 49s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-release/e5b2e32/ : SUCCESS in 2h 07m 19s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-release-python2/86f6e50/ : SUCCESS in 3h 01m 41s\n- networking-ovn-rally-task https://logs.opendev.org/69/660369/5/check/networking-ovn-rally-task/b496a41/ : SUCCESS in 1h 49m 30s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-functional-py27/deef49f/ : SUCCESS in 19m 24s\n- networking-ovn-dsvm-functional https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-functional/0a59591/ : SUCCESS in 18m 04s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/660369/5/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/d46be73/ : SUCCESS in 2h 24m 29s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-grenade/994e9ab/ : SUCCESS in 2h 06m 16s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/69/660369/5/check/openstack-tox-cover/3f7c7bd/cover/ : SUCCESS in 10m 11s\n- networking-ovn-tempest-slow https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-slow/e47876f/ : SUCCESS in 2h 46m 01s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-full-multinode-ovs-master/f5b4e40/ : SUCCESS in 1h 59m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aef2c0af4bc0e22a88278f76be5f65d35f673c4e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-12 07:09:00.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b20c7ea7733a1dca39e64fd2f8f9bb4dce804f2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 10:12:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/69/660369/5/check/openstack-tox-pep8/6b69787/ : SUCCESS in 4m 53s\n- openstack-tox-py27 https://logs.opendev.org/69/660369/5/check/openstack-tox-py27/d2ea239/ : SUCCESS in 8m 58s\n- openstack-tox-docs https://logs.opendev.org/69/660369/5/check/openstack-tox-docs/5b727bd/ : POST_FAILURE in 4m 15s\n- build-openstack-releasenotes https://logs.opendev.org/69/660369/5/check/build-openstack-releasenotes/1487fb5/html/ : SUCCESS in 4m 09s\n- requirements-check https://logs.opendev.org/69/660369/5/check/requirements-check/953838f/ : SUCCESS in 2m 05s\n- openstack-tox-py36 https://logs.opendev.org/69/660369/5/check/openstack-tox-py36/f796db8/ : SUCCESS in 9m 33s\n- openstack-tox-py37 https://logs.opendev.org/69/660369/5/check/openstack-tox-py37/6057cdf/ : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/69/660369/5/check/openstack-tox-lower-constraints/91a9295/ : SUCCESS in 13m 41s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-master/d093290/ : SUCCESS in 2h 12m 16s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-release/ebf8a77/ : SUCCESS in 2h 07m 59s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-dsvm-ovs-release-python2/39f650f/ : SUCCESS in 2h 52m 40s\n- networking-ovn-rally-task https://logs.opendev.org/69/660369/5/check/networking-ovn-rally-task/96e0e1f/ : SUCCESS in 1h 38m 52s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-functional-py27/609386b/ : SUCCESS in 20m 39s\n- networking-ovn-dsvm-functional https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-functional/a79dd0f/ : SUCCESS in 23m 02s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/660369/5/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/18da463/ : SUCCESS in 2h 23m 27s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/69/660369/5/check/networking-ovn-dsvm-grenade/4bcb373/ : FAILURE in 51m 59s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/69/660369/5/check/openstack-tox-cover/67848c5/cover/ : SUCCESS in 8m 57s\n- networking-ovn-tempest-slow https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-slow/ec2447b/ : SUCCESS in 2h 32m 28s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/69/660369/5/check/networking-ovn-tempest-full-multinode-ovs-master/b549b07/ : FAILURE in 2h 06m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"81c3b964650ac087eccfd37a9c5a7e7a86915ea3","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-12 10:21:04.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"964f6ef0f5725e27dd04a8c09151420f6bf44ad3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 13:06:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/69/660369/6/check/openstack-tox-pep8/1d7b4eb/ : SUCCESS in 4m 34s\n- openstack-tox-py27 https://logs.opendev.org/69/660369/6/check/openstack-tox-py27/52ed63b/ : SUCCESS in 9m 41s\n- openstack-tox-docs https://logs.opendev.org/69/660369/6/check/openstack-tox-docs/371b474/html/ : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://logs.opendev.org/69/660369/6/check/build-openstack-releasenotes/ea8b8d9/html/ : SUCCESS in 4m 16s\n- requirements-check https://logs.opendev.org/69/660369/6/check/requirements-check/3c4e7db/ : SUCCESS in 2m 02s\n- openstack-tox-py36 https://logs.opendev.org/69/660369/6/check/openstack-tox-py36/c21e9e1/ : SUCCESS in 7m 56s\n- openstack-tox-py37 https://logs.opendev.org/69/660369/6/check/openstack-tox-py37/54998c0/ : SUCCESS in 8m 06s\n- openstack-tox-lower-constraints https://logs.opendev.org/69/660369/6/check/openstack-tox-lower-constraints/48b5b20/ : SUCCESS in 9m 02s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/69/660369/6/check/networking-ovn-tempest-dsvm-ovs-master/3065665/ : SUCCESS in 2h 43m 43s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/69/660369/6/check/networking-ovn-tempest-dsvm-ovs-release/3f6086b/ : SUCCESS in 2h 29m 31s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/69/660369/6/check/networking-ovn-tempest-dsvm-ovs-release-python2/9cbc643/ : SUCCESS in 2h 32m 58s\n- networking-ovn-rally-task https://logs.opendev.org/69/660369/6/check/networking-ovn-rally-task/5e5322f/ : SUCCESS in 1h 34m 44s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/69/660369/6/check/networking-ovn-dsvm-functional-py27/a43ff36/ : SUCCESS in 19m 08s\n- networking-ovn-dsvm-functional https://logs.opendev.org/69/660369/6/check/networking-ovn-dsvm-functional/4048ebe/ : SUCCESS in 21m 31s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/660369/6/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/f55d86c/ : SUCCESS in 2h 01m 39s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/69/660369/6/check/networking-ovn-dsvm-grenade/9a6dc90/ : SUCCESS in 2h 03m 53s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/69/660369/6/check/openstack-tox-cover/179b43b/cover/ : SUCCESS in 9m 50s\n- networking-ovn-tempest-slow https://logs.opendev.org/69/660369/6/check/networking-ovn-tempest-slow/2dac5b8/ : SUCCESS in 2h 11m 31s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/69/660369/6/check/networking-ovn-tempest-full-multinode-ovs-master/8e2d067/ : SUCCESS in 1h 56m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5dca59fbf64e489ba7a49da1397836809dd652e1","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-08-22 08:56:52.000000000","message":"Patch Set 6: Code-Review+2\n\nNice! Glad to see algorithm type has been included in Octavia","accounts_in_message":[],"_revision_number":6},{"id":"b4a9d63cdb7ac2738aeae26985f24bd4fc47a9b8","author":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"date":"2019-08-22 08:57:42.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"106b98da93833ebcf51a5e25b1e95392f2fd4691","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-22 08:59:36.000000000","message":"Patch Set 6: Code-Review-1\n\nPlease wait with merging it unit octavia will be ready (all related patchsets in the topic will be merged).","accounts_in_message":[],"_revision_number":6},{"id":"682523a617f965fee3ffd3dbfbec6c23222b8ee2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-16 07:23:42.000000000","message":"Patch Set 6: Code-Review+1\n\nThe SOURCE_IP_PORT algorithm has been merged to octavia [1].\nWe can unblock this patch and merge it :)\n\nhttps://review.opendev.org/#/c/672463/","accounts_in_message":[],"_revision_number":6},{"id":"0e2f0da8d4c67a313df9f69c16d878c173a8a05b","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-19 10:20:51.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1585690df3bf84632b383951058f97776e40594c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 10:21:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1ddccd508210b2d133bdd544d1873b6fde075a64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:05:51.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa3aa136c9a4779b08ab12eb1a32bfd : SUCCESS in 8m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1e7ceb24bc34a94be5908a1460ec00a : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cb94d295d88413895b4fb20704ddc17 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76361f86974848f5badd1068e4a34479 : SUCCESS in 4m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/162636dce3264c389105e0d014a1a75f : SUCCESS in 2m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/136fd15872a144e89e9089b9694169f2 : SUCCESS in 11m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6778ad37b71e4c5eabebfe960bd3e0a3 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aad3fb127ff84c628dc07783aedba79c : SUCCESS in 14m 03s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/39a4e581810e41cea8facda4cb293b0f : SUCCESS in 2h 36m 01s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ccc07fe8d3d14d1caee9a85ae96b9078 : SUCCESS in 2h 18m 55s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c3e9f0054a4248fbbdf03e9417ab5ce4 : SUCCESS in 22m 26s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5475c659a8f5494da303e2c79cea2185 : FAILURE in 23m 54s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7c57e3388e904c6d8a706f1a6ae72f1d : SUCCESS in 3h 40m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c167b226cdd84ba5b4902e2fe55a9643 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":6},{"id":"ec586f397f789c8451476921c0af4ccbb05ef796","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-20 07:16:04.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5a137be271a723cd72fee03d9d2d98132d10ea50","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-20 11:32:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5d055be9d98ee1a2b80206ee26d12ce179604979","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:47:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4486c0730043f3af28f83f051add5e : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/988318aaf1db41deb79e05c7383602b3 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a45efbb8f5428dbeb0f4712e47b3a7 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59271fa19a804a5a9870d012f06a5ae6 : SUCCESS in 5m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/333e2158798f442ea6f135e3cfd3c4df : SUCCESS in 2m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8e7b9f666184a2e888c08bb41eaf2e8 : SUCCESS in 9m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2410133b4c6422fb6c113694af69893 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad3d78637a0d4a69813effa66d324bf0 : SUCCESS in 10m 55s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/85d285bceedd483387a4b574a82c777f : SUCCESS in 2h 04m 32s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/42a3ff4f03f74bfd8f05a7429c7e6e52 : SUCCESS in 2h 18m 19s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d06a6af32fa346ad89857d09a7cf04e9 : SUCCESS in 2h 45m 08s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5878e61e3354de9b844c1d392aee8d2 : SUCCESS in 1h 48m 58s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/12c0f5ce102c4e21a9c31189d5904e01 : SUCCESS in 22m 04s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a58314d94e54bcbacdd445fb3fa01b3 : SUCCESS in 30m 01s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3c4c797074e34d7cbcb2b1400f688d34 : SUCCESS in 3h 18m 02s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/ba38284218f7433197e148a1c70f9977 : SUCCESS in 1h 58m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91497b45510b450b81814be53276d425 : SUCCESS in 10m 26s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f7f544df199e4df096f4e0cffffba5de : SUCCESS in 2h 22m 12s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/43ef20188c674c4da67f8d3338e03482 : SUCCESS in 2h 24m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b76bd36977689abc57adfaf89cf03a19ef8480cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:47:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"95d385fc09f5616d0758a6ccb92c0687a9a965e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:15:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11517eadf9e4d7086cc5290315da8af : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9511b12ecafd45ae895dff0edcb2b663 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ee0b05ee5a4b74bfc2959b56ea3cd4 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec50dae767c042ff9987f97f571a0d68 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6c94e1e5f174f7db4e68bca69ed4598 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5be678c6d4646d19daf26b71a49c204 : SUCCESS in 9m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f69504cd9997406faba19a7bc141cbe5 : SUCCESS in 12m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76759aea08114f57bbb314a3d1044ed7 : SUCCESS in 11m 27s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3bbe3042b2ea4661b2665b8f5d65bc5a : SUCCESS in 2h 12m 13s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e7cdede379414892a90de196cbff2f07 : SUCCESS in 2h 27m 49s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/89c04c48cf71425ca11fe12de70db4b4 : SUCCESS in 19m 32s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/99fa58e608ba4164a0f3af2a5b725492 : SUCCESS in 17m 09s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3b8541ee7a024c0cb35837b887a0e323 : SUCCESS in 3h 37m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92c9b45c91b6408cb3567f4deb3f3a09 : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":6},{"id":"55a10578fe2030ea637f5a8294989dc9d73c16fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:15:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d5ca4ef086aee5bec02fd40025acdadf7b2c66ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:16:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba0884314804ebab3f9d059bc5db552 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe252c17845744d2a3385e3ab0d11f32 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"95d5cef5252c2ae5c4c02b0450e3cda2b35f7d3e","revisions":{"689253b9b777347699e76a0cc1fb029848fe496e":{"kind":"REWORK","_number":1,"created":"2019-05-21 12:34:22.000000000","uploader":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"ref":"refs/changes/69/660369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/660369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/1"}}},"commit":{"parents":[{"commit":"25de3d7183420ecd8af8e4d3f73b3f82c2502a3c","subject":"Merge \"Don\u0027t wait for metadata service unnecessarily\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/25de3d7183420ecd8af8e4d3f73b3f82c2502a3c"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-21 12:26:22.000000000","tz":0},"committer":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-21 12:34:13.000000000","tz":0},"subject":"Update OVN LB Algorithm to SOURCE_IP","message":"Update OVN LB Algorithm to SOURCE_IP\n\nAfter investigating, it was observed that core OVN actually\nutilizes a 5 Tuple Hash/RSS Hash in DPDK/Kernel as a Load\nBalancing algorithm. The 5 Tuple Hash has Source IP, Destination\nIP, Protocol, Source Port, Destination Port and a basis value.\nThis maps very closely to SOURCE_IP implementation in Octavia than\nROUND_ROBIN which is the existing supported algorithm.\nThis patch updates the support from ROUND_ROBIN to SOURCE_IP\n\n[0]DP_Hash implementation: https://github.com/openvswitch/ovs/blob/master/ovn/lib/actions.c#L1036\n[1]flow5tuplehash: https://github.com/openvswitch/ovs/blob/74286173f4d7f51f78e9db09b07a6d4d65263252/lib/flow.c#L2219\n[2]Kernel\u0027s Hash implementation: https://github.com/openvswitch/ovs/blob/master/datapath/actions.c#L1100\n[3]RSS Hash: https://github.com/openvswitch/ovs/blob/761398f89abd1bd67514f6dd679e16c548585cb4/lib/dp-packet.h#L504\n\nChange-Id: I6ade0cf6bed6224ecc0cad956bd8efa511811444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/689253b9b777347699e76a0cc1fb029848fe496e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/689253b9b777347699e76a0cc1fb029848fe496e"}]},"branch":"refs/heads/master"},"4359a0b6961e8830ff1c222cdb25abedff119d13":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-23 14:24:13.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/69/660369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/660369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/2"}}},"commit":{"parents":[{"commit":"f92b5aa08708fcb75ebdc864a5a387a27958c17e","subject":"Merge \"Do not set port addresses on LSP while port not bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f92b5aa08708fcb75ebdc864a5a387a27958c17e"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-21 12:26:22.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-23 14:24:13.000000000","tz":0},"subject":"Update OVN LB Algorithm to SOURCE_IP","message":"Update OVN LB Algorithm to SOURCE_IP\n\nAfter investigating, it was observed that core OVN actually\nutilizes a 5 Tuple Hash/RSS Hash in DPDK/Kernel as a Load\nBalancing algorithm. The 5 Tuple Hash has Source IP, Destination\nIP, Protocol, Source Port, Destination Port and a basis value.\nThis maps very closely to SOURCE_IP implementation in Octavia than\nROUND_ROBIN which is the existing supported algorithm.\nThis patch updates the support from ROUND_ROBIN to SOURCE_IP\n\n[0]DP_Hash implementation: https://github.com/openvswitch/ovs/blob/master/ovn/lib/actions.c#L1036\n[1]flow5tuplehash: https://github.com/openvswitch/ovs/blob/74286173f4d7f51f78e9db09b07a6d4d65263252/lib/flow.c#L2219\n[2]Kernel\u0027s Hash implementation: https://github.com/openvswitch/ovs/blob/master/datapath/actions.c#L1100\n[3]RSS Hash: https://github.com/openvswitch/ovs/blob/761398f89abd1bd67514f6dd679e16c548585cb4/lib/dp-packet.h#L504\n\nChange-Id: I6ade0cf6bed6224ecc0cad956bd8efa511811444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4359a0b6961e8830ff1c222cdb25abedff119d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4359a0b6961e8830ff1c222cdb25abedff119d13"}]},"branch":"refs/heads/master"},"1192f950e70e8cb38eb57389a28477b1bff99a38":{"kind":"REWORK","_number":3,"created":"2019-07-23 12:29:28.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/660369/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/660369/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/3"}}},"commit":{"parents":[{"commit":"128e43f16a6eac08c50e194ddf03436aa8fbae51","subject":"Merge \"Maintenance task: Enhance debug logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/128e43f16a6eac08c50e194ddf03436aa8fbae51"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-21 12:26:22.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-23 12:28:46.000000000","tz":0},"subject":"Update OVN LB Algorithm to SOURCE_IP_PORT","message":"Update OVN LB Algorithm to SOURCE_IP_PORT\n\nAfter investigating, it was observed that core OVN actually\nutilizes a 5 Tuple Hash/RSS Hash in DPDK/Kernel as a Load\nBalancing algorithm. The 5 Tuple Hash has Source IP, Destination\nIP, Protocol, Source Port, Destination Port and a basis value.\nThis patch updates the algorithm name from ROUND_ROBIN to SOURCE_IP_PORT.\n\n[0]DP_Hash implementation: https://github.com/openvswitch/ovs/blob/master/ovn/lib/actions.c#L1036\n[1]flow5tuplehash: https://github.com/openvswitch/ovs/blob/74286173f4d7f51f78e9db09b07a6d4d65263252/lib/flow.c#L2219\n[2]Kernel\u0027s Hash implementation: https://github.com/openvswitch/ovs/blob/master/datapath/actions.c#L1100\n[3]RSS Hash: https://github.com/openvswitch/ovs/blob/761398f89abd1bd67514f6dd679e16c548585cb4/lib/dp-packet.h#L504\n\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nDepends-On: I605f44f0f50219aa003df477de9bae4062f3c308\nChange-Id: I6ade0cf6bed6224ecc0cad956bd8efa511811444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1192f950e70e8cb38eb57389a28477b1bff99a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1192f950e70e8cb38eb57389a28477b1bff99a38"}]},"branch":"refs/heads/master"},"97c505ff30e5c9a2677793f55dc9562032c788f6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-09 11:54:26.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/660369/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/660369/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/4"}}},"commit":{"parents":[{"commit":"b2b5e89bca25570a757c62fc6e30277365dae456","subject":"Fix gateway blockers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b2b5e89bca25570a757c62fc6e30277365dae456"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-21 12:26:22.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-09 11:54:26.000000000","tz":0},"subject":"Update OVN LB Algorithm to SOURCE_IP_PORT","message":"Update OVN LB Algorithm to SOURCE_IP_PORT\n\nAfter investigating, it was observed that core OVN actually\nutilizes a 5 Tuple Hash/RSS Hash in DPDK/Kernel as a Load\nBalancing algorithm. The 5 Tuple Hash has Source IP, Destination\nIP, Protocol, Source Port, Destination Port and a basis value.\nThis patch updates the algorithm name from ROUND_ROBIN to SOURCE_IP_PORT.\n\n[0]DP_Hash implementation: https://github.com/openvswitch/ovs/blob/master/ovn/lib/actions.c#L1036\n[1]flow5tuplehash: https://github.com/openvswitch/ovs/blob/74286173f4d7f51f78e9db09b07a6d4d65263252/lib/flow.c#L2219\n[2]Kernel\u0027s Hash implementation: https://github.com/openvswitch/ovs/blob/master/datapath/actions.c#L1100\n[3]RSS Hash: https://github.com/openvswitch/ovs/blob/761398f89abd1bd67514f6dd679e16c548585cb4/lib/dp-packet.h#L504\n\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nDepends-On: I605f44f0f50219aa003df477de9bae4062f3c308\nChange-Id: I6ade0cf6bed6224ecc0cad956bd8efa511811444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/97c505ff30e5c9a2677793f55dc9562032c788f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/97c505ff30e5c9a2677793f55dc9562032c788f6"}]},"branch":"refs/heads/master"},"7116820661ae4bde7a473c4b24f22e6d3eb063aa":{"kind":"REWORK","_number":5,"created":"2019-08-09 14:46:41.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/660369/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/660369/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/5"}}},"commit":{"parents":[{"commit":"b2b5e89bca25570a757c62fc6e30277365dae456","subject":"Fix gateway blockers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b2b5e89bca25570a757c62fc6e30277365dae456"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-21 12:26:22.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-09 14:46:24.000000000","tz":0},"subject":"Update OVN LB Algorithm to SOURCE_IP_PORT","message":"Update OVN LB Algorithm to SOURCE_IP_PORT\n\nAfter investigating, it was observed that core OVN actually\nutilizes a 5 Tuple Hash/RSS Hash in DPDK/Kernel as a Load\nBalancing algorithm. The 5 Tuple Hash has Source IP, Destination\nIP, Protocol, Source Port, Destination Port and a basis value.\nThis patch updates the algorithm name from ROUND_ROBIN to SOURCE_IP_PORT.\n\n[0]DP_Hash implementation: https://github.com/openvswitch/ovs/blob/master/ovn/lib/actions.c#L1036\n[1]flow5tuplehash: https://github.com/openvswitch/ovs/blob/74286173f4d7f51f78e9db09b07a6d4d65263252/lib/flow.c#L2219\n[2]Kernel\u0027s Hash implementation: https://github.com/openvswitch/ovs/blob/master/datapath/actions.c#L1100\n[3]RSS Hash: https://github.com/openvswitch/ovs/blob/761398f89abd1bd67514f6dd679e16c548585cb4/lib/dp-packet.h#L504\n\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nDepends-On: I605f44f0f50219aa003df477de9bae4062f3c308\nChange-Id: I6ade0cf6bed6224ecc0cad956bd8efa511811444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7116820661ae4bde7a473c4b24f22e6d3eb063aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7116820661ae4bde7a473c4b24f22e6d3eb063aa"}]},"branch":"refs/heads/master"},"95d5cef5252c2ae5c4c02b0450e3cda2b35f7d3e":{"kind":"REWORK","_number":6,"created":"2019-08-12 10:21:04.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/660369/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/660369/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/660369/6"}}},"commit":{"parents":[{"commit":"b2b5e89bca25570a757c62fc6e30277365dae456","subject":"Fix gateway blockers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b2b5e89bca25570a757c62fc6e30277365dae456"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-21 12:26:22.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-12 10:21:01.000000000","tz":0},"subject":"Update OVN LB Algorithm to SOURCE_IP_PORT","message":"Update OVN LB Algorithm to SOURCE_IP_PORT\n\nAfter investigating, it was observed that core OVN actually\nutilizes a 5 Tuple Hash/RSS Hash in DPDK/Kernel as a Load\nBalancing algorithm. The 5 Tuple Hash has Source IP, Destination\nIP, Protocol, Source Port, Destination Port and a basis value.\nThis patch updates the algorithm name from ROUND_ROBIN to SOURCE_IP_PORT.\n\n[0]DP_Hash implementation: https://github.com/openvswitch/ovs/blob/master/ovn/lib/actions.c#L1036\n[1]flow5tuplehash: https://github.com/openvswitch/ovs/blob/74286173f4d7f51f78e9db09b07a6d4d65263252/lib/flow.c#L2219\n[2]Kernel\u0027s Hash implementation: https://github.com/openvswitch/ovs/blob/master/datapath/actions.c#L1100\n[3]RSS Hash: https://github.com/openvswitch/ovs/blob/761398f89abd1bd67514f6dd679e16c548585cb4/lib/dp-packet.h#L504\n\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nDepends-On: I605f44f0f50219aa003df477de9bae4062f3c308\nChange-Id: I6ade0cf6bed6224ecc0cad956bd8efa511811444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/95d5cef5252c2ae5c4c02b0450e3cda2b35f7d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/95d5cef5252c2ae5c4c02b0450e3cda2b35f7d3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
