)]}'
{"id":"openstack%2Fnetworking-ovn~668397","triplet_id":"openstack%2Fnetworking-ovn~master~If10016325517e4a1fc5f7cfac7f7bd58deb09b54","project":"openstack/networking-ovn","branch":"master","topic":"bug/1794260","hashtags":[],"change_id":"If10016325517e4a1fc5f7cfac7f7bd58deb09b54","subject":"Add a new Event Handler for Octavia","status":"MERGED","created":"2019-07-01 09:20:22.000000000","updated":"2019-07-29 11:08:31.000000000","submitted":"2019-07-15 18:23:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668397-1563215019682-ef2d6f53","meta_rev_id":"980149af975b6e28b41b5275c5afdcf6854def15","_number":668397,"virtual_id_number":668397,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2019-07-15 13:20:55.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-07-15 18:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2019-07-15 13:59:39.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-07-15 14:51:31.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-07-15 15:57:22.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 09:08:15.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-07-12 14:06:50.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-07-15 13:59:39.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-07-15 15:57:22.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-07-15 18:23:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57b79cb8008a11d06d2eeabcff43be69fab24c8d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-01 09:20:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fac3158e205c8d6f28db403b46bb279ceaba4f1","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-01 09:42:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a56aa0435a5bc0eac0b30c489f757e70e820e1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:49:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/1/check/openstack-tox-pep8/545edc7/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/1/check/openstack-tox-py27/23b7107/ : FAILURE in 9m 16s\n- openstack-tox-docs http://logs.openstack.org/97/668397/1/check/openstack-tox-docs/daa61c7/html/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/1/check/openstack-tox-py36/3852be4/ : FAILURE in 10m 21s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/1/check/openstack-tox-py37/7822352/ : FAILURE in 15m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/1/check/openstack-tox-lower-constraints/a1f1bfc/ : FAILURE in 10m 46s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/1/check/networking-ovn-tempest-dsvm-ovs-master/7d1456f/ : FAILURE in 2h 21m 18s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/1/check/networking-ovn-tempest-dsvm-ovs-release/158fb81/ : SUCCESS in 2h 19m 45s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/1/check/networking-ovn-tempest-dsvm-ovs-release-python2/a5e9ea7/ : SUCCESS in 2h 26m 39s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/1/check/networking-ovn-rally-task/fdccc35/ : SUCCESS in 1h 57m 09s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/1/check/networking-ovn-dsvm-functional-py27/c307764/ : SUCCESS in 24m 55s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/1/check/networking-ovn-dsvm-functional/82c9d70/ : SUCCESS in 25m 14s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/1/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/cfd1286/ : FAILURE in 1h 00m 46s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/1/check/networking-ovn-dsvm-grenade/33f03bf/ : FAILURE in 2h 14m 51s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/1/check/openstack-tox-cover/a4d2df7/ : FAILURE in 11m 44s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/1/check/networking-ovn-tempest-slow/ad8c403/ : SUCCESS in 2h 16m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eab490517183b6384909440ea2d426557082038c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-02 08:58:14.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4fbba3a0d7116432bf9f1a5a6dc0df4d54ea42d6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-02 11:02:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d750a1f17e8dc5c06bc560f33605acfce2d288cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 13:45:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/2/check/openstack-tox-pep8/0337df2/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/2/check/openstack-tox-py27/0de1811/ : SUCCESS in 10m 05s\n- openstack-tox-docs http://logs.openstack.org/97/668397/2/check/openstack-tox-docs/09b57ed/html/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/2/check/openstack-tox-py36/21ac314/ : SUCCESS in 7m 44s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/2/check/openstack-tox-py37/1d10e24/ : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/2/check/openstack-tox-lower-constraints/d3d38da/ : SUCCESS in 10m 03s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/2/check/networking-ovn-tempest-dsvm-ovs-master/acabeaa/ : SUCCESS in 2h 09m 50s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/2/check/networking-ovn-tempest-dsvm-ovs-release/be019ef/ : SUCCESS in 2h 13m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/2/check/networking-ovn-tempest-dsvm-ovs-release-python2/fa95e2b/ : SUCCESS in 2h 35m 15s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/2/check/networking-ovn-rally-task/0c248cf/ : SUCCESS in 1h 56m 19s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/2/check/networking-ovn-dsvm-functional-py27/8bd2460/ : SUCCESS in 18m 44s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/2/check/networking-ovn-dsvm-functional/e712886/ : SUCCESS in 19m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/2/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/1824d05/ : SUCCESS in 2h 14m 18s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/2/check/networking-ovn-dsvm-grenade/3648913/ : SUCCESS in 2h 20m 48s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/2/check/openstack-tox-cover/957b305/cover/ : SUCCESS in 10m 28s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/2/check/networking-ovn-tempest-slow/cf16423/ : SUCCESS in 2h 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abafd789bf0af1ead9c352a58e51c39a3314bd45","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-03 14:37:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3db2655fd860f8f57e862f25d86803443513de88","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-03 14:38:40.000000000","message":"Patch Set 3: Workflow-1\n\nI\u0027m in the middle of creating unittests.","accounts_in_message":[],"_revision_number":3},{"id":"34868970208aa6ac6260f7bd3fbaa545c37b0530","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:34:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/3/check/openstack-tox-pep8/20c74ab/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/3/check/openstack-tox-py27/3216c8a/ : SUCCESS in 8m 35s\n- openstack-tox-docs http://logs.openstack.org/97/668397/3/check/openstack-tox-docs/1cbce13/html/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/3/check/openstack-tox-py36/f436446/ : SUCCESS in 9m 06s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/3/check/openstack-tox-py37/340ade9/ : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/3/check/openstack-tox-lower-constraints/3c2a93a/ : SUCCESS in 11m 33s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/3/check/networking-ovn-tempest-dsvm-ovs-master/a9e2445/ : SUCCESS in 2h 46m 54s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/3/check/networking-ovn-tempest-dsvm-ovs-release/8217a44/ : SUCCESS in 2h 41m 42s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/3/check/networking-ovn-tempest-dsvm-ovs-release-python2/becc689/ : SUCCESS in 2h 30m 35s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/3/check/networking-ovn-rally-task/7102001/ : SUCCESS in 1h 54m 52s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/3/check/networking-ovn-dsvm-functional-py27/84956e9/ : FAILURE in 22m 22s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/3/check/networking-ovn-dsvm-functional/2298ad6/ : SUCCESS in 18m 37s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/3/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/4c29f3b/ : SUCCESS in 2h 01m 20s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/3/check/networking-ovn-dsvm-grenade/8ad88ae/ : SUCCESS in 1h 55m 55s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/3/check/openstack-tox-cover/62eb835/cover/ : SUCCESS in 10m 17s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/3/check/networking-ovn-tempest-slow/1a2de36/ : SUCCESS in 2h 24m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1414abb96e0cdce00334ec2de5d9db3766726f4e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-05 12:59:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"17f6df0e19ca3cf60309b189e0c79730b8f405ce","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-05 13:02:47.000000000","message":"Patch Set 4: Workflow-1\n\nStill working on this. Code itself is ok, but missing a few unittests (set as pass), and looks like functional tests are flickering.","accounts_in_message":[],"_revision_number":4},{"id":"2b773ad5368514363c241347a9a89cb930c9a92c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-05 15:02:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aa40788c2fd1dfa75110d710631fb136919d4278","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-05 15:05:04.000000000","message":"Patch Set 5: Code-Review-1\n\n4 methods to be tested.","accounts_in_message":[],"_revision_number":5},{"id":"fc4ccf3a553a742b15f95d632514ad39107d78ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 17:41:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/5/check/openstack-tox-pep8/ad407d5/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/5/check/openstack-tox-py27/5d59488/ : SUCCESS in 9m 04s\n- openstack-tox-docs http://logs.openstack.org/97/668397/5/check/openstack-tox-docs/b8762af/html/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/5/check/openstack-tox-py36/e2625d0/ : SUCCESS in 9m 52s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/5/check/openstack-tox-py37/72f6506/ : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/5/check/openstack-tox-lower-constraints/dc5bd34/ : SUCCESS in 10m 26s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/5/check/networking-ovn-tempest-dsvm-ovs-master/90a1a46/ : SUCCESS in 2h 22m 26s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/5/check/networking-ovn-tempest-dsvm-ovs-release/78d4cc7/ : SUCCESS in 2h 31m 43s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/5/check/networking-ovn-tempest-dsvm-ovs-release-python2/6b8b16d/ : SUCCESS in 2h 21m 17s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/5/check/networking-ovn-rally-task/651a4cf/ : SUCCESS in 1h 49m 38s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/5/check/networking-ovn-dsvm-functional-py27/348abc0/ : SUCCESS in 18m 54s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/5/check/networking-ovn-dsvm-functional/5b50d0a/ : SUCCESS in 19m 47s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/5/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/d0b02f5/ : SUCCESS in 2h 18m 06s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/5/check/networking-ovn-dsvm-grenade/4c06f5d/ : SUCCESS in 2h 04m 33s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/5/check/openstack-tox-cover/a28e651/cover/ : SUCCESS in 12m 04s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/5/check/networking-ovn-tempest-slow/46f190e/ : SUCCESS in 2h 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a751dbf99f73c859700cac313d984658cd78798","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-08 10:59:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"32dabdc74e23b680685b1ec38a36c40bcaac3610","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-08 10:59:50.000000000","message":"Patch Set 6:\n\nready to review :)","accounts_in_message":[],"_revision_number":6},{"id":"c08e4a3019d90c5b4ac0a7742bd6cf6c2c89b9d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 13:38:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/6/check/openstack-tox-pep8/fcf324f/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/6/check/openstack-tox-py27/500d68d/ : SUCCESS in 9m 26s\n- openstack-tox-docs http://logs.openstack.org/97/668397/6/check/openstack-tox-docs/ad9ceeb/html/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/6/check/openstack-tox-py36/68a6f4f/ : SUCCESS in 8m 57s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/6/check/openstack-tox-py37/63c43cb/ : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/6/check/openstack-tox-lower-constraints/a88da99/ : SUCCESS in 10m 36s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/6/check/networking-ovn-tempest-dsvm-ovs-master/0611f39/ : SUCCESS in 2h 30m 09s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/6/check/networking-ovn-tempest-dsvm-ovs-release/a9d2d04/ : SUCCESS in 2h 14m 35s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/6/check/networking-ovn-tempest-dsvm-ovs-release-python2/9af5d6a/ : SUCCESS in 2h 24m 52s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/6/check/networking-ovn-rally-task/3f53549/ : SUCCESS in 1h 41m 49s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/6/check/networking-ovn-dsvm-functional-py27/87eec75/ : SUCCESS in 19m 49s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/6/check/networking-ovn-dsvm-functional/819a378/ : SUCCESS in 20m 14s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/6/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/fe2a4d9/ : SUCCESS in 2h 01m 17s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/6/check/networking-ovn-dsvm-grenade/7f33bda/ : SUCCESS in 1h 53m 04s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/6/check/openstack-tox-cover/9590dd0/cover/ : SUCCESS in 9m 58s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/6/check/networking-ovn-tempest-slow/34d0ad6/ : SUCCESS in 2h 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c02d19b22a3717f499b44f671e6e85df0ae8dc28","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-09 08:08:57.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)\n\nThanks Maciej!\n\nIt LGTM, the +1 is just about the lock question, I\u0027m not sure it\u0027s been called or if that\u0027s the desired behavior. Other than that consider it a +2!","accounts_in_message":[],"_revision_number":6},{"id":"ff18b6e4db0abccd7bf5a7ac0ff8210494c3fc26","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-09 08:09:31.000000000","message":"Patch Set 6:\n\nCCing Terry to see if he knows about that lock","accounts_in_message":[],"_revision_number":6},{"id":"3f9c8f495f91c616dba7c48e56ede584cb7e3a52","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-09 08:42:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d48433bbb2c1a0f4d311c7ecefb1a9ad872f8e8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-09 08:57:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3369cef72aa18602fb3d523dd7a88602683649b2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-09 12:27:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"23556dbaf6447e3c3fe17e83a88202b43c32bb5c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-09 13:10:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a235932f7d3faa2b195c87b83e561727312a048d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-10 10:07:03.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)\n\nWe need to find a way of proper locking.","accounts_in_message":[],"_revision_number":6},{"id":"e5e0c5cd83775574b399f14112576469d6db17c9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-10 13:41:10.000000000","message":"Patch Set 6: -Workflow\n\nAs we know the source of locking problem and we look forward for grouping patch as long as solving it in Octavia, may we continue with this patch and do not block it?","accounts_in_message":[],"_revision_number":6},{"id":"6026a2b38930ad2fd7fb958b0c937344e4808402","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-12 08:00:50.000000000","message":"Patch Set 6:\n\nLucas, I\u0027ve checked devstack with 2 Octavia APIs and while router-interface is attached this event is processed by both Octavia APIs, but everything is set fine (LR and LS references). I think we can go with this patch and then apply hashring group on it.","accounts_in_message":[],"_revision_number":6},{"id":"b6d06e6029a7c9e136077074a0966b8c9897bc46","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-07-12 09:08:15.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"aaf5586fb57723b16b36a6a0e3e1b45e88f98e5a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-12 09:21:02.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n\nSure, I\u0027ll drop it from here. wait a minute","accounts_in_message":[],"_revision_number":6},{"id":"d13e7f78f799d9ef856516620f87f53b198e4b8c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-12 09:29:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"328c1f560d5d9696c2fc15cf74f9d1e7da689d83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 13:12:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/7/check/openstack-tox-pep8/9b1d3aa/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/7/check/openstack-tox-py27/d4058de/ : SUCCESS in 8m 50s\n- openstack-tox-docs http://logs.openstack.org/97/668397/7/check/openstack-tox-docs/29e0018/html/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/7/check/openstack-tox-py36/4b9a554/ : SUCCESS in 8m 28s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/7/check/openstack-tox-py37/9032ffc/ : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/7/check/openstack-tox-lower-constraints/7a47a95/ : SUCCESS in 12m 19s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-dsvm-ovs-master/0ef539d/ : SUCCESS in 2h 48m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-dsvm-ovs-release/f04b498/ : SUCCESS in 2h 21m 51s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-dsvm-ovs-release-python2/3d5a42e/ : FAILURE in 2h 23m 44s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/7/check/networking-ovn-rally-task/ae7c3d4/ : SUCCESS in 1h 40m 44s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/7/check/networking-ovn-dsvm-functional-py27/014973a/ : SUCCESS in 20m 12s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/7/check/networking-ovn-dsvm-functional/1e54005/ : SUCCESS in 19m 01s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/7/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/367dc07/ : SUCCESS in 2h 09m 14s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/7/check/networking-ovn-dsvm-grenade/e282a86/ : FAILURE in 2h 16m 47s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/7/check/openstack-tox-cover/0484d15/cover/ : SUCCESS in 10m 59s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-slow/dc3d1a7/ : SUCCESS in 2h 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"16f9250f3a22e5c55e57bbcf0797cf02a3fbdbd5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-12 14:06:30.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"291a218c44b97b9b5a410810f9bdc5b40237ab76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 16:52:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/7/check/openstack-tox-pep8/a2cf4b4/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/7/check/openstack-tox-py27/d07f4fd/ : SUCCESS in 9m 23s\n- openstack-tox-docs http://logs.openstack.org/97/668397/7/check/openstack-tox-docs/c9e2caf/html/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/7/check/openstack-tox-py36/8ef6611/ : SUCCESS in 9m 07s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/7/check/openstack-tox-py37/958a178/ : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/7/check/openstack-tox-lower-constraints/f138e7a/ : SUCCESS in 9m 38s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-dsvm-ovs-master/f52acbf/ : SUCCESS in 2h 19m 24s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-dsvm-ovs-release/36fa497/ : SUCCESS in 2h 43m 01s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-dsvm-ovs-release-python2/6a1315a/ : SUCCESS in 2h 19m 54s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/7/check/networking-ovn-rally-task/7bc2afd/ : SUCCESS in 1h 48m 20s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/7/check/networking-ovn-dsvm-functional-py27/45bc011/ : SUCCESS in 20m 33s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/7/check/networking-ovn-dsvm-functional/1edbe77/ : SUCCESS in 19m 05s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/7/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/0c63be1/ : SUCCESS in 2h 15m 59s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/7/check/networking-ovn-dsvm-grenade/f7e9098/ : FAILURE in 10m 11s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/7/check/openstack-tox-cover/353d20b/cover/ : SUCCESS in 11m 26s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/7/check/networking-ovn-tempest-slow/9de94e9/ : SUCCESS in 2h 22m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e72dc6869a8c721495818d8a80f01d292ae125d3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-07-12 18:54:14.000000000","message":"Patch Set 7: Code-Review-1\n\n(9 comments)\n\nA bunch of useless nits, a couple of questions, and just one thing I think might be a bug.","accounts_in_message":[],"_revision_number":7},{"id":"98745f1ec94fd5565cdd786e5da632f074a88b48","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-15 09:18:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"60abc83421b1d1fd9ac04cf597da03882b5abd09","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-15 09:19:06.000000000","message":"Patch Set 8:\n\n(9 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f825260a0dbbcbdddf927530f91e6c87c3816e92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 13:06:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/8/check/openstack-tox-pep8/6e58e56/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/8/check/openstack-tox-py27/9dd556c/ : SUCCESS in 16m 11s\n- openstack-tox-docs http://logs.openstack.org/97/668397/8/check/openstack-tox-docs/b03f828/html/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/8/check/openstack-tox-py36/b9d047f/ : SUCCESS in 10m 16s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/8/check/openstack-tox-py37/f35611c/ : SUCCESS in 15m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/8/check/openstack-tox-lower-constraints/3afe900/ : SUCCESS in 10m 56s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-dsvm-ovs-master/d26983a/ : FAILURE in 2h 06m 11s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-dsvm-ovs-release/c8b2ae6/ : TIMED_OUT in 3h 10m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-dsvm-ovs-release-python2/2ecaba6/ : TIMED_OUT in 3h 14m 28s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/8/check/networking-ovn-rally-task/66acdb9/ : SUCCESS in 1h 45m 14s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/8/check/networking-ovn-dsvm-functional-py27/c2ed507/ : SUCCESS in 21m 04s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/8/check/networking-ovn-dsvm-functional/a882f56/ : SUCCESS in 20m 03s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/8/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/251a820/ : TIMED_OUT in 3h 46m 04s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/8/check/networking-ovn-dsvm-grenade/4eddc6d/ : SUCCESS in 2h 06m 22s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/8/check/openstack-tox-cover/7091be4/cover/ : SUCCESS in 10m 43s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-slow/74f381b/ : SUCCESS in 2h 22m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"23e7850d00d1307e7dc31ddb4d2632975cfcea01","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-15 13:20:55.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"682c192b57617263e17f08893b615b177827809f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-15 13:54:14.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b4019ebc691d1bcaa60021d0b78b27fb5d03fd8f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-15 13:59:39.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks Maciej!\n\nThe patch looks good. I am aware of the problems we have right now with the design in octavia but, I don\u0027t think we can do much within this patch either.\n\nSo let\u0027s get this in and then we can focus on solving the design problem as a follow up.","accounts_in_message":[],"_revision_number":8},{"id":"dd3376cfe5c477b75ae6b331a4136823eb8b5cff","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-07-15 14:51:31.000000000","message":"Patch Set 8: Code-Review+2\n\nwaiting for zuul for w+1","accounts_in_message":[],"_revision_number":8},{"id":"188b97d1147ca6ffb6f534bcad01dd8d40cbbff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 15:53:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/8/check/openstack-tox-pep8/9e334f8/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/8/check/openstack-tox-py27/cab4ae8/ : SUCCESS in 10m 57s\n- openstack-tox-docs http://logs.openstack.org/97/668397/8/check/openstack-tox-docs/c19cb06/html/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/8/check/openstack-tox-py36/2594a5b/ : SUCCESS in 10m 36s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/8/check/openstack-tox-py37/4738cb6/ : SUCCESS in 14m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/8/check/openstack-tox-lower-constraints/f3b9d6b/ : SUCCESS in 11m 23s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-dsvm-ovs-master/243d4ce/ : SUCCESS in 2h 20m 09s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-dsvm-ovs-release/7cf02dd/ : SUCCESS in 2h 19m 42s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-dsvm-ovs-release-python2/9fe8c36/ : SUCCESS in 2h 16m 01s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/8/check/networking-ovn-rally-task/a0a87af/ : SUCCESS in 1h 51m 38s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/8/check/networking-ovn-dsvm-functional-py27/27b52d3/ : SUCCESS in 25m 00s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/8/check/networking-ovn-dsvm-functional/7169174/ : SUCCESS in 20m 48s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/8/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/f3bc033/ : SUCCESS in 2h 02m 21s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/97/668397/8/check/networking-ovn-dsvm-grenade/4537ff6/ : FAILURE in 2h 22m 12s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/97/668397/8/check/openstack-tox-cover/0f6c78c/cover/ : SUCCESS in 11m 09s\n- networking-ovn-tempest-slow http://logs.openstack.org/97/668397/8/check/networking-ovn-tempest-slow/1eccc08/ : SUCCESS in 2h 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6b333759bad08ed265ad11017b11f59b0e4b762b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-07-15 15:57:22.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"b81eeb9e5e180858aa2a1af156d23eae6b27516c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 15:57:35.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f9055bedb0e14dad2c93ba0d92f372b1b2ba7ceb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 18:23:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"3d40e7ed4371c6db2b17d8f5031bce0cc8721538","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 18:23:39.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668397/8/gate/openstack-tox-pep8/9402095/ : SUCCESS in 6m 20s\n- openstack-tox-py27 http://logs.openstack.org/97/668397/8/gate/openstack-tox-py27/b5490dd/ : SUCCESS in 8m 38s\n- openstack-tox-docs http://logs.openstack.org/97/668397/8/gate/openstack-tox-docs/a1c66cb/html/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/97/668397/8/gate/openstack-tox-py36/9fb1149/ : SUCCESS in 10m 56s\n- openstack-tox-py37 http://logs.openstack.org/97/668397/8/gate/openstack-tox-py37/e6f87c4/ : SUCCESS in 11m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/668397/8/gate/openstack-tox-lower-constraints/629676c/ : SUCCESS in 12m 37s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/97/668397/8/gate/networking-ovn-tempest-dsvm-ovs-release/62766ef/ : SUCCESS in 1h 56m 26s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/97/668397/8/gate/networking-ovn-tempest-dsvm-ovs-release-python2/3bb1867/ : SUCCESS in 1h 57m 23s\n- networking-ovn-rally-task http://logs.openstack.org/97/668397/8/gate/networking-ovn-rally-task/c9b6bf2/ : SUCCESS in 1h 31m 49s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/97/668397/8/gate/networking-ovn-dsvm-functional-py27/d7a4b62/ : SUCCESS in 21m 32s\n- networking-ovn-dsvm-functional http://logs.openstack.org/97/668397/8/gate/networking-ovn-dsvm-functional/61ff67d/ : SUCCESS in 21m 19s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/668397/8/gate/networking-ovn-tripleo-ci-centos-7-containers-multinode/079b58e/ : SUCCESS in 2h 11m 14s\n- openstack-tox-cover http://logs.openstack.org/97/668397/8/gate/openstack-tox-cover/bc6df01/cover/ : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":8},{"id":"9ef205ef981ea1767ce4e21c05044dd5ec8b3c3c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-29 11:08:31.000000000","message":"Patch Set 8:\n\n\u003e Thanks Maciej!\n \u003e \n \u003e The patch looks good. I am aware of the problems we have right now\n \u003e with the design in octavia but, I don\u0027t think we can do much within\n \u003e this patch either.\n \u003e \n \u003e So let\u0027s get this in and then we can focus on solving the design\n \u003e problem as a follow up.\n\nI addressed some locking issues here:\nhttps://review.opendev.org/#/c/673269/","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"aa4268a09470aea28f4a8c9f1062f32f057cb151","revisions":{"9b603461493875201be787c45da11e4253ef17af":{"kind":"REWORK","_number":1,"created":"2019-07-01 09:20:22.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/1"}}},"commit":{"parents":[{"commit":"74998089b9184de46fab147392d1d76e5736507f","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/74998089b9184de46fab147392d1d76e5736507f"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-01 09:17:57.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9b603461493875201be787c45da11e4253ef17af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9b603461493875201be787c45da11e4253ef17af"}]},"branch":"refs/heads/master"},"ae7630c2b5d73d867a59126081c1a4173e0b544a":{"kind":"REWORK","_number":2,"created":"2019-07-02 11:02:11.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/2"}}},"commit":{"parents":[{"commit":"6859783f8329bd837d8e6d57bdc7081d3dc2dbb9","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6859783f8329bd837d8e6d57bdc7081d3dc2dbb9"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-02 11:01:01.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ae7630c2b5d73d867a59126081c1a4173e0b544a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ae7630c2b5d73d867a59126081c1a4173e0b544a"}]},"branch":"refs/heads/master"},"24e35ed44475d4e1bd5418c040c52a66473fbda6":{"kind":"REWORK","_number":3,"created":"2019-07-03 14:37:01.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/3"}}},"commit":{"parents":[{"commit":"6859783f8329bd837d8e6d57bdc7081d3dc2dbb9","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6859783f8329bd837d8e6d57bdc7081d3dc2dbb9"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-03 14:34:54.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/24e35ed44475d4e1bd5418c040c52a66473fbda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/24e35ed44475d4e1bd5418c040c52a66473fbda6"}]},"branch":"refs/heads/master"},"6a98b9619cd2b76e57dd527663a9306e60e22fe9":{"kind":"REWORK","_number":4,"created":"2019-07-05 12:59:56.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/4"}}},"commit":{"parents":[{"commit":"6859783f8329bd837d8e6d57bdc7081d3dc2dbb9","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6859783f8329bd837d8e6d57bdc7081d3dc2dbb9"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-05 12:46:29.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6a98b9619cd2b76e57dd527663a9306e60e22fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6a98b9619cd2b76e57dd527663a9306e60e22fe9"}]},"branch":"refs/heads/master"},"ff68dea93d0502af5eab44d6344cd22e5c9e751f":{"kind":"REWORK","_number":5,"created":"2019-07-05 15:02:56.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/5"}}},"commit":{"parents":[{"commit":"6859783f8329bd837d8e6d57bdc7081d3dc2dbb9","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6859783f8329bd837d8e6d57bdc7081d3dc2dbb9"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-05 14:58:00.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ff68dea93d0502af5eab44d6344cd22e5c9e751f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ff68dea93d0502af5eab44d6344cd22e5c9e751f"}]},"branch":"refs/heads/master"},"72cbe632f9751835989c0e1d35075189d5ceb24b":{"kind":"REWORK","_number":6,"created":"2019-07-08 10:59:05.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/6"}}},"commit":{"parents":[{"commit":"48cac11283b32e829d34f57c9f87bc207d184347","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/48cac11283b32e829d34f57c9f87bc207d184347"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-08 10:53:28.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/72cbe632f9751835989c0e1d35075189d5ceb24b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/72cbe632f9751835989c0e1d35075189d5ceb24b"}]},"branch":"refs/heads/master"},"98407e9327242e68bf93c01f676c0e91974651aa":{"kind":"REWORK","_number":7,"created":"2019-07-12 09:29:21.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/7"}}},"commit":{"parents":[{"commit":"48cac11283b32e829d34f57c9f87bc207d184347","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/48cac11283b32e829d34f57c9f87bc207d184347"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-12 09:29:19.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/98407e9327242e68bf93c01f676c0e91974651aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/98407e9327242e68bf93c01f676c0e91974651aa"}]},"branch":"refs/heads/master"},"aa4268a09470aea28f4a8c9f1062f32f057cb151":{"kind":"REWORK","_number":8,"created":"2019-07-15 09:18:29.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/97/668397/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/97/668397/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/97/668397/8"}}},"commit":{"parents":[{"commit":"48cac11283b32e829d34f57c9f87bc207d184347","subject":"Add Octavia OVN Driver\u0027s UT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/48cac11283b32e829d34f57c9f87bc207d184347"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-06-27 14:12:56.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-15 09:13:37.000000000","tz":0},"subject":"Add a new Event Handler for Octavia","message":"Add a new Event Handler for Octavia\n\nThis patch adds a new event handler for Octavia so that in\nthe event that a network is added to a router, the loadbalancer\non the network are also applied to the router and vice versa.\n\nThe process involves:\n- Adding a new connection class for Octavia to communicate with\n  OVN NB DB, setting up the connection and communicating with\n  the IDL.\n- Setting up events for Router interface addition and deletion\n- Updating Loadbalancer tables and Router/Switch tables based\n  upon the generated events.\n- Add network UUID to LRP external_ids\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If10016325517e4a1fc5f7cfac7f7bd58deb09b54\nCloses-Bug: #1794260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/aa4268a09470aea28f4a8c9f1062f32f057cb151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/aa4268a09470aea28f4a8c9f1062f32f057cb151"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
