)]}'
{"id":"openstack%2Fnetworking-ovn~673236","triplet_id":"openstack%2Fnetworking-ovn~master~I8b99a3481b4460ae32f2258dcf4190db9a8fed89","project":"openstack/networking-ovn","branch":"master","topic":"ovn-org","hashtags":[],"change_id":"I8b99a3481b4460ae32f2258dcf4190db9a8fed89","subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","status":"MERGED","created":"2019-07-29 09:05:26.000000000","updated":"2019-09-18 17:42:14.000000000","submitted":"2019-09-18 17:38:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673236-1568828298657-c3ff88a4","meta_rev_id":"47d00f5b0b22caff930a85df007b6a63e903b958","_number":673236,"virtual_id_number":673236,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-18 11:17:04.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2019-09-18 17:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-17 19:57:08.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":2,"date":"2019-09-16 12:41:53.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-09-17 19:57:08.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 09:46:06.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-08-01 08:58:32.000000000","updated_by":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2019-09-09 15:53:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-16 12:41:53.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-09-17 19:57:08.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-09-18 17:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e34ec9132a8547d57c5cb12f829368afc53fb19","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 09:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e29f6be78e594f234128eee0f98e94c609c3547b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 09:06:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ba53efde88cce1d4c74df9899c1c78e7541139b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-29 09:46:06.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"70c25baec998579b6641c5b7eff1359ef4843a7a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 10:02:29.000000000","message":"Patch Set 2: Workflow-1\n\n(3 comments)\n\nThanks a lot for the review!","accounts_in_message":[],"_revision_number":2},{"id":"14be6c6723eab8e64aacd8b4406f37b192e24f7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 11:43:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/673236/2/check/openstack-tox-pep8/a7b8612/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/36/673236/2/check/openstack-tox-py27/986dfa6/ : SUCCESS in 10m 47s\n- openstack-tox-docs http://logs.openstack.org/36/673236/2/check/openstack-tox-docs/6acad7e/html/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/36/673236/2/check/openstack-tox-py36/256545e/ : SUCCESS in 10m 00s\n- openstack-tox-py37 http://logs.openstack.org/36/673236/2/check/openstack-tox-py37/e6d8abd/ : SUCCESS in 11m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/673236/2/check/openstack-tox-lower-constraints/e788d1e/ : SUCCESS in 11m 48s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/36/673236/2/check/networking-ovn-tempest-dsvm-ovs-master/5c02c58/ : SUCCESS in 2h 11m 35s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/673236/2/check/networking-ovn-tempest-dsvm-ovs-release/fcef7ff/ : SUCCESS in 2h 18m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/36/673236/2/check/networking-ovn-tempest-dsvm-ovs-release-python2/6e48b27/ : SUCCESS in 2h 27m 53s\n- networking-ovn-rally-task http://logs.openstack.org/36/673236/2/check/networking-ovn-rally-task/bf2f565/ : SUCCESS in 1h 45m 36s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/36/673236/2/check/networking-ovn-dsvm-functional-py27/8a2f9e9/ : SUCCESS in 21m 12s\n- networking-ovn-dsvm-functional http://logs.openstack.org/36/673236/2/check/networking-ovn-dsvm-functional/0bf2670/ : SUCCESS in 16m 34s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/36/673236/2/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/728db21/ : SUCCESS in 2h 19m 41s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/36/673236/2/check/networking-ovn-dsvm-grenade/81971ff/ : FAILURE in 2h 22m 48s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/36/673236/2/check/openstack-tox-cover/37c0b6b/cover/ : SUCCESS in 10m 57s\n- networking-ovn-tempest-slow http://logs.openstack.org/36/673236/2/check/networking-ovn-tempest-slow/99aa33e/ : SUCCESS in 2h 25m 59s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master http://logs.openstack.org/36/673236/2/check/networking-ovn-tempest-full-multinode-ovs-master/7e5eb20/ : FAILURE in 2h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e8382f026dc7fc54e517f85b30252ee1f38234f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 12:47:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba24bac8e9f87c205b5816dcd6436373579e3e76","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 12:53:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4bbb2df84533d0de052db4e5bb9ff803e8c6aebe","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 13:19:50.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"93f85c20067869b91004229079d1ba7af8e48c8a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 13:57:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a432569a48b508ee143e08ba8f56d28f36ed1331","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 14:27:00.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"327b786c66536156339913c22f064909d77be416","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-29 15:22:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"004086977633f29a9225e12d19cca4ff7132a6b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:06:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/673236/6/check/openstack-tox-pep8/c35c4f5/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/36/673236/6/check/openstack-tox-py27/5c90b73/ : SUCCESS in 10m 55s\n- openstack-tox-docs http://logs.openstack.org/36/673236/6/check/openstack-tox-docs/051b9ac/html/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/36/673236/6/check/openstack-tox-py36/8deabff/ : SUCCESS in 10m 03s\n- openstack-tox-py37 http://logs.openstack.org/36/673236/6/check/openstack-tox-py37/449b71c/ : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/673236/6/check/openstack-tox-lower-constraints/5206b5f/ : SUCCESS in 14m 36s\n- networking-ovn-tempest-dsvm-ovs-master http://logs.openstack.org/36/673236/6/check/networking-ovn-tempest-dsvm-ovs-master/a090e2b/ : SUCCESS in 2h 35m 18s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/36/673236/6/check/networking-ovn-tempest-dsvm-ovs-release/fbcd8ff/ : SUCCESS in 1h 59m 49s\n- networking-ovn-tempest-dsvm-ovs-release-python2 http://logs.openstack.org/36/673236/6/check/networking-ovn-tempest-dsvm-ovs-release-python2/9354cb9/ : SUCCESS in 2h 15m 03s\n- networking-ovn-rally-task http://logs.openstack.org/36/673236/6/check/networking-ovn-rally-task/3eb9d71/ : SUCCESS in 1h 40m 37s\n- networking-ovn-dsvm-functional-py27 http://logs.openstack.org/36/673236/6/check/networking-ovn-dsvm-functional-py27/6a387f1/ : SUCCESS in 24m 55s\n- networking-ovn-dsvm-functional http://logs.openstack.org/36/673236/6/check/networking-ovn-dsvm-functional/47436f9/ : SUCCESS in 21m 10s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/36/673236/6/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/6131f57/ : SUCCESS in 2h 23m 09s\n- networking-ovn-dsvm-grenade http://logs.openstack.org/36/673236/6/check/networking-ovn-dsvm-grenade/22a1151/ : SUCCESS in 2h 17m 08s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/36/673236/6/check/openstack-tox-cover/05d06ed/cover/ : SUCCESS in 12m 34s\n- networking-ovn-tempest-slow http://logs.openstack.org/36/673236/6/check/networking-ovn-tempest-slow/a893f3f/ : SUCCESS in 2h 25m 53s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master http://logs.openstack.org/36/673236/6/check/networking-ovn-tempest-full-multinode-ovs-master/18f2968/ : SUCCESS in 1h 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2d1794548b78cb8829dac5013e5cd198e3f4904b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-08-14 13:50:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"db7defe5c6979c7b0177dab6e9b39a110098b009","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-08-14 13:50:36.000000000","message":"Patch Set 7: Workflow-1\n\nLatest patch-set won\u0027t work until we get [0] merged\n\n[0] https://patchwork.ozlabs.org/patch/1144884/","accounts_in_message":[],"_revision_number":7},{"id":"be12eb73dea2b7b4b483ed217cc32a73a71e0437","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 16:19:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3368a1630e094094afc2a90b8a96e11d : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54377d52d3ea450284d11476e8c62a5f : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c275436a98f84ecdb7ba46d47a08d0f4 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d989910186c244d092c31005d7e35daf : SUCCESS in 9m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2363abebc0f049019080977963898f71 : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3f6f8cb4ac14689a5e2c14a34c67b6c : SUCCESS in 10m 20s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/4c1ed85ef2da46ab9555d99508075115 : FAILURE in 35m 00s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d3d348666b7a46139533e10294cbaa88 : FAILURE in 27m 39s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ad5a6dd12eb7492596f1006f8d3b474f : FAILURE in 35m 03s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e1c11eb2c9644dca7e4a77c0ac71ef9 : FAILURE in 31m 16s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/9f7ed14173e740b99882223d9479e999 : SUCCESS in 21m 17s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c5fdc2c73364d308ee84111b2f60fee : SUCCESS in 21m 41s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c5fa46dc9fdf4bbfbfe941af99cfefc9 : SUCCESS in 2h 21m 46s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/8aa55882205647629d6ebd1b4f5685d7 : SUCCESS in 1h 58m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fbc403f98da434eb693ddaf91d4950a : SUCCESS in 10m 22s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ecc27e02df814fa1a576e9e01572428c : FAILURE in 43m 11s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/421278bb283a4d89afd7ee8a2d4c18cf : FAILURE in 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e7b8ac9aa0c52d1ceb1ecedabe5a8397a1a3543a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 00:58:55.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"2ae5718bc1b593ae8e3e311072f32d4d970ef661","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-09 12:39:06.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"593d3f7f13cb8e7a575438fe010ad2dcbf65963f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-09 12:39:32.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5e5e34f2deb10172652a4fbc9693eb2e34e52ce7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-09 14:00:02.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"37b172d8d091c4a0c036c4b25ff26a4785550c46","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-09 15:53:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"158b2fdd9463e42429a507f853921796b2a8b557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 17:04:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db68e00d3874decb4508161e7d9b81b : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6dafdf7665ea4e48b402faa81ff534f3 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d036f44f3f7c4ecd809f0cbb62496677 : SUCCESS in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8ca8d1121ef4570bd94f0ca18a1a0f8 : SUCCESS in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9241c69406c64bd0ab3a01e4f8d69fa0 : SUCCESS in 14m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3f45950778d46eda619c0108431b2c2 : SUCCESS in 17m 54s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/d274fcfb0a2a483788b2b6b142dd643d : FAILURE in 15m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6632d6af95fe459fac175dcec73c851c : FAILURE in 28m 03s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d4520036c1b74cfb9e2a0eb4a971cf13 : FAILURE in 27m 04s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c9aaf127f3d14e50865f5bee9bece13b : FAILURE in 15m 28s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/4c222b34c5564d9ba8b3edf714b8f0a4 : FAILURE in 12m 44s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/fded1100e6fa4211bca2b86144128231 : FAILURE in 14m 13s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ef51392358e84aee879fa60a80375d8b : SUCCESS in 2h 48m 12s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/5f8a9a7b084a4dc2a475e974e31a5604 : FAILURE in 1h 58m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8ea9d065da34ce396dd276e5cd8eadb : SUCCESS in 13m 21s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b10c0797fc744361a6ccc9602d016d64 : FAILURE in 27m 10s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/221256c7217945b1bbb7a1886bcac61c : FAILURE in 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"69ca0617f2845ecd33dbe047573380d6b1999661","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-12 18:04:46.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4f409208c09661fd9f3854aa37e60c52998acb8c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-12 18:08:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f7b2c529bafed530a413f1f1cca64cce467164cd","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-12 18:31:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"95c36c00d533e9e5134fc8e8171455b4e3b35123","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-12 18:54:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"512751c9872b3fba492253a35764a4e38d12d8af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 22:27:06.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d4385cd8cc4212baee9bb073be4380 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6ca8fc8e9da47cdb5aad464d211aced : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e835bd1ff2b44837b7dd23bd4f2118ad : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad1a4e6690df40af843693919ac98f80 : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce78280e25de42ccac10603b4cb5db19 : SUCCESS in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c30ae397accc445499a43a5e233ef882 : SUCCESS in 10m 20s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f08a0a645da5425881665afd9c66c59f : SUCCESS in 2h 16m 17s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/521d62dd65f94341b93b9fd7af6fb8a1 : FAILURE in 34m 13s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/07bc67f0994d4c70aa6298971d5e3d55 : FAILURE in 25m 05s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a12b83b79c2a4ff4b7c025a58732402a : FAILURE in 35m 42s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/75695985bb27477a8ade4acd531ab996 : FAILURE in 16m 41s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a7b81f9856a4c0a8173573b0ab4fe9d : FAILURE in 15m 54s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0beec22cf7b2404a998401d08ebf5b42 : TIMED_OUT in 3h 26m 02s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/3ac91f025bbe496ca8586ea0f74fb43c : SUCCESS in 2h 12m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95fbb44865c54466bdabae74d5112d54 : SUCCESS in 9m 59s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6d6b3d43bcbc47358740fa383e1ae4ca : SUCCESS in 2h 05m 53s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9b9bc66e0df94af4a3385737e89684c5 : FAILURE in 2h 26m 57s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"873dd63d8e207a4c2168b474e98981073a0e1026","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-13 12:36:44.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"96b0de7db7a3b0d7dabc0d1a36d84e1099ac5f16","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-13 14:23:01.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ed2314be57a60bd18433dc126c5c08916b98cff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 18:47:19.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02bc82777e79456cb7b81a4ba50eb1bb : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/769c78e2158b495ea2b43459f4c4486a : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0b91535453454e94aeca95ad734ced : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31be0c71c22142efb417ed3b5b085419 : SUCCESS in 9m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b903ece2f2e4c439dab5564fc85c07d : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b305d5ecdaa4a77a81125a8f648de66 : SUCCESS in 14m 52s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5620bb1ed3dc401b898a57c1688608f3 : SUCCESS in 1h 56m 13s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/82494db9ee96429abc63fab28df97010 : SUCCESS in 2h 25m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f0ae92394c054f63ba4ef1bcd8bd108f : SUCCESS in 2h 16m 53s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7e43bcc742b14f8698f0a3f31a6c5df8 : FAILURE in 31m 28s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c996a3f9b4b3408ca595d6c0555866a7 : FAILURE in 38m 46s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/69f89f12befc4dd29da33f2de59bf6eb : SUCCESS in 25m 07s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0921ccdbc7c8466594b9c67024e9147b : SUCCESS in 3h 21m 58s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/2d57de7589ab45658744b544cb85af00 : SUCCESS in 2h 12m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fa50b48d4c943078dfd774e85d88a20 : SUCCESS in 10m 47s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/af22b04c2bba48d0b1b1f71ba441b843 : SUCCESS in 2h 47m 01s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/37da73fd8bfb45bc811814d19edae441 : SUCCESS in 2h 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"82b14ea1e844c32fbe93b46916619c378d50f815","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-13 18:59:13.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"f733fb43dd0be36956d6310c30136423fe1833b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 22:39:09.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e744eb8b55304e4dbc843efb7468e4b7 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/369d89f2e58f4abf89d7496afdbfdaf5 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77b3852790448c3bcc3f87ada2d79f4 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf52f74aac9a49dabf167fb9ecbe5964 : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e7850b60e994ba7ba351b4a8205d6c2 : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a49c34c70f024e72a16f8976b056f53a : SUCCESS in 11m 26s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/415297200d97463f8696e0e3fd1434b8 : SUCCESS in 2h 14m 16s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8484e696290c485ba3da054d114d4104 : SUCCESS in 2h 09m 16s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/caa750757c604835a1db50be3fb715fb : SUCCESS in 3h 06m 59s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ac19db933bd4524ac951f69db74a99f : FAILURE in 56m 06s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/7780657c90f34c948e5e54ec1a5c23b9 : SUCCESS in 19m 02s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4f181397c644c1fbbb55ea70b3031b3 : SUCCESS in 17m 04s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1fc3389c9daa431f9a8f4a81de8202f6 : TIMED_OUT in 3h 23m 40s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/ff02a7506b2744a9ac3c7fb3d3b37cd8 : SUCCESS in 2h 07m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67c0174d61874a77aa9417a9e823a5be : SUCCESS in 10m 06s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/04270f0de1444ee6b9ba07e635c4f222 : SUCCESS in 2h 29m 43s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e08550e948064de4926048cdbb14c910 : SUCCESS in 2h 23m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9e2816f7e462eb3bcfd565f36173c6204e7fbcb3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-14 12:30:33.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"046bd3792efad09de1d6f3d316672644b4a2e0ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 15:40:28.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371b44c48924404d88901008b20d7ff8 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/898da34ae66a4b7291433080b543f075 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6731eb398a964e2ab3b0ca73a18fc6e2 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d6a196e24bc4ed58fda3fa92d7e6995 : SUCCESS in 8m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/842b45831c1342fab4b432cf92a782e9 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7da679c9ca42496299c1c1d9906e00a6 : SUCCESS in 10m 37s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/8dbbb091e0844665816039feed441246 : SUCCESS in 2h 08m 02s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/81db876973714fb3b5bf15d53e64e289 : SUCCESS in 2h 09m 26s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a5f2ffb95c2c4c54a6302d20a00b6ec5 : SUCCESS in 2h 14m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb138661f72d42719217b208d62acb68 : FAILURE in 32m 39s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ffc6488542164d6dbb5ab26602e30b88 : SUCCESS in 15m 56s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/504fb7612971431886f405ca7542b7db : SUCCESS in 23m 44s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fda48ea8d1714adaaf2a8a4cc6023fa7 : SUCCESS in 3h 03m 43s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/927963955c05492a8117cf04b232256a : SUCCESS in 1h 53m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5380ca571e944b0afc70276a895a8ea : SUCCESS in 10m 21s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/37e81fd600a247b9a11cff053dbfd39d : SUCCESS in 2h 20m 08s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/48709797c8c84f2fb190108339eb6ff7 : SUCCESS in 2h 07m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f3e25267d46d846a0708d6289f00a810d8fc4ece","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 11:24:25.000000000","message":"Patch Set 15:\n\nThis is ready for review :-)","accounts_in_message":[],"_revision_number":15},{"id":"bce90e78cd473778dd7ed70eea5a3a60634ee004","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-16 11:58:45.000000000","message":"Patch Set 15:\n\n(11 comments)\n\nnit-pickah is back :)","accounts_in_message":[],"_revision_number":15},{"id":"85da70114f03670d2ed4c3f27edf3104d690558a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 12:17:09.000000000","message":"Patch Set 15:\n\n(10 comments)\n\nthanks a lot kuba for the review!","accounts_in_message":[],"_revision_number":15},{"id":"73ab1921a0e20e592304ab0b83fec356fb10a637","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 12:17:13.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"198c22cc65571026a58f437ba8923dc0b6c9fed8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 12:31:07.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"30d111b423f6fe1323220a3947a042e12f92a545","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 12:34:16.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"9ad88f53c7144d53aa9ca11d3bca6e9199848108","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-16 12:35:03.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3e93605e12ab2ed4eeea672370e3a6e9bbd9822c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-16 12:41:53.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)\n\nThanks :)","accounts_in_message":[],"_revision_number":16},{"id":"126c30391fdc1501a33fe29ee604b078971727e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 16:35:30.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a29c88915f42f2a9f4b207a3a3d0fc : SUCCESS in 9m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e8e1a0859bb424ab44a066d0313622e : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae62a61bb2c6426584cec086c1763455 : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f016e480037145f1b528c0b0ff0696a5 : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/287950cf4f8c470fb8cc86e97be930f6 : SUCCESS in 12m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/445ed963d68f4dc3a37b10e821fba85e : SUCCESS in 11m 52s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5a1ff277ffb24109b21a2f77b4e86cdb : SUCCESS in 2h 23m 36s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c190ef2ca203499a9d962d0180c74955 : SUCCESS in 2h 26m 20s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/4eca50854b79407197ee53ee901a7139 : SUCCESS in 2h 55m 04s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a027cb6b8ca44bfbb02cda69fd63c84 : SUCCESS in 2h 08m 09s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ca3332f0a92843ebb16ab8b04a1c1597 : SUCCESS in 16m 51s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/681a4677226f4a8a91a71b2c0c00cf5b : FAILURE in 23m 14s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/42040ac15a8441378f0da738dfb4ab3e : SUCCESS in 3h 14m 37s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/38f6d3f55a5d4bf181bb3997c302cc1b : SUCCESS in 2h 05m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7beb6b1658564ac585abd4e764d9725d : SUCCESS in 10m 10s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a08896e14f9d4031bff130471a991539 : SUCCESS in 2h 25m 50s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e89a64fe4d204a55b365853f0fcf8530 : SUCCESS in 2h 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4e05e784c2deaf9b0ab3466e600d1093e673cbb8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 16:46:34.000000000","message":"Patch Set 16:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"d149eddaf7f9ce60c9b2f1b51a6721bb03639005","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 21:25:45.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"a9f9e5be9db2f0afeea275fc37d3b7d2ecb57d3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 00:55:44.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9826c582aba4bf79e11cb874f41a10d : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d9d2fede1f6483aa216801323f78ec9 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2ba91d9fc343c596d37750ae67e057 : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0a43f44551141cc9d453b40accb27d0 : SUCCESS in 9m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2dbf01d18e446c99b3e16e466329bc6 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76fc0a86bf53484b8ac4be82a446c219 : SUCCESS in 11m 37s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/86a50b87cbea4131a2bdd73de8bfba25 : SUCCESS in 2h 16m 33s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6c80e6aa500640c198a335ca3517f64c : SUCCESS in 2h 36m 22s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/41e96a71c7f74ef4a2feebc6438f001d : SUCCESS in 2h 32m 51s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/669269157fa54e1badc0d5dbf28f04b4 : SUCCESS in 1h 29m 54s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/21a70bccb0f74ed2af4d0a099b76e6df : SUCCESS in 21m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5fbf8eb36a754a19b047882d42a4fd36 : FAILURE in 22m 23s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/568905c04df34a0e830d3d3178efef4f : SUCCESS in 3h 03m 21s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/6124a978647d4ba187685ac89a1f077e : FAILURE in 2h 24m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b34784c4e26f448a9f53aa5efd180ffe : SUCCESS in 12m 32s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c80bdf750d994a7d980cb78d72a1e92d : SUCCESS in 2h 23m 07s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b62cd62a837f4dfe9350bf4dfbb82a07 : SUCCESS in 2h 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"340947fbdba1e167e1857fb1a89afd80bc6b16cf","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-17 11:20:43.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"16bd3bff17520a75f6342eb31cf8b897ee5c5485","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:40:45.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad4c6134852476da5e967ff257bcfbe : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4475908d8eaf47129f2fd1dc34f4c893 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6b7c8ce8d84501970f62f0ec08ae10 : SUCCESS in 11m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/816cb69e801a4116a52f22415d391f64 : SUCCESS in 12m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc4e0927361949b3a3a27dd0857b2cb1 : SUCCESS in 15m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9717a97b57eb43878dcc1d1002c6cdcb : SUCCESS in 18m 01s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/6f41f7461e5b403bb5e01c4c33a1ccb3 : SUCCESS in 2h 28m 00s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/83a945ce2d3c479f9af6c1d2ecadcc58 : SUCCESS in 2h 48m 27s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/11f2eb506e2d4633bcbea531db1ac391 : SUCCESS in 2h 35m 05s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f7c03f2510c4a9d80a659def875199d : SUCCESS in 1h 48m 31s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/2bb371a832344cd682f51c5f436a3570 : SUCCESS in 31m 59s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/4fbd852be6804d6d95e7b5a9c5a12f59 : SUCCESS in 19m 22s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2602ff7bb352426c8ccde51a61c258ea : POST_FAILURE in 2h 05m 14s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/cb8118768e4d411b88438a6a00bee4f2 : SUCCESS in 2h 14m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d99565d4d47f42f6853349a5e0b17c69 : SUCCESS in 15m 29s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f89cf6222ffc4179bdf932d7a3ad3023 : SUCCESS in 2h 31m 37s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d58a014bcc534e07a0a1f88c680db493 : SUCCESS in 2h 07m 08s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4f552f19f07f6b1be1dd8c1050d215500b2b6f7b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-17 16:12:13.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"27cf145c0f162dd495ddfd9e837e1325a60d0119","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-17 19:57:08.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"ab93f1faaf07855d165d56fede8e8d5ceec639c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 23:42:00.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c6366487d8a443da556a6211a7c5816 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41230efec14548d4a3962c0647b7717a : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe6f5e311474815a9a0d146805eb20b : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d74eeda5f1f421a88e83f35ffbb7124 : SUCCESS in 10m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bdcaf8ca94642f7ad00360fee7df386 : SUCCESS in 14m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c41564c566bc43bfbc57f88032574d79 : SUCCESS in 13m 28s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/e258d0aa9aa04273a2960df1f416d16f : SUCCESS in 2h 18m 34s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9baaccb2208840e3bbde85631cdb4266 : SUCCESS in 2h 21m 34s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/7eca4f1d7b6c414db347225ae061b027 : SUCCESS in 2h 09m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b73dae59d53e4c9abd11cbb0d76e6409 : SUCCESS in 1h 34m 01s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/70728221c869440785493095ce70c4a7 : SUCCESS in 19m 15s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1f0fd1f2e9a4bf1ba2cb5a3a16bfff8 : SUCCESS in 19m 24s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b54b283af0ea412e966c430ee7b2ac18 : TIMED_OUT in 3h 22m 28s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/f9cfdb3db56b424abde076c5d4bd6508 : SUCCESS in 2h 14m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4b63018553241c9a41fc93160c953c8 : SUCCESS in 10m 53s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a3f309de5cc545718cb17e28e6644d7a : SUCCESS in 2h 22m 15s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/da5d1743fe124486814f11cdc952d969 : SUCCESS in 1h 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"16723baff38e8f093b017ad4ef33f53bdb207f6f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-18 11:17:04.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"1d51670d012fccdf6b445c9d02eeac1b1faf6450","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:32:05.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fb4ebe2f794bfb9acd2abd03aa9568 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6558531e46b74ea0841feca98cabac06 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44ca4317c3844aecb3f9bce8eaac0a48 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/168c61e4a6a64520ae611470718376a7 : SUCCESS in 8m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6070ee7153b4422cae1603b850abcb4e : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b77930ccf184da499a7fe2f1228e646 : SUCCESS in 13m 19s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/abcf3a4f77664254918c1a76c7a22c5a : SUCCESS in 1h 52m 19s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0c79586e45194ad89d890844037ea7ae : SUCCESS in 1h 53m 16s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3db7139971ff44fa8b6833afe95bfb73 : SUCCESS in 2h 16m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/88af5f401e594762a5686cd99bfe3afe : SUCCESS in 1h 42m 37s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b8aea2d487b340e3a97c37e6b6e381b3 : SUCCESS in 23m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/04a54b4ea7c242d992b7302a21c1cdba : SUCCESS in 23m 17s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fd06ffe931a347b298a1fe2078dcd13e : SUCCESS in 3h 02m 06s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/963a9f4617424008b93b2ea096a49089 : SUCCESS in 2h 01m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a881bc6fa9b4426aa602611efaf4793 : SUCCESS in 9m 04s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fdf9f62d472d455cbbbea3e39d0e7f19 : SUCCESS in 2h 21m 57s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/28803e4dc71840f69541e7acfe1dbe57 : FAILURE in 1h 59m 19s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"be3f6dbcecc65b239f65a655c98959700e978a0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:32:34.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"d2eb5ff13a0bd48c22d76b0cb6e9fef53eafd2c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:38:18.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94736f7cc7a418a8eb88fff4e4eb730 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c7a2a40ba15469799d7801c724abc5a : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d2eb8c00834f1aafaa3a72534ec063 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf12b0d3d7be4ac2821570354627b832 : SUCCESS in 11m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0501513cfc04ec9bdbd3c25df7f78c2 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78a57d758d6b468882121068c863fa33 : SUCCESS in 10m 46s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7a7387e71bc4419f9ee96d79f2053ab8 : SUCCESS in 2h 08m 04s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/71e97ac2c68549d08f1ca9e9857398fd : SUCCESS in 2h 28m 42s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b2a763d92b944ffb9a912faafaa0cf86 : SUCCESS in 25m 48s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dddeb6048974e8aa89dc35066eada5c : SUCCESS in 21m 40s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/45ce04d9846d4800980662acca9029bd : SUCCESS in 3h 03m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d71e14b23b6a45aebc6e166ca5c8bbc5 : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":16},{"id":"7791407475dd106272f38346ed22d970c124ac6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:38:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"d956a66921efe28fc2094cb55ffc759e11b0e8bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:42:14.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2861e0f8c545f4a560757298c961d5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"0337b2e582fa96c8937794bc5a72ec744e979ad1","revisions":{"c6721e92f618ca6e344c2e64658d7c89215b6c7f":{"kind":"REWORK","_number":1,"created":"2019-07-29 09:05:26.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/1"}}},"commit":{"parents":[{"commit":"31919d6c894ed9bee42d435cd25e91f4e587f6d2","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/31919d6c894ed9bee42d435cd25e91f4e587f6d2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 09:04:06.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch.\n\nThe patch includes new functions based on the \"compile_ovs\" one changing\nit to OVN.\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master, which will be the 2.12\nversion).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c6721e92f618ca6e344c2e64658d7c89215b6c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c6721e92f618ca6e344c2e64658d7c89215b6c7f"}]},"branch":"refs/heads/master"},"9a7729cf43cd88e31bfdabebb178adc7dbbed484":{"kind":"REWORK","_number":2,"created":"2019-07-29 09:06:59.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/2"}}},"commit":{"parents":[{"commit":"31919d6c894ed9bee42d435cd25e91f4e587f6d2","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/31919d6c894ed9bee42d435cd25e91f4e587f6d2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 09:06:52.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch.\n\nThe patch includes new functions based on the \"compile_ovs\" one changing\nit to OVN.\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master, which will be the 2.12\nversion).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9a7729cf43cd88e31bfdabebb178adc7dbbed484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9a7729cf43cd88e31bfdabebb178adc7dbbed484"}]},"branch":"refs/heads/master"},"5995f0df2905d4f5aeeae27739d23fa5b9574979":{"kind":"REWORK","_number":3,"created":"2019-07-29 12:47:00.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/3"}}},"commit":{"parents":[{"commit":"31919d6c894ed9bee42d435cd25e91f4e587f6d2","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/31919d6c894ed9bee42d435cd25e91f4e587f6d2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 12:46:36.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5995f0df2905d4f5aeeae27739d23fa5b9574979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5995f0df2905d4f5aeeae27739d23fa5b9574979"}]},"branch":"refs/heads/master"},"e20ae3ca2e69692fd2f879be330496d10b03344c":{"kind":"REWORK","_number":4,"created":"2019-07-29 12:53:09.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/4"}}},"commit":{"parents":[{"commit":"31919d6c894ed9bee42d435cd25e91f4e587f6d2","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/31919d6c894ed9bee42d435cd25e91f4e587f6d2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 12:52:54.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e20ae3ca2e69692fd2f879be330496d10b03344c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e20ae3ca2e69692fd2f879be330496d10b03344c"}]},"branch":"refs/heads/master"},"baefa23dc70e7d7a60dd27511f6986c4e7826704":{"kind":"REWORK","_number":5,"created":"2019-07-29 13:57:55.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/5"}}},"commit":{"parents":[{"commit":"31919d6c894ed9bee42d435cd25e91f4e587f6d2","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/31919d6c894ed9bee42d435cd25e91f4e587f6d2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 13:38:30.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/baefa23dc70e7d7a60dd27511f6986c4e7826704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/baefa23dc70e7d7a60dd27511f6986c4e7826704"}]},"branch":"refs/heads/master"},"139839e3d30a7a66c5f2fba7b70b4ee6e88b5b56":{"kind":"REWORK","_number":6,"created":"2019-07-29 15:22:32.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/6"}}},"commit":{"parents":[{"commit":"31919d6c894ed9bee42d435cd25e91f4e587f6d2","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/31919d6c894ed9bee42d435cd25e91f4e587f6d2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 14:29:43.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/139839e3d30a7a66c5f2fba7b70b4ee6e88b5b56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/139839e3d30a7a66c5f2fba7b70b4ee6e88b5b56"}]},"branch":"refs/heads/master"},"abfea3d28258da480bdb2cd294547e8001325bf1":{"kind":"REWORK","_number":7,"created":"2019-08-14 13:50:01.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/7"}}},"commit":{"parents":[{"commit":"642739c0eb0c46278effe843f6a9c509660875f1","subject":"Merge \"Support for Router Scheduling on addition/removal of chassis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/642739c0eb0c46278effe843f6a9c509660875f1"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-08-14 13:25:52.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/abfea3d28258da480bdb2cd294547e8001325bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/abfea3d28258da480bdb2cd294547e8001325bf1"}]},"branch":"refs/heads/master"},"53dfa11bc88fe3e91486d00acf05bd8cd087c8c4":{"kind":"REWORK","_number":8,"created":"2019-09-09 12:39:06.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/36/673236/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/8"}}},"commit":{"parents":[{"commit":"642739c0eb0c46278effe843f6a9c509660875f1","subject":"Merge \"Support for Router Scheduling on addition/removal of chassis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/642739c0eb0c46278effe843f6a9c509660875f1"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2019-09-09 12:39:03.000000000","tz":0},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/53dfa11bc88fe3e91486d00acf05bd8cd087c8c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/53dfa11bc88fe3e91486d00acf05bd8cd087c8c4"}]},"branch":"refs/heads/master"},"834f992d15cdf58a4a5b5b37f5d99cce67f15b67":{"kind":"REWORK","_number":9,"created":"2019-09-09 14:00:02.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/36/673236/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/9"}}},"commit":{"parents":[{"commit":"642739c0eb0c46278effe843f6a9c509660875f1","subject":"Merge \"Support for Router Scheduling on addition/removal of chassis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/642739c0eb0c46278effe843f6a9c509660875f1"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-09 13:55:27.000000000","tz":0},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/834f992d15cdf58a4a5b5b37f5d99cce67f15b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/834f992d15cdf58a4a5b5b37f5d99cce67f15b67"}]},"branch":"refs/heads/master"},"823209ea3037cfc445f666ad6ac435a99765b6ce":{"kind":"REWORK","_number":10,"created":"2019-09-12 18:04:46.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/10"}}},"commit":{"parents":[{"commit":"cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0","subject":"Merge \"Handle events in separate IDL instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-09-12 18:04:00.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/823209ea3037cfc445f666ad6ac435a99765b6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/823209ea3037cfc445f666ad6ac435a99765b6ce"}]},"branch":"refs/heads/master"},"ff08a349a4a59d928aeb5dbfbc0e9a6c6295a030":{"kind":"REWORK","_number":11,"created":"2019-09-12 18:08:42.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/11"}}},"commit":{"parents":[{"commit":"cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0","subject":"Merge \"Handle events in separate IDL instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-09-12 18:08:11.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ff08a349a4a59d928aeb5dbfbc0e9a6c6295a030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ff08a349a4a59d928aeb5dbfbc0e9a6c6295a030"}]},"branch":"refs/heads/master"},"0e9f73fdf9b202e2764d8297ceee241d5e59ac46":{"kind":"REWORK","_number":12,"created":"2019-09-12 18:31:51.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/12"}}},"commit":{"parents":[{"commit":"cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0","subject":"Merge \"Handle events in separate IDL instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-09-12 18:31:43.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0e9f73fdf9b202e2764d8297ceee241d5e59ac46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0e9f73fdf9b202e2764d8297ceee241d5e59ac46"}]},"branch":"refs/heads/master"},"9dad929389483a5fa2365ffc680199ead6608ff4":{"kind":"REWORK","_number":13,"created":"2019-09-12 18:54:26.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/13"}}},"commit":{"parents":[{"commit":"cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0","subject":"Merge \"Handle events in separate IDL instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-09-12 18:54:15.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9dad929389483a5fa2365ffc680199ead6608ff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9dad929389483a5fa2365ffc680199ead6608ff4"}]},"branch":"refs/heads/master"},"f9a9b98d234f6f13f0f5e5c5a8c1f833ef189d55":{"kind":"REWORK","_number":14,"created":"2019-09-13 12:36:44.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/14"}}},"commit":{"parents":[{"commit":"cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0","subject":"Merge \"Handle events in separate IDL instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-09-13 12:36:35.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nWhen building the kernel modules and/or using the \"ovs\" python library\nfrom source is enabled, they will be compiled from the OVS repository,\nnot from the new OVN one (although the new OVN repository has an ovs/\nand python/ submodule, it will be later deleted. This patch is one step\nahead for when it happens).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f9a9b98d234f6f13f0f5e5c5a8c1f833ef189d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f9a9b98d234f6f13f0f5e5c5a8c1f833ef189d55"}]},"branch":"refs/heads/master"},"6fc452b943d1309677ef5ef00a2cc1ac99000664":{"kind":"REWORK","_number":15,"created":"2019-09-13 14:23:01.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/15"}}},"commit":{"parents":[{"commit":"cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0","subject":"Merge \"Handle events in separate IDL instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-09-13 14:21:58.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6fc452b943d1309677ef5ef00a2cc1ac99000664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6fc452b943d1309677ef5ef00a2cc1ac99000664"}]},"branch":"refs/heads/master"},"0337b2e582fa96c8937794bc5a72ec744e979ad1":{"kind":"REWORK","_number":16,"created":"2019-09-16 12:34:16.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/36/673236/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/36/673236/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/36/673236/16"}}},"commit":{"parents":[{"commit":"cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0","subject":"Merge \"Handle events in separate IDL instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cb1d0d91d8cba40bb4f7617ab2d995b50d70fbe0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-29 08:32:55.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-09-16 12:34:06.000000000","tz":60},"subject":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)","message":"DevStack: Use OVN from ovn-org/ovn (the new OVN repository)\n\nThis patch is modifying the DevStack scripts to use the new OVN\nrepository for our jobs targeting the master branch (or \u003e\u003d 2.12, when\nit\u0027s released).\n\nIf the OVN_BRANCH is set to stable branches the code will fallback to\nusing the \"compile_ovs\" function like before (since those branches does\nnot exist in the new OVN repository, only master and \u003e\u003d 2.12 when\nreleased).\n\nChange-Id: I8b99a3481b4460ae32f2258dcf4190db9a8fed89\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0337b2e582fa96c8937794bc5a72ec744e979ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0337b2e582fa96c8937794bc5a72ec744e979ad1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
