)]}'
{"id":"openstack%2Fnetworking-ovn~674261","triplet_id":"openstack%2Fnetworking-ovn~master~Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a","project":"openstack/networking-ovn","branch":"master","hashtags":[],"change_id":"Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a","subject":"Add missing unittests to OVN provider driver","status":"MERGED","created":"2019-08-02 10:46:06.000000000","updated":"2019-11-13 08:11:59.000000000","submitted":"2019-10-28 18:41:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674261-1572288088209-2bb4bfd8","meta_rev_id":"caf5fdbc04cad8f7d6a824b84a5fbd3a1d1545c3","_number":674261,"virtual_id_number":674261,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 18:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2019-10-26 11:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2019-10-26 11:03:24.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-10-26 11:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-28 15:37:31.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2019-10-26 11:03:24.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-28 15:37:31.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 10:44:19.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-10-22 14:50:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-24 19:52:35.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2019-10-25 06:26:50.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-10-28 15:37:31.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-10-28 18:41:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c641dfeac5abc44c233e2855ba3df27c1a354eda","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-02 10:46:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5d2faa1e86a4caa63e8673a89af4797e283eda1","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-02 10:46:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"68d5141b3d839e3fb19e0d3a73714dc3b3123b29","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-02 13:29:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e1467c1e2de823149ff20700e0050e4c103a9029","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-02 13:29:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cdcba0d8db9046d0b15d1526cd2d20ea8b63836d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 16:36:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/61/674261/3/check/openstack-tox-pep8/453f50c/ : SUCCESS in 6m 17s\n- openstack-tox-py27 https://logs.opendev.org/61/674261/3/check/openstack-tox-py27/ba53fa9/ : SUCCESS in 10m 36s\n- openstack-tox-docs https://logs.opendev.org/61/674261/3/check/openstack-tox-docs/162e02a/html/ : SUCCESS in 5m 00s\n- openstack-tox-py36 https://logs.opendev.org/61/674261/3/check/openstack-tox-py36/c3856c9/ : SUCCESS in 10m 08s\n- openstack-tox-py37 https://logs.opendev.org/61/674261/3/check/openstack-tox-py37/e6c3301/ : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://logs.opendev.org/61/674261/3/check/openstack-tox-lower-constraints/2aa75aa/ : SUCCESS in 12m 30s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/61/674261/3/check/networking-ovn-tempest-dsvm-ovs-master/31a9b88/ : SUCCESS in 2h 16m 26s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/61/674261/3/check/networking-ovn-tempest-dsvm-ovs-release/5c4b2ef/ : SUCCESS in 2h 57m 39s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/61/674261/3/check/networking-ovn-tempest-dsvm-ovs-release-python2/cbc2903/ : SUCCESS in 2h 16m 14s\n- networking-ovn-rally-task https://logs.opendev.org/61/674261/3/check/networking-ovn-rally-task/59aafa7/ : SUCCESS in 1h 46m 47s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/61/674261/3/check/networking-ovn-dsvm-functional-py27/a7fd650/ : SUCCESS in 21m 13s\n- networking-ovn-dsvm-functional https://logs.opendev.org/61/674261/3/check/networking-ovn-dsvm-functional/3f418d8/ : SUCCESS in 19m 21s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/674261/3/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/95d95d9/ : SUCCESS in 2h 31m 06s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/61/674261/3/check/networking-ovn-dsvm-grenade/4b7c017/ : SUCCESS in 2h 09m 51s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/61/674261/3/check/openstack-tox-cover/409d3ef/cover/ : SUCCESS in 10m 42s\n- networking-ovn-tempest-slow https://logs.opendev.org/61/674261/3/check/networking-ovn-tempest-slow/337334b/ : SUCCESS in 2h 22m 37s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/61/674261/3/check/networking-ovn-tempest-full-multinode-ovs-master/1dd91c1/ : SUCCESS in 1h 50m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e118e428896cc44abccf7d347e16eaa24ea2ab2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-05 14:46:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"725fbf0b4cda4e1d63e6b7414adc562717207ac9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:25:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/61/674261/4/check/openstack-tox-pep8/14a54f8/ : FAILURE in 5m 08s\n- openstack-tox-py27 https://logs.opendev.org/61/674261/4/check/openstack-tox-py27/475ef7b/ : TIMED_OUT in 41m 43s\n- openstack-tox-docs https://logs.opendev.org/61/674261/4/check/openstack-tox-docs/8ff6b5c/html/ : SUCCESS in 4m 39s\n- openstack-tox-py36 https://logs.opendev.org/61/674261/4/check/openstack-tox-py36/fa1db6f/ : TIMED_OUT in 41m 15s\n- openstack-tox-py37 https://logs.opendev.org/61/674261/4/check/openstack-tox-py37/6f328c2/ : TIMED_OUT in 41m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/61/674261/4/check/openstack-tox-lower-constraints/88acaed/ : FAILURE in 14m 00s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/61/674261/4/check/networking-ovn-tempest-dsvm-ovs-master/01bdb58/ : SUCCESS in 2h 18m 15s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/61/674261/4/check/networking-ovn-tempest-dsvm-ovs-release/6fee06f/ : SUCCESS in 2h 12m 29s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/61/674261/4/check/networking-ovn-tempest-dsvm-ovs-release-python2/a668203/ : SUCCESS in 2h 09m 50s\n- networking-ovn-rally-task https://logs.opendev.org/61/674261/4/check/networking-ovn-rally-task/ebfa6d5/ : SUCCESS in 1h 47m 04s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/61/674261/4/check/networking-ovn-dsvm-functional-py27/b40dcb3/ : FAILURE in 28m 23s\n- networking-ovn-dsvm-functional https://logs.opendev.org/61/674261/4/check/networking-ovn-dsvm-functional/8211325/ : FAILURE in 25m 55s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/674261/4/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/6ea71b9/ : SUCCESS in 2h 32m 43s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/61/674261/4/check/networking-ovn-dsvm-grenade/cce29d0/ : SUCCESS in 2h 07m 06s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/61/674261/4/check/openstack-tox-cover/7021da6/ : FAILURE in 50m 35s\n- networking-ovn-tempest-slow https://logs.opendev.org/61/674261/4/check/networking-ovn-tempest-slow/c441fd4/ : SUCCESS in 2h 30m 19s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/61/674261/4/check/networking-ovn-tempest-full-multinode-ovs-master/9946e0e/ : SUCCESS in 2h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8b040bf49c0be01593e246705fa329507c46716c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-06 16:41:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4fbbbc06b7b6def183125fb4605aee4aeea1b35a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 19:46:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/61/674261/5/check/openstack-tox-pep8/3970d70/ : SUCCESS in 5m 02s\n- openstack-tox-py27 https://logs.opendev.org/61/674261/5/check/openstack-tox-py27/e205a6c/ : TIMED_OUT in 41m 02s\n- openstack-tox-docs https://logs.opendev.org/61/674261/5/check/openstack-tox-docs/1c584c2/html/ : SUCCESS in 4m 46s\n- openstack-tox-py36 https://logs.opendev.org/61/674261/5/check/openstack-tox-py36/1cf1039/ : TIMED_OUT in 41m 08s\n- openstack-tox-py37 https://logs.opendev.org/61/674261/5/check/openstack-tox-py37/baa663f/ : TIMED_OUT in 41m 21s\n- openstack-tox-lower-constraints https://logs.opendev.org/61/674261/5/check/openstack-tox-lower-constraints/21d8fe7/ : SUCCESS in 10m 24s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/61/674261/5/check/networking-ovn-tempest-dsvm-ovs-master/99db0f1/ : SUCCESS in 2h 14m 57s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/61/674261/5/check/networking-ovn-tempest-dsvm-ovs-release/ba753f9/ : SUCCESS in 2h 02m 47s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/61/674261/5/check/networking-ovn-tempest-dsvm-ovs-release-python2/0253a94/ : SUCCESS in 1h 59m 44s\n- networking-ovn-rally-task https://logs.opendev.org/61/674261/5/check/networking-ovn-rally-task/5d4b841/ : SUCCESS in 1h 25m 28s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/61/674261/5/check/networking-ovn-dsvm-functional-py27/9815c7e/ : FAILURE in 27m 09s\n- networking-ovn-dsvm-functional https://logs.opendev.org/61/674261/5/check/networking-ovn-dsvm-functional/78ff175/ : FAILURE in 30m 39s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/674261/5/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/67bcd16/ : SUCCESS in 2h 20m 48s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/61/674261/5/check/networking-ovn-dsvm-grenade/43d0858/ : SUCCESS in 2h 16m 28s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/61/674261/5/check/openstack-tox-cover/161c444/ : TIMED_OUT in 51m 12s\n- networking-ovn-tempest-slow https://logs.opendev.org/61/674261/5/check/networking-ovn-tempest-slow/f99abb5/ : SUCCESS in 2h 20m 41s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/61/674261/5/check/networking-ovn-tempest-full-multinode-ovs-master/186bd09/ : FAILURE in 2h 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"32ec890ba9d180bb0a8e6a24205792b3595b6733","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-07 10:42:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9a524fcc741cc34d6e228ab0aed03c1cdfe768ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:40:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/61/674261/6/check/openstack-tox-pep8/db80777/ : SUCCESS in 5m 19s\n- openstack-tox-py27 https://logs.opendev.org/61/674261/6/check/openstack-tox-py27/a1a0e12/ : TIMED_OUT in 41m 05s\n- openstack-tox-docs https://logs.opendev.org/61/674261/6/check/openstack-tox-docs/7b60304/html/ : SUCCESS in 4m 10s\n- openstack-tox-py36 https://logs.opendev.org/61/674261/6/check/openstack-tox-py36/711a4d5/ : TIMED_OUT in 41m 26s\n- openstack-tox-py37 https://logs.opendev.org/61/674261/6/check/openstack-tox-py37/6a013a4/ : TIMED_OUT in 41m 35s\n- openstack-tox-lower-constraints https://logs.opendev.org/61/674261/6/check/openstack-tox-lower-constraints/7028bf8/ : FAILURE in 9m 19s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/61/674261/6/check/networking-ovn-tempest-dsvm-ovs-master/c2e7b34/ : SUCCESS in 2h 07m 51s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/61/674261/6/check/networking-ovn-tempest-dsvm-ovs-release/a2950cc/ : SUCCESS in 2h 00m 02s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/61/674261/6/check/networking-ovn-tempest-dsvm-ovs-release-python2/ee42c21/ : SUCCESS in 2h 08m 44s\n- networking-ovn-rally-task https://logs.opendev.org/61/674261/6/check/networking-ovn-rally-task/aa15a9f/ : SUCCESS in 1h 47m 39s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/61/674261/6/check/networking-ovn-dsvm-functional-py27/d8a8219/ : FAILURE in 31m 11s\n- networking-ovn-dsvm-functional https://logs.opendev.org/61/674261/6/check/networking-ovn-dsvm-functional/942154e/ : FAILURE in 28m 03s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/674261/6/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/059bfa6/ : SUCCESS in 2h 48m 51s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/61/674261/6/check/networking-ovn-dsvm-grenade/62611c0/ : FAILURE in 2h 29m 15s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/61/674261/6/check/openstack-tox-cover/f2562c4/ : TIMED_OUT in 52m 04s\n- networking-ovn-tempest-slow https://logs.opendev.org/61/674261/6/check/networking-ovn-tempest-slow/aa77344/ : SUCCESS in 2h 18m 16s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/61/674261/6/check/networking-ovn-tempest-full-multinode-ovs-master/3d06760/ : SUCCESS in 1h 58m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"016f8dca42c02ff236e8753232eca83277bdf024","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-09 12:22:29.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b9b6448d0692699a34b1eebeabf8eaef2c9cf261","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:22:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/61/674261/7/check/openstack-tox-pep8/740d143/ : SUCCESS in 5m 31s\n- openstack-tox-py27 https://logs.opendev.org/61/674261/7/check/openstack-tox-py27/9437a47/ : SUCCESS in 11m 13s\n- openstack-tox-docs https://logs.opendev.org/61/674261/7/check/openstack-tox-docs/f223334/html/ : SUCCESS in 3m 20s\n- openstack-tox-py36 https://logs.opendev.org/61/674261/7/check/openstack-tox-py36/38df7b0/ : SUCCESS in 10m 06s\n- openstack-tox-py37 https://logs.opendev.org/61/674261/7/check/openstack-tox-py37/7509a4e/ : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/61/674261/7/check/openstack-tox-lower-constraints/fa59d5c/ : SUCCESS in 10m 04s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/61/674261/7/check/networking-ovn-tempest-dsvm-ovs-master/43a571b/ : SUCCESS in 2h 03m 52s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/61/674261/7/check/networking-ovn-tempest-dsvm-ovs-release/4d05320/ : SUCCESS in 2h 12m 33s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://logs.opendev.org/61/674261/7/check/networking-ovn-tempest-dsvm-ovs-release-python2/eedde65/ : SUCCESS in 2h 08m 55s\n- networking-ovn-rally-task https://logs.opendev.org/61/674261/7/check/networking-ovn-rally-task/22b0662/ : SUCCESS in 1h 44m 35s\n- networking-ovn-dsvm-functional-py27 https://logs.opendev.org/61/674261/7/check/networking-ovn-dsvm-functional-py27/f20cfd6/ : FAILURE in 23m 33s\n- networking-ovn-dsvm-functional https://logs.opendev.org/61/674261/7/check/networking-ovn-dsvm-functional/8e0e70c/ : FAILURE in 22m 06s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/61/674261/7/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/fde2dc4/ : SUCCESS in 2h 26m 03s\n- networking-ovn-dsvm-grenade https://logs.opendev.org/61/674261/7/check/networking-ovn-dsvm-grenade/1a60c3a/ : SUCCESS in 2h 05m 20s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/61/674261/7/check/openstack-tox-cover/2620c3e/cover/ : SUCCESS in 12m 02s\n- networking-ovn-tempest-slow https://logs.opendev.org/61/674261/7/check/networking-ovn-tempest-slow/b69cce4/ : SUCCESS in 2h 26m 45s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://logs.opendev.org/61/674261/7/check/networking-ovn-tempest-full-multinode-ovs-master/5c3ba76/ : SUCCESS in 1h 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cbf9cfaaf9e0bd3557b3e013bb37181900e619ed","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-22 10:31:38.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"c900ea53f376d6987b5a63f059501157a141a89c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-22 10:36:41.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"e24cd6e72cc7790c0ac1a0e8b3b3732031b3f6ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 13:47:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49194c2d7c7c431a9cb73ce9d074be1f : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0355f4f90ad4c20bb73bfab81aa2104 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20909476a88942c3af5c6816dc376d69 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c232f2bfb404ef5bd1da791ef75f025 : SUCCESS in 10m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b25206aa9c3943b0b8a04b0ae7e8db7e : SUCCESS in 12m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac67c12ed0ae491b956b1fa00eba0e14 : SUCCESS in 11m 22s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/e99a196e678f41808e56105db80447a7 : SUCCESS in 2h 16m 17s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/df5ea5694edd4d3ea574ef1da81bb6ad : SUCCESS in 2h 06m 36s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/49a2bb4f7ac14559be8f5f4b748f002d : SUCCESS in 2h 59m 14s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2a17cb9523c14e0eb8de5d21d1c00ccf : SUCCESS in 1h 40m 20s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d78d404bb43c4a4892ed269f25a55e82 : FAILURE in 25m 46s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/05fc974f24ed46e59c90e77639125361 : FAILURE in 21m 04s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/97389012e00b47f99804b09bbaa5580d : SUCCESS in 2h 46m 33s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/5c4b44d1fbb0496fa486d8ce632fa4e7 : SUCCESS in 1h 42m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e263f4166c747d9bba2b34df7b4f9a4 : SUCCESS in 12m 53s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e924eb386cb2463d9a57eeec287040e6 : SUCCESS in 2h 29m 28s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/13cead38780248b08e7a5b880690958f : SUCCESS in 1h 55m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6134082e9d2a5e82ea7eb7f25b654a886391085b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-24 13:03:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"30c8f2bc4d7bd8686d60c05ea9a0ff39c3c24158","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-24 13:04:10.000000000","message":"Patch Set 10:\n\nrebased/fixed a few nits","accounts_in_message":[],"_revision_number":10},{"id":"8c471c3737d65960504db8dcc9d6477e75334fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 16:58:11.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c57a728aefd42169312aa6017b36aaa : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d386727a5d434ff79469dfc843adc658 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5863d0939c64e2098df1e0671938064 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aef34ccd7b854cefa7604bb4cbf15ae0 : SUCCESS in 10m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df17fde3a4b4424595016c7450513012 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/634858a6b82c445a9127677cd4a08dbe : SUCCESS in 14m 29s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/1ac7cb474afb40b789962ea03d8cd0e3 : SUCCESS in 2h 30m 04s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0ee787425fb34332be2b8af2dfde5931 : SUCCESS in 2h 32m 51s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e119dcba36754734b833312c21bdc6ad : SUCCESS in 2h 45m 30s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/32f09675b2424a2099fa1fc5fc85ff28 : SUCCESS in 2h 12m 46s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d23fab152b9442b594d4cd100607ac31 : FAILURE in 21m 14s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f55e8759dd348afbbce50589a50e8af : FAILURE in 20m 39s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5d3a06660b8c45fc8e210eeda1cfc433 : SUCCESS in 3h 43m 08s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/3f0a1c6c52da4651822ef6513085c016 : FAILURE in 2h 06m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e5533125e0648bc81a02619e532c255 : SUCCESS in 10m 16s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/abc0821112e6456d899fd744c2500654 : SUCCESS in 2h 09m 26s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a2449031a2e24a95851c0b52ee0602eb : SUCCESS in 2h 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bfba945cb9bb33f0ecdc8f468563bff9645769b6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-01 14:14:18.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"225ff7a029013cd0fe937ae86394aca8a8a18fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:49:53.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c75d1f0050458c8aa2d6a417863db0 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d0240296ace4870bd377b239ac75e99 : FAILURE in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0d2833989a43f4904e4f7cebaf5224 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/070d815e6b7a43dca257f55124b8a996 : FAILURE in 10m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e88e4e822df640bb86ca7a50d17d5219 : FAILURE in 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4d2979049a74de68eb11a79d1c47b20 : FAILURE in 11m 43s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/d54134b5c0bd463cad14861f568947ed : SUCCESS in 2h 53m 40s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/81b19cf2ff2e47329bf675f2b751d156 : SUCCESS in 2h 32m 03s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/9c931ef573d64437bb7df390d86c7508 : SUCCESS in 2h 24m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9603a1cdb32041c6bea911f92dbfdb2e : SUCCESS in 1h 50m 48s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d26b965c544f467db687702c2ffc77b2 : FAILURE in 23m 25s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e06247e849745008b7b1cbdac1e006e : FAILURE in 21m 05s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/adb2adbf13024de9a1d77d8675fbbcdf : SUCCESS in 3h 21m 28s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/2f81a610b2c74ba48fa580bab6ec3411 : SUCCESS in 2h 04m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e5e1ec97b034b0d8e701e63a57d6b47 : FAILURE in 11m 11s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/044e1e153c70434ab9dd8e494c296119 : SUCCESS in 2h 03m 17s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/69ce133a3542428bac0163e78f345b52 : FAILURE in 1h 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"aa70ebc05a24774c386eb08d73de851db9213d72","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-02 14:07:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f43ddbb55d9d3712f0ff037abe2f1ae4a85dc9d2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-02 14:41:23.000000000","message":"Patch Set 12: Workflow-1\n\nCheck this one networking_ovn.tests.functional.octavia.test_ovn_driver.TestOctaviaOvnProviderDriver.test_pool","accounts_in_message":[],"_revision_number":12},{"id":"58392c228c17405a33e7104e6f8a9d07feb8ea10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:48:58.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4691b74656fd4000841f76dd50aa6999 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c390be4cf54e4435bbf47fc3c2b17433 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2b3cae358c4f8da27b9d229fa57f8d : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41bb77f40d704b6a9b50b7b7a1be8a69 : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efb3f319818f45be8757f101c59e9516 : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d783dcf7b074aa684cd4a58d0c4a04e : SUCCESS in 9m 45s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/fd35ab31e99f44b1936dd3a41bdc4887 : SUCCESS in 2h 12m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/34cb1d19a3c14d59811b7df359605359 : SUCCESS in 2h 11m 49s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/06c3dbe1afbf4672a521cd5f18c7c679 : SUCCESS in 2h 37m 17s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f156020161fe490eb7509d84e815214f : SUCCESS in 1h 53m 47s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c9e5cc09150d40b2864f064922e04815 : FAILURE in 23m 30s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1a1a8cf0201445fa4c581d1f6739ab7 : FAILURE in 19m 41s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9dbac2fefe5142f6944aef8724e9ff86 : SUCCESS in 3h 35m 50s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/2e0b3112bd09459ea96655b982f9aca2 : SUCCESS in 1h 57m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/583f73e9b91a4ee79ca0053705d882a7 : SUCCESS in 8m 57s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/27403cd75c224a008069ab39e18872f1 : SUCCESS in 2h 34m 13s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6e870b68379649a8a4dcb076f1706205 : SUCCESS in 1h 54m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3b8d58791db19d38449743e5d9dfb80380ff9df0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-03 11:06:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d786a177a799ccd8e4059a35f7cd1032ae45a814","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-03 14:56:32.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b0a518bb7ed30d3c53d9bfcef7bda0689742c88c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 18:37:33.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e619f7e7ff934f2eb8cd7b78061a97af : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce67412c7fa1476695a91871f6cb3174 : FAILURE in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8efc181322684a77863003d3a120ba03 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c51f772770a47ee921f6f4933e0a9ad : FAILURE in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/956a1547a2554c8083115b36b5aba90b : FAILURE in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04ea72ec002748e6975371818cc5628d : FAILURE in 13m 26s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/cf2903a27e594c1fb2c7552e9a06a1d1 : SUCCESS in 2h 14m 21s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4ae059a0192d48f8b821e996da759cdb : SUCCESS in 2h 47m 33s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/24eab27e6bbb4448bbcca45de1e8dbc7 : SUCCESS in 2h 16m 50s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dfa9d6c36b1045d6a6aa1a875f85260e : TIMED_OUT in 2h 14m 35s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/745f2a66ddbb4418a7a80ce6580ebc64 : SUCCESS in 20m 50s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/69da02edd67746d58485af0484c0734b : SUCCESS in 30m 04s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0e290d3f13cc486d8258bbaddd7a988b : SUCCESS in 3h 34m 46s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9846f4a567db49faaac559e5b3ffb482 : FAILURE in 2h 26m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b874c91c79bd49b4822e8d91d0a80460 : FAILURE in 9m 50s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/020f48f9552c4f279e27314069af6983 : SUCCESS in 2h 25m 21s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bed3d5951ebb4371a521ba721376cc8f : SUCCESS in 2h 07m 49s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"970dfda6b7e3d90f15ac192ae1e7f5ac14a70460","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-04 11:35:35.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"5499c08bda9da573b6c346a8f7d5b9a5f0a4d90f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-04 11:35:45.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"64dc55ea9964ea27e23d6626e2bf5b189301d043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:49:23.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c3dec575c9b4b27b4c8ca54e5021071 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba3048eba0244b80ac74e6b61d3397d5 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdde5efbd66e4218a4f714f631196df5 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dd7f5f95dde4f78b3b0a981e2a93f0c : SUCCESS in 10m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9d1e9c280604dd59cf46aa12fa9cc23 : SUCCESS in 10m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/496d01b89de3407cbedeceada742897b : SUCCESS in 10m 28s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/b782352e54b94d8ba02e109de68a374b : SUCCESS in 2h 33m 13s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/39f63572696f4598a2949f9b8c003cc8 : SUCCESS in 2h 45m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8dc9ae491c9340d89724ecbdf6d748e0 : SUCCESS in 2h 31m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09c715b409104c7b8ef0953180409351 : SUCCESS in 2h 08m 01s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/6a910635760444d7a0aceea34ed7b66c : SUCCESS in 25m 01s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8858f6da1f5344ffaab4c595026196c1 : SUCCESS in 24m 59s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6decbe6cc44c4fe1860f8b28369a3537 : SUCCESS in 3h 08m 58s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/32019a5ee19146868c0e7b85c45b90c7 : FAILURE in 1h 59m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/174932952bcf49b5bb53fe6a6cfc0489 : SUCCESS in 9m 46s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0c007d82fa4e4d6986ec8dfd4581d8fc : SUCCESS in 2h 26m 23s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/daa897a69fec46ce983b0f5b08a95c51 : SUCCESS in 2h 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ea183f9e1f74a33b8227590556e75d4b3a5c8e7a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-04 15:51:25.000000000","message":"Patch Set 16:\n\n(5 comments)\n\nWith this patch you actually increase test coverage to 77% :)  You should change tox.ini to increase the --fail-under limit from 65% to 75% maybe?  And remove the --omit networking_ovn/octavia ?  I tried that locally and it still passed.","accounts_in_message":[],"_revision_number":16},{"id":"10e183f7c04d44009ec01887d9f911765465deef","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-07 09:07:30.000000000","message":"Patch Set 16:\n\n\u003e (5 comments)\n \u003e \n \u003e With this patch you actually increase test coverage to 77% :)  You\n \u003e should change tox.ini to increase the --fail-under limit from 65%\n \u003e to 75% maybe?  And remove the --omit networking_ovn/octavia ?  I\n \u003e tried that locally and it still passed.\n\nHey!\n\nThanks for the reviews!\n\nFrom the report I see it bumped up to 88%:\nhttps://99955c5911b1b1e76812-7d520a13c5c9e187dc2d0cbaab8b9870.ssl.cf5.rackcdn.com/674261/16/check/openstack-tox-cover/1749329/cover/\n\nYes, I\u0027ll bump-up the limit also :)","accounts_in_message":[],"_revision_number":16},{"id":"b582af3b6dcd57e31f4460933af4bdb2df6b716d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-10 08:23:54.000000000","message":"Patch Set 16:\n\n\u003e \u003e (5 comments)\n \u003e \u003e\n \u003e \u003e With this patch you actually increase test coverage to 77% :) \n \u003e You\n \u003e \u003e should change tox.ini to increase the --fail-under limit from 65%\n \u003e \u003e to 75% maybe?  And remove the --omit networking_ovn/octavia ?  I\n \u003e \u003e tried that locally and it still passed.\n \u003e \n \u003e Hey!\n \u003e \n \u003e Thanks for the reviews!\n \u003e \n \u003e From the report I see it bumped up to 88%:\n \u003e https://99955c5911b1b1e76812-7d520a13c5c9e187dc2d0cbaab8b9870.ssl.cf5.rackcdn.com/674261/16/check/openstack-tox-cover/1749329/cover/\n \u003e \n \u003e Yes, I\u0027ll bump-up the limit also :)\n\nAh yea, 77% the whole networking-ovn, and 88% for ovn provider driver :)","accounts_in_message":[],"_revision_number":16},{"id":"97f4c42ee1ca1a028d4981b5cd103190b779ee75","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-17 21:00:16.000000000","message":"Patch Set 16:\n\n(1 comment)\n\nI\u0027ll push and update for the ovn_driver.py nits, just to keep this going forward","accounts_in_message":[],"_revision_number":16},{"id":"b66198a32387dc7f6e24975c5b92717ab473febe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-17 21:00:25.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c1d6dd9fb6a9c11e2750864e9a43386dc7d75aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:25:33.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7091e7da92704f37ad18a4cba3536c13 : SUCCESS in 7m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbc478760c5f45ac8e580505db6feb0f : SUCCESS in 15m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73da425367264c059c8894eadda00083 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b05abbc62d294308a7e9cc49b6b57271 : SUCCESS in 14m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/073f519cfedd46d3a345adcff2fc19d0 : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a301205d40e746bd9a668f9255278103 : SUCCESS in 19m 15s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/7e4f87cfcc444f56aa76b7eb69d9b3a8 : SUCCESS in 2h 08m 38s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/2d4d122b980748bf937cfbfeed4410c1 : SUCCESS in 2h 56m 54s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d36372b118184ecf8201c6b3cefb95dc : TIMED_OUT in 2h 14m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/914e1a12c83c4ec6b2d0679bdd4b95fc : FAILURE in 27m 13s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1079de9f503d4619a73c5d1afa348f80 : SUCCESS in 3h 22m 55s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/6597786c20a24d989d68f2b151470050 : FAILURE in 2h 25m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb665d0229064da4815cee544901844b : SUCCESS in 16m 31s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c3c17cce75d2459fbe1224390ba8b786 : SUCCESS in 2h 48m 25s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/277986c7f1d04484b6e29223b1d67be2 : FAILURE in 1h 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b662e71192fe9491c50d78a1e54e7a92fb9c6449","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-22 08:02:44.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"b5ac88e0d08e6bae7092df1e9a646cf8f5946feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 10:38:36.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db0932887614a2bbcd1a57cc93c8834 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f524d48035c842cfaf64cd49a3968374 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/250e7feacc02429caabfc35e2548dc5c : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92cdfd9be235496a97dc368f1485e852 : SUCCESS in 9m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81de9a2c1cf647bea46e9f2341d37271 : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/297362bdd7254da79e790d0ed31f93bb : SUCCESS in 11m 17s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/7cec0a612e004cd89b6d198399ef8f8b : SUCCESS in 2h 28m 37s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b738781bb0124f99b8a71c16ecb690e9 : SUCCESS in 2h 13m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/01a7cb1055f2428f98f49a5ec956970f : SUCCESS in 2h 10m 48s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2987d1ed006a4ea38f6a76accddec4c8 : SUCCESS in 20m 26s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6239b118d78e4c29835667d0ea086baf : SUCCESS in 2h 27m 59s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/ecc46fd7aa69409097999d1befc45cab : FAILURE in 2h 26m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c4403d4d64c4f86be7dc8d491b4aa93 : SUCCESS in 11m 24s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0118cde165c24882b080c47094bab98c : SUCCESS in 2h 22m 25s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/be9cb58b8eeb4617bf2718f818d97234 : FAILURE in 2h 08m 14s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0ad8169bae8793988e4aab11353a91533281bed7","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-10-22 10:44:19.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"b7a9505e8e9ba57b45af6a960238a138528362ce","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-22 14:50:25.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"d3865a70051ab93834f44a182c5776300ceb397f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-24 19:52:35.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"8363fae584a816c49e73d6cc5e64f415de2a4a21","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-25 06:26:50.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"6f9dec85686615abe3cc3b441769b294f9c6024f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 06:27:03.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"1fb0037deec0c903a07749ee91d2408b0f44b89c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 09:16:45.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc40a1507d1b45f6ba3bbdf84f809ba0 : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6105ee16f42f4364ab83ae5f34ce6fa2 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e84aed8b464ae49ab5ff94b3c5f7e6 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53ef5a7107ec4b62b73df5e292005dcd : SUCCESS in 10m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97e1e0656d554ded8cbd3c22b9a9e413 : SUCCESS in 12m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e89c1a433504b2e9af085078d969d9a : SUCCESS in 12m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/30c231cb170445daa8281927b6b5b491 : SUCCESS in 2h 29m 45s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/700c2f27ebf14b98847aeec710896a7f : SUCCESS in 1h 51m 17s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bd22890eb154a24bcafcf0c77c06b9a : FAILURE in 12m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fb66b5adf4644455a5eadc22ddf7265e : SUCCESS in 2h 47m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06caa7985c5844d9a60194b44876ac20 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":17},{"id":"de108213ea9a06ca96531246f1e7052a97fc2574","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-26 11:03:24.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"43a7fe742396437873badd02f43180bbbd0f7a51","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-26 11:06:11.000000000","message":"Patch Set 18:\n\nRebased to pick up fixes from https://review.opendev.org/#/c/685122/","accounts_in_message":[],"_revision_number":18},{"id":"ce3065e32c925f2c0db5eba2e3ea4382fd0a208b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 13:34:44.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faee483ca70f46089b140048dea489dd : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88e892698d4d47f69120455dfcb29e99 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82afa7bdd6b3487394f50e960e1e9529 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55560e2de0ab4f50810624cbd2c23f2a : SUCCESS in 9m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdce2f20f0304a66b15d63814cecbbb4 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7b0c5856d1b49febdf7e681fbff91b9 : SUCCESS in 10m 47s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5231bfe1a6dc4f2eaa1132c63c21ab8e : SUCCESS in 1h 53m 45s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f4592fcf131d4232a896a013f34f8c1d : SUCCESS in 2h 29m 09s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27d747e0aa8a4a93ad9c2a6ddf90d9d8 : SUCCESS in 2h 01m 55s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/10451ba6cec04908b4c53527a3404887 : SUCCESS in 20m 54s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/28b22c9d45ab4adca1926ac0031adf3b : SUCCESS in 2h 29m 40s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/bd9fc7a246b64e67a5689bdd7a619d6e : FAILURE in 2h 01m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d641b3037a14769a56b0ca99ef7bcd7 : SUCCESS in 10m 40s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/289d0b77f6054a8991e24416c01a696a : SUCCESS in 2h 17m 44s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/04958a18595e4302bfd809c32878521b : FAILURE in 2h 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2fa8cddbd177fa5c49d7c54babec5a6086c68eec","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-10-28 15:37:31.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"0e4dd7f17907d957cd544d3fbfb10eb29dcd2c71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 15:37:44.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"69c08311b63a1535179aee430e1b7ccf070bbe76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:41:28.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7012336890114bb6a3d55fd994ab4677 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b3c7c10b51c416a8c95037e5ca571c6 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a34515b0fe4999a904a6119c920a45 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284f30a4140c488da5f0e8d5fbcb054e : SUCCESS in 9m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/220751d0537d4435b9cab5198a72309f : SUCCESS in 12m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff6909f6a8ea4443a1c8f3c7d55fb03d : SUCCESS in 10m 50s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/bcecdad0cc8e4576b101d83b80ad3e10 : SUCCESS in 2h 10m 41s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b73826b174243d4b45275a4a657f8b3 : SUCCESS in 2h 10m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/77d6baf01f1643fd962b17d3ec202309 : SUCCESS in 26m 21s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8615cbec504e41eaa614e7eaa175281d : SUCCESS in 2h 58m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9446fa8d2c6d46ba94ba1a42c2173de2 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":18},{"id":"237347c8d5ef16f1c01e6f92b1e09ca78250869e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:41:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"aede0e807011e400fee45d430c7673dfea8d6cfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:42:50.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9c05c1e9b0419cadb0b9af4fe73d77 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":18},{"id":"ba2a9ecac0bb3c9963138cc335c7efed5a059618","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-13 08:11:51.000000000","message":"Patch Set 18: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 050b14255f1a304d01e91e2aba10242e1a7d26b1","accounts_in_message":[],"_revision_number":18},{"id":"caf5fdbc04cad8f7d6a824b84a5fbd3a1d1545c3","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-13 08:11:59.000000000","message":"Patch Set 18: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 68ae720bb42bf9a9efc528158d4643c08daeeda7","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"77d03f6f17081e0f8c15cab0444261bc4e517055","revisions":{"dc1722bf7277e442b13332fa8a5f719a0cc7f6b5":{"kind":"REWORK","_number":1,"created":"2019-08-02 10:46:06.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/1"}}},"commit":{"parents":[{"commit":"e05c8a663f3f9ac2926b23cdb8aa29b0362cf782","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e05c8a663f3f9ac2926b23cdb8aa29b0362cf782"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:45:55.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dc1722bf7277e442b13332fa8a5f719a0cc7f6b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dc1722bf7277e442b13332fa8a5f719a0cc7f6b5"}]},"branch":"refs/heads/master"},"6d850e776dc5802daa828063b1e73b361cea916b":{"kind":"REWORK","_number":2,"created":"2019-08-02 13:29:08.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/2"}}},"commit":{"parents":[{"commit":"e05c8a663f3f9ac2926b23cdb8aa29b0362cf782","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e05c8a663f3f9ac2926b23cdb8aa29b0362cf782"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 13:28:56.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6d850e776dc5802daa828063b1e73b361cea916b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6d850e776dc5802daa828063b1e73b361cea916b"}]},"branch":"refs/heads/master"},"16ff2c4f31fbb084985d99fce4fa076814433e73":{"kind":"REWORK","_number":3,"created":"2019-08-02 13:29:54.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/3"}}},"commit":{"parents":[{"commit":"e05c8a663f3f9ac2926b23cdb8aa29b0362cf782","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e05c8a663f3f9ac2926b23cdb8aa29b0362cf782"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 13:29:45.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/16ff2c4f31fbb084985d99fce4fa076814433e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/16ff2c4f31fbb084985d99fce4fa076814433e73"}]},"branch":"refs/heads/master"},"10ea58c637d9298e04dd971b7ee8df98bc4989a5":{"kind":"REWORK","_number":4,"created":"2019-08-05 14:46:15.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/4"}}},"commit":{"parents":[{"commit":"e05c8a663f3f9ac2926b23cdb8aa29b0362cf782","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e05c8a663f3f9ac2926b23cdb8aa29b0362cf782"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-05 10:30:24.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/10ea58c637d9298e04dd971b7ee8df98bc4989a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/10ea58c637d9298e04dd971b7ee8df98bc4989a5"}]},"branch":"refs/heads/master"},"ebe1758b049d3cf6fbbaef409b601cb361538fbc":{"kind":"REWORK","_number":5,"created":"2019-08-06 16:41:08.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/5"}}},"commit":{"parents":[{"commit":"e05c8a663f3f9ac2926b23cdb8aa29b0362cf782","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e05c8a663f3f9ac2926b23cdb8aa29b0362cf782"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-06 16:39:31.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ebe1758b049d3cf6fbbaef409b601cb361538fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ebe1758b049d3cf6fbbaef409b601cb361538fbc"}]},"branch":"refs/heads/master"},"4716a5ea0e0ab006c50a20c800985bc5d18dbc21":{"kind":"REWORK","_number":6,"created":"2019-08-07 10:42:33.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/6"}}},"commit":{"parents":[{"commit":"e05c8a663f3f9ac2926b23cdb8aa29b0362cf782","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e05c8a663f3f9ac2926b23cdb8aa29b0362cf782"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-07 10:42:23.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4716a5ea0e0ab006c50a20c800985bc5d18dbc21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4716a5ea0e0ab006c50a20c800985bc5d18dbc21"}]},"branch":"refs/heads/master"},"2b556c6958bb2c60255380f080bd3ae448e0ef74":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-09 12:22:29.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/7"}}},"commit":{"parents":[{"commit":"137c229a242f85806ddc18ad76e3841dc1535938","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/137c229a242f85806ddc18ad76e3841dc1535938"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-09 12:22:29.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2b556c6958bb2c60255380f080bd3ae448e0ef74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2b556c6958bb2c60255380f080bd3ae448e0ef74"}]},"branch":"refs/heads/master"},"4da0d149b04232a1298310c6620732c9edc60baf":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-22 10:31:38.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/8"}}},"commit":{"parents":[{"commit":"9062adbc84e1d5c1d28161ab52d0b0837972ab70","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9062adbc84e1d5c1d28161ab52d0b0837972ab70"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-22 10:31:38.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4da0d149b04232a1298310c6620732c9edc60baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4da0d149b04232a1298310c6620732c9edc60baf"}]},"branch":"refs/heads/master"},"2fff5331c6d7a5b212bcc8cac2520326583ea667":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-08-22 10:36:41.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/9"}}},"commit":{"parents":[{"commit":"84ef9e15f797a4ef0f64765d788145f2a9e95b18","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/84ef9e15f797a4ef0f64765d788145f2a9e95b18"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-22 10:30:39.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2fff5331c6d7a5b212bcc8cac2520326583ea667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2fff5331c6d7a5b212bcc8cac2520326583ea667"}]},"branch":"refs/heads/master"},"0466a23ba8d5ebf2b38ae19f30ec9d6cbcd0406d":{"kind":"REWORK","_number":10,"created":"2019-09-24 13:03:58.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/10"}}},"commit":{"parents":[{"commit":"561ca74b8a6806909b8192196188fa591cdd2f0c","subject":"Merge \"Fix container configuration file path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/561ca74b8a6806909b8192196188fa591cdd2f0c"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-24 13:03:50.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0466a23ba8d5ebf2b38ae19f30ec9d6cbcd0406d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0466a23ba8d5ebf2b38ae19f30ec9d6cbcd0406d"}]},"branch":"refs/heads/master"},"21d2484368070814a6807cfc2fc0d208a63b1a48":{"kind":"REWORK","_number":11,"created":"2019-10-01 14:14:18.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/11"}}},"commit":{"parents":[{"commit":"e748bd0f09aafa944a120fb47e3b5907e42dcce4","subject":"Disable ip_version checking while updating member without new ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e748bd0f09aafa944a120fb47e3b5907e42dcce4"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-10-01 14:13:18.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/21d2484368070814a6807cfc2fc0d208a63b1a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/21d2484368070814a6807cfc2fc0d208a63b1a48"}]},"branch":"refs/heads/master"},"a7d8fced599e7ebf93fe1a9bebd0378d8e3259a5":{"kind":"REWORK","_number":12,"created":"2019-10-02 14:07:44.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/12"}}},"commit":{"parents":[{"commit":"e748bd0f09aafa944a120fb47e3b5907e42dcce4","subject":"Disable ip_version checking while updating member without new ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e748bd0f09aafa944a120fb47e3b5907e42dcce4"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-10-02 14:07:32.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a7d8fced599e7ebf93fe1a9bebd0378d8e3259a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a7d8fced599e7ebf93fe1a9bebd0378d8e3259a5"}]},"branch":"refs/heads/master"},"67bce37664e2b376ed6248ed44935293a8a72f5e":{"kind":"REWORK","_number":13,"created":"2019-10-03 11:06:30.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/13"}}},"commit":{"parents":[{"commit":"e748bd0f09aafa944a120fb47e3b5907e42dcce4","subject":"Disable ip_version checking while updating member without new ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e748bd0f09aafa944a120fb47e3b5907e42dcce4"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-10-03 11:00:32.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nThis change bumps up test coverage.\nDuring writing those tests I found also a few issues:\n\n1) In method pool_update the operating_status shouldn\u0027t\nbe set with veryfing if the pool has members or not.\nThe only way of setting operating_status to OFFLINE is\nto administratively disable it or via member healt monitor,\nwhich we don\u0027t yet support. Dropping this wrong code.\nhttps://docs.openstack.org/api-ref/load-balancer/v2/?expanded\u003dcreate-pool-detail#create-pool\n\n2) _check_monitor_options() logic was wrong - fixed it\n\n3) In pool_delete() in some coditions the external_ids\ncould be used, but not initlized. I moved it to be on the top\nof function.\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/67bce37664e2b376ed6248ed44935293a8a72f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/67bce37664e2b376ed6248ed44935293a8a72f5e"}]},"branch":"refs/heads/master"},"c76e5552786dcc9074f2a3c339f90522ee95473e":{"kind":"REWORK","_number":14,"created":"2019-10-03 14:56:32.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/14"}}},"commit":{"parents":[{"commit":"e748bd0f09aafa944a120fb47e3b5907e42dcce4","subject":"Disable ip_version checking while updating member without new ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e748bd0f09aafa944a120fb47e3b5907e42dcce4"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-10-03 14:27:58.000000000","tz":0},"subject":"WIP Add missing unittests to OVN provider driver","message":"WIP Add missing unittests to OVN provider driver\n\nThis change bumps up test coverage.\nDuring writing those tests I found also a few issues:\n\n1) In method pool_update the operating_status shouldn\u0027t\nbe set with veryfing if the pool has members or not.\nThe only way of setting operating_status to OFFLINE is\nto administratively disable it or via member healt monitor,\nwhich we don\u0027t yet support. Dropping this wrong code.\nhttps://docs.openstack.org/api-ref/load-balancer/v2/?expanded\u003dcreate-pool-detail#create-pool\n\n2) _check_monitor_options() logic was wrong - fixed it\n\n3) In pool_delete() in some coditions the external_ids\ncould be used, but not initlized. I moved it to be on the top\nof function.\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c76e5552786dcc9074f2a3c339f90522ee95473e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c76e5552786dcc9074f2a3c339f90522ee95473e"}]},"branch":"refs/heads/master"},"af60164b8b2ab7fece01234f3b5a8e39a50dbd6d":{"kind":"REWORK","_number":15,"created":"2019-10-04 11:35:35.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/15"}}},"commit":{"parents":[{"commit":"e748bd0f09aafa944a120fb47e3b5907e42dcce4","subject":"Disable ip_version checking while updating member without new ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e748bd0f09aafa944a120fb47e3b5907e42dcce4"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-10-04 11:31:55.000000000","tz":0},"subject":"Add missing unittests to OVN provider driver","message":"Add missing unittests to OVN provider driver\n\nThis change bumps up test coverage.\nDuring writing those tests I found also a few issues:\n\n1) In method pool_update the operating_status shouldn\u0027t\nbe set with veryfing if the pool has members or not.\nThe only way of setting operating_status to OFFLINE is\nto administratively disable it or via member healt monitor,\nwhich we don\u0027t yet support. Dropping this wrong code.\nhttps://docs.openstack.org/api-ref/load-balancer/v2/?expanded\u003dcreate-pool-detail#create-pool\n\n2) _check_monitor_options() logic was wrong - fixed it\n\n3) In pool_delete() in some coditions the external_ids\ncould be used, but not initlized. I moved it to be on the top\nof function.\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/af60164b8b2ab7fece01234f3b5a8e39a50dbd6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/af60164b8b2ab7fece01234f3b5a8e39a50dbd6d"}]},"branch":"refs/heads/master"},"39e10bd09d883c48254e4477d5f1530b82c0fcf6":{"kind":"NO_CHANGE","_number":16,"created":"2019-10-04 11:35:45.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/61/674261/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/16"}}},"commit":{"parents":[{"commit":"7f802007843bc9177d7f6a3081045f444b9e5968","subject":"Disable ip_version checking while updating member without new ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7f802007843bc9177d7f6a3081045f444b9e5968"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-10-04 11:35:45.000000000","tz":0},"subject":"Add missing unittests to OVN provider driver","message":"Add missing unittests to OVN provider driver\n\nThis change bumps up test coverage.\nDuring writing those tests I found also a few issues:\n\n1) In method pool_update the operating_status shouldn\u0027t\nbe set with veryfing if the pool has members or not.\nThe only way of setting operating_status to OFFLINE is\nto administratively disable it or via member healt monitor,\nwhich we don\u0027t yet support. Dropping this wrong code.\nhttps://docs.openstack.org/api-ref/load-balancer/v2/?expanded\u003dcreate-pool-detail#create-pool\n\n2) _check_monitor_options() logic was wrong - fixed it\n\n3) In pool_delete() in some coditions the external_ids\ncould be used, but not initlized. I moved it to be on the top\nof function.\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/39e10bd09d883c48254e4477d5f1530b82c0fcf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/39e10bd09d883c48254e4477d5f1530b82c0fcf6"}]},"branch":"refs/heads/master"},"7cd45b20539a36bf42c959cfe90a7071c1633d90":{"kind":"REWORK","_number":17,"created":"2019-10-17 21:00:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/61/674261/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/17"}}},"commit":{"parents":[{"commit":"7f802007843bc9177d7f6a3081045f444b9e5968","subject":"Disable ip_version checking while updating member without new ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7f802007843bc9177d7f6a3081045f444b9e5968"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-17 20:59:45.000000000","tz":-240},"subject":"Add missing unittests to OVN provider driver","message":"Add missing unittests to OVN provider driver\n\nThis change bumps up test coverage.\nDuring writing those tests I found also a few issues:\n\n1) In method pool_update the operating_status shouldn\u0027t\nbe set with veryfing if the pool has members or not.\nThe only way of setting operating_status to OFFLINE is\nto administratively disable it or via member healt monitor,\nwhich we don\u0027t yet support. Dropping this wrong code.\nhttps://docs.openstack.org/api-ref/load-balancer/v2/?expanded\u003dcreate-pool-detail#create-pool\n\n2) _check_monitor_options() logic was wrong - fixed it\n\n3) In pool_delete() in some coditions the external_ids\ncould be used, but not initlized. I moved it to be on the top\nof function.\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7cd45b20539a36bf42c959cfe90a7071c1633d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7cd45b20539a36bf42c959cfe90a7071c1633d90"}]},"branch":"refs/heads/master"},"77d03f6f17081e0f8c15cab0444261bc4e517055":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-10-26 11:03:24.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/61/674261/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/61/674261/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/61/674261/18"}}},"commit":{"parents":[{"commit":"8ebfa02f8acc99ae05df7d192afe213a18fcc65e","subject":"Merge \"Functional tests: Use the latest version of OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/8ebfa02f8acc99ae05df7d192afe213a18fcc65e"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-02 10:44:16.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-10-26 11:03:24.000000000","tz":0},"subject":"Add missing unittests to OVN provider driver","message":"Add missing unittests to OVN provider driver\n\nThis change bumps up test coverage.\nDuring writing those tests I found also a few issues:\n\n1) In method pool_update the operating_status shouldn\u0027t\nbe set with veryfing if the pool has members or not.\nThe only way of setting operating_status to OFFLINE is\nto administratively disable it or via member healt monitor,\nwhich we don\u0027t yet support. Dropping this wrong code.\nhttps://docs.openstack.org/api-ref/load-balancer/v2/?expanded\u003dcreate-pool-detail#create-pool\n\n2) _check_monitor_options() logic was wrong - fixed it\n\n3) In pool_delete() in some coditions the external_ids\ncould be used, but not initlized. I moved it to be on the top\nof function.\n\nChange-Id: Ib60ba2b7a958e3d55699988d9d4f2b848e5e1b5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/77d03f6f17081e0f8c15cab0444261bc4e517055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/77d03f6f17081e0f8c15cab0444261bc4e517055"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
