)]}'
{"id":"openstack%2Fnetworking-ovn~675607","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fqueens~I1ad224960173fe896f6eb8049a188d9e82874068","project":"openstack/networking-ovn","branch":"stable/queens","topic":"bug/1839434","hashtags":[],"change_id":"I1ad224960173fe896f6eb8049a188d9e82874068","subject":"Do not modify passed by reference variables in mechanism_driver","status":"MERGED","created":"2019-08-09 12:17:09.000000000","updated":"2019-08-12 18:45:03.000000000","submitted":"2019-08-12 18:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675607-1565635503389-f1476ba1","meta_rev_id":"e13867a5d0070c787b05165833931e7a147abfb2","_number":675607,"virtual_id_number":675607,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2019-08-12 07:08:02.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-08-12 18:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2019-08-12 13:58:47.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2019-08-12 14:37:46.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-08-12 14:37:46.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 13:58:47.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-08-12 14:37:46.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-12 18:45:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80db0af7ed221e3e854b1e1c4e85ab2459160977","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-09 12:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ee6f0eb400e5b14ebf9fe83d99835527946d8d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:15:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/07/675607/1/check/openstack-tox-pep8/ace07aa/ : SUCCESS in 5m 31s\n- openstack-tox-py27 https://logs.opendev.org/07/675607/1/check/openstack-tox-py27/28ce3ec/ : SUCCESS in 11m 40s\n- build-openstack-sphinx-docs https://logs.opendev.org/07/675607/1/check/build-openstack-sphinx-docs/95c1abf/html/ : SUCCESS in 6m 55s\n- openstack-tox-py35 https://logs.opendev.org/07/675607/1/check/openstack-tox-py35/9b40108/ : SUCCESS in 11m 42s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/07/675607/1/check/networking-ovn-tempest-dsvm-ovs-master/879fa32/ : FAILURE in 1h 07m 04s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/07/675607/1/check/networking-ovn-tempest-dsvm-ovs-release/eacfb9c/ : FAILURE in 1h 03m 39s\n- networking-ovn-rally-task https://logs.opendev.org/07/675607/1/check/networking-ovn-rally-task/ae611c8/ : SUCCESS in 1h 42m 10s\n- networking-ovn-dsvm-functional https://logs.opendev.org/07/675607/1/check/networking-ovn-dsvm-functional/41139dc/ : SUCCESS in 19m 01s\n- networking-ovn-dsvm-functional-py35 https://logs.opendev.org/07/675607/1/check/networking-ovn-dsvm-functional-py35/f98502f/ : SUCCESS in 19m 35s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://logs.opendev.org/07/675607/1/check/networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release/c128c75/ : FAILURE in 35m 07s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/07/675607/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/7384a3e/ : SUCCESS in 2h 37m 28s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/07/675607/1/check/openstack-tox-cover/fb0d8e4/cover/ : SUCCESS in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02be0953888cb975002d2db78bc3207293b4bb8a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-12 07:08:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2657fb69ae39145299bc175b7d4b0aa866097fb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 10:21:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/07/675607/1/check/openstack-tox-pep8/b798548/ : SUCCESS in 4m 40s\n- openstack-tox-py27 https://logs.opendev.org/07/675607/1/check/openstack-tox-py27/2d24d12/ : SUCCESS in 8m 06s\n- build-openstack-sphinx-docs https://logs.opendev.org/07/675607/1/check/build-openstack-sphinx-docs/f2dfd3b/html/ : SUCCESS in 4m 39s\n- openstack-tox-py35 https://logs.opendev.org/07/675607/1/check/openstack-tox-py35/7472003/ : SUCCESS in 10m 40s\n- networking-ovn-tempest-dsvm-ovs-master https://logs.opendev.org/07/675607/1/check/networking-ovn-tempest-dsvm-ovs-master/f140128/ : SUCCESS in 1h 02m 24s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/07/675607/1/check/networking-ovn-tempest-dsvm-ovs-release/1800008/ : SUCCESS in 1h 17m 37s\n- networking-ovn-rally-task https://logs.opendev.org/07/675607/1/check/networking-ovn-rally-task/0e58857/ : SUCCESS in 1h 46m 25s\n- networking-ovn-dsvm-functional https://logs.opendev.org/07/675607/1/check/networking-ovn-dsvm-functional/c55fb61/ : SUCCESS in 23m 30s\n- networking-ovn-dsvm-functional-py35 https://logs.opendev.org/07/675607/1/check/networking-ovn-dsvm-functional-py35/e95fa83/ : SUCCESS in 23m 28s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://logs.opendev.org/07/675607/1/check/networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release/be116fc/ : FAILURE in 50m 07s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/07/675607/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/d48500f/ : SUCCESS in 3h 06m 56s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/07/675607/1/check/openstack-tox-cover/8cc8c3e/cover/ : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d4a029032b2ec3e9a53f9c91c7a8188e6269c7c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-08-12 13:58:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a441b68deed29453216018ab5783881945fb8f76","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-12 14:37:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6d8f7fc74236cbab5a78822816ea0a26910f95c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:37:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9deea0cf13d8ab2dd8efceae4a3901d09651e18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 18:45:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b60d9db6fdc4bbcb1a269e5c7666cd9 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f20196127a1444f69854ffae2d0dd0d1 : SUCCESS in 12m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/10bf1e79814e4bedbfede637436b98c6 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0ae82f37addd460eb09b4b859207ec23 : SUCCESS in 13m 48s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a398febd90ca43dca305e225c969fc9e : SUCCESS in 54m 42s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a2eec5379cd40cea422b32c61ce25b7 : SUCCESS in 1h 36m 57s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e36512e9d48b460591d866ae7774a698 : SUCCESS in 23m 06s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/47703111e0e247a6867f2c52521877cd : SUCCESS in 20m 29s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/08a3b47de2074d53816ddf65e551bb34 : SUCCESS in 2h 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43cbeaa525b46eb1a0dfab7438165dfe6bbb722e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 18:45:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b382bbfe75897697620b37ae5bc5de62fb087d6","revisions":{"5b382bbfe75897697620b37ae5bc5de62fb087d6":{"kind":"REWORK","_number":1,"created":"2019-08-09 12:17:09.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/07/675607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/07/675607/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/675607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/675607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/07/675607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/07/675607/1"}}},"commit":{"parents":[{"commit":"e541246631b503074d77bc400e5b45bb399b2c75","subject":"Merge \"Maintenance task: Enhance debug logging\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e541246631b503074d77bc400e5b45bb399b2c75"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-08 12:25:13.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-09 12:12:47.000000000","tz":0},"subject":"Do not modify passed by reference variables in mechanism_driver","message":"Do not modify passed by reference variables in mechanism_driver\n\nTest test_update_port_with_empty_data [0] from TestOVNMechanismDriverPortsV2\nclass in neutron is runned agains:\nneutron.tests.unit.plugins.ml2.drivers.mechanism_test.TestMechanismDriver\nwhich in fact during update_port_postcommit() does nothing with\nthe port, only validates it [1].\n\nIn networking-ovn this test is runned against real mechanism driver:\nnetworking_ovn.ml2.mech_driver.OVNMechanismDriver. This ends with\nlittle difference - to port dict \u0027network\u0027 information is added\nduring call of update_port_postcommit(). The port data itself\nremains the same.\nWe shouldn\u0027t modify passed by reference variables there. So doing\ndeepcopy on all provided data.\n\nHere comes also the question if this test inheritance is the\nright way.\n\n[0] https://review.opendev.org/#/c/673486\n[1] https://github.com/openstack/neutron/blob/master/neutron/tests/unit/plugins/ml2/drivers/mechanism_test.py#L208\n\nCloses-Bug: #1839434\nChange-Id: I1ad224960173fe896f6eb8049a188d9e82874068\n(cherry picked from commit b2b5e89bca25570a757c62fc6e30277365dae456)\n(cherry picked from commit 36bd132b409b82416cc49311099a8db7fbcbcda3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5b382bbfe75897697620b37ae5bc5de62fb087d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5b382bbfe75897697620b37ae5bc5de62fb087d6"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
