)]}'
{"id":"openstack%2Fnetworking-ovn~677698","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fqueens~Ib9fe6e1bfea1d5f62b2f2b6fdb12d16878108c3f","project":"openstack/networking-ovn","branch":"stable/queens","hashtags":[],"change_id":"Ib9fe6e1bfea1d5f62b2f2b6fdb12d16878108c3f","subject":"Update port_status to ACTIVE during live-migration","status":"MERGED","created":"2019-08-21 10:55:24.000000000","updated":"2019-08-29 16:52:14.000000000","submitted":"2019-08-29 16:52:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677698-1567097534308-91f773fe","meta_rev_id":"5af987560aeb489e41da4449551b349fef0eabc1","_number":677698,"virtual_id_number":677698,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-08-29 16:52:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2019-08-29 13:55:23.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2019-08-26 19:17:59.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2019-08-29 13:55:23.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 19:17:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-08-29 13:55:23.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-08-29 16:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40c509ef0910f2511be61ec9b663a9155c1cf87a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-21 10:55:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bccffeeb7a06df3f972536821bc9efac3d4b83e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 13:37:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1849fc1040e34eb0a719f99a0e30dece : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/920b8f7042a14896b4c78ac08254b3cf : SUCCESS in 12m 22s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e98466f56e3d44d88c6469f7e3898e95 : SUCCESS in 6m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fd39ce319b1149889f42db2d14502d5a : SUCCESS in 11m 41s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/1e31e5a9a20949019614552c2bebc8fe : SUCCESS in 55m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ea7f37e7db4c4bf09c38625bd1307d7b : SUCCESS in 1h 01m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a21f6b7753bb4599916ae131cd8d5bd1 : SUCCESS in 1h 36m 52s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d408c50cbde4bef8b52aaff851f4487 : SUCCESS in 19m 00s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/e1cf16af8aef415897f1e1d48b65a561 : SUCCESS in 17m 37s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/d9e74a1ed2494537bbefc94da82f8a1a : FAILURE in 35m 37s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bb4e3bf219444dd9989d5302014848d6 : SUCCESS in 2h 37m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/799a95908d9e4b568bb0006a307426be : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c60a56c0a64a65adf47d1a340e3edd332258789","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-26 19:17:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46b4186cf5e7a55eb83fd2792502ad9a3735bed7","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-08-29 13:55:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"868e87d084bb8d87df4a087d7d2402db9edfc634","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 13:55:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9aa703a638628c3b6cfb5860d30bc269415c0c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:52:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5af987560aeb489e41da4449551b349fef0eabc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:52:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f7a50a825a4d8ba13afd2ec957758b : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/720426af4797412298cbbf8de2fe03a7 : SUCCESS in 8m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9490f8307b204b93bc4e7068ecec3dc2 : SUCCESS in 17m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f3431cf4eaf8481e91e80f1deb726267 : SUCCESS in 12m 04s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/bf6692a9eeb2445db2b39739b620b624 : SUCCESS in 1h 16m 30s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/748fa6dae60a48e7866ecbe333583296 : SUCCESS in 1h 42m 56s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad603cfc7c574c8abbe2e4bed6bf1c26 : SUCCESS in 21m 32s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/d0d805ec93e84e03b91332a70fc21cfa : SUCCESS in 22m 00s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/846f81590a2a4445b4f1252ebc22a696 : SUCCESS in 2h 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3166e8206c7b5f5e2ae64f66c960c5800aa65e75","revisions":{"3166e8206c7b5f5e2ae64f66c960c5800aa65e75":{"kind":"REWORK","_number":1,"created":"2019-08-21 10:55:24.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/98/677698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/98/677698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/98/677698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/98/677698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/98/677698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/98/677698/1"}}},"commit":{"parents":[{"commit":"5b382bbfe75897697620b37ae5bc5de62fb087d6","subject":"Do not modify passed by reference variables in mechanism_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5b382bbfe75897697620b37ae5bc5de62fb087d6"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-31 12:21:42.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-21 10:54:54.000000000","tz":120},"subject":"Update port_status to ACTIVE during live-migration","message":"Update port_status to ACTIVE during live-migration\n\nDuring live-migration nova waits for \u0027network-vif-plugged\u0027 event\nin order to proceeed with live-migration [1]. While using OVN\nit never happened because the original design assumes usage\nof neutron agents.\n\nThis patch updates port status from DOWN to ACTIVE during\npre_live_migration state, which in fact for Neutron/ML2\nis done by neutron-ovs-agent, just only to emit\n\u0027network-vif-plugged\u0027 notification and allow nova\nto perform live-migration.\n\n[1] https://review.opendev.org/#/c/558001/\n\nChange-Id: Ib9fe6e1bfea1d5f62b2f2b6fdb12d16878108c3f\nRelated-Bug: 1834045\n(cherry picked from commit 229f894ee6aff6b5da8cc399575537feed1c6a49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3166e8206c7b5f5e2ae64f66c960c5800aa65e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3166e8206c7b5f5e2ae64f66c960c5800aa65e75"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
