)]}'
{"id":"openstack%2Fnetworking-ovn~679545","triplet_id":"openstack%2Fnetworking-ovn~master~I6ed2573b71dc91a60d860537512bfcef7a212e17","project":"openstack/networking-ovn","branch":"master","topic":"bug/1841913","hashtags":[],"change_id":"I6ed2573b71dc91a60d860537512bfcef7a212e17","subject":"Active wait to retrieve the Agent stats","status":"MERGED","created":"2019-08-31 18:18:13.000000000","updated":"2020-01-10 13:03:34.000000000","submitted":"2019-09-18 13:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679545-1568815075480-7b169225","meta_rev_id":"a864c628aa7f3ed90bc9e78b64073b2b25d07789","_number":679545,"virtual_id_number":679545,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-18 06:06:21.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-09-18 13:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-09-17 18:35:46.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-17 12:09:57.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-09-17 18:35:46.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 13:10:16.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-09-17 12:09:57.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-09-17 18:35:46.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-09-18 13:57:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6e9879c7d492ef89ddf074a7fe46fae395bbafd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-08-31 18:18:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f3d2d8eef4a135c8d0e955c5994645b8b7c92c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 20:52:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b78f92c8015492696c388ebb669c79a : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a663fa6536b4953ba987041f2fec820 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dada0349d4224af5898bc2cc187705e1 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb0c16f9a9124021b30e6242fed4f27a : SUCCESS in 8m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10815ca9cb074a82933abb9a94a5383e : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0869f450dd9a473aa35f18900d3d8c04 : SUCCESS in 10m 30s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/361a079db69a4a37857ce37113759e80 : SUCCESS in 2h 17m 11s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c41dd5592c1e4761817670fb424c06d1 : SUCCESS in 2h 04m 24s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f756fb9249d04a7c8bf6c74e2559aeca : SUCCESS in 2h 11m 14s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7bd50041679c475785d0ac9010cf8258 : SUCCESS in 1h 33m 02s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/7f5af072b2754f64833f4971c5942b90 : SUCCESS in 21m 11s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/50db679a0ded41e9b886402e5c0cac54 : SUCCESS in 19m 48s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d7b116e5a0ca460e9d4ae5ef35b63c14 : FAILURE in 15m 05s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/3e17ef69cd6643e3b70dc3ee2a1f2d52 : SUCCESS in 2h 09m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b5c6022860743f184bd99399bd8e3f8 : SUCCESS in 9m 05s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6c816b6df7124cb4a62b776f43eac601 : SUCCESS in 2h 24m 54s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5826c30e99244fb1a0e610dea0502c10 : SUCCESS in 2h 08m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c63e24df1794b0c9f60b0a8c50a0991aee1f37d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-01 00:17:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eec805c521dfa5d5b682545f712d0fca8b3b5f68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 02:44:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060f1be86638429c9c84ff1d4870ac2c : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db2390f6297a4bdfbf70518addc508d2 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c941c9948814482396c8accb743b42db : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ebc9df3f5e4340bd1db53bb7740eb4 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/968e3f3e087946e89a2bf52cd00d80c0 : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3a1529be17f424da5bbd3a1502e7541 : SUCCESS in 10m 19s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/65c2549934da42e593b06bc0bc569031 : SUCCESS in 2h 18m 19s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/26051a673c0848ae8dc949c46abfc9eb : SUCCESS in 2h 14m 30s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/b49781e2c80248049726d9bb0c56b74f : SUCCESS in 2h 12m 26s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe14ae0283b1465bbd2f407a4104f5eb : SUCCESS in 1h 36m 00s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a14d885085c44b6591f8d864ba71f171 : SUCCESS in 19m 56s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a367ffcc6748437290024ef7a5fbf0aa : SUCCESS in 18m 53s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0f3c36eaf9e64e7abb8d0175eb0ac493 : FAILURE in 15m 27s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/32bf5e3f208b4893be71383f924a004e : SUCCESS in 1h 59m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8796be6409b42389d93cb52c321cc74 : SUCCESS in 10m 47s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e042b0918eac4d8b86d35a123e3fa119 : SUCCESS in 2h 20m 32s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/39f0009e6869414d8d30983b7f04f0f5 : SUCCESS in 2h 04m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f741b6e964d377a9f9f0fb6e59f2b82cf0ea2622","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-01 12:33:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"697cb9303f0f970c1c9e8b3936f288a5a273303c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 14:58:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34aaee00066c4e9e9a2d17e3c103cfea : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e562d9419a34d56bc28af61c90fc3dd : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c46dc427c1dd4eb1abcfcc7a0974e92d : SUCCESS in 9m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de4fe1efb10c45799738209c424e7179 : SUCCESS in 11m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5f13d2c568c40ee932316c58dd93f3b : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/101175978f6d4286ad63e01a3c538192 : SUCCESS in 13m 10s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/b716f6e6478344a89bcb880cfd3dcac4 : SUCCESS in 2h 04m 36s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/993ea29fe1934945aece11ba2e46f6af : SUCCESS in 2h 06m 19s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/117e4f9b4cf245e5bdb83b1425a47f9f : SUCCESS in 2h 12m 15s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc961a95ad994833a86bfa95f38bc802 : SUCCESS in 1h 40m 48s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/34145a78f6f14821bea19594754c691a : SUCCESS in 23m 51s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7725f5daea5e4045ba4cfb669d307db1 : SUCCESS in 21m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d0150382613d4c2a88dc32aaf8b2af61 : FAILURE in 14m 57s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/d7407dce5d114ce7a1e1fbbba8aa49a1 : SUCCESS in 1h 58m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/722fa116e0a041bf97db88b2f13afb42 : SUCCESS in 10m 03s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7cb14b16a2ae4263bff07a5fb77a4a7d : SUCCESS in 2h 19m 25s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b05bd001deda4c6eaa147d7824cd8cf3 : SUCCESS in 1h 52m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a31b64901c8f5f092ca3b97be7acaf0b4562e49d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-02 08:28:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de7c52a4bd8caa898d717b3f8153d1a5b1d2d521","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-02 09:42:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cd013482d93d6a6cd7502305fedf1bd4500e5c9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-02 09:53:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8c7bb8a982535727505ea728243e56bdcfc9eec9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 12:21:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9adbc1f8cd74b5db0cc3eeb80121935 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/687b81610b8b47f7aa427d6133c4c10b : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f867bc6b4445908d7e9efa7ac26892 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/847a6dcdc3b340d0b6bed988a33a9e54 : SUCCESS in 9m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50cabe43468747be8039bece1a654a63 : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ae61176877a484b866bd01e7fa3fedc : SUCCESS in 11m 01s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/c3153cd3af5045aa91fe329b99a22fef : SUCCESS in 2h 09m 02s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/04c8ba33fc8d4db6a4b81d0c15e78cf8 : SUCCESS in 2h 26m 09s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/fdae36d946064b15a5d3dd05f8a98253 : SUCCESS in 2h 18m 45s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c174c4cb6f64cec9891b740b863c66b : SUCCESS in 1h 58m 17s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c10f549028844fdaa1c801654e7c6e65 : FAILURE in 22m 42s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e10e771486d34c5ebeefeab88e62a394 : SUCCESS in 18m 26s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/17cdd66d0666436aac1f5fffb296fc24 : FAILURE in 15m 46s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/abc5f4fa7f2c483aa02ecf36a285e94f : SUCCESS in 2h 02m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d4df5ccdee945ad8ddb251cdb112dcd : SUCCESS in 12m 07s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f99c563d0c964d73a6ec8fba307f3fdc : SUCCESS in 2h 19m 00s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/28a34dbc03bf46ad9798699fa1476005 : SUCCESS in 1h 53m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2eab7f4710b796f94352c9723dabe42d1aac6a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-02 12:39:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"383859a91905f9d542a1a2c1c8c0725cbcad706d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 15:14:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a4e739be6fe48f4b099b206d4768823 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1503080878c4f848602432e19290831 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2273d1dbcf8491882ce71e43b789531 : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb99533c52314f0a8f257e716d1b2e96 : SUCCESS in 9m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b77fb01282ed401e808946fe9f1306f4 : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/190e6f8bb43c489f907504886d7a4eae : SUCCESS in 12m 31s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/43e42f7fcec948668331e6e1cf0efb2f : SUCCESS in 2h 22m 59s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/cd82dc597c054a8d87839659059a1c04 : SUCCESS in 2h 23m 36s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/4c17aa0abdfe42a1aa4dc3c1becdb9ac : SUCCESS in 2h 19m 06s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5022fa903e294155b85be7c09e322783 : SUCCESS in 1h 37m 49s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f17b462f5a064b33a629dc0d634a7b24 : SUCCESS in 25m 16s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4ae6f2e66a54409acd91634e9acbad9 : SUCCESS in 19m 43s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d7620fad9cc948879ab3e5346ca4b6d9 : FAILURE in 15m 36s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/7d999a1140894d828878bea826891128 : SUCCESS in 2h 01m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/998c39fff8cf463584d247c648e11715 : SUCCESS in 11m 25s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5d431096d3e04e34b6509de48f3f2199 : SUCCESS in 2h 21m 52s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f36ab84918fb417f9d33754c8d624720 : SUCCESS in 2h 07m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23f3f54d27d8cfd653c885d2a430a7bd440ee43e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-02 15:15:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"65ed85733de5c7a7fdaee5fe5eedc131b1d483ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 18:12:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e5e41ce2bc54b12b745f564e047f8bb : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87e5379981754250814cccd9ebb0ec3d : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab7f8d3583448999c0e1a1a3c2df042 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b2b12a2441341cabd02a5fa7c6b9eae : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/829e783cf9f34c52ba60cb7f71c39d5b : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da6da4575a07496a8fb6465612681b31 : SUCCESS in 13m 29s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/636f8849ea78476684adafc41b3073d8 : SUCCESS in 2h 16m 18s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e8e444762fb54a01a0f2f3e5b67eddd4 : FAILURE in 2h 01m 10s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/c1a104e3a56f4e7b8ecb596dbafbbc52 : SUCCESS in 2h 22m 46s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d85d699a4e524d7fb64c90de14be7b22 : SUCCESS in 1h 42m 16s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/3be80206f94c47b29b060234ce5d1848 : SUCCESS in 20m 01s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5639ec5588340249dcb51916913c6c0 : SUCCESS in 21m 31s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a5fac2dc37d74f1c9dfc5fde633a27ea : FAILURE in 17m 18s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/5630fc98df4c4749a9578eab1c60f7eb : SUCCESS in 1h 43m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adeaa72e4fb7411dab139b8c3b6905b8 : SUCCESS in 10m 48s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c8295b96370b4bdda4560b064de81874 : SUCCESS in 2h 26m 05s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/48b1d728352540ce9cd3db5d80037127 : SUCCESS in 1h 58m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76f7f61debc1153d17cd1b5621b1a89d3f034b5f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-02 21:24:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a599a5a8aca727d7f6af884d012edd4fce6d1c5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 23:50:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/959e4fd86c3f440288fa70688d8fa254 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9d8688e0608413b9d0d50bcc2098278 : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a08cfaad2f4e26bb9040dcbdb07772 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fabb0dac82a94fe98cf65afa8d801c80 : SUCCESS in 10m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ebbe84435a741c0bdc653512b345180 : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98f6c0920e4947018317575c02ced28a : SUCCESS in 11m 03s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/e1acca7b25fe40939a9256247223196e : SUCCESS in 1h 50m 38s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/08fd6e05ed554c029252bfaf713886aa : SUCCESS in 2h 07m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/94250b70ca16422a9e980a8491d9acf5 : SUCCESS in 2h 16m 38s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51325ba1b4e44f41bd63180830301fc8 : SUCCESS in 1h 38m 04s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/5e8562871bbe4247b702a0b46c40d6ca : SUCCESS in 18m 32s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/535af501942547aead4f1016eb40b02b : SUCCESS in 18m 59s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c576852fea0e4d7a90750c1921e73b70 : FAILURE in 13m 42s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/3fa89066359b4fa6a50bc42fe39b28b2 : SUCCESS in 1h 55m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b9de01a57184b028e45d4815b04d397 : SUCCESS in 10m 54s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/28d250b15a764d39b7f98a4efe29272d : SUCCESS in 2h 23m 22s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/25c9f3301b9348bb92e40864c15b92a1 : SUCCESS in 2h 00m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ca8859cec642cb788f3ef39179518d9e55fd903","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-03 13:13:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3fcc0a33d73366352087038921280818bd6ad475","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 16:15:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c906207d47a84a1c8a3ae87d5ba7ab28 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/898e0987fac8444bbfd35f7bbf64dcdb : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824b2f6255e643959e68e741cae09e2f : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc88836c8de14adfbd809685e33cbb4b : SUCCESS in 14m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c976a3635b56443b9a028836066898b1 : SUCCESS in 12m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ad477365dce4f73bc17c6574b08afeb : SUCCESS in 15m 18s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/fe76848d3c9845c8bc9f16e273c46d33 : SUCCESS in 2h 22m 19s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/55ad027974a4457b8043f34f7597de5b : FAILURE in 25m 52s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/9a25322c074245d0b4e6b618fe5d46d2 : FAILURE in 29m 44s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e80baf37035144438166ba3cec962665 : SUCCESS in 1h 38m 04s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/da716e55258b4c2ea95cc1be0c4d10d9 : SUCCESS in 24m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5755efd85cfa4bf79ae29555f78fb8ee : FAILURE in 20m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9375af4fc784483fb62c81649c7a9c20 : SUCCESS in 2h 57m 04s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9a794752f737488b9a59e3f38279dd59 : FAILURE in 29m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06de33eea3c440069b76cf823194ace3 : SUCCESS in 11m 10s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b98d634d481b4c0f9e68d515e1c816ca : SUCCESS in 2h 35m 51s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bfddb1941e21481286b66a0c3d5c56d8 : SUCCESS in 2h 09m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"162a830fe3e7a53dc7d72a475636aca218d06861","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-04 08:25:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"aa067c2087adfea635269aefda3f04bc346b3938","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 11:25:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d98d40d5c994c818fe8e5c54d44301e : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd3fa0740bcb4b4aaa5db56ee9b313bd : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc884f6b3e9c42a48b69568e359a0190 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e2e25bf3ab64f4db405ad18fa26e906 : SUCCESS in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51794124928b434e8f9cf9b5a2426cd3 : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53aecf1b0bcb46cd9879b654c03e2f1b : SUCCESS in 10m 45s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/7a42e4cb74f4411890f815a477aff64e : SUCCESS in 2h 07m 25s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/dc2a2f30eef343a9b9312950156a8a40 : FAILURE in 20m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/4b83964ee6e44e378f524548eab8c7df : FAILURE in 21m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc22170c1d3a42a5b81e4b6f31a21e47 : SUCCESS in 1h 45m 14s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f94960d5c1d545c9ae07a046587390ed : SUCCESS in 20m 24s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/c398d7475a754e1aa26590dc91cd1986 : SUCCESS in 21m 33s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ec7b9f4dcb124b94addbfb4733c7c775 : SUCCESS in 2h 54m 03s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/f0bd11ce8302424990459f949521765d : FAILURE in 24m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/227685bf1d104cffa03433ab2825ccff : SUCCESS in 11m 11s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bf282e50b8214ce2a8ea016989d64808 : SUCCESS in 2h 20m 38s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b7a2a3b8410644bea2059c2bc2d0d590 : FAILURE in 2h 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e619e093075a5b643292d379b3e0f361ea2687ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-06 08:10:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"71b9accb1238de14e7d9bd10851077c79feef4f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:27:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced3468737424efbb90de3379814bbb9 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba50a36b6a934cdead786b73439b674f : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5063171a7d82455ab3f26b91129a62a7 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71b55c0fe6f440d69dbf6d46c43e454a : SUCCESS in 11m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af1c30d33f81461aa7ab5ff3d9c67c7c : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30f5fa5b4dca4bf4bcbdb8ea398b0c04 : SUCCESS in 11m 00s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/c3bb06b732224dfa8469b399afb82dff : SUCCESS in 2h 15m 20s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/cc00f3f807f840948e30085a734e7c24 : SUCCESS in 2h 34m 32s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/31036e9c2e6d415da6e2228642c9a28f : SUCCESS in 2h 43m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9c454c64ff847b58f0547089ff121fc : SUCCESS in 1h 55m 17s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/2aa4254041cd48538ef225017436e4d5 : SUCCESS in 20m 24s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e77980491c584b03b3624ea3112da180 : FAILURE in 25m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9e5dbbeb68fd4d7eaf546a92b410ccd2 : SUCCESS in 2h 52m 17s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/695af967b8764f58ab3246a06606b27f : SUCCESS in 1h 55m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bf5325dd2ee4ac7870d65eca3f0d13b : SUCCESS in 13m 59s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/583c6424a0a4483ba9e5b2ebc7350bc8 : SUCCESS in 2h 38m 19s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ed463dbd35704f57ba86e3415e1388be : SUCCESS in 2h 07m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5d00e0685ed3831fa36e84d9d1cfd8807edd317","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-06 15:14:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9f683db4c00b1b11014093ebcd34e55371ff4a2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:20:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda5cb31f89148dda337c21cc0fec5fb : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62d968db79ad42c59eb5ab33b7a91b4b : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9ff78f87ef4cb4852001d14c4d812f : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75d3df100d8644cca150e8d7c0e9bdc4 : SUCCESS in 9m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/813e606726b84b5598aeb58a52e6e076 : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5a85cf350074fd6a10d015610428bfe : SUCCESS in 9m 36s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/c5b0dc6453374f92a3134eb1ee1c585b : SUCCESS in 2h 34m 25s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e79e891279f448bcb03650c06db1434b : SUCCESS in 2h 00m 14s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/6b958e1c2cfc476699024d7d82ca6088 : SUCCESS in 2h 36m 10s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5e2ea75179d45c1af6d99958e43bd13 : SUCCESS in 1h 46m 55s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/827fc863e48a46eeae1c0888fbf21104 : SUCCESS in 20m 20s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/17b4e34e18d24bb190b209064f1c722c : SUCCESS in 26m 13s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4316fb7288094574b39f0a0908f382e5 : SUCCESS in 2h 58m 33s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/dc85ec755d234c6bbdf7a24034644069 : SUCCESS in 2h 13m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1693d82928c44aa396d339520cceb7dc : SUCCESS in 12m 18s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f612316389174e3baa28f92381a70ec5 : SUCCESS in 2h 26m 59s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6312dcbd5385415d9abe449c3056a009 : FAILURE in 2h 03m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"529ec194e966b6fa03d946c007e2d20633012fcf","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-10 13:10:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8a8093755abb69b95adb3557d034528d05d7552c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-17 09:05:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84ea615d6de7cb0353db08dea8947adb24d1a42c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-17 12:07:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbfaa2c3a32ddb5f97e053f487bda2d7003b91b6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-17 12:08:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9abd2b58044a8768d12e69b83ab2b4f7c324ad5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-17 12:09:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"08093d3be31225017ac47f5cb141e955bc2b1212","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:16:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8553c91eb54e8eb1d0384651f53d39 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7db28d138be45bbaf36bd728816fd2b : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bef27d3721d4e14a9b751fcc7565cd3 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19168fe599404bd2a0c9e277435e4cee : SUCCESS in 9m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c59a4faf4eca4b54a3e66c53524564d1 : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b032d9c7e884d73a460cba46cac680f : SUCCESS in 12m 40s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/675dfabf2e914456b25eb3b1b203733b : SUCCESS in 2h 03m 17s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/34be7607360c4572a42952149c1007d3 : SUCCESS in 2h 07m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/2d8aec729dbf42f6999933f0eb3bd5f8 : SUCCESS in 2h 31m 33s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2049d90f579d4fc2b551d15babb3015f : SUCCESS in 1h 39m 26s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/3b54f8acc58f4fc99da9e4ca93f43634 : SUCCESS in 19m 59s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce1f6199143643fc96d9b0a5117ab812 : SUCCESS in 24m 53s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5e0b819070cc43cb96939976faeb5d43 : SUCCESS in 3h 12m 32s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/446fc26846a74bd084702776dd50e526 : SUCCESS in 2h 10m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef49a094a894417788a7286a7e1c094d : SUCCESS in 9m 59s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/34d42ef62b5641d3bb2e11533d4aa96c : SUCCESS in 2h 35m 07s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f47944fcf8e8402081eb02c1dba7e051 : SUCCESS in 2h 06m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb1904c9b4b05b8e24ce6e79658fc065bdb4c3e4","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-17 18:35:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nGood stuff!","accounts_in_message":[],"_revision_number":3},{"id":"b7d130f83d74b59dc6d7b304a91a5590008232b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:35:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a707e1fa96fd76fca40d13391510f1cc7c466a62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:02:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872db136ef7041cc9944c9e24f53deac : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/041ca92d68044af98efb339f1725db64 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692086b04e4a47cc8f8a697773a494e7 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4dc52975483438d8620ddced23f84f0 : SUCCESS in 10m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8c1ae4537ff47dc8641b4528739d7c1 : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f9517a0acd0497baaf5a1417f4eb8fe : SUCCESS in 10m 25s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0f3b8c904dd54c08a33f66bd54f796b1 : SUCCESS in 2h 11m 07s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d1adc35d54fb4b7da111a18093c84444 : FAILURE in 2h 20m 40s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/901e50e0a6a642d8b2de7a37da0e1b54 : SUCCESS in 19m 41s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/38139879733e4e5ca8288c5e4b077470 : SUCCESS in 23m 09s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/340b4221e982479090fddd3668094ab9 : FAILURE in 2h 05m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4ab8fb1970e44a4bd3d9db3b026a017 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ba77c82ab44006cf59408c07f1f8963b0e22d998","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-17 22:51:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"04f5739d6e11b2af86775478a135eb94d91af47d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 02:39:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c3cc77be444e65875600871c7b7b0a : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c674ba839bb4ee49cb54dc3310aa5ed : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311a5003f0d84332a6d8116e251d0d62 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8dd4e9841a94b2c894f52955634ab77 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f64a0655ce44821963dc5c1a415798e : SUCCESS in 8m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/182cf7e1861a441bbbd62ff73e3957eb : SUCCESS in 12m 27s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/187295c2a82345f0a6241a03c412159d : FAILURE in 2h 47m 39s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3781c9db40154b689381dc8a06dc9e46 : SUCCESS in 2h 27m 45s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/74aa2fcc7d5e41f496922b0f2f8c8a58 : SUCCESS in 2h 53m 02s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/136f7dc8f5244d719512773146b89cdb : SUCCESS in 1h 41m 49s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/30098ef6042544b1b4e03ea174b289a0 : FAILURE in 29m 57s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a9d11fe13f240e08a08082bed148645 : SUCCESS in 28m 25s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e653a9f0e1194ec8a20f8499b4fb3af1 : SUCCESS in 3h 13m 35s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/1f5804462e5b4489af02c598cb0e679b : SUCCESS in 2h 00m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/123b2ac0cbca4e6884353fa7dbb017b0 : SUCCESS in 12m 07s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/970f401fe7964e9e89363d29d981e994 : SUCCESS in 2h 21m 39s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cd2a3b81a01e4d7eb621be478e91d21c : FAILURE in 1h 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31efd1344c9aade5e2780efa73efe8baf41af287","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-18 06:06:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"52041c31710b11beaf3af12d325e9bb09d5e26bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:02:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c59d87bda86f40d4b4ce165f141ae4c6 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6c849867fbf4fea8df4d9c7a7f16b40 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5fd14bfa2246d6ba0c77c00c656fa8 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d595e1d1b034bc59f2a6e0f83887842 : SUCCESS in 9m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e5a4f37e8db467aa6c05452584f8b1c : SUCCESS in 11m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11a51a7a7a0d497f919639e21489cd11 : SUCCESS in 12m 09s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/de458ae84ed647cd8f392b72bf5d1015 : SUCCESS in 2h 08m 38s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5882ba99dc63459ba228846e69b019db : SUCCESS in 2h 43m 07s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/49619d8383e74ca48b53a2068a9db07b : SUCCESS in 2h 25m 03s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/52570300bd764f9ebade7cad0d4771ee : SUCCESS in 1h 36m 15s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/7c911b3a62c9415793ba5cbf0357945d : SUCCESS in 22m 31s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d59a7254e184b429a4916c2a1f11340 : SUCCESS in 19m 04s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/25331d1f32a14f8fbb0ecb3dd27b42bc : SUCCESS in 3h 01m 09s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9663d543073c4a8b992acbf3000a06f5 : SUCCESS in 2h 15m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f150648180448b99251d233c02d5a2e : SUCCESS in 10m 42s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f6dae4d86b204594ace8db62f5243bc9 : SUCCESS in 2h 12m 21s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/158e0cfedd2f45b88004f1e0cd76e721 : FAILURE in 2h 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8312990a83730cf6a398fd982aad20252246f09d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:02:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28446d20744d088a1067e8714de11b82ad8237f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:57:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506446ff15814633a7e3e456dbeeb581 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/caae034c4c174f63828078a828478a09 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee31ac35a047415ab30f2294a0bc46a5 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d654bab88e1c45fb8b10f44356cf77f9 : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/186c3a0e71a84471bb6c516b3fde30c5 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9de40b930af04fbcabecd1ccd6d3c8c5 : SUCCESS in 11m 47s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/067a6e830b904aeb8733ff42977645ed : SUCCESS in 2h 30m 39s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d0a1ec3a5bf14abd92c4be34d8384137 : SUCCESS in 2h 20m 41s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/5945a2ae74d649dc8fb4f629683f0840 : SUCCESS in 20m 50s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/83b76a9900354e82b61f200d5df76d35 : SUCCESS in 21m 55s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cbfe1e134e4747ce9d3e46e82d9ef8f6 : SUCCESS in 2h 52m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de6e7ef58091402299d911bf0c754dcd : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":3},{"id":"04cb672a9c2453175107ad18d57e1a196c65bbc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:57:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c9af896f021fa99e566aabdcae6e6672e907477f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:59:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd7bf8f7450468f8459aab6ccc19535 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3},{"id":"86c0c1c9bf0d403da57bf017d1811dcabc64fd80","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-01-10 13:03:34.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d5e6e0d265d2214c7e9130f4588abc76211f52db","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ae5975c2a98f6c3bd0a617316e86a79270af29c7","revisions":{"b9041013417f909a8225859e1bd997d5141c0554":{"kind":"REWORK","_number":1,"created":"2019-08-31 18:18:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/679545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/45/679545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/1"}}},"commit":{"parents":[{"commit":"d4fea5793f55933b0a62d3d17c08af184ae6b021","subject":"Merge \"Only set \"unknown\" in LSP that makes sense\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d4fea5793f55933b0a62d3d17c08af184ae6b021"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-08-31 18:12:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-08-31 18:12:52.000000000","tz":0},"subject":"Active wait to retrieve the Agent stats","message":"Active wait to retrieve the Agent stats\n\nIn functional test\n\"test_updating_metadata_doesnt_update_controller_stats\", do an active\nwait to retrieve the Agent stats until \"ChassisMetadataAgentEvent\" is\ngenerated and a new stat is added to the global instance \"AgentStats\".\n\nChange-Id: I6ed2573b71dc91a60d860537512bfcef7a212e17\nCloses-Bug: #1841913\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b9041013417f909a8225859e1bd997d5141c0554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b9041013417f909a8225859e1bd997d5141c0554"}]},"branch":"refs/heads/master"},"81774005c7cff7b2c17efc2ae47c5a572c039f5c":{"kind":"REWORK","_number":2,"created":"2019-09-02 09:53:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/679545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/45/679545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/2"}}},"commit":{"parents":[{"commit":"d4fea5793f55933b0a62d3d17c08af184ae6b021","subject":"Merge \"Only set \"unknown\" in LSP that makes sense\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d4fea5793f55933b0a62d3d17c08af184ae6b021"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-08-31 18:12:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-09-02 09:53:14.000000000","tz":0},"subject":"Active wait to retrieve the Agent stats","message":"Active wait to retrieve the Agent stats\n\nIn functional test\n\"test_updating_metadata_doesnt_update_controller_stats\", do an active\nwait to retrieve the Agent stats until \"ChassisMetadataAgentEvent\" is\ngenerated and a new stat is added to the global instance \"AgentStats\".\n\nChange-Id: I6ed2573b71dc91a60d860537512bfcef7a212e17\nCloses-Bug: #1841913\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/81774005c7cff7b2c17efc2ae47c5a572c039f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/81774005c7cff7b2c17efc2ae47c5a572c039f5c"}]},"branch":"refs/heads/master"},"ae5975c2a98f6c3bd0a617316e86a79270af29c7":{"kind":"REWORK","_number":3,"created":"2019-09-17 12:08:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/679545/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/45/679545/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/45/679545/3"}}},"commit":{"parents":[{"commit":"d4fea5793f55933b0a62d3d17c08af184ae6b021","subject":"Merge \"Only set \"unknown\" in LSP that makes sense\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d4fea5793f55933b0a62d3d17c08af184ae6b021"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-08-31 18:12:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-09-17 12:07:39.000000000","tz":0},"subject":"Active wait to retrieve the Agent stats","message":"Active wait to retrieve the Agent stats\n\nIn functional test\n\"test_updating_metadata_doesnt_update_controller_stats\", do an active\nwait to retrieve the Agent stats until \"ChassisMetadataAgentEvent\" is\ngenerated and a new stat is added to the global instance \"AgentStats\".\n\nChange-Id: I6ed2573b71dc91a60d860537512bfcef7a212e17\nCloses-Bug: #1841913\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ae5975c2a98f6c3bd0a617316e86a79270af29c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ae5975c2a98f6c3bd0a617316e86a79270af29c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
