)]}'
{"id":"openstack%2Fnetworking-ovn~679965","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~I01e95ceb10bd6b4e0d9562c9a2abbefa85cdf1d7","project":"openstack/networking-ovn","branch":"stable/stein","hashtags":[],"change_id":"I01e95ceb10bd6b4e0d9562c9a2abbefa85cdf1d7","subject":"Pass Domain Name to dhcp options","status":"MERGED","created":"2019-09-04 11:11:49.000000000","updated":"2019-09-18 12:13:27.000000000","submitted":"2019-09-18 12:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679965-1568808726694-08dddfb7","meta_rev_id":"d496bdd893e630c8572c18098869adefcdfc74cc","_number":679965,"virtual_id_number":679965,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2019-09-18 06:04:17.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2019-09-18 12:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2019-09-16 06:59:21.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-09-16 06:59:21.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-09-16 06:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2019-09-16 10:14:07.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 20:20:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-09 07:45:08.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-09-16 10:14:07.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-18 12:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8adc3d05b36f16e0e9640b28879f6030d39efea4","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-04 11:11:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fafe0c1326e8367ce9c01358dbc8b5c811dda1ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:01:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c60b3520f44c4987b6bd6ecc51d6c5 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbf15c9a74eb45849be0c73477d8f835 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60955f9516a940d0a03c46d76ad08fc1 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f60ae20069b48099d9f36e8660cb814 : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ce06181924b4c03ad23e1085661788b : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63ce7387f34a4926a071004771142406 : SUCCESS in 11m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48b8a132329342ae90e3079041611ae4 : SUCCESS in 25m 03s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/61161a2635be4b7a82fb4a4dbf7c1560 : SUCCESS in 1h 58m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3fc0c1fb2be544698b11c6b78213d2e2 : FAILURE in 19m 25s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/5ab7067d47a744db8b1232c436885c65 : FAILURE in 21m 01s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bc28af8129984b82af6f8d1485a6ef2a : SUCCESS in 1h 51m 51s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d6ff3e7d074f41ab9d11785371798b99 : SUCCESS in 17m 09s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/91d390b7164b45e688f8fbec1ab96023 : SUCCESS in 23m 44s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/07efb5de36c94729b5a0c15db4070fef : SUCCESS in 2h 41m 46s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/4a22730654bd413c968a34d6023a8450 : FAILURE in 23m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6ea4af0e3d14393a8e84fe63cc17853 : SUCCESS in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"167212839f14cfc3eb53b48112b16d4a98f0fb83","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-04 20:20:22.000000000","message":"Patch Set 1: Code-Review+2\n\nSuspect this is having the same issue as the master branch is with the tempest job","accounts_in_message":[],"_revision_number":1},{"id":"5664c7ac24c4eb36733d83b8240ebe8f77e2766c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-06 05:58:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b1479af5679f8d3cf4bb6147119f2a4ec0dfa357","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 12:58:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8024ffdadc014328abbd908f5050b766 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c9ae0ff98e343a0a1e8109a92226ae8 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fb81e3266647e4a7c5af1e067c6f43 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ccc7963cf19d414e9cd3c46bea7274b4 : SUCCESS in 9m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da733b8e401a427faa1b3b24f4a8b37e : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17741789ba574a7a86161701efbc26f5 : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4add3d0a9ee04cb9951f78f25e1603ae : SUCCESS in 10m 59s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/60f3665e7e244844ba040609502c1b94 : SUCCESS in 2h 04m 30s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f5f576c55e154b8dac82746f0b56c08a : SUCCESS in 2h 03m 25s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/6050a67159154ec88cd2e88206c3353d : SUCCESS in 2h 25m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e69ad48222ab4ad6a8bf783a936ec7ff : SUCCESS in 1h 40m 50s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f32dd2e2969344b88265e00d5e4b8297 : SUCCESS in 21m 05s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2867d20006442f29be11a4c40ed6de9 : SUCCESS in 20m 10s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b31c2ae0c9164658a36acde28e58227e : SUCCESS in 2h 39m 45s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/534e2c5ad18642089aa16063aedecad7 : SUCCESS in 1h 56m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12c8f79042654147bed2b03348413479 : SUCCESS in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2217c011668abebb9b9055fc963a03ff09d4b9a5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-09 07:42:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab1fd8f319ae67bb361aa97a5dd39711e102d419","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 07:42:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75742896cfee888beb263e2c55a7fcf3a88db9ae","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-09 07:45:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"192172cad6532cd87e78c213eb8ec6eddd967e5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 10:27:19.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"2de884d442b32a14307da8da716a24a3e3488652","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-10 09:37:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"94e500dd6cb803fe34a7cb27e28c07910ef9708c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:29:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171ada9f8e2f4a72887d1cc7aea093d4 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0290e7fd3dba42ad994e095cc6a3da53 : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7480bd0f9b7a40e5a62c371265a26de3 : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a6974fe7e2834ba493600e2ff10db3b5 : SUCCESS in 9m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b719e7a5f527409db53abcc0653dbe28 : SUCCESS in 8m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09c1286fa21d41c5ab1ffb75d743bb12 : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c0cda271ef14199bf707e9b3bf69313 : SUCCESS in 9m 54s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/14fb65f17cf243b89a2dfb9afacdc83d : FAILURE in 30m 22s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6667b21840b7490084b69df5d86fa4e5 : SUCCESS in 1h 49m 07s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/44685f6439a74c8fada1d3e4f67b1186 : SUCCESS in 2h 02m 50s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/04ba019e3bda42ff8abddfa6ff44516f : FAILURE in 29m 03s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c4c1e0344f514a8d91b411a286fe1120 : FAILURE in 15m 57s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7291b1ac12be4a8f83ea881befcadb4c : FAILURE in 15m 13s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/61c883effb504799847fd3d076f368c0 : SUCCESS in 2h 32m 37s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/519fa3fce2ef4965a2788324ae0da69d : FAILURE in 43m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dce9b0ce5a84273a5bdf2392155aa8e : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef31fe77bc1a518d012d8ecaa0adab85eef08c38","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-10 14:00:20.000000000","message":"Patch Set 2:\n\nWe merged this on master yesterday to unblock the gate: https://review.opendev.org/#/c/680995/\n\nPerhaps we need to do something to avoid running against OVS master branch in stable branches jobs","accounts_in_message":[],"_revision_number":2},{"id":"6c2f9b54acf49b4ffe2f78d637b7f312e8840c34","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-11 07:30:06.000000000","message":"Patch Set 2:\n\n\u003e We merged this on master yesterday to unblock the gate:\n \u003e https://review.opendev.org/#/c/680995/\n \u003e \n \u003e Perhaps we need to do something to avoid running against OVS master\n \u003e branch in stable branches jobs\n\nLets give it a chance: https://review.opendev.org/681394","accounts_in_message":[],"_revision_number":2},{"id":"8b61aa92610f0fcdc66c0dc745adec5076036f8b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-16 06:59:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ac90c0f7beb55f86879ee2f5bec78965e32e338b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 09:54:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c674c94ff074595bc1f9923737e9153 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0987577a932540b784985ce907c4595e : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0f52ea368345638681de80a0fb7d87 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29523bd307424d44bf8295f67dedcc19 : SUCCESS in 10m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7d4b18657bb4c7990321c53f9fb1cb1 : SUCCESS in 10m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09f08fcedb57471d84ef523944453d73 : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0232d1907252434bbdc1dbdf14b8d0c4 : SUCCESS in 10m 26s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/2c589ece282646a8a830e05055c9232e : SUCCESS in 2h 03m 34s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/83e810e45c644f6e9894de59e6a60e50 : SUCCESS in 2h 17m 58s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/da6f68f367ad4c9fa544f59815da4673 : SUCCESS in 1h 36m 21s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d9d6c3df26984222a3b37b87e70be0e1 : SUCCESS in 21m 56s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/64b696108326411eb961dcba4d5edccb : SUCCESS in 22m 03s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/85e1a0b0e1f14706ba07ab6e367d7d4c : SUCCESS in 2h 43m 08s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/5ccb899e4cdf4d799e7e2ffedb0ae1ff : SUCCESS in 1h 52m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f82c8f26df6b49f9ad13d4e8a4b2aa2d : SUCCESS in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"284c9785001a3782ffc38317b29c2bc94b442508","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-16 10:14:07.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"37d824ff3c0e050dae6c42774e72c711d7cd08b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 10:15:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"55e846cb911002cc8c9bd119a473eac64076108b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 13:15:22.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2fbd5f8769474a81fd93df4bb465bf : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3083738e761f4c9e8e88536d30243d19 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2d0ce4c0c6453dad52b639b9d078f7 : SUCCESS in 3m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a2644db9486142d9849ee737f839e182 : SUCCESS in 10m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05d20d0e3d224016829c8354e99042bc : SUCCESS in 10m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d64c91dc97d4125967fc7edfc0b5350 : SUCCESS in 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f66d1f3a184d4ad7bebcbc9b2e232075 : SUCCESS in 10m 26s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5fab242f384c48d885f617a3f7ad3980 : SUCCESS in 2h 05m 19s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/557ea60b8a3044d180ee284dddccbba5 : FAILURE in 2h 24m 48s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa2a2930def84445bc4d6cda45e331e5 : SUCCESS in 1h 40m 40s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/8f6220e9f65f467ab295d09b06cf3608 : SUCCESS in 23m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f123c9c69e14d1aa9922696c997f339 : SUCCESS in 22m 05s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/026919c54eea469e9023a099b43665e7 : SUCCESS in 2h 30m 55s","accounts_in_message":[],"_revision_number":3},{"id":"cc1d820e502b1c8a48c0d47d932a6056d078e4db","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-16 13:18:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"769243b232fb1d00aa175fd4ea0b404829bb3938","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 16:26:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c117ec2364430dbe04cf99c7b9fd23 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09082a61a255413e8f6bf1eae78e34eb : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cab002f60eb4dd6ac919cb328a0fc13 : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4096627d0b2249fb90241ebce7cf7497 : SUCCESS in 9m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f852544f72b043caba136f95f7fe52e8 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/710addaf98aa46428b809052dbfe7d8f : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eda5acec422a410391c45aec2fc66e62 : SUCCESS in 10m 58s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/fd453d7d38544a76b6bfed08d18a37b1 : SUCCESS in 2h 12m 41s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/9743add9dfd447a0a55b648461f13078 : SUCCESS in 2h 00m 27s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e60addd2591b4c8889f40721e0dd0d3d : FAILURE in 1h 40m 54s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/bd382aaf9da54265aefbc7e0d1d8fd6d : SUCCESS in 23m 47s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/691deb8a70c840c5810eb3903ebff8de : SUCCESS in 22m 45s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ff896e0d2e0f49aeb12776f2e2aa69e1 : SUCCESS in 2h 58m 37s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/2d03d7c5b4004cec98430d5321f297e2 : SUCCESS in 1h 49m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e343f547e9424f78b9f57e28e127dc03 : SUCCESS in 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"091770ca0cb47906cdbae8fed00ef046ef751c3e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-16 16:28:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7ee014c158b3bc9efc0723842176fe446fe93494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:48:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76402329824a45fe9a8203aea97d4add : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74375336439f4aea8e8dfeb62b45453e : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a7788d78c04160a2490b2d70c68396 : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/861f544312594a9c88a43d3ab524ef49 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c41ebb39cede42ba88dc425b2ff545af : SUCCESS in 11m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2fae9d4722b405fbba6ef5a701edf91 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd88bb3f6fd04582ad3def2c334bbf08 : SUCCESS in 11m 04s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/55ba6cf08d824d7ba7588c855eaa6193 : SUCCESS in 2h 20m 08s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a227441253644c9e9e17d35a0fb3c458 : SUCCESS in 2h 29m 22s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2a424fb13e44088895ad0c39f37a3c0 : SUCCESS in 1h 43m 21s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/56f2e5b111a443e3bcbca0f07bbf94a4 : SUCCESS in 20m 11s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb66337d1a694534a309e66f3988cee0 : SUCCESS in 24m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/03055925fc8e49faa8c7a70eaa0b2ba4 : SUCCESS in 2h 43m 48s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/5e04da28401e437f946159cc1c29c6b0 : SUCCESS in 1h 49m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0673a37764dc4f2cb536efe86ef78e2d : SUCCESS in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d198611f961911a0126dc204795598baa7d9942","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:48:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b78e6e2d0760a4bc97bcf9f9ec86fb7eee6f148a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 23:15:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea056b2991cd42ff8ae14c2da910e0b9 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25da5c9f556a4ba2bf60d9f6598d619f : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db35c18523ec4a6c912d4b71f5d2aaa8 : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6260cc52b80d47f3aefba6cd22ec47c9 : SUCCESS in 12m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af6b5dcb7f5d4f8b8de1c4025b51458d : SUCCESS in 9m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/beb1246de7904af09e910fffde32acb1 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5d398b57e7c43ff9d91ba7b5edf57b8 : SUCCESS in 12m 45s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/73cd2f2483814212ab54f045c7ee77c6 : SUCCESS in 2h 08m 51s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e850d2833878419e8f0ad229fe695da2 : SUCCESS in 2h 04m 14s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/da260c3413024a239617b3ebc7694c8d : SUCCESS in 1h 41m 31s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e960c6e9e60b4903b08f555baf14bdef : SUCCESS in 22m 06s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/477952cd00014e79b1daffcf2b8052ad : FAILURE in 19m 36s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/337fe35a6c7a44c2ad6008c5917422a1 : SUCCESS in 3h 07m 45s","accounts_in_message":[],"_revision_number":3},{"id":"6ec0c4376938725bc4f900fa98e62e713445444c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-17 06:21:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b156005cc830c6ee024b7262541b886c3cbd7754","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:16:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826366f2c4464723833dc5a2565d2b93 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6dd4435dd30a4878bde4392e790fbc0f : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f7a891744148cca2a082701d90be3e : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2cf0cbb1e0bb49979cb3be92539d677e : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdf9ffef498044df9affd132f5dd24db : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04d7814fc2f149b587012fbad4b84214 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d469e78775e4c4996d0a8e6737562ce : SUCCESS in 13m 24s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ecadfed4789547bc8e160c8afbb0b5a7 : SUCCESS in 2h 06m 25s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/88a3be7d43dd4784bfd4fb16c524d52a : SUCCESS in 2h 25m 07s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3976efc391fc453f8580087572e262bf : SUCCESS in 1h 46m 52s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c33b12ac22c54004b7bebfa21941571d : SUCCESS in 20m 56s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/950126388a024d66a67f980cc9b8128e : SUCCESS in 24m 37s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/962e12f789024b56b774a1fd95144fea : SUCCESS in 2h 53m 22s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/827c8fb0e30c4b5e888f83c9b8ab462b : SUCCESS in 1h 48m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/312c395d72e04bfbadebeaa664b9c806 : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36e8da936f2af0b71c1b6ed8c1a4b8baf96b5442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:16:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"48a3602973bc7ade05760cf2f5ad4ec724d8a820","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:00:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72465fdd84645bd855277c16d2f50b0 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dabb545d4407481689afda4dffee3ca1 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2bd0736737465683e387678b3137b9 : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d22b3c077e146498dfc122f5171e9da : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490ef436e29e4878aea814dbcb1d72db : SUCCESS in 9m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4bf4d2eed8748418753a40c1b72b006 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4aace3b93934f09be486020ef0c99ed : SUCCESS in 9m 18s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/93ba75f410f24f50b469b28e39bc5ef7 : SUCCESS in 1h 49m 25s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/1906a60aa7a74f6483bd06054d8c9873 : SUCCESS in 2h 09m 32s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5fa2d91f3c3a4b3e9de0167e9bea44da : RETRY_LIMIT in 5m 17s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/83157d2a920a492ab6f47baf61a485f0 : SUCCESS in 17m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d94e79d26acb44069d9e202332734ca2 : SUCCESS in 21m 13s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0228d408c91b428c9beccf280d88bb3a : TIMED_OUT in 3h 26m 02s","accounts_in_message":[],"_revision_number":3},{"id":"8fe538e7173a8c0555bba0a2c6e8c48957f7a562","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-17 14:01:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9d92d0861bcf520b926ad343f56a1a3f2944868c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:50:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538e37183432478d9c3e6134583cdc77 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ad895e697004544aa87c6f37af575c3 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0877a1b19f2e4f7c9dcbbe7adcbdccee : SUCCESS in 5m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9596cc0c095141e6948638f1c6331488 : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57fefe890c904c2fa56aae5d3525908c : SUCCESS in 11m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d81838568584e799c843fe55ffea93f : SUCCESS in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f290e46f6f844e6aca50083ec6a9fab : SUCCESS in 12m 13s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/1792091be5b84931bacd4537eaacd45c : SUCCESS in 1h 59m 11s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/02b93103f0d6464885e60455ab39e407 : SUCCESS in 1h 55m 02s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5f27f94730b94449b2d3df62b4b2080e : SUCCESS in 1h 45m 39s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/1d4ce3cf1e024cc89d77cfd4200c0816 : SUCCESS in 28m 58s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ad5dae2e3d949b8849d843489b824a0 : SUCCESS in 19m 08s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fb7a1998e84f4088a2f619b72968c84b : SUCCESS in 2h 41m 42s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/53abe84788ab4f65919aa2dfc8e3e614 : SUCCESS in 1h 51m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6730f7ccb19c4e75a371d4c80068d863 : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c3d267665c5d541c80e09fcf1042b55bd16b9fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:51:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"704ed1c4c4c140d29b30a7c4e780cb6313f23437","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:36:42.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d0a3c80e90483e8661e0d8f6f461e5 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cba9f2a2c3da4d75a3b2add78c75e489 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c5ec3371f442118cf50c43bfea3084 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2bd6b8ee5df42af8c824ae00f946a9e : SUCCESS in 10m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a4de0f22d684a95b944a702edcf5a20 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12a17aeafe074766a4c68b3fb335abaf : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6c8d70cb81a47b1a8e165d3ba775c9c : SUCCESS in 11m 10s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/83a39ca0087f4fe4b8ede152aaa26e88 : SUCCESS in 1h 57m 18s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/597729e7c925439dad37424a60bec6e2 : SUCCESS in 2h 06m 17s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c2e757044d7f4f15af48c5885b647ee6 : SUCCESS in 1h 41m 24s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f42dc2e4d991434eb68d9aed4686b7a1 : SUCCESS in 29m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/74e4386ddfbd4e9ea635708be8b013fd : FAILURE in 27m 34s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/27904ddb795543198768563d227e319d : SUCCESS in 2h 42m 34s","accounts_in_message":[],"_revision_number":3},{"id":"0ae0435f575bfe02c45ad472af4c50625a93b768","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-18 06:04:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"58f3cc73b147196c4b9649dd4cc3cfd42c355e56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:08:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a780b856e734a14b8c77ab269bffec9 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf4cf51f23a9420c8d6ed1b7274331bd : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5badced83914db28b99f324999f9188 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/40f40dc50c47485991fa489707ec16cc : SUCCESS in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac13144f33974bf7a381991de8f4af49 : SUCCESS in 10m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70cb63d4025e491d93ac64f4949755f1 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56d607d0445f4ff8b3f4718beca152c8 : SUCCESS in 10m 35s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a2aa051f93974bc5b8335a40b9933ef6 : SUCCESS in 2h 07m 37s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a036283c862d4108a9bf6b106c4cae90 : SUCCESS in 2h 02m 01s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/93b778a14f6748a0a248d904abfcd286 : SUCCESS in 1h 54m 54s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/1e2afd631f264f72a1bda3d7856954c4 : SUCCESS in 20m 47s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7284215368e2419699925cf9d4e8453a : SUCCESS in 21m 38s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/48437a81306d41b2868370b245095293 : SUCCESS in 2h 36m 57s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/f4ee2a413b644e7e8ecd33b257e10df7 : SUCCESS in 1h 58m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74f8bece4008422398731957d43df035 : SUCCESS in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"755c1ee052161fdf5af7fbd0419a7a9ac42432a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:08:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"769836685447a90d7d1377de19134b7ec5e305ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:12:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"998dbfff0839e4d250d0d5b669f2c36ccf4ee5c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:12:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0df82bf8fc19468a86085b6ac6e6c9ad : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f809cb762587447c9287bef2ed562128 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3639dddefef2400ab6b38fd632cbd0a1 : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8291476da228456e9908a064b1109f20 : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d589a2c7bb34abd9fd0e7e2bc07f230 : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e079bfbc27d94b79af860e384225257a : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55c55fab52d94591a6437f3ba0251067 : SUCCESS in 12m 28s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/04b159822d2743aea37cef3056abd7c9 : SUCCESS in 1h 59m 56s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/03a3d7e043614f938c37cd841fc7e8c9 : SUCCESS in 2h 05m 27s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d27038ea3f4e40f1ac22d53673df2564 : SUCCESS in 1h 45m 37s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d9d98f6dfc07475c8ba0f1a1a174e5ec : SUCCESS in 24m 02s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4d1768b865b47f2ad3abc58d085bb52 : SUCCESS in 23m 03s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/aaa655c7be71422d838d88b9178dd49e : SUCCESS in 2h 37m 44s","accounts_in_message":[],"_revision_number":3},{"id":"d496bdd893e630c8572c18098869adefcdfc74cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:13:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070003713c334dd9b65c15503c33e908 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"300c2e09311a4336ce155c701d5c824745b92f80","revisions":{"fa81351a0759b15e290787928c042fe74a2f3048":{"kind":"REWORK","_number":1,"created":"2019-09-04 11:11:49.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/65/679965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/65/679965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/1"}}},"commit":{"parents":[{"commit":"c4367485a18947bd204ce6ef72cc7d43890bfb3d","subject":"migration: Use python3 packages when needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c4367485a18947bd204ce6ef72cc7d43890bfb3d"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-05-21 15:07:33.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-04 11:11:49.000000000","tz":0},"subject":"Pass Domain Name to dhcp options","message":"Pass Domain Name to dhcp options\n\nOn subnet create dns_options are passed to OVN.\nThis patch adds missing domain_name field, which\nis needed to support InternalDNS functionality.\n\nChange-Id: I01e95ceb10bd6b4e0d9562c9a2abbefa85cdf1d7\n(cherry picked from commit 8c9501f6d74141af0127630c420c756c5f885c1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa81351a0759b15e290787928c042fe74a2f3048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa81351a0759b15e290787928c042fe74a2f3048"}]},"branch":"refs/heads/stable/stein"},"221c8b0f5395d9fe37df5ec1b31871ce21a96fbb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-06 05:58:31.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/65/679965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/65/679965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/2"}}},"commit":{"parents":[{"commit":"c49c67e70bb0db81dfeef46b72037d68db532a0a","subject":"Fix OVS build issue on Fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c49c67e70bb0db81dfeef46b72037d68db532a0a"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-05-21 15:07:33.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-06 05:58:31.000000000","tz":0},"subject":"Pass Domain Name to dhcp options","message":"Pass Domain Name to dhcp options\n\nOn subnet create dns_options are passed to OVN.\nThis patch adds missing domain_name field, which\nis needed to support InternalDNS functionality.\n\nChange-Id: I01e95ceb10bd6b4e0d9562c9a2abbefa85cdf1d7\n(cherry picked from commit 8c9501f6d74141af0127630c420c756c5f885c1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/221c8b0f5395d9fe37df5ec1b31871ce21a96fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/221c8b0f5395d9fe37df5ec1b31871ce21a96fbb"}]},"branch":"refs/heads/stable/stein"},"300c2e09311a4336ce155c701d5c824745b92f80":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-16 06:59:21.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/65/679965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/65/679965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/65/679965/3"}}},"commit":{"parents":[{"commit":"1db61810937b622f8c786ca6e46cad54430df061","subject":"Fix OVS build issue on Fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1db61810937b622f8c786ca6e46cad54430df061"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-05-21 15:07:33.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-16 06:59:21.000000000","tz":0},"subject":"Pass Domain Name to dhcp options","message":"Pass Domain Name to dhcp options\n\nOn subnet create dns_options are passed to OVN.\nThis patch adds missing domain_name field, which\nis needed to support InternalDNS functionality.\n\nChange-Id: I01e95ceb10bd6b4e0d9562c9a2abbefa85cdf1d7\n(cherry picked from commit 8c9501f6d74141af0127630c420c756c5f885c1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/300c2e09311a4336ce155c701d5c824745b92f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/300c2e09311a4336ce155c701d5c824745b92f80"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
