)]}'
{"id":"openstack%2Fnetworking-ovn~680363","triplet_id":"openstack%2Fnetworking-ovn~master~I815928e1a67e34c3aa577ee081541b67b2c43ddb","project":"openstack/networking-ovn","branch":"master","topic":"bug/1842901","hashtags":[],"change_id":"I815928e1a67e34c3aa577ee081541b67b2c43ddb","subject":"Fix LogicalSwitchPortUpdateEvent in Octavia provider driver","status":"MERGED","created":"2019-09-05 10:10:38.000000000","updated":"2019-09-19 10:32:09.000000000","submitted":"2019-09-18 20:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680363-1568837451647-cb0ff890","meta_rev_id":"c248cc30aed8e86ab05b5109986ebd1c98526064","_number":680363,"virtual_id_number":680363,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2019-09-18 14:01:32.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2019-09-12 07:06:17.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2019-09-18 20:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2019-09-18 14:54:52.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-09-18 12:42:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 11:24:19.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2019-09-18 14:54:52.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 07:06:17.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-09-18 11:24:19.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-09-18 12:42:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-18 14:54:52.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-09-18 20:10:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff97ff59d7f7157a41e6b434c05d419db3668950","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-05 10:10:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"116b628ddba64ecf16aa0d3708c2e9bbfee6e2cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 13:29:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50a7e62004545ff856648ec97fa74c2 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4baa56737d434305972cbb2892d849de : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651b5d6edf4245778ad52c8cfd2b66d2 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbf8c57ce19f4b3bb4c62a2532da6b56 : SUCCESS in 10m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7d0c0d1af7b4f05ad3281fb715581e1 : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fdf2a76924a42298fb6c0e85110f751 : SUCCESS in 12m 21s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/50730916dc67490798c8d67389a42c3b : SUCCESS in 2h 17m 18s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9f99db0cbb9e48278f6dc5f2fa8904f1 : FAILURE in 19m 16s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ae39f81fa3884070a18e3c86f67420bb : FAILURE in 24m 02s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb8d7a48269b4f3bb37a9553dd3c3b44 : SUCCESS in 1h 40m 04s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/66dc3cad5d1a4091b20c322a4a176536 : SUCCESS in 29m 23s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e7e698af4b84365838e9e0447e80266 : SUCCESS in 19m 51s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3b9ab68587a84480940a90c545db3f7d : SUCCESS in 3h 13m 52s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/97120afaf2c546289b186d415a5bf474 : FAILURE in 20m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c618f4d5b0d4e638b9d377726162583 : SUCCESS in 10m 38s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/067c294935c6469287a86d427cfadf5e : SUCCESS in 2h 19m 54s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c5ece2123c1244599a65ba36eba7d7ca : SUCCESS in 2h 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"429cb5aa657ab7755089e74e267069130053ee96","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-06 05:56:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3b7f095ab8ad57eb8790c2682ce53639d6622a85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 08:43:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568a088d3af04db58571e694eccd464e : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b388f4057414a0db30b4f49e949b992 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a6b979ae7f4f8b93bffc8d116c9eee : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/990a5f5c6bb8476a9a5428fc71dba721 : SUCCESS in 9m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e1d0cac0a7c40ebb475ee5ac2aef62c : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e74b904b26142ccbf302d8c97d06115 : SUCCESS in 11m 42s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/80d000fe566942f498cff20dbc970f4a : SUCCESS in 2h 20m 38s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a0ccc893c9994e4687c03bb66d450c38 : SUCCESS in 2h 39m 43s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/5000365cacef4c75a34356742d4c893c : SUCCESS in 2h 19m 26s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70355bfb4841489bb22005491ba3e4b7 : SUCCESS in 1h 57m 27s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/46d94bf09c5c4c009c4b855f5fb5e25c : SUCCESS in 18m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bb0611e4bec4a40ae7a2d34ba321199 : FAILURE in 19m 36s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7f90495a4cb2455db683fb3f3aa707f6 : SUCCESS in 2h 39m 45s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/4a0c5e2ed55e4f349e0037b5117ff9d6 : SUCCESS in 2h 08m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30d26204eee340e9a5d7e3b6ee377108 : SUCCESS in 10m 40s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/96db84eefc724d0fb5cc3313be450a76 : SUCCESS in 2h 17m 36s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9392595df0404c8d916d1bc644ffc1d6 : SUCCESS in 1h 58m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec2467457f8bf071d109328c6d7288112084dbc3","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-06 10:32:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"66d00058a2b84e488bec8d5912842cccfbb3ac2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:02:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dadbf023d654987bcd137e217d27eaf : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fee30e6f0884b4895eb5c524e6665fe : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fa027b8b134eca8454eee19f34d065 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbd38cc4a2a74b49b272cc495309e9eb : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8c53327e0434491975977706f6e0790 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0027fc605dd415e8cb05a7c2ce45b1f : SUCCESS in 11m 33s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/2b38b93f2bf94eeeb3c674a9efbcb5ec : SUCCESS in 2h 05m 10s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5dceb5ba3b9243f990a15e3e02b2c190 : SUCCESS in 2h 15m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/34bde31c6fbf419a87b5d92ef952f4a2 : SUCCESS in 2h 38m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b57c32b887004b879f22dc4434a257f3 : SUCCESS in 1h 48m 21s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b020cd258ccc43f9b7764b0236154d85 : SUCCESS in 18m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/78c9ec579663404da3889c8fd42e93fb : SUCCESS in 22m 13s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/987a78d336f2492fb007cbd369583dba : SUCCESS in 2h 57m 28s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/174e823dc11c4a8f8de730b4c10b3916 : FAILURE in 1h 46m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/003b00704bab4105b96d8acca6ab2879 : SUCCESS in 13m 00s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/14e8660162b84d7084b953c445585e4e : SUCCESS in 2h 17m 49s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2fa6e476966f47e59aa26d2f18e69c50 : SUCCESS in 2h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6952f6c8b54f046c74ac12088fc8620bbedabb0","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-12 07:06:17.000000000","message":"Patch Set 2:\n\nIs this fixing some bug that we\u0027ll want to backport?","accounts_in_message":[],"_revision_number":2},{"id":"d8f14f77697d204ce6ec27dbe64abe8c71880e14","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-18 08:02:53.000000000","message":"Patch Set 2:\n\n\u003e Is this fixing some bug that we\u0027ll want to backport?\n\nyes","accounts_in_message":[],"_revision_number":2},{"id":"324f7198ac8a75e232ce615d2b068c0e00272c4e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-18 11:24:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f13dd47e6ed60f8b645e76fdb11ce3225697983b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-18 12:42:22.000000000","message":"Patch Set 2: Code-Review+1\n\nFunny, I had filed a similar bug but a different fix - https://review.opendev.org/#/c/682632/ figuring the test was bad for not sending the value.","accounts_in_message":[],"_revision_number":2},{"id":"625bc5a12c798f3ae81c088d0ddfa15d1784aaa9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-18 14:01:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ad4ddcd3e94262c5ded95b5333cec62944ddd72","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-18 14:54:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nGreat catch!","accounts_in_message":[],"_revision_number":2},{"id":"9705d88393364fb15e92d0ebf3b732d209c329d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:55:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a65e6445987161d2b16c1dc69a61ae4b4d816677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 20:10:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"aea6d53858b1ff8ab4a31a60bdf7d76d187de36e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 20:10:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a7adddec7745aaa37bdd1a4e0c8da5 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2efa9f6e96774da487f747f1ceb2135a : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b95bb731a54eee9f2d83bdda41619d : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b532736cfdf74852ae91eb22cbe2dfd0 : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0d74284769e48c5b65876d36a8a926a : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/151ac4aa325c4871a7772e9f1c5c9812 : SUCCESS in 11m 31s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ad518e49e53a4133b173d4d9e3604d2f : SUCCESS in 2h 38m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e9242dfac78e4926b9693a68c8f03232 : SUCCESS in 1h 58m 30s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/413abe5c9a874396b16202ef2af74a38 : SUCCESS in 20m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d719c222aa24b5f839bacabdd8773b6 : SUCCESS in 20m 54s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6b4c1de976b04419a0b51c9f2aa34a7e : SUCCESS in 3h 11m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afeb25f3840d4b388bdde791f7e86eab : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b77f1e31f73125257847ad7b45f3bc75026ce233","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 20:11:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b87453ee9d44d388f3edd6d4f67f813 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2},{"id":"97c90b5dd728bc92a997c30b1dbe12067c50ccc6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-19 10:32:09.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 28cd4e01c30b41b9626ba2dadacef7faa93d25a6","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c174fe3d3f3b0c9ca1b23c2cc76b5063a3e9436","revisions":{"1859daa1aa6819158773be967f9a6a15945ca25e":{"kind":"REWORK","_number":1,"created":"2019-09-05 10:10:38.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/63/680363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/680363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/1"}}},"commit":{"parents":[{"commit":"d4fea5793f55933b0a62d3d17c08af184ae6b021","subject":"Merge \"Only set \"unknown\" in LSP that makes sense\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d4fea5793f55933b0a62d3d17c08af184ae6b021"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-05 10:10:23.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-05 10:10:23.000000000","tz":0},"subject":"Fix LogicalSwitchPortUpdateEvent in Octavia provider driver","message":"Fix LogicalSwitchPortUpdateEvent in Octavia provider driver\n\nIf port ovn_const.OVN_PORT_NAME_EXT_ID_KEY external_id is not\nset we should return empty string for processing instead\nNone.\n\nChange-Id: I815928e1a67e34c3aa577ee081541b67b2c43ddb\nCloses-Bug: 1842901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1859daa1aa6819158773be967f9a6a15945ca25e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1859daa1aa6819158773be967f9a6a15945ca25e"}]},"branch":"refs/heads/master"},"5c174fe3d3f3b0c9ca1b23c2cc76b5063a3e9436":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-06 05:56:00.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/63/680363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/680363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/680363/2"}}},"commit":{"parents":[{"commit":"4d51416cff1ef8784b57c80e116927ffc90bb300","subject":"Fix OVS build issue on Fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4d51416cff1ef8784b57c80e116927ffc90bb300"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-05 10:10:23.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-06 05:56:00.000000000","tz":0},"subject":"Fix LogicalSwitchPortUpdateEvent in Octavia provider driver","message":"Fix LogicalSwitchPortUpdateEvent in Octavia provider driver\n\nIf port ovn_const.OVN_PORT_NAME_EXT_ID_KEY external_id is not\nset we should return empty string for processing instead\nNone.\n\nChange-Id: I815928e1a67e34c3aa577ee081541b67b2c43ddb\nCloses-Bug: 1842901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5c174fe3d3f3b0c9ca1b23c2cc76b5063a3e9436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5c174fe3d3f3b0c9ca1b23c2cc76b5063a3e9436"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
