)]}'
{"id":"openstack%2Fnetworking-ovn~680749","triplet_id":"openstack%2Fnetworking-ovn~master~If0f15418e9d75a73ca69a058785cef737ee48a15","project":"openstack/networking-ovn","branch":"master","topic":"set-router-ha-dist-attrs","hashtags":[],"change_id":"If0f15418e9d75a73ca69a058785cef737ee48a15","subject":"Fake the L3 DVR and HA extensions out","status":"ABANDONED","created":"2019-09-06 17:31:31.000000000","updated":"2019-09-24 19:02:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a94411014bf11732603377b459b446382748e192","_number":680749,"virtual_id_number":680749,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"date":"2019-09-10 11:57:04.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"date":"2019-09-10 13:07:51.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":-1,"date":"2019-09-09 18:54:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2019-09-10 11:35:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-09 18:54:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-10 11:35:49.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-09-10 13:07:51.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"}],"messages":[{"id":"c790c4621ee90b711a7bc3188a2dcc28b5a4d058","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 17:31:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd045c75fbfbe31e62be98b43ed23278a197bed2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 17:32:54.000000000","message":"Patch Set 1: Workflow-1\n\nHack to see if this would work to pass the extension tests, further discussion required as to if we really want to do it.","accounts_in_message":[],"_revision_number":1},{"id":"403183e50618351fe4f0ad14a640ff2aa71c9551","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:23:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87973d843c447a78068f5b8489c3e8b : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fb13708e0b849aaa03d8b7476827d68 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23be1714c664aeb922ca41e0a772098 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dafbc9aa8f24eaaaa7c14221298413d : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9c3bffee45b43629cbb9fd53539bc81 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce4d5c9f26b94ca298f0753c2c03770c : SUCCESS in 11m 51s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/514d495fcb5c442dae1bb646bc4e5c60 : SUCCESS in 2h 22m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/24546dc7622c474fb1eb6bbd18b0cea1 : SUCCESS in 2h 27m 07s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/defe3a70736b4fbea7661fcb73c79906 : SUCCESS in 2h 24m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd3b6927435b4c1087c972e125bf23dc : SUCCESS in 1h 50m 05s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d245a62187e84597a9ed2558e02f68f4 : SUCCESS in 25m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b1f4cbb99624efa8d8b09ee46d5796c : SUCCESS in 23m 41s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/55335b10f02849269b365bcbfc5398d3 : SUCCESS in 2h 43m 52s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/6cb25dfb4ccc45f888ed281c4cd3054a : SUCCESS in 2h 02m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8792f24c6c0c41fc9e867c571d1c75b6 : SUCCESS in 10m 53s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8be16fa0b3d44f2fa52980e0ab3fde42 : SUCCESS in 2h 28m 48s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/30970cdcc280467c914bb17fdee24169 : SUCCESS in 2h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e14bc4e863174a739dcae01d53fc2cf6fb54a879","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-09 15:38:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4a3d74285d155e46f6cf9fbb7f6f3c8213cd5245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 18:54:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdd0aedd62304591825b706d0b00bb80 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d3477074c0d4e5b8a3f7295a818e166 : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382a0f350f7944cfaba8441a0ddb1bf8 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cba74abc1ad466390dc151a2a2c7473 : SUCCESS in 9m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ebb36b537504c9aa4734a3218b04b67 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54ea439a382b46b3a6bc09c570d9218e : SUCCESS in 10m 25s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/601bd6928e09453893ca9376f750e169 : FAILURE in 27m 59s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e17b5a764bd3494abee345df46e8c487 : FAILURE in 2h 10m 37s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/fee0def9470d4bebbf481e7d61a6cd0c : FAILURE in 2h 24m 13s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77a97459262e48a3819f4a5d976d1e89 : FAILURE in 32m 36s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/30a0991246b9422bb3e0ad26eae30253 : FAILURE in 14m 26s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/be1eb646a8384a16b75837fa109d9410 : FAILURE in 13m 16s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dd5cb62e0b8f47f0bf816da62c93814d : SUCCESS in 3h 10m 11s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9cc2707d8ff14cd1bf3444f5557ba705 : FAILURE in 47m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2acbf31e3c7140149433d88c14865736 : SUCCESS in 10m 19s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9f4f2afefe8246758746e3c7e411eb99 : FAILURE in 31m 57s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8e0aecc10ca04c38a3ca03bc40400744 : FAILURE in 48m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b324af55055b76748b5423796cfebf86cf440147","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-09 21:20:02.000000000","message":"Patch Set 2:\n\nWell, that didn\u0027t work, tempest tests still throw a KeyError for router[\u0027ha\u0027] :(","accounts_in_message":[],"_revision_number":2},{"id":"356bacecb3dd7a9bab92f3f19bce970af6f89e64","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-10 11:34:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"792451d83309de462df7a7081745d1e0dcf86258","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-10 11:35:49.000000000","message":"Patch Set 2: -Code-Review\n\nHmm, a lot of failures :(","accounts_in_message":[],"_revision_number":2},{"id":"2971b12b62581aad67992a38f23e3caafce3fbd8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-10 11:57:04.000000000","message":"Patch Set 2:\n\nYes, I expected failures, was hoping to just fix two tests but it didn\u0027t even do that.  It\u0027s still worth a discussion too - should we be saying we\u0027re ha and distributed or should it just go away for OVN?","accounts_in_message":[],"_revision_number":2},{"id":"cb1dd55eb74b4c2e94c0c46f42110c8e5f9a6ad9","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-10 13:07:51.000000000","message":"Patch Set 2:\n\n\u003e Yes, I expected failures, was hoping to just fix two tests but it\n \u003e didn\u0027t even do that.  It\u0027s still worth a discussion too - should we\n \u003e be saying we\u0027re ha and distributed or should it just go away for\n \u003e OVN?\n\nYes, I think they should go away :)\nFrom an API standpoint there\u0027d be then no way to tell if N/S traffic is also distributed but this extensions will not help much with that either if we want to make N/S vs E/W differentiation.","accounts_in_message":[],"_revision_number":2},{"id":"649fc22049412fa2b985153a704c8913b737c51f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-24 19:02:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad6a43c3d78a464ba368afe2c745f8244bd367fd","revisions":{"06cae59daeac074ab476c6b95c5d2909edfd14f5":{"kind":"REWORK","_number":1,"created":"2019-09-06 17:31:31.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/680749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/680749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/1"}}},"commit":{"parents":[{"commit":"d4fea5793f55933b0a62d3d17c08af184ae6b021","subject":"Merge \"Only set \"unknown\" in LSP that makes sense\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d4fea5793f55933b0a62d3d17c08af184ae6b021"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-09-06 17:29:24.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-09-06 17:29:24.000000000","tz":-240},"subject":"Fake the L3 DVR and HA extensions out","message":"Fake the L3 DVR and HA extensions out\n\nAdd \u0027distributed\u0027 and \u0027ha\u0027 attributes to all OVN routers,\nmimicking what happens when both the DVR and HA extensions\nare loaded in base Neutron.\n\nChange-Id: If0f15418e9d75a73ca69a058785cef737ee48a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/06cae59daeac074ab476c6b95c5d2909edfd14f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/06cae59daeac074ab476c6b95c5d2909edfd14f5"}]},"branch":"refs/heads/master"},"ad6a43c3d78a464ba368afe2c745f8244bd367fd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-09 15:38:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/680749/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/680749/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/680749/2"}}},"commit":{"parents":[{"commit":"4f44eff67e30ccac503dfd28a23ff4140d8602cc","subject":"DNM Check all extensions used in ML2/OVS-DVR scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4f44eff67e30ccac503dfd28a23ff4140d8602cc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-09-06 17:29:24.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2019-09-09 15:38:22.000000000","tz":0},"subject":"Fake the L3 DVR and HA extensions out","message":"Fake the L3 DVR and HA extensions out\n\nAdd \u0027distributed\u0027 and \u0027ha\u0027 attributes to all OVN routers,\nmimicking what happens when both the DVR and HA extensions\nare loaded in base Neutron.\n\nChange-Id: If0f15418e9d75a73ca69a058785cef737ee48a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ad6a43c3d78a464ba368afe2c745f8244bd367fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ad6a43c3d78a464ba368afe2c745f8244bd367fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
