)]}'
{"id":"openstack%2Fnetworking-ovn~680792","triplet_id":"openstack%2Fnetworking-ovn~master~Ia89283c1f6569340a891b40c92405941cbda5800","project":"openstack/networking-ovn","branch":"master","hashtags":[],"change_id":"Ia89283c1f6569340a891b40c92405941cbda5800","subject":"Only notify nova of port status changes if configured","status":"MERGED","created":"2019-09-07 01:47:50.000000000","updated":"2019-09-16 23:16:27.000000000","submitted":"2019-09-16 23:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680792-1568675712568-0a491f22","meta_rev_id":"7eaac364f90b8845069de8730f344ceab9d16335","_number":680792,"virtual_id_number":680792,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-16 23:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2019-09-16 17:14:34.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-09-16 15:19:47.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":1,"date":"2019-09-16 19:32:32.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 01:52:33.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2019-09-10 11:33:50.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-09-10 11:52:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-16 15:09:29.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-09-16 15:19:47.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-09-16 17:14:34.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-09-16 23:15:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"729556975adc682ef629882288ba2b0e9653e0ca","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-07 01:47:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3c95f0cad1a55bfac9aa9de5798650e015b1fdd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-07 04:03:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"947eb340a0a84490a319b2dea938f501315feee6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-07 06:12:31.000000000","message":"Patch Set 2:\n\nI\u0027m assuming that the \"remove OVN\" patch is breaking anything using master, but the unit tests actually pass locally, so I\u0027m a little confused about that.","accounts_in_message":[],"_revision_number":2},{"id":"3877597f263788f899636daba471bff5090fed29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 06:59:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f1f36858e7e4a088b0c0de94bd0428f : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ef918e3f7f94fc49c429817b1d9c4af : FAILURE in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569475161e6447aa963c9d915d620e38 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abc9094c468a4a23a7af0911cbe0d64e : FAILURE in 11m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dedcc88f1ead4775ae7a1e045b493345 : FAILURE in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f917b8acf61a46cbbcee8fd085baf9be : FAILURE in 11m 49s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/424c808d6b0147858c44f868761ed4f8 : FAILURE in 32m 28s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0273fa69a4fa4b0dba4992bbf7d8e2cc : SUCCESS in 2h 21m 35s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3b76774e1cf9403b9cb29781f7785905 : SUCCESS in 2h 08m 51s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/39f535bcd0c04ee28a4f2a2e02d7bc89 : FAILURE in 27m 34s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b0a628796e75438f846c504f9c45af31 : FAILURE in 10m 22s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c720fbbc7c14832a54139825de5bc4c : FAILURE in 10m 49s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7cbade80649e476ebbc20453f2ef3f8e : SUCCESS in 2h 43m 24s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/b9de4f01edad421ca9d5676aa0390d21 : FAILURE in 47m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ed372151123464b8ff0bfe9ada61020 : FAILURE in 12m 11s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2fc48d5cc29849f696a7d9849ffaf1b8 : FAILURE in 32m 40s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d43ac51460654099ac51af0559b62b34 : FAILURE in 39m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f522f4a4538e6d306f2605620e83a507bacee93","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-09 13:36:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6e97df5ab4e9f3f01d202fbbd248fc46a87bbad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-09 13:48:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9113a197ec47e220923e690012228a8a0272ceae","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-09 23:14:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3962a98fa2779cf1909c39aa95f61a2180833b55","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-09 23:14:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71e72d242ca0219ad2ae4f8d6107fc7bbf8e4ee4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-09 23:15:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ac0d61e1b8b4c24867c572d928f25ebc2e2c23dd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-09 23:43:27.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"5c84a4b302489758739fda5dbf652b488747aee9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-09-10 01:52:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6410ea1c1abd41133d394a73ed7d0628d160687b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 03:04:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876273e5c2a14cc8a479964f898bdb55 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f730beaf3fbf4f34969f2d566d5b7074 : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771ffaffae38424984c96eb896a0f02b : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30934071cfa641ff9c8914bfff26b906 : SUCCESS in 9m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b66102a318044a8ca78b48398d4ff3e7 : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d778d87578d04ae2ab17ede7e5a90d67 : SUCCESS in 10m 45s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/7847efbaaee94212901c5f16574ede03 : SUCCESS in 2h 04m 35s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6bdc96224fed41d8baff09499d9c9037 : SUCCESS in 2h 43m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ed41af62789540039e6bc7b63931dea0 : SUCCESS in 2h 05m 12s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3bf12af38623409ea574b9a0e76c7f7e : FAILURE in 30m 57s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/132be8eb1b21429ab1739dd172ffd43a : SUCCESS in 20m 22s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e22555fc6b5441a2b6ccc45a070cb646 : SUCCESS in 21m 01s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ba3f989a4f3944d4b0dd3b7f11c5f59c : SUCCESS in 3h 15m 23s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/15a1d2bbe20547fc9f339ba33612740a : SUCCESS in 2h 08m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c1b6cb7e664f0daa47e6bb5f38762a : SUCCESS in 9m 21s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/656377b5ae8d4368898814559eb28908 : SUCCESS in 2h 28m 47s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b5d0e9455565452fb78f5e8031d4f768 : SUCCESS in 2h 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c3b6a4a414e91468582d982efd50e42e111b7f6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-10 11:33:50.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7493ec699d8a1cd1b75e7078663b3aaf08df4168","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-10 11:52:46.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2fe9497602aa53c69f58312b4e64e7f871909b8e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-10 11:58:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c72eccb1848e4492d567b98da84c5c8089ee8884","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-16 13:08:06.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d49a675edeaca089fbf926b62b418bec4bc4dc8b","accounts_in_message":[],"_revision_number":5},{"id":"b5d298323a00acc257812b828dce847b6e9a60eb","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-16 15:05:26.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5ab6cf777699bf7aacee33cd156a47eb75289921","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 15:06:28.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cae0fca5c7e0ca0d6e3ae4bb5ad23f60fbb7cc0e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 15:08:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"623f4fe75203934465c3c6ae3117d9800a16654e","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-16 15:09:29.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2fac99fe0547368536324ff686426c9c67190f5d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-16 15:10:36.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"5e08158ff01a316a1919899fcaabf0d14c9034ea","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-16 15:11:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"08c7fed3e35e6b734ed4d6bb8851f391947440ca","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-16 15:19:47.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"121ca7cab4c6abf2f2f484d8345f46aa9a116b3a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-16 17:14:34.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"202bf92c93abce6b83b775ffd97274cf5aacb8ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 18:08:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815c3431e5014036870ab9f31bbb7d67 : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a63cf71c590472db9c8db459964eb29 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1556478d994c378b7dbeca264e769b : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5610b5149ab492d861b661552e8faea : SUCCESS in 12m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdcb880cd3c54019ab205cd85489116e : SUCCESS in 11m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/673c7914d27444328d115452d04362dd : SUCCESS in 12m 11s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/086220c064824cc786b109c1c208c79e : SUCCESS in 2h 15m 57s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7b71995fff764d8094f28d86019b79e0 : SUCCESS in 2h 23m 26s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a28debc70d624f8eb8abed0fd1223e94 : SUCCESS in 2h 19m 38s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3a2a43ae2db14812a701a4bdabf91bbe : SUCCESS in 1h 45m 15s (non-voting)\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/9b9137c33dab4a868f604f56f686965f : SUCCESS in 22m 11s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/045de1dd3c594b7b8069aa5bb89a59b5 : SUCCESS in 19m 35s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1a3e1e990028407a812ea8c2bfc0a329 : SUCCESS in 2h 42m 51s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/0e82bcb6903841aa948759e3d6260b45 : SUCCESS in 1h 44m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38f4d2706083491e8a66c9855123a3fc : SUCCESS in 9m 30s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6c728e95ed43438ca0c413ee1868d2b7 : SUCCESS in 2h 19m 11s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8c774d3a14c74c5097a75b100a3d144b : SUCCESS in 2h 05m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0e18c4f1b8c08a643ad8bf85aa97b80704ceacf7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-16 19:32:32.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a5fd779a0e32547bf3e8b4ee6cfffc640d29df6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:32:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5c34cccf8cf3df595dd177e181351c59b330df4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 23:15:12.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d553f89c7949b38c0d320b1c95d3c5 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fb69b62f1f344239f12d1515a407b12 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76850e221244142a6ac3d133ab3ff7d : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4deb5da16f7c491db41f03ff0b33de73 : SUCCESS in 13m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60d00df38aa348f0acc24d3d5af6df1d : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f36821082c54ee9870436fa2e365b70 : SUCCESS in 14m 53s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a0046885bad040e9bf1793f596ddbbea : SUCCESS in 2h 27m 10s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ae31aada33f54a078619a928d7f8dd9f : SUCCESS in 2h 33m 11s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/796890ff82d14d3d8e15c8bd1d809a06 : SUCCESS in 19m 06s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/bdbfb8bd3fec41a88ddf23b3530a527c : SUCCESS in 25m 20s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fb96bd45eec042e6a9b422ec96c79ebb : SUCCESS in 3h 23m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54fd64795fd544bfa66ede00c91537f7 : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":6},{"id":"e3efca9b24d434d5dfac39cc5ca63918eadf23e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 23:15:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"7eaac364f90b8845069de8730f344ceab9d16335","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 23:16:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d829cf7d0e8446fac74cd3f2ecd697f : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e039fb1bcdf58d7fb26640089f8260abce7e3a71","revisions":{"bfe76dcd0d70d934224fb304c89ec87dbb282bd0":{"kind":"REWORK","_number":1,"created":"2019-09-07 01:47:50.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/680792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/92/680792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/1"}}},"commit":{"parents":[{"commit":"4d51416cff1ef8784b57c80e116927ffc90bb300","subject":"Fix OVS build issue on Fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4d51416cff1ef8784b57c80e116927ffc90bb300"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 01:39:38.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 01:47:43.000000000","tz":-300},"subject":"Only notify nova of port status changes if configured","message":"Only notify nova of port status changes if configured\n\nAlthough notify_nova_on_port_status_changes defaults to true, it\ncould be to false, making the nova_notifier attribute unsafe to\nuse without checking. In fact, networking-ovn functional tests\ninherit from the neutron base *unit* test which disables this\noption.\n\nThis patch checks both the config option and that the attribute\nexists, since the config could be changed after the plugin is\nalready initialized without the nova_notifier attribute being set.\n\nChange-Id: Ia89283c1f6569340a891b40c92405941cbda5800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/bfe76dcd0d70d934224fb304c89ec87dbb282bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/bfe76dcd0d70d934224fb304c89ec87dbb282bd0"}]},"branch":"refs/heads/master"},"2ce49d5ec2f4ba626dc5e546d83197cc30f1e5fb":{"kind":"REWORK","_number":2,"created":"2019-09-07 04:03:03.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/680792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/92/680792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/2"}}},"commit":{"parents":[{"commit":"4d51416cff1ef8784b57c80e116927ffc90bb300","subject":"Fix OVS build issue on Fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4d51416cff1ef8784b57c80e116927ffc90bb300"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 01:39:38.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 04:02:57.000000000","tz":-300},"subject":"Only notify nova of port status changes if configured","message":"Only notify nova of port status changes if configured\n\nAlthough notify_nova_on_port_status_changes defaults to true, it\ncould be to false, making the nova_notifier attribute unsafe to\nuse without checking. In fact, networking-ovn functional tests\ninherit from the neutron base *unit* test which disables this\noption.\n\nThis patch checks both the config option and that the attribute\nexists, since the config could be changed after the plugin is\nalready initialized without the nova_notifier attribute being set.\n\nChange-Id: Ia89283c1f6569340a891b40c92405941cbda5800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2ce49d5ec2f4ba626dc5e546d83197cc30f1e5fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2ce49d5ec2f4ba626dc5e546d83197cc30f1e5fb"}]},"branch":"refs/heads/master"},"553fb6e980aaed7f70e51752176600708ee0cb6a":{"kind":"REWORK","_number":3,"created":"2019-09-09 23:14:42.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/680792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/92/680792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/3"}}},"commit":{"parents":[{"commit":"4d51416cff1ef8784b57c80e116927ffc90bb300","subject":"Fix OVS build issue on Fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4d51416cff1ef8784b57c80e116927ffc90bb300"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 01:39:38.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-09 23:13:51.000000000","tz":-300},"subject":"Only notify nova of port status changes if configured","message":"Only notify nova of port status changes if configured\n\nAlthough notify_nova_on_port_status_changes defaults to true, it\ncould be to false, making the nova_notifier attribute unsafe to\nuse without checking. In fact, networking-ovn functional tests\ninherit from the neutron base *unit* test which disables this\noption.\n\nThis patch checks both the config option and that the attribute\nexists, since the config could be changed after the plugin is\nalready initialized without the nova_notifier attribute being set.\n\nChange-Id: Ia89283c1f6569340a891b40c92405941cbda5800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/553fb6e980aaed7f70e51752176600708ee0cb6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/553fb6e980aaed7f70e51752176600708ee0cb6a"}]},"branch":"refs/heads/master"},"881813505360778e01f8fd640b96f72ec4ef8132":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-09 23:15:41.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/680792/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/92/680792/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/4"}}},"commit":{"parents":[{"commit":"6d540e49e6c483990d3ae87d35992a9d67e670be","subject":"Temporarily switch master jobs to OVS 2.12 branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6d540e49e6c483990d3ae87d35992a9d67e670be"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 01:39:38.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-09 23:15:41.000000000","tz":0},"subject":"Only notify nova of port status changes if configured","message":"Only notify nova of port status changes if configured\n\nAlthough notify_nova_on_port_status_changes defaults to true, it\ncould be to false, making the nova_notifier attribute unsafe to\nuse without checking. In fact, networking-ovn functional tests\ninherit from the neutron base *unit* test which disables this\noption.\n\nThis patch checks both the config option and that the attribute\nexists, since the config could be changed after the plugin is\nalready initialized without the nova_notifier attribute being set.\n\nChange-Id: Ia89283c1f6569340a891b40c92405941cbda5800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/881813505360778e01f8fd640b96f72ec4ef8132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/881813505360778e01f8fd640b96f72ec4ef8132"}]},"branch":"refs/heads/master"},"5f47f0ed977be6a4ca5a16de1b87f55ae203f40a":{"kind":"REWORK","_number":5,"created":"2019-09-09 23:43:27.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/680792/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/92/680792/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/5"}}},"commit":{"parents":[{"commit":"6d540e49e6c483990d3ae87d35992a9d67e670be","subject":"Temporarily switch master jobs to OVS 2.12 branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6d540e49e6c483990d3ae87d35992a9d67e670be"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 01:39:38.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-09 23:43:20.000000000","tz":0},"subject":"Only notify nova of port status changes if configured","message":"Only notify nova of port status changes if configured\n\nAlthough notify_nova_on_port_status_changes defaults to true, it\ncould be to false, making the nova_notifier attribute unsafe to\nuse without checking. In fact, networking-ovn functional tests\ninherit from the neutron base *unit* test which disables this\noption.\n\nThis patch checks both the config option and that the attribute\nexists, since the config could be changed after the plugin is\nalready initialized without the nova_notifier attribute being set.\n\nChange-Id: Ia89283c1f6569340a891b40c92405941cbda5800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5f47f0ed977be6a4ca5a16de1b87f55ae203f40a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5f47f0ed977be6a4ca5a16de1b87f55ae203f40a"}]},"branch":"refs/heads/master"},"e039fb1bcdf58d7fb26640089f8260abce7e3a71":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-09-16 15:10:36.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/92/680792/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/92/680792/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/92/680792/6"}}},"commit":{"parents":[{"commit":"6d540e49e6c483990d3ae87d35992a9d67e670be","subject":"Temporarily switch master jobs to OVS 2.12 branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6d540e49e6c483990d3ae87d35992a9d67e670be"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-07 01:39:38.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-16 15:10:01.000000000","tz":0},"subject":"Only notify nova of port status changes if configured","message":"Only notify nova of port status changes if configured\n\nAlthough notify_nova_on_port_status_changes defaults to true, it\ncould be set to false, making the nova_notifier attribute unsafe to\nuse without checking. In fact, networking-ovn functional tests\ninherit from the neutron base *unit* test which disables this\noption.\n\nThis patch checks both the config option and that the attribute\nexists, since the config could be changed after the plugin is\nalready initialized without the nova_notifier attribute being set.\n\nPartial-Bug: 1843269\nChange-Id: Ia89283c1f6569340a891b40c92405941cbda5800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e039fb1bcdf58d7fb26640089f8260abce7e3a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e039fb1bcdf58d7fb26640089f8260abce7e3a71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
