)]}'
{"id":"openstack%2Fnetworking-ovn~685206","triplet_id":"openstack%2Fnetworking-ovn~master~I63acf0c809aa158aeba1c79c7c9083966ffd58a3","project":"openstack/networking-ovn","branch":"master","topic":"bug/1834637","hashtags":[],"change_id":"I63acf0c809aa158aeba1c79c7c9083966ffd58a3","subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","status":"MERGED","created":"2019-09-27 06:46:40.000000000","updated":"2019-10-02 19:34:15.000000000","submitted":"2019-10-02 19:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685206-1570044713650-6aa585c6","meta_rev_id":"0d8b2d7d0bd2239d2280859f58197bb081515699","_number":685206,"virtual_id_number":685206,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 19:31:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2019-09-30 22:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2019-09-30 18:35:56.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-10-01 06:44:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-09-30 18:35:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-30 18:35:56.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-10-01 08:55:05.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 06:46:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-27 10:17:55.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-09-27 12:54:43.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-09-30 22:15:38.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2019-10-01 06:44:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-01 08:55:05.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-10-02 19:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5af038a19d7cb0678a56ffb90cfacff3c9a4fc2f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-27 06:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2c0c11730ac548399cc36dfeb2d4187ca4e3c38","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-27 09:45:46.000000000","message":"Patch Set 1: Code-Review-1\n\nI think we\u0027ll need a related-bug and then we\u0027ll close it again?","accounts_in_message":[],"_revision_number":1},{"id":"d2d002db62559f569bcbf3f2fe1f208c4464fe82","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-27 09:59:49.000000000","message":"Patch Set 1:\n\n\u003e I think we\u0027ll need a related-bug and then we\u0027ll close it again?\n\n+1 for this -1 comment","accounts_in_message":[],"_revision_number":1},{"id":"427636d5a21c8db3930b70fca401f0d2ac921044","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-27 10:02:56.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"60e0dec36af4d4d0bf4862704d52995d3f59a975","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-27 10:06:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f6821b2db8a90c4e585c795a05e35d85f7829031","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-27 10:17:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bf6f6b1e2534103733212078743fb47ad865f7db","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-27 12:54:43.000000000","message":"Patch Set 3: Code-Review+2\n\n:-(\n\nWaiting on the gate","accounts_in_message":[],"_revision_number":3},{"id":"ae1e108f1600a2b4fb4f0c69744d0ecf7a11c942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 13:49:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4543cb9ae8e9404982bdcdcbe10ab63f : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f96cb9da2af04f5593e68db7bb5af0b6 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5583e3b8dfa14905a14111679384b25d : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f66d7f53844f12904655ff3e15e5f0 : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df48edc31c4b49ebbf796da76a33a3cd : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4061910b7824dae87c5ae4ae929f4b1 : SUCCESS in 14m 09s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f6995ed917a04be19f2cebad8df23d45 : SUCCESS in 2h 22m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d015709f8a9e46eaadcd5fe986861436 : SUCCESS in 2h 35m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/92ade8b7466843aab04d9c3e777755a1 : SUCCESS in 2h 24m 56s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/00b0a9ea11744615bcd1f983b6a0c0c8 : TIMED_OUT in 2h 15m 43s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/291ec3393eb34c8f9960e7fa3b02cb75 : SUCCESS in 20m 36s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/89a88a17c8b64ff4bb294f20b0378f12 : SUCCESS in 33m 31s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7423c00399564794b9c66f9e41b16649 : SUCCESS in 3h 36m 10s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/0fd9d019a1ce4ac1a9ddcffba04120b8 : SUCCESS in 2h 03m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c460394c9f34432bbd5351f8a0deb7c9 : SUCCESS in 12m 03s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a4a93f46fb8d402889c8a243fbeedc17 : SUCCESS in 2h 20m 56s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dd22584680e54a7bb120e375b3df591f : SUCCESS in 1h 52m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"671ae42e045162744ab7321abe0fc5a3e2f5e87d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-09-27 14:10:26.000000000","message":"Patch Set 3: Workflow+1\n\nrecheck\n\nrally timeout","accounts_in_message":[],"_revision_number":3},{"id":"ea8cf7663e0ba1365aa9b3a53fc67248364f1426","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-09-27 14:32:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"679adbd90d5902e9037f96aca4443736e81d683f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 18:06:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b66f2883b844eda5a7cd43e3869dd4 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a94191217944f8eb37039b16cfaa188 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0573636bfd44843a52564fe8b0a61e2 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d979a22b0fff4059b6b6d6a84abeebf8 : SUCCESS in 8m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e994b3e6ebcf415caee493c10fef7e92 : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a66208e58364a3da560e22691de8479 : SUCCESS in 9m 26s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/c900b8b34c9a473581cd7ef31bc97a8e : SUCCESS in 2h 13m 51s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/483215f8c0ea4ed491baa378f9bc06e9 : SUCCESS in 2h 12m 42s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/71d112b6e5db4682b1bb839694ad1418 : SUCCESS in 2h 59m 02s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef53a9786ff94418bfd1efaef73ae6fb : TIMED_OUT in 2h 13m 57s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/dcff6dc1c9c24be787785ec95f3bcd79 : SUCCESS in 26m 35s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/087f008dd0ab4318aee0d02f51e759b4 : SUCCESS in 27m 52s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b0834c9aac394dceb9e54bd238bdd7c0 : TIMED_OUT in 3h 33m 33s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/8b061cc202944924b3f9ca69b4dc70ee : SUCCESS in 1h 50m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b41048f37f94ed5a842fbd9289b7c96 : SUCCESS in 13m 26s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/50ad8cbc56a0404fa468c18827128e52 : SUCCESS in 2h 16m 12s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6c0cd21ccdb24667935a655a0dbd9908 : SUCCESS in 2h 09m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0565609d8855a61be0b947ca3cecd3bb4fa21ed8","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-09-27 19:34:10.000000000","message":"Patch Set 3:\n\n\u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e http://docs.openstack.org/infra/manual/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b66f2883b844eda5a7cd43e3869dd4\n \u003e : SUCCESS in 4m 41s\n \u003e - openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a94191217944f8eb37039b16cfaa188\n \u003e : SUCCESS in 11m 00s\n \u003e - openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0573636bfd44843a52564fe8b0a61e2\n \u003e : SUCCESS in 6m 53s\n \u003e - openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d979a22b0fff4059b6b6d6a84abeebf8\n \u003e : SUCCESS in 8m 58s\n \u003e - openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e994b3e6ebcf415caee493c10fef7e92\n \u003e : SUCCESS in 8m 51s\n \u003e - openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a66208e58364a3da560e22691de8479\n \u003e : SUCCESS in 9m 26s\n \u003e - networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/c900b8b34c9a473581cd7ef31bc97a8e\n \u003e : SUCCESS in 2h 13m 51s (non-voting)\n \u003e - networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/483215f8c0ea4ed491baa378f9bc06e9\n \u003e : SUCCESS in 2h 12m 42s\n \u003e - networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/71d112b6e5db4682b1bb839694ad1418\n \u003e : SUCCESS in 2h 59m 02s\n \u003e - networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef53a9786ff94418bfd1efaef73ae6fb\n \u003e : TIMED_OUT in 2h 13m 57s\n \u003e - networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/dcff6dc1c9c24be787785ec95f3bcd79\n \u003e : SUCCESS in 26m 35s\n \u003e - networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/087f008dd0ab4318aee0d02f51e759b4\n \u003e : SUCCESS in 27m 52s\n \u003e - networking-ovn-tripleo-ci-centos-7-containers-multinode\n \u003e https://zuul.opendev.org/t/openstack/build/b0834c9aac394dceb9e54bd238bdd7c0\n \u003e : TIMED_OUT in 3h 33m 33s\n \u003e - networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/8b061cc202944924b3f9ca69b4dc70ee\n \u003e : SUCCESS in 1h 50m 42s (non-voting)\n \u003e - openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b41048f37f94ed5a842fbd9289b7c96\n \u003e : SUCCESS in 13m 26s\n \u003e - networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/50ad8cbc56a0404fa468c18827128e52\n \u003e : SUCCESS in 2h 16m 12s (non-voting)\n \u003e - networking-ovn-tempest-full-multinode-ovs-master\n \u003e https://zuul.opendev.org/t/openstack/build/6c0cd21ccdb24667935a655a0dbd9908\n \u003e : SUCCESS in 2h 09m 25s (non-voting)\n\nhttps://youtu.be/mJZZNHekEQw?t\u003d84    :^(","accounts_in_message":[],"_revision_number":3},{"id":"cc4420827c57c8d1b8dd61dbb68d477671f961a8","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-09-27 19:45:09.000000000","message":"Patch Set 3:\n\nrecheck rally timeout","accounts_in_message":[],"_revision_number":3},{"id":"8d777ef27180fc155055f070b9d5c9bc42fc01ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:37:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc09b673a5084187a306af8d5638630c : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b6aa31eead244b79ed758ead578faa8 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f17f5e07d743ae884991721cd7620b : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0adc85e20fe4467ba7a9ec81722a06db : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4be6e70c442545e0a255758a6243d2f7 : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aedd62369d52424189fbd1cef8256a12 : SUCCESS in 9m 31s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/3c9dffd442bb4bda8c7dabce54a8e36c : SUCCESS in 2h 10m 42s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a8dea32661dc4196aebff0a685e351c4 : SUCCESS in 2h 08m 39s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/85506f5a68f1410a86f77979ada2666a : SUCCESS in 2h 10m 30s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8bd4a2139f4f47c28f9e3e6059adc3da : FAILURE in 2h 14m 12s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/22c7eeb584b44599869042078b61e238 : SUCCESS in 20m 21s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/03cdd5536841477986d3a79d084c458b : SUCCESS in 21m 53s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2438d390b53648a48a5bc30b7a4be551 : SUCCESS in 3h 47m 57s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/beb7e4d0b4a844cdbf175a7af3364fa4 : SUCCESS in 1h 54m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f715974d64d4fabb248bcd947121e34 : SUCCESS in 8m 40s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/19a4ce3d39ef4b49bce6da0e65ad7233 : SUCCESS in 2h 32m 39s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e18d799ec58b4cbc88f0d01f4f556e23 : SUCCESS in 2h 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c7c40c625bbe2fc925fb335be6966aaf094d5eff","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-28 06:27:28.000000000","message":"Patch Set 3:\n\nrecheck rally","accounts_in_message":[],"_revision_number":3},{"id":"e90832c0e80b8e01a595be388503f9de559e28e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 09:48:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf48fd56e1044a9858bf9b52752a534 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/523d78566c7145babf578c1652b76240 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c846ba5b974999996a0413166ee817 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36dd946cb7b947ef988ee25f128bb279 : SUCCESS in 9m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01dfc474d7e3465990fe6a5fbd0afdbb : SUCCESS in 11m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f38f47be10df4a898d77da54ae14c444 : SUCCESS in 12m 25s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/095e9441e9e24b81abefe93ad5a2c4c5 : SUCCESS in 2h 09m 11s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4b74cc65e54d40859073873732b263b9 : SUCCESS in 2h 15m 39s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/635cc8b4a7034cc9b024198659ca7a8a : SUCCESS in 2h 12m 21s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/03bf4dd88c5e40cbbd4840ea9f3b36cf : FAILURE in 2h 02m 34s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e88ee681a43748b5974c6ccf230fb8e7 : SUCCESS in 31m 45s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5915f5fa04c84730af10d8ac23ec509a : FAILURE in 9m 53s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1d11c61ba96444cb8e508bdb3ed539f2 : SUCCESS in 3h 15m 35s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/c6ebe6450ebf42ed85d9791ca76a98bf : FAILURE in 2h 24m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b2bfb42afb34df5adc0ac1a29eaac4d : SUCCESS in 10m 21s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c5471f8eee8b462f83d3c407c1a22650 : SUCCESS in 2h 38m 14s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dbce93e9ac1e47bc9800947fd614b873 : SUCCESS in 2h 07m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b37d951336e0ba7f20ec63b4f9acc1e7c53e8e7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-28 10:11:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"25a78bf1320ce208fee03f4dfbeca3af28fd2ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 13:30:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698532422a2b433a9dcdf7fa19dd2ff3 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/859d256771a24314bc90004f6d5d2c04 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093629ca6bd34034abf630022ef82d4b : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19ba1aa9167b4f43851d7762089c0788 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d20a82dff8d40cc9e841fb0c1bf0929 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7275626e877d4edc9c8f4c690496f448 : SUCCESS in 10m 17s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/9793caac7dce4712ad4bf3b7c8db3ec2 : SUCCESS in 2h 14m 10s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/886012321d064c8abb333efd176a6757 : SUCCESS in 2h 32m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/9143189c5c5a4e59ae6e8477fa73e496 : SUCCESS in 2h 21m 14s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6c12a65b627f4b738b6747f5494be7fd : FAILURE in 1h 53m 25s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/5c8f624f9d54485293ae272723624fb8 : SUCCESS in 21m 16s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/cfbdb0cc147d4252b304ba7d22e1fded : SUCCESS in 21m 30s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cc48a74894b140619d4b52e6d946a597 : SUCCESS in 3h 14m 18s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/afbd0974b1bf4502bf93737a0bc1608f : SUCCESS in 2h 08m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fc32528a2b44d6098a9b260f6711710 : SUCCESS in 9m 32s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1fd4a9675bf5437484e1898208e21d8e : SUCCESS in 2h 31m 29s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c0022d9365094af89993f622e081a434 : SUCCESS in 2h 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"250663c706141bac7d40161721664d21ff7f609f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-28 15:30:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9e3454a7ffd46bf0d6543ec7e6469561fe9cc21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 19:04:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33950a710eba4a04a4ee8d1578022a69 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e38782306ecc497ca667277d96c33a44 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5d1256186c42958cb1419398e85146 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e53f1dc5b87b4b35a7f87758eb24cac3 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/945d39001c97440880e7ed944982ec63 : SUCCESS in 10m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe00b6ae016c449f9854c109f887c6a9 : SUCCESS in 12m 05s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f908c199a8c341d392770edd27809595 : SUCCESS in 2h 16m 21s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f9a028e0fa324972a1da6197e80fb208 : SUCCESS in 2h 19m 03s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/fb9c1a1fb3004270b464ad9898ca8ca7 : SUCCESS in 2h 26m 43s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/30ea82fa67634f11a87063c3ea4370f9 : FAILURE in 2h 11m 34s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/58f8dac4251b4e348be9a81d4883397b : SUCCESS in 19m 58s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/44889be1fbec421b990601b61e4b47d3 : SUCCESS in 19m 23s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/22f7fa4f34b44ccc8401321d0a593c02 : SUCCESS in 3h 22m 54s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/413834af8c714e38958f16bf482a7f45 : SUCCESS in 2h 08m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c65d4171a764ecf9472d7dcd95a210b : SUCCESS in 8m 57s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d58b61358ca34680bb1ba8d5b0b00bf1 : SUCCESS in 2h 04m 00s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4088e87c8e0a4d5b804f05525371d602 : SUCCESS in 1h 57m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d87f463e55c9d31b016e21cbce3c362b85759e2d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-09-29 11:12:27.000000000","message":"Patch Set 3: -Code-Review\n\nhmm... so reverting this change causes SLA of rally to fail\nand that is expected since this commit is all about making rally faster, right? And if we do not revert, all other gates fail. Rock and a hard place? Can SLA for rally be relaxed until the real fix is put in place?","accounts_in_message":[],"_revision_number":3},{"id":"538638c49a64f752d01eaf2f20eaaf1adf0d37d0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-30 08:38:10.000000000","message":"Patch Set 3:\n\n\u003e hmm... so reverting this change causes SLA of rally to fail\n \u003e and that is expected since this commit is all about making rally\n \u003e faster, right? And if we do not revert, all other gates fail. Rock\n \u003e and a hard place? Can SLA for rally be relaxed until the real fix\n \u003e is put in place?\n\nWe can update rally config to not check trunk related jobs first. Agreed? Others voice?","accounts_in_message":[],"_revision_number":3},{"id":"d150833d743beb50ca777b4c8423d06d4f5cdd25","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-30 11:02:47.000000000","message":"Patch Set 3: Code-Review-1\n\nAgree with Maciej, let\u0027s either make it non-voting or disable trunk test there","accounts_in_message":[],"_revision_number":3},{"id":"a5c59a5041125716af58a1b34b47f25d54fcfd96","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-30 11:03:46.000000000","message":"Patch Set 3: Code-Review+1\n\nChanging to +1. I think it\u0027d best to leave this patch as is (like the revert it is) and rebase on top of another one that handles rally stuff. wdyt?","accounts_in_message":[],"_revision_number":3},{"id":"1dec289afa6a8427b6d56a76a6f5bf4963deb25c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-30 13:47:18.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"cd2de0b891870b5b5edbc49858957277f13f6ffc","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-30 13:47:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"50525d601d6d207bc40ec472fb4ce993e0087c11","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-30 13:54:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"092822f5848b1ba722872363e5e714c149089335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 17:23:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff01d60a3bc24cafbf42c1e0d0227ab8 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2fbb0046a6bc409ea273d7ad16365985 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9bc1aee92ae418b9570488e14027699 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8354002528cd46d29a74e2694ee689ef : SUCCESS in 9m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbaff0993a93481c844f33f30541d589 : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/623661f293eb4569b79cd2cc99de7232 : SUCCESS in 11m 05s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5fdd7965fece4a098add89ab2b241aff : SUCCESS in 2h 16m 18s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9729ac6e9c944f43abb50ede64f0ad54 : SUCCESS in 2h 31m 43s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8c0a00c53f6a419994274da079b7780f : FAILURE in 2h 40m 38s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/798dc3ec01654e38afb43f85a37b8a0b : SUCCESS in 1h 50m 58s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b6082daff42148a8ad306170688b0e26 : SUCCESS in 25m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d24a85d2916f4a0c833b9ed6e579091c : SUCCESS in 22m 55s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/25ca6a937a434040ba5e0493b10e7a0b : SUCCESS in 3h 07m 52s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/044b25258fc04b4abd19542931cda21b : SUCCESS in 2h 11m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e79b65525104634ba32a2ba23994a34 : SUCCESS in 10m 19s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ea39f7454f364ae293481d232bf8f220 : SUCCESS in 2h 40m 06s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/730a01e24d1a42efa33b210e55fac655 : SUCCESS in 2h 04m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9882f124ea92f0a4e892db060bf9b953d4e8ed26","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-30 18:35:56.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"24699d060abfe153e537ff845c5c469130f65dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 22:08:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda55952894e46a2944d750f967701c4 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f627783558c6489f98f2e930e70a70cb : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb78fc9256b4ce782cfa653ecdc3805 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/316b638f2a064540aa5afa74b0b4d353 : SUCCESS in 10m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30e9f7f383124dee8263501b71fac22a : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97a98198d6c842f9962520bcf603ad8d : SUCCESS in 9m 57s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/faf97f29b3034b1ba07f5d09bacd3bae : SUCCESS in 2h 15m 19s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/178ec52d412540488f28e2ef7f73d393 : SUCCESS in 2h 12m 59s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/39601d657bb0435db19b501dd655124f : SUCCESS in 2h 17m 19s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e22bb491ea846c9ad15f76d3bd0deaa : SUCCESS in 2h 05m 56s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/3b883984883c4847a130e4d8789b1244 : SUCCESS in 19m 30s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e3a4abbaa5e43c3a4333b7991c6d5e3 : SUCCESS in 20m 11s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/02363014d0b8499fb605293bd6c39e52 : SUCCESS in 3h 26m 06s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/99e130e4ab9d4714a8321b966000d2d1 : SUCCESS in 1h 58m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7a5689d944b4329b7a432afa63f2563 : SUCCESS in 10m 53s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ae96836ef1384468a91d34b6ae3bb243 : SUCCESS in 2h 21m 02s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/73cd4592e25e4b4ba2cd185e88f4737f : SUCCESS in 2h 00m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"411183a84c6422b90e60fb3f91d033369872622c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-09-30 22:15:38.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"61fec4dd2115a019718e1b1192938349e6d885ee","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-01 06:44:42.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"7efc7019e6cb5512579217cdc0e8ec527c8860fb","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-10-01 08:55:05.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cf79eb0201f77840a1589f1227f04308f36a2fab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:55:52.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"dbc1d7e5a4369046fce09a2c30a743144bf69540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:19:22.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"b4b4d256259356fe97d3fc845323b175c26d85eb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-01 12:30:12.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e26ca4c707a6c19e6046235d63021dfb823df3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:21:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a85f1d873f124137885a276732b2cdc4 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fccd417377f04c78acc73d82401275c7 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c71f9e47f9e4cc69a14f1b9908dcf9d : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b91cd9dc78cd4bd58602e1d3d92cba9c : SUCCESS in 11m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dc09b5f122b43458f3410d36e348897 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0822f8b4b1a44408daa8c4829063804 : SUCCESS in 10m 42s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/fb293cda5ad3401c8829923455055ab2 : SUCCESS in 2h 15m 47s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6e66e4dc3e1f446d8cdb613e794b0fa6 : SUCCESS in 2h 21m 42s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/1065adf2f21e40abab97fa2628fda869 : SUCCESS in 2h 51m 56s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b5dc1e52901245a8934910654d6cd24c : SUCCESS in 2h 04m 03s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a5ed6cb2e0624d8591459f273371588c : SUCCESS in 26m 16s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8221ed03e7ce482b8f3b805a31255fe3 : SUCCESS in 26m 26s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/482b4dee017c4964aac1235039ee2412 : SUCCESS in 3h 30m 47s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/c6ffb85d39c94895948ce88bdd01def5 : FAILURE in 2h 09m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db4ec03dbee84387aba139cb018b2b0c : SUCCESS in 12m 53s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d5d68bf0d1734d9991be3a59a51730ef : SUCCESS in 2h 24m 09s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/67710a26a2a646e6b4252a3e3876871a : SUCCESS in 2h 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"51f9eddc7c918a1d803ab0c7996716ce5b6f51d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 03:10:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"833bd746e17037a3125f5a134c42d1d7b5cf9a1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 06:40:17.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"d4739a0b41789c6e3ac1a772d6c011bc820d8a4c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-02 08:25:31.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"45e2d0d946e6b04756ad707efc63a6e63bb7b6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:31:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b771b5069ef246b189d230981b3f6312 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9ee55c7eb5f47948b8ab5c7c0b47dda : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b6d17215394b7b962ecf4d2948bf9c : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2debbb318634473da16320279d7a5eee : SUCCESS in 9m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05cc5ef9c70a4e658a64cbf309d48e4c : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02e16d82efcb48e881433c0163e570ae : SUCCESS in 8m 19s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/d940c82f72ce47f2af8dabb8d4b27fec : FAILURE in 2h 02m 43s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/591cf6e517b64b7da50cef5617d95a03 : SUCCESS in 2h 01m 04s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f5c56b621ad448d79e07ca00fb06805e : SUCCESS in 2h 24m 20s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5d29fa712fa4720a8926977c7a16262 : SUCCESS in 1h 56m 08s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/be9ec32b5cba40f997087ed70836c4b3 : SUCCESS in 17m 26s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a21a6026cea4e66abb640667f19f39c : SUCCESS in 18m 50s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0d1d0a84183c44dbb1a472c6f06951e5 : SUCCESS in 2h 55m 58s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/724ac6c5cbb1408d9546e7cdfd3668c7 : SUCCESS in 2h 18m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78be1d02fe6244b98d8ec9f5f2fefd3f : SUCCESS in 12m 02s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/73d908d470304d2c9513b67028d5694b : SUCCESS in 2h 33m 15s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/66346ab5c2424935a5a462f7f065b52d : SUCCESS in 2h 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bb4677f62ad509668bbc880398e47713ec813516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:36:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4e46620cb6e2a019a7991c5071cfa66e29575865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:31:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d5a58b43473049a1ffcacd8592d9160836febc08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:31:53.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/435bf217a3444491a56a29d8e6bcc2f5 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74629ba7cd7a4d9ab056f3896b1c3045 : SUCCESS in 12m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff228d5581cf4819a15934ef92ed6b71 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f3f43b710014d3fa7d2f566e18dd8be : SUCCESS in 10m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c94b229764b45499f421a215261e597 : SUCCESS in 12m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24aff1d5f48d47f5b3b92b5bcc4682e6 : SUCCESS in 13m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/939346bb794145548689550878095bd0 : SUCCESS in 2h 49m 12s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/c45a90d258004265bde2ddc32e79cf2b : SUCCESS in 2h 07m 00s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67b7828e9e5341b897c012153cdee439 : SUCCESS in 1h 45m 51s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e3c2628846804f98935c38b7fa508758 : SUCCESS in 25m 11s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2e75b0baa20464f9fc1cea139a2d389 : SUCCESS in 21m 46s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fbce95949a7d463080b04ce55246ac3a : SUCCESS in 3h 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42833ebc0bac4016a2d3ea74f315392a : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":7},{"id":"62808ed268538e9d68a392ce616c97f5fe35ac1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:34:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc3917da0644cba9b9bcb5bbb6305d3 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2e0832f7b8bfc31780b657aa0abda4e8b244fbbd","revisions":{"ceb9cefa22586021422d8282db342a197b8716a0":{"kind":"REWORK","_number":1,"created":"2019-09-27 06:46:40.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/06/685206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/06/685206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/1"}}},"commit":{"parents":[{"commit":"41f6d622abc65ca4c10b08f8c022915df2809255","subject":"Set binding profile directly from OVNTrunkDriver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/41f6d622abc65ca4c10b08f8c022915df2809255"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","message":"Revert \"Set binding profile directly from OVNTrunkDriver\"\n\nDuring recent days we many observed failures of neutron_tempest_plugin.scenario.test_trunk.TrunkTest.test_trunk_subport_lifecycle test. It started to fail on a different asserts. Lets revert this patch, as it is a regression, and work further on this issue.\n\nThis reverts commit 41f6d622abc65ca4c10b08f8c022915df2809255.\n\nChange-Id: I63acf0c809aa158aeba1c79c7c9083966ffd58a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ceb9cefa22586021422d8282db342a197b8716a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ceb9cefa22586021422d8282db342a197b8716a0"}]},"branch":"refs/heads/master"},"5e6cb5d1a1473f7b822996428ead6ea521af0537":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-27 10:02:56.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/06/685206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/06/685206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/2"}}},"commit":{"parents":[{"commit":"41f6d622abc65ca4c10b08f8c022915df2809255","subject":"Set binding profile directly from OVNTrunkDriver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/41f6d622abc65ca4c10b08f8c022915df2809255"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 10:02:52.000000000","tz":0},"subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","message":"Revert \"Set binding profile directly from OVNTrunkDriver\"\n\nDuring recent days we many observed failures of neutron_tempest_plugin.scenario.test_trunk.TrunkTest.test_trunk_subport_lifecycle test. It started to fail on a different asserts. Lets revert this patch, as it is a regression, and work further on this issue.\n\nThis reverts commit 41f6d622abc65ca4c10b08f8c022915df2809255.\n\nRelated-Bug: #1834637\nRelated-Bug: #1845479\n\nChange-Id: I63acf0c809aa158aeba1c79c7c9083966ffd58a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5e6cb5d1a1473f7b822996428ead6ea521af0537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5e6cb5d1a1473f7b822996428ead6ea521af0537"}]},"branch":"refs/heads/master"},"a6961f85e36d7ee756496baf974226af5ad02447":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-27 10:06:01.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/06/685206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/06/685206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/3"}}},"commit":{"parents":[{"commit":"41f6d622abc65ca4c10b08f8c022915df2809255","subject":"Set binding profile directly from OVNTrunkDriver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/41f6d622abc65ca4c10b08f8c022915df2809255"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 10:05:27.000000000","tz":0},"subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","message":"Revert \"Set binding profile directly from OVNTrunkDriver\"\n\nDuring recent days we many observed failures of \nneutron_tempest_plugin.scenario.test_trunk.TrunkTest.test_trunk_subport_lifecycle test. \nIt started to fail on a different asserts. \nLets revert this patch, as it is a regression, and work further on this issue.\n\nThis reverts commit 41f6d622abc65ca4c10b08f8c022915df2809255.\n\nRelated-Bug: #1834637\nRelated-Bug: #1845479\n\nChange-Id: I63acf0c809aa158aeba1c79c7c9083966ffd58a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a6961f85e36d7ee756496baf974226af5ad02447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a6961f85e36d7ee756496baf974226af5ad02447"}]},"branch":"refs/heads/master"},"6a0a4a6b045e85343cefb4f04e65e59dcbc9b9d1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-30 13:47:18.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/06/685206/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/06/685206/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/4"}}},"commit":{"parents":[{"commit":"129ff8e8a0e11293d89fb625f242bc04e725b500","subject":"Sync rally config from Neutron.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/129ff8e8a0e11293d89fb625f242bc04e725b500"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-30 13:47:18.000000000","tz":0},"subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","message":"Revert \"Set binding profile directly from OVNTrunkDriver\"\n\nDuring recent days we many observed failures of \nneutron_tempest_plugin.scenario.test_trunk.TrunkTest.test_trunk_subport_lifecycle test. \nIt started to fail on a different asserts. \nLets revert this patch, as it is a regression, and work further on this issue.\n\nThis reverts commit 41f6d622abc65ca4c10b08f8c022915df2809255.\n\nRelated-Bug: #1834637\nRelated-Bug: #1845479\n\nChange-Id: I63acf0c809aa158aeba1c79c7c9083966ffd58a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6a0a4a6b045e85343cefb4f04e65e59dcbc9b9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6a0a4a6b045e85343cefb4f04e65e59dcbc9b9d1"}]},"branch":"refs/heads/master"},"3796d059a229ce88546cb3a8dede9eace1d40120":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-09-30 13:47:51.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/06/685206/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/06/685206/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/5"}}},"commit":{"parents":[{"commit":"9758c4a99eca4801ded0754edf9047c63d3a9493","subject":"Temporary set non-voting for Trunk rally tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9758c4a99eca4801ded0754edf9047c63d3a9493"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-30 13:47:51.000000000","tz":0},"subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","message":"Revert \"Set binding profile directly from OVNTrunkDriver\"\n\nDuring recent days we many observed failures of \nneutron_tempest_plugin.scenario.test_trunk.TrunkTest.test_trunk_subport_lifecycle test. \nIt started to fail on a different asserts. \nLets revert this patch, as it is a regression, and work further on this issue.\n\nThis reverts commit 41f6d622abc65ca4c10b08f8c022915df2809255.\n\nRelated-Bug: #1834637\nRelated-Bug: #1845479\n\nChange-Id: I63acf0c809aa158aeba1c79c7c9083966ffd58a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3796d059a229ce88546cb3a8dede9eace1d40120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3796d059a229ce88546cb3a8dede9eace1d40120"}]},"branch":"refs/heads/master"},"80992e0b748e790462b63c422eb433dd05509dad":{"kind":"NO_CHANGE","_number":6,"created":"2019-09-30 13:54:04.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/06/685206/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/06/685206/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/6"}}},"commit":{"parents":[{"commit":"06c321d33203feaecdb7b9f121f223e41ed68804","subject":"Temporary allow to fail Trunk rally tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/06c321d33203feaecdb7b9f121f223e41ed68804"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-30 13:54:04.000000000","tz":0},"subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","message":"Revert \"Set binding profile directly from OVNTrunkDriver\"\n\nDuring recent days we many observed failures of \nneutron_tempest_plugin.scenario.test_trunk.TrunkTest.test_trunk_subport_lifecycle test. \nIt started to fail on a different asserts. \nLets revert this patch, as it is a regression, and work further on this issue.\n\nThis reverts commit 41f6d622abc65ca4c10b08f8c022915df2809255.\n\nRelated-Bug: #1834637\nRelated-Bug: #1845479\n\nChange-Id: I63acf0c809aa158aeba1c79c7c9083966ffd58a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/80992e0b748e790462b63c422eb433dd05509dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/80992e0b748e790462b63c422eb433dd05509dad"}]},"branch":"refs/heads/master"},"2e0832f7b8bfc31780b657aa0abda4e8b244fbbd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-09-30 18:35:56.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/06/685206/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/06/685206/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/06/685206/7"}}},"commit":{"parents":[{"commit":"733e1d672648943f5d0307e1fdb3eb4abb381b53","subject":"Temporary allow to fail Trunk rally tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/733e1d672648943f5d0307e1fdb3eb4abb381b53"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-27 06:46:40.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-30 18:35:56.000000000","tz":0},"subject":"Revert \"Set binding profile directly from OVNTrunkDriver\"","message":"Revert \"Set binding profile directly from OVNTrunkDriver\"\n\nDuring recent days we many observed failures of \nneutron_tempest_plugin.scenario.test_trunk.TrunkTest.test_trunk_subport_lifecycle test. \nIt started to fail on a different asserts. \nLets revert this patch, as it is a regression, and work further on this issue.\n\nThis reverts commit 41f6d622abc65ca4c10b08f8c022915df2809255.\n\nRelated-Bug: #1834637\nRelated-Bug: #1845479\n\nChange-Id: I63acf0c809aa158aeba1c79c7c9083966ffd58a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2e0832f7b8bfc31780b657aa0abda4e8b244fbbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2e0832f7b8bfc31780b657aa0abda4e8b244fbbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
