)]}'
{"id":"openstack%2Fnetworking-ovn~686372","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~I6d4a4b396eaa0d4be2a2c7afeccdb336020a40b3","project":"openstack/networking-ovn","branch":"stable/stein","topic":"ovn-provider-driver-stein-cherry-picks","hashtags":[],"change_id":"I6d4a4b396eaa0d4be2a2c7afeccdb336020a40b3","subject":"Add _find_ovn_lb_by_id() in OvnProviderHelper","status":"MERGED","created":"2019-10-03 10:08:37.000000000","updated":"2019-10-16 23:13:20.000000000","submitted":"2019-10-16 23:12:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686372-1571267527731-d25b5e01","meta_rev_id":"601fdf270684de8bb5433434482d339b9970fb5d","_number":686372,"virtual_id_number":686372,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2019-10-16 13:43:19.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 23:12:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2019-10-16 09:32:46.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-10-03 12:44:31.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2019-10-16 09:32:46.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 12:44:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-16 09:32:46.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-10-16 23:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5471af2db99cfd6b21555e39f84c420e85305f31","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-03 10:08:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2cfaf8272c444c8a6af6d31c7ff767919f24b46","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-03 10:08:55.000000000","message":"Topic set to ovn-provider-driver-stein-cherry-picks","accounts_in_message":[],"_revision_number":1},{"id":"6066d1b1bacafd7d3ee9d4437c73995667ecf073","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-03 12:44:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a5fa04d7f21bad6ab8205a1fa0306d3ff74248c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 13:32:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3ef13e61644a4cbde901a9110b3964 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5698cdfa4504803ac257f4d300a669a : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b45f08243eae457686076b5e9b1598bc : SUCCESS in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ec2812d467284bb3997b2877b97e2142 : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e0dfd55305e40e5ac05f7926f09587d : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af4fc67595d24f29b38d2b353748dfb5 : SUCCESS in 15m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/374311c4fb1041998f7ca8b381e28399 : SUCCESS in 9m 50s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8623cebfb2ec4f93a911d3826074001f : SUCCESS in 2h 09m 18s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/15cdae55cfb14153a43d7c2c5fa1b990 : SUCCESS in 2h 09m 31s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a80b10c8163448ed8f44db960c16fc97 : SUCCESS in 1h 37m 38s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/44db0742cb5f459094f7ebb5f07f9bfc : SUCCESS in 24m 44s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/68f6546d4bb3460fa22d83cfe7b23fc9 : SUCCESS in 20m 01s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/21e82b4d547a43c1b9137b62dd6e159d : SUCCESS in 3h 00m 29s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/1b2ebedb1bea403ba122aaa26abe53f0 : SUCCESS in 1h 53m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e9cc520cb684f29a1855f6125d12fe0 : SUCCESS in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4ad9d181e456c748f54845757111b83e08c7ea5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-10-16 09:32:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abaa52a40aff995f92eb9b119ef3bcca26302027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 09:33:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"717dfb3aceb40f2bafc3429cbd18405d0012f128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 13:39:32.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"42d5dc378463fa995dbbc3e317566733a630e215","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-16 13:43:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64381a0a867f2a43c3bb1174eaa2ea6ea637d34a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:44:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f393cee26b46e682d3fa03881d2252 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef5a522d3bf040ee8b59ab2236678d20 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da06c5559a8e429bb046a3e4a63c3635 : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7d84063bd8a945dc84996d1a0c7c43bd : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7ae0270a48a4e39805ce1a576c4d5bb : SUCCESS in 10m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43a52f3dbd0f4b90aa0a17fa55f6c597 : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bf1ae76f8834737a88101170b6bdcca : SUCCESS in 9m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/eca743566ea84a4c833ebe405f2b47eb : SUCCESS in 1h 48m 54s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/56fd8a61b98a474e9c42aee649906fd5 : SUCCESS in 2h 14m 57s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2bb09012fef44db91d64945b28d4fc7 : SUCCESS in 2h 02m 18s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/3576f08e4fd34c2b957d975e5fb1af16 : SUCCESS in 20m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/07b9b57ae3c94e809a4610dfeb51fd8d : SUCCESS in 20m 43s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7ba6c4f8e1fc44159cdddf467163cb13 : SUCCESS in 2h 49m 57s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/087cca1aad654c0881e814f493622683 : SUCCESS in 1h 46m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d883620b8425429aadcece0947033e97 : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52edef63d45099d8fe699459f7af593fc4ef9a6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:44:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c606e704cfde20564de8a1b1f50e4993050d85f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 20:00:08.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ce560a3645e40536956f26caaea2ce468da7005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 23:12:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aabdb80e3b3ec5e2a3335b39a2a59e4548a6953f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 23:12:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23130d122374db09f3751d02b0d07cd : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b8bee2eab4b43da95c10155593c77ba : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5859c91af9814420b985847a90f6a1f0 : SUCCESS in 5m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3775016445cd4e4f91cf7f614508773b : SUCCESS in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69e142eb236e49ada8561cb93229e091 : SUCCESS in 11m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ee2651fc6ec4d75aa346c727cb1071e : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d249462604f44adb61a25efa5355303 : SUCCESS in 10m 30s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e908772b28b142328453e33006c0e6e5 : SUCCESS in 2h 09m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/400abf00443c44828fabb19a6ebf7c9f : SUCCESS in 2h 12m 09s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c2dfabf3c1745b19a29c18a35cde664 : SUCCESS in 1h 57m 25s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/05c4647094494668b45eeae61565382d : SUCCESS in 17m 10s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a14b6adfcecb4162a1706c4ecebfe577 : SUCCESS in 16m 51s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3b4186bdbe154236b5e829c776c98dba : SUCCESS in 2h 58m 41s","accounts_in_message":[],"_revision_number":1},{"id":"601fdf270684de8bb5433434482d339b9970fb5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 23:13:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dcdb5f5444e41c48783c68bb42500a4 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f545f516d5da62e2f1512096bd660af6c10bc615","revisions":{"f545f516d5da62e2f1512096bd660af6c10bc615":{"kind":"REWORK","_number":1,"created":"2019-10-03 10:08:37.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/72/686372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/686372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/686372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/686372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/686372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/686372/1"}}},"commit":{"parents":[{"commit":"e5721859badac2ff3ffb83ef3e39d428771019a5","subject":"Don\u0027t allow mixing IPv4/IPv6 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e5721859badac2ff3ffb83ef3e39d428771019a5"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-09-11 17:15:37.000000000","tz":-240},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-10-03 10:08:34.000000000","tz":120},"subject":"Add _find_ovn_lb_by_id() in OvnProviderHelper","message":"Add _find_ovn_lb_by_id() in OvnProviderHelper\n\nThere were four code paths doing the same thing, created\na helper to lookup a loadbalancer given a pool id.\n\nChange-Id: I6d4a4b396eaa0d4be2a2c7afeccdb336020a40b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f545f516d5da62e2f1512096bd660af6c10bc615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f545f516d5da62e2f1512096bd660af6c10bc615"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
