)]}'
{"id":"openstack%2Fnetworking-ovn~687972","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fqueens~I6b659cbede25f271fa3b6a1c9e72019694ab6608","project":"openstack/networking-ovn","branch":"stable/queens","topic":"bug/1834637","hashtags":[],"change_id":"I6b659cbede25f271fa3b6a1c9e72019694ab6608","subject":"Set binding profile directly from OVNTrunkDriver (redo)","status":"MERGED","created":"2019-10-10 20:02:34.000000000","updated":"2019-10-18 02:34:53.000000000","submitted":"2019-10-18 02:34:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687972-1571366093159-9e252c24","meta_rev_id":"d7fe0d2ac49f1d5f7f96eaf3f4cbeac5a0dbb0fe","_number":687972,"virtual_id_number":687972,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-17 21:14:37.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 02:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2019-10-15 09:46:32.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2019-10-17 11:14:12.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-10-15 09:40:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-16 12:43:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-10-17 11:14:12.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"date":"2019-10-15 09:45:00.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 09:40:53.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-10-15 09:45:00.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-10-15 09:46:32.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-10-16 12:43:28.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-10-17 11:14:12.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-18 02:34:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5a03d3ae56c2e71f122bd9bd9e9e87d14af264f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-10 20:02:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b0d411bd4c63cfd05e4e7b962812921b858043e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-10 21:03:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29771134283de4c09e3c79f766847d68c6a78048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 23:53:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455ad2d6c93b4c4eb9fe413737889c28 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/974745eec71a46cb8a3090ac7b0fada7 : SUCCESS in 8m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/cdf6fabeea954c13b8c6167edf2e5518 : SUCCESS in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5e64c916743f4f1c99081569617d883e : SUCCESS in 10m 07s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d6c85d2882d84b949665d4f6d1d9c63b : SUCCESS in 53m 36s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1ed27acb57994d348c2a03090c5e85b9 : SUCCESS in 1h 41m 39s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c9a4c5f78ba410caada99420bdd2d48 : FAILURE in 20m 14s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/1b3b5fbe11b84165b80bbf8f6f2eb4b6 : FAILURE in 19m 46s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/7c9c7c2e734647ed8c27bd527ee4341f : FAILURE in 31m 23s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c32f825e9a014375b9df3d44347acba3 : SUCCESS in 2h 44m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6cc65d5e7484fb9b354de36696f9871 : SUCCESS in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b694d79142d416a2322fb4a1123da800eb374f7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-11 20:15:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8629e4141242d62f45cd3f24859cc679cca3855a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-11 20:54:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"08aafa6b90d577eedf1535eba5dd7f09f2d7d63f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-11 21:45:38.000000000","message":"Patch Set 2: Workflow-1\n\nNeed to look into funtional test failures... WIP","accounts_in_message":[],"_revision_number":2},{"id":"8553de20900b092b3b07ab72694f3a5ed705fc71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 00:07:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6689df27b4c484d87e26439662878d9 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1adcc94025e04dca8f2d44d4ff9f605a : SUCCESS in 10m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/327764debb9f4eb3873de90dbf8a37f9 : SUCCESS in 6m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/932e9a567c754600a4bf7c8946387212 : SUCCESS in 8m 49s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/de88a2097d6a48f0b6c3666b54eba2a8 : SUCCESS in 1h 04m 56s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7db34ee6b3ce4ebd8b1b40a5cc5c9f66 : SUCCESS in 1h 33m 11s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d06233d8fc84e81bb62b819203aa394 : FAILURE in 24m 25s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/0f0c79d5ccb545be9f26a7250f332d90 : FAILURE in 23m 15s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/ee4e16aa65b2467dbb437cbda72f04be : FAILURE in 34m 33s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b17466c1ad9f4cc782ccb631b0f4925b : SUCCESS in 3h 04m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/126b6e5209a74a61bfc2f848a16cc2a4 : SUCCESS in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b68182f5577255f7a6f27ba7d87784ad32819726","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-14 10:53:30.000000000","message":"Topic set to bug/1834637","accounts_in_message":[],"_revision_number":2},{"id":"ae409f4e2711c318b59f49e08436ce5d36dbbaac","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-14 12:12:36.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"364f831366cc2ceca92e9dcd5d19ec775848fb24","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-14 12:14:50.000000000","message":"Patch Set 3: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":3},{"id":"7e9a11737867da683e3ac71abf99fa7d41852d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 14:42:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d6c6465d3f4b27aac12135503f09c7 : SUCCESS in 8m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba049d2a144542e6bb3feec2734ba100 : SUCCESS in 10m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/adf98e8726bb4fccb43e9f63c707f288 : SUCCESS in 6m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/621793dc55344cf68116efa66e0fc747 : SUCCESS in 9m 47s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/91012ac66eaf4bc785c642156e123e02 : SUCCESS in 1h 07m 05s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b17230db2c4642acb3c1b4580abafa09 : SUCCESS in 1h 46m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/63fdee90b3ed43609a4aea01f147e7e2 : FAILURE in 20m 32s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/6346434b49ac48ab8ca41db2205cf20e : FAILURE in 18m 04s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/05270894eeda401c9d3385cb16f478cf : FAILURE in 28m 52s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8f27e6affc514113a5930c18848c39e4 : SUCCESS in 2h 27m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e83b33d5424693a913b6998bb6f94a : SUCCESS in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6cc56f1f287ee81062b5c8fdd982720e74b8a1d0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-14 15:44:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"31d8ff2428b13adc98bd1df47829d7e52a7e6de5","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-14 15:44:43.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"88831b806ff9b408503c02f0b38f5b18f4cede7e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-14 18:15:50.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"0b89161e3689cdf0a6c91dc96224a76bee9b7a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:10:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4669f73b2a4b4174bfcd8c187cdef414 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f14b6583a3b045859df47cba31ff6389 : FAILURE in 9m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fbcc1ddb63ca4caab6c32a86a086edce : SUCCESS in 5m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b653833f01d64db893c22a65dc26178f : FAILURE in 14m 02s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/814bf29349ad47a3a89c2b2bd444d782 : SUCCESS in 56m 10s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf73c56bb712447faead27213444feee : SUCCESS in 1h 40m 31s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6f56e59d37b4da2ac551f3278ebeaf9 : SUCCESS in 20m 44s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/bc2be8ade8514b3a8fd7b9300e93d221 : SUCCESS in 20m 20s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/b4310e9fbb3b44bfb7bbb4a8cc776895 : FAILURE in 36m 53s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/24f0bb1f0d2a4d1a99c1f5381f8c8f99 : SUCCESS in 3h 06m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c15704c2c2ea4265a723cd0b5e61be86 : FAILURE in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fb2576703adae7a1f91e9f952abc03d40ac319f6","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-14 19:31:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"399c27a1c835d7d14f581f4580070da72c9687fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:21:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a119b40107f24bf4a6373d0920a85b93 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9cf0a7a582a4bb09441025d2b541678 : SUCCESS in 9m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a5797f62ae014ccd95668178e044d27e : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55704a2d7a1d4c30998bf5ab79960dfc : SUCCESS in 9m 37s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/00512983bf8b464a8052d2612f456238 : SUCCESS in 1h 12m 17s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43c1f87b3dca4adaa6363493386cae94 : SUCCESS in 1h 37m 36s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7989edcc54be47abbb94aea482155bda : SUCCESS in 28m 46s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/c64e0f9447b7442285a7a68c1f48cdac : SUCCESS in 28m 08s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/b1d53633c6164e999454ee8c0f28ffe7 : FAILURE in 33m 28s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6f08daf7b8da49bbbae7fcd607f007ff : SUCCESS in 2h 42m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f4b1e80a4544b5a92a76c20f0dcab9d : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8d2884cb7dd666eeed7179623296030916477b08","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-10-15 09:40:53.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM Thanks for fixing this!","accounts_in_message":[],"_revision_number":6},{"id":"0a12101faf61e47e7f48e6f6d2219438cc31019b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-10-15 09:46:32.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d5720562a136733212a72a964b095731529d3224","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-16 12:43:28.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"ef52efa438bc04ff707f77f3e3ba3776ccab26de","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-17 11:14:12.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4d08fe34a5925873cf43004b582254139d2f2b48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 11:14:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f8cde965bb38e92f65c45dda6c51834c07a33839","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 16:16:04.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8b6ddc092b4a3ab8a62215aaa2e2de : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4641d4ea61bc48fbac9a57351d8c5f04 : POST_FAILURE in 16m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/91e79aa72af648528c7cba7b8b75799b : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5b250913cb5e4fb498fa75cf8f66a6fd : SUCCESS in 11m 21s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/04005b229c024c26a5d7bcec792c34cb : POST_FAILURE in 7m 10s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/041148a585d9400797a8f87c1f420f32 : SUCCESS in 1h 49m 25s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/30d7a54b307e4e229e3d8d0156eefd29 : FAILURE in 5m 22s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/cfba06b919d147048ceb37172d57c3bf : SUCCESS in 19m 00s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a4ecb99db3cb41d7b383939cf743ea3e : SUCCESS in 3h 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"638c0c258cafee47a85ed514c70f29dcd361a484","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-17 21:14:37.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"fa44f02d472de24b56678488af351f9b5c875e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 23:59:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4227d06d294463799f2086f80623817 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/667436a276814d96b3c8e8d22b77c5e7 : SUCCESS in 11m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/22b3cf1291ac48109beffcfd8c0477c1 : SUCCESS in 6m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3332c21f7ea2442590def094554e1c2a : SUCCESS in 10m 13s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/bb90d667178e44679166347a31c22845 : SUCCESS in 1h 05m 19s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4413226214e14dd482c3c963cefad41b : SUCCESS in 1h 34m 58s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/70e4fca181b54e5e832a87a1fe6894fb : SUCCESS in 18m 13s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/fc8353177a744760a1c12fceb18f2b2f : SUCCESS in 21m 41s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/90d868aafea9416bbbaaeb1e4812dab9 : FAILURE in 38m 55s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/30c255dfc68c4707ba6997039036aa9d : SUCCESS in 2h 39m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6be31695ca6a497abb7b1621860175eb : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"639cb3222954e48170d2a6914a549ab2b7aa97ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 23:59:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7507a59e83f632b2f41bc550de8bb5001ff8a244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 02:34:52.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c4289ba5274bb286634fbd2e1b40e5 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdddebbaace74810912b5698da9db2bc : SUCCESS in 9m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ff85db6d001543059b5ac2b9da112470 : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0bde3edf5d60496183c6aae0adc0c95c : SUCCESS in 8m 51s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f0c7d3a410ba408f999a60d499f8e22d : SUCCESS in 56m 21s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e99954a1c4c541dea99a6d39849899ef : SUCCESS in 1h 27m 17s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/fedb110320064942833abc5295fcca88 : SUCCESS in 19m 28s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/bcd08282a2334962a1a86d29309bec89 : SUCCESS in 18m 42s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/acf101350e39499f9533a890f0b28198 : SUCCESS in 2h 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d7fe0d2ac49f1d5f7f96eaf3f4cbeac5a0dbb0fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 02:34:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"01d86c3c97b196279769522532ccdef2c3bf52ec","revisions":{"523cf15b698227ad590c8395e731a99c137bd585":{"kind":"REWORK","_number":1,"created":"2019-10-10 20:02:34.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/687972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/687972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/1"}}},"commit":{"parents":[{"commit":"969ec74b2d50f3de2c6a27852ec6162340b171a0","subject":"Merge \"Fix evacuation when host dies uncleanly\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/969ec74b2d50f3de2c6a27852ec6162340b171a0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-09-05 09:29:45.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-10-10 20:02:05.000000000","tz":-240},"subject":"Set binding profile directly from OVNTrunkDriver (redo)","message":"Set binding profile directly from OVNTrunkDriver (redo)\n\nSetting binding profile for Trunk subports takes\ntime - for 125 subports rally CreateAndListTrunks\nscenario [0] takes about 150 seconds. We need to\nbump up the perfomance because large number of\nsubports is widly used in Kuryr deployments.\n\nTo achieve that I changed setting the binding\nprofile to be saved directly to the neutron DB.\nInstead calling port_update I update only related\nfields in OVN NorthBound DB rows. That gave performance\nimprovement in trunk port creation:\n\nfrom 101 sec to 35.6 for 95%ile\nfrom 99 sec to 34.2 for 50%ile\n\nThe same thing has been done for Trunk deletion.\n\nThis reverts commit 2e0832f7b8bfc31780b657aa0abda4e8b244fbbd\n\n[0] https://github.com/openstack/rally-openstack/blob/master/rally_openstack/scenarios/neutron/trunk.py#L37\n\nChange-Id: I6b659cbede25f271fa3b6a1c9e72019694ab6608\nCloses-Bug: #1834637\nRelated-Bug: #1845479\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 7bdf2eb824083b6785d78dd67b3effe071e3f7a4)\n(cherry picked from commit e14b6cd76bade8291764a88b6a2fa851a3a86b87)\n(cherry picked from commit e63c5c97ce662656c84988ff2f47616a0e36ca81)\n(cherry picked from commit c5d97dab248ca5c5cc576f70d4e819bc582bb65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/523cf15b698227ad590c8395e731a99c137bd585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/523cf15b698227ad590c8395e731a99c137bd585"}]},"branch":"refs/heads/stable/queens"},"9bf08665dd4c04eb64537ff020e6fc7374710b0b":{"kind":"REWORK","_number":2,"created":"2019-10-10 21:03:26.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/687972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/687972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/2"}}},"commit":{"parents":[{"commit":"969ec74b2d50f3de2c6a27852ec6162340b171a0","subject":"Merge \"Fix evacuation when host dies uncleanly\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/969ec74b2d50f3de2c6a27852ec6162340b171a0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-09-05 09:29:45.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-10-10 21:03:06.000000000","tz":-240},"subject":"Set binding profile directly from OVNTrunkDriver (redo)","message":"Set binding profile directly from OVNTrunkDriver (redo)\n\nSetting binding profile for Trunk subports takes\ntime - for 125 subports rally CreateAndListTrunks\nscenario [0] takes about 150 seconds. We need to\nbump up the perfomance because large number of\nsubports is widly used in Kuryr deployments.\n\nTo achieve that I changed setting the binding\nprofile to be saved directly to the neutron DB.\nInstead calling port_update I update only related\nfields in OVN NorthBound DB rows. That gave performance\nimprovement in trunk port creation:\n\nfrom 101 sec to 35.6 for 95%ile\nfrom 99 sec to 34.2 for 50%ile\n\nThe same thing has been done for Trunk deletion.\n\nThis reverts commit 2e0832f7b8bfc31780b657aa0abda4e8b244fbbd\n\n[0] https://github.com/openstack/rally-openstack/blob/master/rally_openstack/scenarios/neutron/trunk.py#L37\n\nChange-Id: I6b659cbede25f271fa3b6a1c9e72019694ab6608\nCloses-Bug: #1834637\nRelated-Bug: #1845479\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 7bdf2eb824083b6785d78dd67b3effe071e3f7a4)\n(cherry picked from commit e14b6cd76bade8291764a88b6a2fa851a3a86b87)\n(cherry picked from commit e63c5c97ce662656c84988ff2f47616a0e36ca81)\n(cherry picked from commit c5d97dab248ca5c5cc576f70d4e819bc582bb65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9bf08665dd4c04eb64537ff020e6fc7374710b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9bf08665dd4c04eb64537ff020e6fc7374710b0b"}]},"branch":"refs/heads/stable/queens"},"a62c62e0c432871669986cbef5c8227234348ca5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-14 12:12:36.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/687972/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/687972/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/3"}}},"commit":{"parents":[{"commit":"969ec74b2d50f3de2c6a27852ec6162340b171a0","subject":"Merge \"Fix evacuation when host dies uncleanly\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/969ec74b2d50f3de2c6a27852ec6162340b171a0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-09-05 09:29:45.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-10-14 12:11:33.000000000","tz":-240},"subject":"Set binding profile directly from OVNTrunkDriver (redo)","message":"Set binding profile directly from OVNTrunkDriver (redo)\n\nSetting binding profile for Trunk subports takes\ntime - for 125 subports rally CreateAndListTrunks\nscenario [0] takes about 150 seconds. We need to\nbump up the perfomance because large number of\nsubports is widly used in Kuryr deployments.\n\nTo achieve that I changed setting the binding\nprofile to be saved directly to the neutron DB.\nInstead calling port_update I update only related\nfields in OVN NorthBound DB rows. That gave performance\nimprovement in trunk port creation:\n\nfrom 101 sec to 35.6 for 95%ile\nfrom 99 sec to 34.2 for 50%ile\n\nThe same thing has been done for Trunk deletion.\n\nThis reverts commit 2e0832f7b8bfc31780b657aa0abda4e8b244fbbd\n\n[0] https://github.com/openstack/rally-openstack/blob/master/rally_openstack/scenarios/neutron/trunk.py#L37\n\nChange-Id: I6b659cbede25f271fa3b6a1c9e72019694ab6608\nCloses-Bug: #1834637\nRelated-Bug: #1845479\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 62eb828186f9248e99fa12cc9c0c8397e1520514)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a62c62e0c432871669986cbef5c8227234348ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a62c62e0c432871669986cbef5c8227234348ca5"}]},"branch":"refs/heads/stable/queens"},"5a942fe240afdb0ccfbefdc835066159085ff690":{"kind":"REWORK","_number":4,"created":"2019-10-14 15:44:01.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/687972/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/687972/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/4"}}},"commit":{"parents":[{"commit":"969ec74b2d50f3de2c6a27852ec6162340b171a0","subject":"Merge \"Fix evacuation when host dies uncleanly\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/969ec74b2d50f3de2c6a27852ec6162340b171a0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-09-05 09:29:45.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-10-14 15:41:10.000000000","tz":-240},"subject":"Set binding profile directly from OVNTrunkDriver (redo)","message":"Set binding profile directly from OVNTrunkDriver (redo)\n\nSetting binding profile for Trunk subports takes\ntime - for 125 subports rally CreateAndListTrunks\nscenario [0] takes about 150 seconds. We need to\nbump up the perfomance because large number of\nsubports is widly used in Kuryr deployments.\n\nTo achieve that I changed setting the binding\nprofile to be saved directly to the neutron DB.\nInstead calling port_update I update only related\nfields in OVN NorthBound DB rows. That gave performance\nimprovement in trunk port creation:\n\nfrom 101 sec to 35.6 for 95%ile\nfrom 99 sec to 34.2 for 50%ile\n\nThe same thing has been done for Trunk deletion.\n\nThis reverts commit 2e0832f7b8bfc31780b657aa0abda4e8b244fbbd\n\n[0] https://github.com/openstack/rally-openstack/blob/master/rally_openstack/scenarios/neutron/trunk.py#L37\n\nNote: While this is a bac-merge from Rocky, special care was\nneeded to account for the fact that in Queens release there\nis only one port binding for a db_port. Thus no iteration over\ndb_port.bindings is used.\n\nChange-Id: I6b659cbede25f271fa3b6a1c9e72019694ab6608\nCloses-Bug: #1834637\nRelated-Bug: #1845479\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 62eb828186f9248e99fa12cc9c0c8397e1520514)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5a942fe240afdb0ccfbefdc835066159085ff690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5a942fe240afdb0ccfbefdc835066159085ff690"}]},"branch":"refs/heads/stable/queens"},"440ea6ef9489272979f390adbe5ccf88e8d1694d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-10-14 15:44:43.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/687972/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/687972/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/5"}}},"commit":{"parents":[{"commit":"969ec74b2d50f3de2c6a27852ec6162340b171a0","subject":"Merge \"Fix evacuation when host dies uncleanly\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/969ec74b2d50f3de2c6a27852ec6162340b171a0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-09-05 09:29:45.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-10-14 15:44:31.000000000","tz":-240},"subject":"Set binding profile directly from OVNTrunkDriver (redo)","message":"Set binding profile directly from OVNTrunkDriver (redo)\n\nSetting binding profile for Trunk subports takes\ntime - for 125 subports rally CreateAndListTrunks\nscenario [0] takes about 150 seconds. We need to\nbump up the perfomance because large number of\nsubports is widly used in Kuryr deployments.\n\nTo achieve that I changed setting the binding\nprofile to be saved directly to the neutron DB.\nInstead calling port_update I update only related\nfields in OVN NorthBound DB rows. That gave performance\nimprovement in trunk port creation:\n\nfrom 101 sec to 35.6 for 95%ile\nfrom 99 sec to 34.2 for 50%ile\n\nThe same thing has been done for Trunk deletion.\n\nThis reverts commit 2e0832f7b8bfc31780b657aa0abda4e8b244fbbd\n\n[0] https://github.com/openstack/rally-openstack/blob/master/rally_openstack/scenarios/neutron/trunk.py#L37\n\nNote: While this is a back-merge from Rocky, special care was\nneeded to account for the fact that in Queens release there\nis only one port binding for a db_port. Thus no iteration over\ndb_port.bindings is used.\n\nChange-Id: I6b659cbede25f271fa3b6a1c9e72019694ab6608\nCloses-Bug: #1834637\nRelated-Bug: #1845479\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 62eb828186f9248e99fa12cc9c0c8397e1520514)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/440ea6ef9489272979f390adbe5ccf88e8d1694d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/440ea6ef9489272979f390adbe5ccf88e8d1694d"}]},"branch":"refs/heads/stable/queens"},"01d86c3c97b196279769522532ccdef2c3bf52ec":{"kind":"REWORK","_number":6,"created":"2019-10-14 19:31:47.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/687972/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/687972/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/687972/6"}}},"commit":{"parents":[{"commit":"969ec74b2d50f3de2c6a27852ec6162340b171a0","subject":"Merge \"Fix evacuation when host dies uncleanly\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/969ec74b2d50f3de2c6a27852ec6162340b171a0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-09-05 09:29:45.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-10-14 19:31:39.000000000","tz":-240},"subject":"Set binding profile directly from OVNTrunkDriver (redo)","message":"Set binding profile directly from OVNTrunkDriver (redo)\n\nSetting binding profile for Trunk subports takes\ntime - for 125 subports rally CreateAndListTrunks\nscenario [0] takes about 150 seconds. We need to\nbump up the perfomance because large number of\nsubports is widly used in Kuryr deployments.\n\nTo achieve that I changed setting the binding\nprofile to be saved directly to the neutron DB.\nInstead calling port_update I update only related\nfields in OVN NorthBound DB rows. That gave performance\nimprovement in trunk port creation:\n\nfrom 101 sec to 35.6 for 95%ile\nfrom 99 sec to 34.2 for 50%ile\n\nThe same thing has been done for Trunk deletion.\n\nThis reverts commit 2e0832f7b8bfc31780b657aa0abda4e8b244fbbd\n\n[0] https://github.com/openstack/rally-openstack/blob/master/rally_openstack/scenarios/neutron/trunk.py#L37\n\nNote: While this is a back-merge from Rocky, special care was\nneeded to account for the fact that in Queens release there\nis only one port binding for a db_port. Thus no iteration over\ndb_port.bindings is used.\n\nChange-Id: I6b659cbede25f271fa3b6a1c9e72019694ab6608\nCloses-Bug: #1834637\nRelated-Bug: #1845479\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 62eb828186f9248e99fa12cc9c0c8397e1520514)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/01d86c3c97b196279769522532ccdef2c3bf52ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/01d86c3c97b196279769522532ccdef2c3bf52ec"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
