)]}'
{"id":"openstack%2Fnetworking-ovn~693270","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fqueens~I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c","project":"openstack/networking-ovn","branch":"stable/queens","topic":"bug/1834637","hashtags":[],"change_id":"I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c","subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","status":"MERGED","created":"2019-11-06 23:40:38.000000000","updated":"2020-01-16 20:21:42.000000000","submitted":"2020-01-16 20:21:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693270-1579206102648-0e1c537b","meta_rev_id":"f51b128ef396f447e895aeb7fb91e52a4ee67420","_number":693270,"virtual_id_number":693270,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 20:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-01-15 15:08:30.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-01-16 15:48:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-01-16 18:01:52.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 15:46:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-01-16 18:01:52.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 20:37:58.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-01-15 15:08:30.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-16 15:46:53.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-01-16 15:48:43.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-01-16 18:01:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-16 20:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f96231b48f84f3d9475bfbde87d5dd1dc2671097","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-06 23:40:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63726019f85ed6fd31aee00053755dd3f7d41a20","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-06 23:41:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"02a00194b9c2e1cfb03023380bc771cf530782c1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-06 23:41:51.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"45cc444cad532d94ed24ccc936f1ec43a3e3b1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 01:51:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62272ac0c164a98bc8fe0e1ff16886d : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad3ce62216fe448cb6df8a09db84e40d : SUCCESS in 9m 07s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/72315832922144dc9427c02bd7858d06 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c70b213e692a4905babf3effa534e4a5 : SUCCESS in 10m 38s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c0d79d369ad4463ebebe29cdf4d079c8 : SUCCESS in 55m 34s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/24f12e7990a148178578fa39aff8b87b : SUCCESS in 1h 32m 48s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d83bef64f5f48f5af37222c169cd6dd : SUCCESS in 22m 30s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/cdb5a1b40e704dc7bb6c7eb67f04a23b : SUCCESS in 15m 11s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/a1cb84484d9a4b9bbc8f678613f09729 : FAILURE in 27m 13s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9ed143addc0f46f9992f89a1cacc8aea : SUCCESS in 2h 09m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e8fe04b242048359d956ac766f30968 : SUCCESS in 10m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d9088bdb27c354d4cff14d8fe1ee69594b3d6a4","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-22 20:27:45.000000000","message":"Topic set to bug/1834637","accounts_in_message":[],"_revision_number":2},{"id":"9f304f1ad7ed366159f4fa4dedab2d8861636818","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-22 20:37:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a66aa15e72d47d5a4448b7ae5b5732a62c9d9d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 23:02:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a949d6c4e6417db8c7c282191b60ab : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03e9b6ea20da4803b99f73baea5dcb93 : FAILURE in 4m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ac0ec3d49d754098b8d9e1aa3629f4ff : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fa0c37c6d9454051a14703777e6dcb50 : FAILURE in 4m 44s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4e7fd7158d214dd4948e7c1b23c5b028 : FAILURE in 25m 28s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a761fbe9cb314243b0996a3cbd3e301b : FAILURE in 30m 11s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff8eaf82d17a411697717966654f8f73 : FAILURE in 11m 24s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/f23a5dc3d26a433dbd8817628a0e988d : FAILURE in 11m 38s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/f90973654d774887a6504fbc5325a878 : FAILURE in 25m 11s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f0d31759fc9d48a4b0f93c06f765ae6e : FAILURE in 2h 23m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d85449974bf94d039b704021beac3a77 : POST_FAILURE in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ccc6978ba0b9f96368dda87f5d27f4ede589f4fa","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-25 10:40:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"42e02e58a09d1ca8d7b8ed56b607d1e44f4fcbf2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-25 10:59:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d86b289c304b158f574562cc354988960ebb169d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 13:20:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a705122d4147d3994c485147ab7a67 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa0d5b761a804b96ad18471e2fb5e724 : SUCCESS in 17m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/74dd3ba510204fd3b31ac2e6058a8a18 : SUCCESS in 14m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/094f761c724247c1856f180211f2961e : SUCCESS in 18m 49s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4c392bc9e888400d8bf3826cd3bd79e8 : SUCCESS in 1h 20m 43s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8debb4fd54fc48c1b35be24710cf6745 : SUCCESS in 1h 55m 25s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a27e234ad1554af7b78b9c8103b9ab9b : FAILURE in 28m 53s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/5904860a00b1471ba343f3474f1230c5 : SUCCESS in 31m 03s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/a010da7d9a0b4f18ad253c168de2707d : FAILURE in 52m 00s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9980842a7af6428585b02f02e3c69efc : FAILURE in 2h 19m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac8400e82aca412481f303e42c276c1e : SUCCESS in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"de52b604867ba633266f4dced7e38bf8e03de3c5","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-25 19:44:04.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c940da42e65d234dcbd1be9114c8fa0e607548f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 22:26:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef65ea599944d8493db42b011449f6e : SUCCESS in 11m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/733bc9cdc2184db2a027ce9bfc0d9836 : SUCCESS in 10m 51s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1ed6362c3f4c4ec89b14166f27eb42c7 : SUCCESS in 7m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b50c726e02874e128955cdc6cad075c2 : SUCCESS in 9m 52s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3fefcaa5edc7425c8482c40f3c90fa71 : SUCCESS in 58m 51s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c38d687c0b3c4967aba2e1fdbf8e7aa7 : SUCCESS in 1h 39m 13s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/567699e6ce8942ccbe72abab8b7d08a0 : SUCCESS in 21m 37s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/e7e8532aa0bc40e48a8f3b1be54a364c : SUCCESS in 19m 03s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/e8c21e8c8c034fb4b0679e4aceb50d91 : FAILURE in 34m 38s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/84f3465ee11e4c0ea91c2c155ffbf2d6 : SUCCESS in 2h 34m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1a58b37fec648b3ba57498d1401c75d : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5c7286dc06462b4c872cd31edf93a5acde68f117","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-04 11:11:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3612ee3ed883e23ddc06263a05b162f831fc5b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 13:48:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf63e1cf693647d78cd51127b8c0e2e3 : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e86b8b1c957c49e79c2308b3f95a3541 : SUCCESS in 10m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e7d6c9cb0fb8403e8104681daccd3e9a : SUCCESS in 5m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a145010adfe409db5c424dcf39fa91c : SUCCESS in 9m 44s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6531bd838fa441929080996492658db7 : SUCCESS in 1h 04m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70addcc8d9544250a88f1be1215ed2ea : FAILURE in 1h 49m 24s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/01f9e10f17f34717a9b0879eb043568d : SUCCESS in 19m 23s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/85fa793ce3d841aa84586bc761589a1f : SUCCESS in 19m 28s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/d3f5024625dd4824a1fbe37e1bfdcdbe : FAILURE in 25m 53s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/dceb036a45894bd7bfa201d095bb6d5c : SUCCESS in 2h 33m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9414693ff4d4618bf7ee945706d3874 : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ac5824428e8b3648d6820b916e8fb88dd04a2995","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-04 16:15:47.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"243485ea6253b1118d4460680ea3add4362ef0fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 18:59:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b32a8fd9dd4ec785b50690e1718264 : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa52eeafe96f485ba7efd30f1b4b8576 : SUCCESS in 10m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/de69c0e4b552490f9c8b813720b8ae83 : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/671162fc88d0461ba8ad3a82ea51ff9e : SUCCESS in 13m 20s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0df19c4e6f2346f790f1752425e543c4 : SUCCESS in 1h 07m 45s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2d70ff4ab4884f59a82cd35de5c4ad54 : SUCCESS in 1h 45m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/87eeac0060524637a8ed3e111fad0e31 : SUCCESS in 23m 47s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/cccf5ae5b011482e815108770ff294eb : SUCCESS in 20m 09s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/85e2ebb6e3ce448cb67067137eec13ec : FAILURE in 34m 19s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/337484c0cad5421fb58a960281619dd7 : SUCCESS in 2h 37m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a1cd537f4b44dd5b2527ae46a80a538 : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f61c50dd17b518b162417aa1b42415e31f7f8b61","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-04 19:52:39.000000000","message":"Patch Set 6:\n\n\u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e http://docs.openstack.org/infra/manual/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf63e1cf693647d78cd51127b8c0e2e3\n \u003e : SUCCESS in 6m 24s\n \u003e - openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e86b8b1c957c49e79c2308b3f95a3541\n \u003e : SUCCESS in 10m 45s\n \u003e - build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e7d6c9cb0fb8403e8104681daccd3e9a\n \u003e : SUCCESS in 5m 56s\n \u003e - openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a145010adfe409db5c424dcf39fa91c\n \u003e : SUCCESS in 9m 44s\n \u003e - networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6531bd838fa441929080996492658db7\n \u003e : SUCCESS in 1h 04m 47s\n \u003e - networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70addcc8d9544250a88f1be1215ed2ea\n \u003e : FAILURE in 1h 49m 24s\n \u003e - networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/01f9e10f17f34717a9b0879eb043568d\n \u003e : SUCCESS in 19m 23s\n \u003e - networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/85fa793ce3d841aa84586bc761589a1f\n \u003e : SUCCESS in 19m 28s\n \u003e - networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release\n \u003e https://zuul.opendev.org/t/openstack/build/d3f5024625dd4824a1fbe37e1bfdcdbe\n \u003e : FAILURE in 25m 53s (non-voting)\n \u003e - tripleo-ci-centos-7-scenario007-multinode-oooq-container\n \u003e https://zuul.opendev.org/t/openstack/build/dceb036a45894bd7bfa201d095bb6d5c\n \u003e : SUCCESS in 2h 33m 32s (non-voting)\n \u003e - openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9414693ff4d4618bf7ee945706d3874\n \u003e : SUCCESS in 11m 37s (non-voting)\n\nthis is why \nRally failed:  http://pastebin.test.redhat.com/819285","accounts_in_message":[],"_revision_number":6},{"id":"33f85b32eac11901bcc6959ee953abed9b8025da","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-05 10:09:15.000000000","message":"Patch Set 6:\n\nThis should be first merged in master and then here. \nI dunno what would be the destination for master, networking-ovn or neutron. For now I would stay with not merging it for a while.","accounts_in_message":[],"_revision_number":6},{"id":"905c4160e96a92f58be7156e531aa5b5448d8cd7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-09 12:28:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2bc24ed1e904edebc4a32b24d0e0ddb1f1cd549e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 15:22:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb63774099649039328c57fa3506b5b : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/852b0ff559624af896117f1e8a8b0647 : SUCCESS in 7m 54s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2d05c4e07ee847a5b23795879a6391c7 : SUCCESS in 6m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f108cafcd6874ec0aadc5b06744c8a6d : SUCCESS in 11m 16s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/96ed46d83df84190838a1a3d1f254de7 : SUCCESS in 1h 19m 45s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6289c659d1d545c9be52a1ebd14f5f7b : SUCCESS in 1h 31m 40s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/24ff381fe422483f831445c14cfcb124 : SUCCESS in 29m 01s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/1a1f79674aaf453896a0e8562c2e2c1e : FAILURE in 26m 55s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/614e46e16c17470d9099da2b8b9971b5 : FAILURE in 51m 09s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e3414d01d4554333a6c4d35a26b4bbff : SUCCESS in 2h 41m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e272b6528b7a49d08da84387573db094 : SUCCESS in 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e0c29c6b1ccd0abdf55983944caf4f639184025f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-09 15:48:34.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7956378154dd11ec5a3f57e989aefc13a2859bcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:53:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83feabfa091a496d88c669e6cccb1e4b : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1858ca688e744389cff70df06765982 : SUCCESS in 10m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6099e299a5ac42ab825b26d46419644a : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/97cdd46c7fdb42ba98ea814fc1c56bd1 : SUCCESS in 9m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/fb95323cf0874209b5cb476d6f797a4f : SUCCESS in 58m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d9024ef5e0ce4027a86b90eea775d68d : SUCCESS in 1h 36m 18s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/53d6e8e127c047b08188c9ef294a260a : SUCCESS in 17m 27s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/dd61d0efd8bb48b4921bd2ec03e7896b : SUCCESS in 16m 46s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/745299ae480b48829833fa273f53b3dd : FAILURE in 36m 42s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/01394848fea6487f93f0c1425c4e9c2d : SUCCESS in 2h 44m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f370d06f6a2411d817cfd3011daea2f : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0c77943243d5301982497fc63db9d64371279ec3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-13 11:21:17.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"ff36394482122bb40da646d3d1bb1c30e3cd4c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:59:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5121f8ca124545299eaa11c8f844f211 : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0486c08c7b044c2af2eeeb4000b8cb1 : SUCCESS in 9m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/742cd26003ca4210b7df78d2ae72078c : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/247c9aadee3f4908a02b97e2970fa80c : SUCCESS in 11m 17s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a5bd970fc2144572af8c36b9febcc38a : SUCCESS in 1h 08m 10s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d904658b3921457992a316988611be36 : SUCCESS in 1h 40m 12s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd63a6ee9abc46e3b86d571221b86290 : SUCCESS in 21m 35s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/040ffaf91f944d7d995e75a0a069f908 : SUCCESS in 20m 03s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/582957eaf48547ca95c749327c861a14 : FAILURE in 34m 29s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/14e7920aeda146548df76bd18aedc02f : SUCCESS in 2h 33m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a520fe847e1343849cfd2300a1dc3d4a : SUCCESS in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bb2c4c337873053b5184bc24a69fc4bb94d8b5ed","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-18 10:34:32.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"5ca012d49b5e77dc7c20fc8004558b8d5b145649","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-18 10:35:18.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"69c357d77f3e4f7f6ada89be4098cdc20fe8551c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:00:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d075ce5a983457ab9650f0a576ee37b : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae8f2b78c3a14f269627713cba8da02a : SUCCESS in 8m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/36218d88ce444baa8024e1186ab24b09 : SUCCESS in 5m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d2b0f921180241b8bf62ba7eb0c7da57 : SUCCESS in 8m 50s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ff8b7fac632043bd805cc79bbe9aa776 : SUCCESS in 52m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91bcdb679ce04ecda45b57fe2563fb2a : SUCCESS in 1h 34m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d9114d26f0148fb80b808e6178ae79b : SUCCESS in 15m 50s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/cc7864fd4d34471299605c4aef057e3c : SUCCESS in 17m 19s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/3abcef464a144d20851fe444ab82eab3 : FAILURE in 28m 46s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ae6fac3b7f444618a6aa61f3dfb1315d : SUCCESS in 2h 22m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/550cfb896b08440d9bc4e473c6603a69 : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e82e65681abe23f7209381f6837df81a61fe4c5e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-14 17:39:58.000000000","message":"Patch Set 10:\n\nSame comments as stein/rocky + conflict on trunk_driver.py too?","accounts_in_message":[],"_revision_number":10},{"id":"220be3230840afe3b38dab42d4a1b0178f1883ba","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-14 18:32:06.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a2d3c83647d48c14be1e814185bd508e718964ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 21:31:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fa9aab47214e86aeb41bbfeb0d00a0 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b42d9b1ba20546338954adfad9476cf8 : SUCCESS in 10m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c73892543d364ed1a29e5ac05d8de766 : SUCCESS in 6m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/214480a4543447d8a45b593aeea9c423 : SUCCESS in 12m 42s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/193487a15936456b9618cc80ae10a22f : SUCCESS in 1h 12m 00s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0056c78f292e4978ae255838f64425ab : SUCCESS in 1h 37m 19s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ffc148711d694dfdaec3d4f4edbc7526 : SUCCESS in 18m 12s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/957ee32fddb542fdac0845a1c52446d9 : SUCCESS in 22m 05s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/4a8084a0b8314249b6543c02f1e7872d : FAILURE in 33m 05s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ff84c00773454a46a3adc6f3768e14a8 : SUCCESS in 2h 55m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26643f9ff01049198a480dee3fc2447b : SUCCESS in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f2decc65b96843b81aad2605cad437e98e6f6a27","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-15 15:08:30.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"c7201514551a4b78162bb8bd0a3c02cf785a69ce","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-01-16 15:46:53.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"1346419ef2a6ddb4acc836d28ca3fddd5bfec424","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-16 15:48:43.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"abe27777dee785140927f0dc92de752f89e08520","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-16 18:01:52.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"9ec5ba2b7e7a84e4cb904aa9d6c3d5ed0233ff3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:02:05.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"fbc38e6c8ea3b7b5730eb138fa9181f6f3d53087","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:21:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"f51b128ef396f447e895aeb7fb91e52a4ee67420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:21:42.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94fde1c1e51e4e5ca43dafd7e9484b8d : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8455c553a3d440e6b8037648ad12de7a : SUCCESS in 13m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3d7656c6ca3d47fbb94c7643461f4006 : SUCCESS in 7m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f115276165b433786da276b5191af9d : SUCCESS in 10m 12s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/120263ef60ec44fa9f37e0ade9c1ff81 : SUCCESS in 1h 11m 56s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56f3f49e91c44a229020c29a4cfc9ac4 : SUCCESS in 1h 43m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8ec2d2da8d042a0ab7d46954b9c4c9b : SUCCESS in 28m 25s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/98505fd62f4245d3b9c324daac8bb2d4 : SUCCESS in 24m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f83f15052fd645f8ba32a982f9f884ed : SUCCESS in 2h 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4bed17daa762f9787391484945a445eabd617c04","revisions":{"e09b00d60211fd34bdfc242daca24e18f4df0566":{"kind":"REWORK","_number":1,"created":"2019-11-06 23:40:38.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/1"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-06 23:40:16.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e09b00d60211fd34bdfc242daca24e18f4df0566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e09b00d60211fd34bdfc242daca24e18f4df0566"}]},"branch":"refs/heads/stable/queens"},"7c0718eba4c8b0a37b2abf69ddb63bdad273a54c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-06 23:41:25.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/2"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-06 23:41:12.000000000","tz":-300},"subject":"WIP: Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"WIP: Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7c0718eba4c8b0a37b2abf69ddb63bdad273a54c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7c0718eba4c8b0a37b2abf69ddb63bdad273a54c"}]},"branch":"refs/heads/stable/queens"},"843867cb01b600cee7897e8d1daa074b43958e07":{"kind":"REWORK","_number":3,"created":"2019-11-22 20:37:15.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/3"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-22 20:36:12.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/843867cb01b600cee7897e8d1daa074b43958e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/843867cb01b600cee7897e8d1daa074b43958e07"}]},"branch":"refs/heads/stable/queens"},"fba109a2155d0213375903f20601c96f38d2aa73":{"kind":"REWORK","_number":4,"created":"2019-11-25 10:40:04.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/4"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-25 10:38:58.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fba109a2155d0213375903f20601c96f38d2aa73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fba109a2155d0213375903f20601c96f38d2aa73"}]},"branch":"refs/heads/stable/queens"},"64e3dd8eaeb302177685145c25d72e56634b1a38":{"kind":"REWORK","_number":5,"created":"2019-11-25 10:59:24.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/5"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-25 10:59:15.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/64e3dd8eaeb302177685145c25d72e56634b1a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/64e3dd8eaeb302177685145c25d72e56634b1a38"}]},"branch":"refs/heads/stable/queens"},"624334cb37a26aad1adfeb689a6378606c248203":{"kind":"REWORK","_number":6,"created":"2019-12-04 11:11:24.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/6"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-04 11:04:31.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nAnother cleanup done by this change is to fold an additional\nupdate on neutron db into ovn trunk driver. That saves\nupdate from doing another database transaction, thus making\nit faster. The no longer needed function in mech_driver was\ncalled _update_subport_host_if_needed\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to TODO_TBD for 50%ile\n  from 35.6 sec to TODO_TBD for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/624334cb37a26aad1adfeb689a6378606c248203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/624334cb37a26aad1adfeb689a6378606c248203"}]},"branch":"refs/heads/stable/queens"},"6246c3165210d1b55c3575599b7c248979e6a5e4":{"kind":"REWORK","_number":7,"created":"2019-12-09 12:28:40.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/7"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-09 12:27:59.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nAnother cleanup done by this change is to fold an additional\nupdate on neutron db into ovn trunk driver. That saves\nupdate from doing another database transaction, thus making\nit faster. The no longer needed function in mech_driver was\ncalled _update_subport_host_if_needed\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6246c3165210d1b55c3575599b7c248979e6a5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6246c3165210d1b55c3575599b7c248979e6a5e4"}]},"branch":"refs/heads/stable/queens"},"4c1a2104520879255f663efecbbe82a1aee5a3d6":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-12-13 11:21:17.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/8"}}},"commit":{"parents":[{"commit":"f5d20263a54c5fe3d8f68affe0518eec1bd83be0","subject":"Merge \"Avoid port group creation race\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5d20263a54c5fe3d8f68affe0518eec1bd83be0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 11:20:01.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nAnother cleanup done by this change is to fold an additional\nupdate on neutron db into ovn trunk driver. That saves\nupdate from doing another database transaction, thus making\nit faster. The no longer needed function in mech_driver was\ncalled _update_subport_host_if_needed\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 82efc6197a469e963d8a8105efbae0ff5a4dc82d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4c1a2104520879255f663efecbbe82a1aee5a3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4c1a2104520879255f663efecbbe82a1aee5a3d6"}]},"branch":"refs/heads/stable/queens"},"17946a43f141df4ad05556badf1b5ffae269ac9a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-12-18 10:34:32.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/9"}}},"commit":{"parents":[{"commit":"dc9da4e8d8269365a7af28aeebeaa7579382a132","subject":"Merge \"Metadata agent: fetch ovn-bridge from OVSDB and not from config\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dc9da4e8d8269365a7af28aeebeaa7579382a132"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-18 10:34:32.000000000","tz":0},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nAnother cleanup done by this change is to fold an additional\nupdate on neutron db into ovn trunk driver. That saves\nupdate from doing another database transaction, thus making\nit faster. The no longer needed function in mech_driver was\ncalled _update_subport_host_if_needed\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit 82efc6197a469e963d8a8105efbae0ff5a4dc82d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/17946a43f141df4ad05556badf1b5ffae269ac9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/17946a43f141df4ad05556badf1b5ffae269ac9a"}]},"branch":"refs/heads/stable/queens"},"24794226c1696b995fd23a96c7d96c69f3f4c16e":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-12-18 10:35:18.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/10"}}},"commit":{"parents":[{"commit":"dc9da4e8d8269365a7af28aeebeaa7579382a132","subject":"Merge \"Metadata agent: fetch ovn-bridge from OVSDB and not from config\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dc9da4e8d8269365a7af28aeebeaa7579382a132"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-18 10:34:56.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nAnother cleanup done by this change is to fold an additional\nupdate on neutron db into ovn trunk driver. That saves\nupdate from doing another database transaction, thus making\nit faster. The no longer needed function in mech_driver was\ncalled _update_subport_host_if_needed\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit c418dd720b9be9047d779f32407c474e90cb0002)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/24794226c1696b995fd23a96c7d96c69f3f4c16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/24794226c1696b995fd23a96c7d96c69f3f4c16e"}]},"branch":"refs/heads/stable/queens"},"4bed17daa762f9787391484945a445eabd617c04":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2020-01-14 18:32:06.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/693270/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/70/693270/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/70/693270/11"}}},"commit":{"parents":[{"commit":"63ff1f433ab9fbd56fe82e6ef48da714a3aeef0c","subject":"Only set \"unknown\" in LSP that makes sense","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/63ff1f433ab9fbd56fe82e6ef48da714a3aeef0c"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-01 20:33:12.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-14 18:31:37.000000000","tz":-300},"subject":"Set binding profile directly from OVNTrunkDriver (redo cont.)","message":"Set binding profile directly from OVNTrunkDriver (redo cont.)\n\nThis is a tweak to changes done to fix bug 1834637. Specifically,\nthis change addresses scaling. The previous gerrit change had\nmodifications to all OVN sub-ports performed as a single\ntransaction. That did not account for race-condition on neutron\nDB queries, which leads to timeouts under heavy loads.\n\nAnother cleanup done by this change is to fold an additional\nupdate on neutron db into ovn trunk driver. That saves\nupdate from doing another database transaction, thus making\nit faster. The no longer needed function in mech_driver was\ncalled _update_subport_host_if_needed\n\nBy breaking the iteration into multiple transactions, the\nchange in time is marginal:\n\n  Service-level agreement\n  NeutronTrunks :: neutron.create_trunk\n\n  from 34.2 sec to 35.6 for 50%ile\n  from 35.6 sec to 36.1 for 95%ile\n\nThis patch doesn\u0027t go to master networking-ovn. It\nhas been  migrated to master Neutron [1].\n\n[1]: https://review.opendev.org/#/c/701646/\n\nDepends-On: https://review.opendev.org/#/c/695693/\nChange-Id: I7de3ac2a7cf8869ead8ab5fbb34a9861a96d3a0c\nCloses-Bug: #1834637\nCo-authored-by: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n(cherry picked from commit c418dd720b9be9047d779f32407c474e90cb0002)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4bed17daa762f9787391484945a445eabd617c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4bed17daa762f9787391484945a445eabd617c04"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
