)]}'
{"id":"openstack%2Fnetworking-ovn~694085","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~I50ae263d49bb512a33c0202107d97970bcbaf6ae","project":"openstack/networking-ovn","branch":"stable/stein","hashtags":[],"change_id":"I50ae263d49bb512a33c0202107d97970bcbaf6ae","subject":"Enable InternalDNS related tests","status":"ABANDONED","created":"2019-11-13 14:28:23.000000000","updated":"2019-12-10 08:20:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4311442076a7721a87404d265bb3419168a5cc7a","_number":694085,"virtual_id_number":694085,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-13 17:52:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"value":-1,"date":"2019-11-14 17:18:33.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 17:52:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52e7fd32ec1a0c917318b041c66893a930d7e25d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-13 14:28:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13a7c06fae387c6e23f7bcaf333b6a76930c8ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 17:52:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91b6f1d806af4c1e99e5c663c23d167b : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/164098dd4dc847668a72a21f20eb324f : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6f5162024743ce8e659e65ea22a0d5 : SUCCESS in 6m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fc96fb96af14493cafdd9db99479dda3 : SUCCESS in 13m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/031229f16c204d79931bfcc0850d7512 : SUCCESS in 9m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4403922852749cd829ef912315e1a7e : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/320d6bd127ae4689b56623cc3506524d : SUCCESS in 12m 35s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3ad6f4cd9f9c4de68df9b23b7d95b07d : TIMED_OUT in 3h 06m 14s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/2833f0e76e4b4166a81220bda5d86e52 : TIMED_OUT in 3h 08m 50s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f790bf05ecd43ef98035a84b85a7961 : SUCCESS in 1h 34m 02s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/cd2d249796ee45929f68a87240b82d0e : SUCCESS in 21m 44s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ffc91a7a69741849cd9ae27abd725b8 : SUCCESS in 22m 45s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5b2bd30dec024b69855f939d62c5f90e : SUCCESS in 2h 46m 50s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/f59563d492a141978453228ff575f031 : SUCCESS in 1h 38m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec4065c8bc09429faa10ff3d7386e9ad : SUCCESS in 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f82d3fc3170d676d2c42e9b19dc6ccf64463ccba","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-14 17:18:33.000000000","message":"Patch Set 1: Workflow-1\n\nWe don\u0027t have branch-2.12 in the stable/stein, that means we don\u0027t have required dhcp parameters. So this code will not work.\n\nWe have two options:\n1) abandon this\n2) bump-up ovn to 2.12","accounts_in_message":[],"_revision_number":1},{"id":"4311442076a7721a87404d265bb3419168a5cc7a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-10 08:20:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b69f91674a2a48474d4be71fb6dcac06398df835","revisions":{"b69f91674a2a48474d4be71fb6dcac06398df835":{"kind":"REWORK","_number":1,"created":"2019-11-13 14:28:23.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/85/694085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/85/694085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/694085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/694085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/694085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/85/694085/1"}}},"commit":{"parents":[{"commit":"88427c8622043648c6186a5682c51e7bb6ed4878","subject":"Change passed domain_name parameter value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/88427c8622043648c6186a5682c51e7bb6ed4878"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-25 07:51:01.000000000","tz":120},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-13 14:29:08.000000000","tz":60},"subject":"Enable InternalDNS related tests","message":"Enable InternalDNS related tests\n\nPreviously this test needed OVN 2.12 release.\nAs it is shipped lets enable the test.\n\nThe test test_internal_dns.InternalDNSTest is configured\nto use hardcoded domain_name value [0]. So in order\nto have it passing we need to change our gates configuration\naccordingly.\n\nIn addition we need to configure parameter ovn.dns_servers for\ngate, because otherwise dns_server: 127.0.0.53 will be passed\nfor VM while using Ubuntu 18.04 on gates [1].\n\nOPTION:  60 ( 12) Vendor class identifier   udhcp 1.23.2\n---------------------------------------------------------------------------\n\n  TIME: 2019-09-30 10:14:12.519\n    IP: 10.1.0.1 (fa:16:3e:23:d4:4f) \u003e 10.1.0.3 (fa:16:3e:82:52:f4)\n    OP: 2 (BOOTPREPLY)\n HTYPE: 1 (Ethernet)\n  HLEN: 6\n  HOPS: 0\n   XID: c9c04b3a\n  SECS: 0\n FLAGS: 0\nCIADDR: 0.0.0.0\nYIADDR: 10.1.0.3\nSIADDR: 0.0.0.0\nGIADDR: 0.0.0.0\nCHADDR: fa:16:3e:82:52:f4:00:00:00:00:00:00:00:00:00:00\n SNAME: .\n FNAME: .\nOPTION:  53 (  1) DHCP message type         5 (DHCPACK)\nOPTION: 121 ( 14) Classless Static Route    20a9fea9fe0a0100  .......\n                                            02000a010001     ......\nOPTION:   6 (  4) DNS server                127.0.0.53\nOPTION:  15 ( 19) Domainname                openstackgate.local\nOPTION:  51 (  4) IP address leasetime      43200 (12h)\nOPTION:  26 (  2) Interface MTU             1342\nOPTION:   1 (  4) Subnet mask               255.255.255.240\nOPTION:   3 (  4) Routers                   10.1.0.1\nOPTION:  54 (  4) Server identifier         10.1.0.1\nOPTION:   0 (  0) pad\nOPTION:   0 (  0) pad\nOPTION:   0 (  0) pad\nOPTION:   0 (255) pad\n\n[0] https://github.com/openstack/neutron-tempest-plugin/blob/af04e1db7d6e5d7a2170a16a168f29bdbea42fe5/neutron_tempest_plugin/scenario/test_internal_dns.py#L75\n[1] https://github.com/openstack/networking-ovn/blob/0337b2e582fa96c8937794bc5a72ec744e979ad1/networking_ovn/common/ovn_client.py#L1612\nChange-Id: I50ae263d49bb512a33c0202107d97970bcbaf6ae\n(cherry picked from commit fd50c88347e738ad4b2476b36073dd1ec372682a)\n(cherry picked from commit 714e8ca35ffda71d0b8f20b1043a5d59df8f2426)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b69f91674a2a48474d4be71fb6dcac06398df835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b69f91674a2a48474d4be71fb6dcac06398df835"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
