)]}'
{"id":"openstack%2Fnetworking-ovn~695885","triplet_id":"openstack%2Fnetworking-ovn~master~I604e2f515ace0ade2888fe700eb4264e72ba9eb9","project":"openstack/networking-ovn","branch":"master","topic":"bug/1853829","hashtags":[],"change_id":"I604e2f515ace0ade2888fe700eb4264e72ba9eb9","subject":"snat can not be disabled while router update","status":"ABANDONED","created":"2019-11-25 11:26:43.000000000","updated":"2019-12-03 06:26:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"89c498c4d7926fe43785c450f2a899ec292b495d","_number":695885,"virtual_id_number":695885,"owner":{"_account_id":27079,"name":"Frank Wang","email":"wangpeihuixyz@126.com","username":"FrankWang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-25 14:20:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 14:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f523f213ab76a8b146dd9fb7dfe9638bb4121293","author":{"_account_id":27079,"name":"Frank Wang","email":"wangpeihuixyz@126.com","username":"FrankWang"},"date":"2019-11-25 11:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2ce8a5395174671d92211619e788c6a447f41ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:20:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/441a532dd6ed4a0ba7790b83e48c16ff : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa13e53618e4d319d70f037c266c2ed : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec27365c45b445528738953d83e8b848 : SUCCESS in 9m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/069e8149912a4ef6ba5c57e0be11f0b0 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f8a3580ba714cdf816772f414b92624 : SUCCESS in 10m 13s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5422b6b0142f40c7a3409e2a1ced9157 : SUCCESS in 2h 18m 45s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9b1ea3c2637a4b388a429363622ba53f : SUCCESS in 2h 25m 13s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f73402d127742588338433bd26c0e77 : SUCCESS in 2h 09m 00s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e45a5c79b6248a0b71b9e9e3e0011be : FAILURE in 21m 32s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/39905ac4ba7a4556a8d5d3ad607c787a : SUCCESS in 2h 52m 50s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/f264f6243eff465a994fa1bc36405557 : FAILURE in 57m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3239af39c4743159ed1d6a4ee9afbb5 : SUCCESS in 11m 54s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b674cbe7dc954cc18ec8518ca2ede73c : FAILURE in 2h 29m 30s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c0ded45039ef4446897833a06a056547 : FAILURE in 2h 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d71ac26022195d0ca7ff1a081a1fe4853be14315","author":{"_account_id":27079,"name":"Frank Wang","email":"wangpeihuixyz@126.com","username":"FrankWang"},"date":"2019-12-03 06:26:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30982aab66ef90994946d0918fb31dee87aa6635","revisions":{"30982aab66ef90994946d0918fb31dee87aa6635":{"kind":"REWORK","_number":1,"created":"2019-11-25 11:26:43.000000000","uploader":{"_account_id":27079,"name":"Frank Wang","email":"wangpeihuixyz@126.com","username":"FrankWang"},"ref":"refs/changes/85/695885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/85/695885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/695885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/695885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/695885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/85/695885/1"}}},"commit":{"parents":[{"commit":"635cd5e10f72caff3c3774883ab50931ffb1afb6","subject":"Merge \"Add support for virtual port type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/635cd5e10f72caff3c3774883ab50931ffb1afb6"}]}],"author":{"name":"Frank Wang","email":"wangpeihuixyz@126.com","date":"2019-11-25 11:19:20.000000000","tz":480},"committer":{"name":"Frank Wang","email":"wangpeihuixyz@126.com","date":"2019-11-25 11:25:04.000000000","tz":480},"subject":"snat can not be disabled while router update","message":"snat can not be disabled while router update\n\nrouter snat can not be disabled while invoking router rest API to update\nsnat flag from true to false, Ovn is still forwarding external traffic\n\nChange-Id: I604e2f515ace0ade2888fe700eb4264e72ba9eb9\nCloses-Bug: #1853829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/30982aab66ef90994946d0918fb31dee87aa6635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/30982aab66ef90994946d0918fb31dee87aa6635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
