)]}'
{"id":"openstack%2Fnetworking-ovn~696495","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Ic4c8e81b0d1ab7491bc317c02464bd6853c109fa","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1853810-stable/train","hashtags":[],"change_id":"Ic4c8e81b0d1ab7491bc317c02464bd6853c109fa","subject":"Fallback to default network_driver value if octavia not present","status":"MERGED","created":"2019-11-28 10:56:39.000000000","updated":"2019-12-05 06:30:44.000000000","submitted":"2019-12-05 06:28:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696495-1575527339811-82587bb4","meta_rev_id":"37b8f62840b83c4b481c91ff8eadc2ea52e20b13","_number":696495,"virtual_id_number":696495,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2019-12-04 07:54:20.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-05 06:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2019-12-03 08:55:15.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-12-04 03:46:02.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-12-04 03:46:02.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 08:55:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-12-04 03:46:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-05 06:28:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"872a2d96829f84dff119e0e6e0d34e267bf14e32","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-28 10:56:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4e5383d191a7998873edeee62bf4a7782c29555a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 13:50:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a16e6a4d4ac45e7b3d895d7f23ad4d6 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41d2a8c43fb44264b568c74016877329 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de649b95cfe42cda8c2e3f4d9d10819 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2316510925a54cf493fc0f6c0e2169c8 : SUCCESS in 11m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c65311277ae443fcaa7127cb0d56c8c8 : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9b5f38f841e42b19c85b5036dda8c78 : SUCCESS in 9m 55s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/6775d4144a7d491fb69f382b1ecee6df : SUCCESS in 2h 01m 45s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/051d195c7b9d44d3a5969045aa034066 : SUCCESS in 2h 48m 49s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/1d2b102758474be79675738a3aa9953c : SUCCESS in 2h 04m 27s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ad59ef9df6e4de993adb62e40e4e85b : SUCCESS in 2h 05m 54s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/46299905b6cf43c5a049f232e6d95a79 : SUCCESS in 17m 38s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f3d6ee929c64721b8a4d4a7e64e7acb : SUCCESS in 19m 58s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7a60a81c3a7444f3844ffe6b3bcb3fe6 : SUCCESS in 2h 41m 42s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/0cd337cf371d4bf7a9421c3cf038a82f : FAILURE in 1h 47m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeb269191d164e518e8ebca9280b7bee : SUCCESS in 9m 18s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c4047702582541bd822d8da94777d020 : FAILURE in 2h 47m 54s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b9d4775b175d4272a3c74fa0c74f102b : SUCCESS in 2h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80b9891092c37fe920596dec391805fae6f3a9c0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-03 08:55:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"022f03e05fffd5b86adb579608d8fda684dc00a2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-04 03:46:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9e7a053a41f0b06fdfdf814c091eade462e26a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:46:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f56c649a1f8905883c31553c21f27c209ca2ce8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 06:47:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78bc0c04d6e343a39c4ba09af05c0ecf : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41ffcbc6d16c4818a627bea1b32db606 : SUCCESS in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f793b36ff3b445bd862100a5c49a8d75 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfe8a7321de84cf7b82e51a908cfb5ec : SUCCESS in 12m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99ca885325d644a3bae37a50f806f49f : SUCCESS in 12m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e7c7aaf07c2420db6253ff606a078f2 : SUCCESS in 11m 08s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ce25fffa768941e2a895003bd304eccf : SUCCESS in 2h 33m 58s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/748b068ed97f4326a734961f9fbdd6ae : FAILURE in 2h 53m 35s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77602b1bf12643edb7148efd8b538424 : TIMED_OUT in 2h 15m 02s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/0bf7a401d3c944da992c9ad88bba0dc0 : SUCCESS in 23m 14s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/af2c925a9fd945efaa3138cd6685763c : SUCCESS in 21m 24s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0552819065834247add1129f2c848494 : SUCCESS in 2h 55m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5914f2b7ac843de897f04d19359bbc5 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7ce59f7a0f545611fb4116a47c843155e562406d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-04 07:54:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd388501a59fed335d9b9d6c628d3645fb540138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 10:53:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef162a43c594ecd94fc2993945b091d : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e74a876b9c544a009514f44f2ace12e8 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f56fc2116f4aa395d236e7e710ba30 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4065bee1e2c24a59aedde7d7e66bfbf0 : SUCCESS in 9m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a25c2d6eb10344a7a39176fddb23f255 : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95b9736e51a14eac83321dc374a4e0ac : SUCCESS in 13m 13s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/3555e69b644a4a1c9882ba91829b5bda : SUCCESS in 2h 34m 32s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ef6d4f169edb4090b053da997ae379cd : SUCCESS in 2h 36m 20s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f1261f346fa4408dbedef6a5a99166d3 : SUCCESS in 2h 51m 19s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7daf4dcc826a4d67ae5dc5e54738b197 : TIMED_OUT in 2h 14m 16s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/80555a50933e49a49b042f0484f7f48b : SUCCESS in 24m 56s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/164d9c85185f447bb2b4b503bf47a31c : SUCCESS in 26m 06s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5d482466c89b4f80b38d296c49d0ad6c : SUCCESS in 2h 21m 31s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/a1e9f50dfd1e4213911a080d09f8b45d : FAILURE in 2h 23m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8b42299e7bd4e3ba00400d5ad486584 : SUCCESS in 9m 55s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f1f664f8359046a6a2b1447822dcb0a6 : FAILURE in 2h 43m 50s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cd530c37eda5469085298212ecff5f64 : SUCCESS in 1h 57m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"674bdcdd5bddd0bc83b4fd4d135eb7e30fa5acdb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-04 12:30:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f98b87c5cdf7fcae0328932245d44e54ca06fdf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:41:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb372f1c5214ee5bde98907325e0677 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a062814bc7fa4bccaefdd44557a4af05 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ceb9b2f6664c89860380e1834c7168 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbc2e9d4cc784ca383a781a311df23cc : SUCCESS in 11m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb5f2baa27d149998f8904b16cb0e47c : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/727495decf554c0e8973122f0b730462 : SUCCESS in 14m 01s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/6f79193ee74a4f6d8b8f13b4c567b8a9 : SUCCESS in 2h 35m 58s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c7f644dfdb4a44439c62c326eb02d80a : SUCCESS in 2h 05m 04s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/87768bba5b954f90a5a6f7ece02c095b : SUCCESS in 2h 49m 35s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/75e6de96e02643e98a07c9cc85cd32aa : SUCCESS in 2h 05m 45s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/0f36db0f971347cc88f2e7f20ebf361c : SUCCESS in 21m 08s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/49e9b722e67c46b5aaf7e3360334a80f : SUCCESS in 21m 24s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d1509d4e66ba49bca62047d10b575118 : SUCCESS in 2h 47m 49s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/83ad1f9cc5d045578d8d7388d72b5f87 : FAILURE in 2h 22m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf9b5b624f564f9ab79d1fc63fddb5c5 : SUCCESS in 10m 39s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8972c4c40058456d994cf300df0037eb : FAILURE in 3h 04m 59s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ca3cbce5d0fc44b7b8f561ac7d8081cf : SUCCESS in 2h 07m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42e1f1bf6211e26cb4dde81ca1196fcfb3e3a527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:42:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3378d950d2be40395618aae0f05a0d19487053b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 18:36:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353ba9a7dc394c5b8a54b5579648a632 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49d0e4df85bb41eeb1ae34384d6aaf65 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2130b82d28b9479a92df357dc2c3560e : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fbdc663c5b342ffbd9ba469078542b2 : SUCCESS in 11m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1ae993ea5db446dab55e1ce28e1829a : SUCCESS in 11m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d62d99bcb3ef4265b8626b8e9ef36dfd : SUCCESS in 11m 33s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4a8c867702674ee88eb12cf8457d9708 : SUCCESS in 2h 34m 24s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/83047de40414429d866c3b1b0f5313bf : SUCCESS in 2h 48m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce8530d76752407b9661fb1024ecf99c : TIMED_OUT in 2h 18m 10s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/39b06b6a614a402985d22b2472a6408f : SUCCESS in 30m 25s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f9c40adf6c849d18213eb2fe3cd78ec : SUCCESS in 18m 22s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e6061ac8b05f487b81981b02850bbe9e : SUCCESS in 2h 40m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74e219e8d5e443c5a18249b2697d0313 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"048a347e42dacd8d563fe5a69f75705f44baf259","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-05 00:51:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d2a6297e1c323977610e838d9f392f2295dd8765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 04:00:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ec6afef00a4ada9f2b7c286b38691f : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b5e345437d5485bb3e010cf1036dd16 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe40cc84d314ddf8f053979899f0fc6 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef73e95bead84488b36d45c8f4fa53ac : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b9d58b990fd47bd83724d536769595d : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/062cebebd0cd4a38afb0789d8113e6b7 : SUCCESS in 10m 41s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/1b5f3cbb5117491d96695e39b4a811ac : SUCCESS in 2h 29m 39s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/526c3178143442ab919b3dea1211a835 : SUCCESS in 2h 40m 19s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/5cebc4ed7d0b466c8d40e92b8cdc4362 : SUCCESS in 2h 27m 48s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3a3c687057a941daa7e2b76f255d40ca : SUCCESS in 1h 57m 12s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/4137e311dd804e3f9bea23406045df3d : SUCCESS in 21m 34s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5275ff14f4854227adf0baf6b38e6ca8 : SUCCESS in 23m 16s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b979fc29cd79475e99163e3478a1c23e : SUCCESS in 3h 02m 19s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/519b21dac428425d94af502c62955dd7 : FAILURE in 1h 58m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b3053613e44eeeb05aca4a1c0bf8b3 : SUCCESS in 10m 05s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3e8bb001b8684656bda3e82620c7e6e1 : FAILURE in 3h 06m 23s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cb28267a098c4503bd80220ca67a4de0 : SUCCESS in 1h 57m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a18b0462ec733d3f57bdf659954e64535e329035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 04:00:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"095925cb201c03257932813e82c51526c20f94a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 06:28:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"67d15a5a88370d588c62ce404be2885443749fa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 06:28:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9db9f8ac2245b1a1bf38b8a159d92c : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/726dfe784944498486da0ff1ca900306 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720e0de7047a4d7c8edda87736171197 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/804895d2f458472c878bade42a802394 : SUCCESS in 9m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b84bdc3179d74632a1c7494e4c6d4ab1 : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e644cc6dc5cb4d1a8d1dac5e638c040c : SUCCESS in 10m 43s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/93f6432502b0428486d24a97fe30e824 : SUCCESS in 2h 22m 44s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ded1424434574addb50010c1f87aa07b : SUCCESS in 2h 22m 16s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cffe1942eaf44c7ca2f16cd2acece712 : SUCCESS in 1h 40m 16s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f1e257e8c0274470af2b40277c023202 : SUCCESS in 25m 28s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/845be28193ac4ad9b10d545f2e97a989 : SUCCESS in 17m 22s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fdd6f276e0bc4c34825a58470ef67f1a : SUCCESS in 2h 23m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63b34f8f060d442a8e20f298bb83748d : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"18e2cd877a204d7e8a7ecb0353e460aed0cef462","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 06:30:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9790a6b89ade497d9ebd42824031c807 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cd28b83e27704931dfe11a572cb0e001e59a734","revisions":{"8cd28b83e27704931dfe11a572cb0e001e59a734":{"kind":"REWORK","_number":1,"created":"2019-11-28 10:56:39.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/95/696495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/95/696495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/95/696495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/95/696495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/95/696495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/95/696495/1"}}},"commit":{"parents":[{"commit":"0351f5f83614505c7e0a37dafbd533111b9ab30d","subject":"Remove escape characters from container command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0351f5f83614505c7e0a37dafbd533111b9ab30d"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-25 09:09:30.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-28 10:56:39.000000000","tz":0},"subject":"Fallback to default network_driver value if octavia not present","message":"Fallback to default network_driver value if octavia not present\n\nIn OVN Octavia driver we don\u0027t fallback to default network_driver\nconfiguration option value in case of missing octavia module.\n\nWe wouldn\u0027t add this module to requirements, so better is to\nfallback. This will prevent functional test failures.\n\nChange-Id: Ic4c8e81b0d1ab7491bc317c02464bd6853c109fa\nCloses-Bug: 1853810\n(cherry picked from commit 5804e2bb0d03b0a7eeb5705d6bd5983c056cded0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/8cd28b83e27704931dfe11a572cb0e001e59a734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/8cd28b83e27704931dfe11a572cb0e001e59a734"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
