)]}'
{"id":"openstack%2Fnetworking-ovn~697039","triplet_id":"openstack%2Fnetworking-ovn~master~Id625b0abab09fda022be23954ffca453499566bd","project":"openstack/networking-ovn","branch":"master","topic":"skip-db-table-creation","hashtags":[],"change_id":"Id625b0abab09fda022be23954ffca453499566bd","subject":"Skip DB table creation and model definition","status":"MERGED","created":"2019-12-03 09:38:55.000000000","updated":"2019-12-04 21:09:13.000000000","submitted":"2019-12-04 21:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697039-1575493451007-dc25f70e","meta_rev_id":"f93231a317ebb2f4fef3bdc1bec42347d35b3dcf","_number":697039,"virtual_id_number":697039,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 21:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2019-12-03 21:26:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-12-04 09:59:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-12-04 11:03:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-12-04 15:50:21.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-04 10:17:17.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-12-04 15:50:21.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 17:04:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-03 21:26:41.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2019-12-04 09:59:05.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-12-04 10:17:17.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-12-04 11:03:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-04 15:50:21.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-12-04 21:04:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb9a63749e3ac38d5abc191e2cd18e055c45cbaf","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-03 09:38:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b39e449d972305dd6b07d140c181f186df93ec57","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-03 10:13:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"17594cc718d9017af7b7c08e08385f95d53ba842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 12:06:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e23b6c7c3c481790d9058278b99518 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06f6e3b0b143487c9ed2ecf97f6e51f9 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/945aa9756de7443e9cfc5731a88a6384 : FAILURE in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0888e2d930af4274948bc3cd05a27ca6 : FAILURE in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9017f005a0e4480dab39d7684b84a1f6 : FAILURE in 5m 28s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5f7a2a61b84e490cb571cb63ebe81211 : FAILURE in 23m 26s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8746bf0e03494e47a49ebddef537decc : FAILURE in 23m 03s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e46c1e976906422e8cef0aa1aa13449d : FAILURE in 35m 23s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/c33c7ff27906478c9ca461813351ae58 : FAILURE in 20m 29s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1ac232a44e154cea82f0448ecfe0feb1 : SUCCESS in 2h 26m 25s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/3c7a2d2fad09456b8cd833c0d1a20d0d : FAILURE in 49m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b8718974bb64fc086ffc71957f243e1 : SUCCESS in 8m 37s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/12874f2bd3c84d5c86b065954cc61b8f : FAILURE in 30m 26s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/99b996d2d8514e9ca99cc61c76f3b4cd : FAILURE in 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a017ebfa579235fc11335d0b6e5599a92314ae9","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-03 13:19:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"371ffca8b18252e66b55bfbedebbeab7efd01de5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-12-03 13:34:01.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI still think we need to solve the models themselves because they get registered to global metadata object outside of alembic. At least that\u0027s how I understood the error in the unittests.","accounts_in_message":[],"_revision_number":2},{"id":"5dfd447869aa1275b642985328f29f01e3acdd8d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-03 13:43:14.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e I still think we need to solve the models themselves because they\n \u003e get registered to global metadata object outside of alembic. At\n \u003e least that\u0027s how I understood the error in the unittests.\n\nOh absolutely, this is more for me to test the changes in neutron: https://review.opendev.org/#/c/696347/\n\nBut, yeah, I am with you regarding solving the models","accounts_in_message":[],"_revision_number":2},{"id":"79f8f8a9d36c98a930eb5c65e9fdedf68631bb62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 16:31:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa6a363c31b4a5e9c92d80a7616f180 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390fcdc8881a4ab4a41bed1907b275ab : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8897e6c51106420cb9a97560a69e87e0 : FAILURE in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72d536acd22d418f8091d45861c4ca67 : FAILURE in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dfdcb35eea94c669c620e89855ad0ec : FAILURE in 7m 06s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/1c06b8abdcf143d59c3ee5e32d608f62 : SUCCESS in 2h 26m 35s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c380139d0cab4b679e873be1ca3d3085 : SUCCESS in 2h 07m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/02e7c80b805c49689bc8308d8d307bea : SUCCESS in 1h 57m 29s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/99e4ee0e2661408b9d36558ce80efd93 : SUCCESS in 21m 59s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ad56c74bc6f141eabedae64235ee6435 : SUCCESS in 2h 52m 27s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/94ae95e5889b4e76aeedd82d7281258e : FAILURE in 50m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/996014dde51c471ea29338d033ad609b : SUCCESS in 13m 20s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7da1e7ea2fd04d819ce26e3ab9918014 : TIMED_OUT in 3h 04m 19s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6a098bf542c44aaeb059285f8f25fdb7 : FAILURE in 2h 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"637be4a3df6907750db9d2f4166f13b527ef151e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-03 17:04:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c42dd9e494325786d9499f03b802a025dfd665e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-03 18:06:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6897a2d4e49223d2e7be87fc672f570532c14117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 21:22:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d327d52f8a477a9eb9e5b4aefc5de8 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c224a6efd40a428f96d789d4fa4cbf96 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb963b477644adea079dc740c570df9 : SUCCESS in 10m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/530b6c0803474300bcbfdfa6b3e725be : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc5779bccfd743069228be379b22f626 : SUCCESS in 11m 46s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/e593965e632c4eb589d31649fbdbfd72 : SUCCESS in 2h 25m 07s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/70ce1499fe4a4e4b93b89aabb6eec526 : SUCCESS in 2h 37m 01s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab453492f33c43798e09f149b1777ec1 : SUCCESS in 2h 16m 42s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a99bc23c69654fc8bd87bbfaca74fd57 : SUCCESS in 23m 21s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/28541d26245244e99d927ab64b4508bf : SUCCESS in 2h 45m 50s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/a6ff37806b2f48acaccac231322a40ab : FAILURE in 52m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/713bf7ad05774defaa73c77f46d7e2d2 : SUCCESS in 12m 21s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0665fb5a8e4b4f4ea33db184364a7961 : TIMED_OUT in 3h 05m 28s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c495be7e22b2415c8d54d2d9127035fd : FAILURE in 1h 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"486a62a73f59fca67ee1b0dea1f62c3a43983ccd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-03 21:26:41.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0bc0b30c5a5cc30cca42b571761f7887508d8526","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-04 09:59:05.000000000","message":"Patch Set 4: Code-Review+1\n\n+1 I\u0027m allright with actual shape of this patch.","accounts_in_message":[],"_revision_number":4},{"id":"4a5a377b0c5d2b5f9e0df7654c9bf12e7da44de2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-12-04 10:17:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"af43defa2e7a2dbfb54b5c7b0b54e11d46e47613","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 11:03:13.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"875c931e1b32b24cdd7cfd02bc547aea98235db2","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-12-04 15:50:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"72632729c7f96e9732b0e2709cc566497702113b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:50:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f9c16c03e1573f7d692f479c18485663ccbf60b4","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-04 20:12:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"584d347682b1825e73a50e6eb753f11ac036c57b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:04:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada9daf7dfa24780873a6885ca2ca52d : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3adf7b1ad74544469fa8e00290e23c13 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a68817c3996481aa5b20c7eb91017ec : SUCCESS in 11m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65fe434a6aa9449eaeeb4847d628f33f : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2891ccc579d944dd92fdc7f2c3977d86 : SUCCESS in 11m 34s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/31821f3c965141cda0b312a5980e23cb : SUCCESS in 2h 49m 33s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a3261b55bb444512b44478333256728c : SUCCESS in 2h 01m 33s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0b1e2387526456c8883a632ac39307b : SUCCESS in 21m 52s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/efdd0a6343e649f9b07eb9aab8fc1288 : SUCCESS in 2h 25m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae9e61834dcb42a8abfc97054086a6b4 : SUCCESS in 12m 09s","accounts_in_message":[],"_revision_number":4},{"id":"ccb920f85d36762c2894c550c14b15136b3cbc6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:04:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4c57fc3d52d2e9bd6df396423c8c5b7142a8b5b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:09:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a3d5f61dd88446384aeb585afc3cb05 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3c0225c994d7386d73a0c451c3bb49c7f6f2bdae","revisions":{"1ac97a56a421934ce697ae993b11b63fd966a995":{"kind":"REWORK","_number":1,"created":"2019-12-03 09:38:55.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/39/697039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/39/697039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/1"}}},"commit":{"parents":[{"commit":"7c8bbd2942a58ccaff0a7040fd5247400144addd","subject":"Merge \"Support QoS by meter action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7c8bbd2942a58ccaff0a7040fd5247400144addd"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-12-03 09:37:14.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-12-03 09:38:52.000000000","tz":0},"subject":"DO NOT REVIEW: Skip DB table creation","message":"DO NOT REVIEW: Skip DB table creation\n\nSince Neutron now creates the database ovn_hash_ring and\novn_revisio_numbers table as part of [0], networking-ovn needs to be\naware of that and skip the creation if the tables are already present in\nthe database.\n\n[0] https://review.opendev.org/#/c/695969\n\nChange-Id: Id625b0abab09fda022be23954ffca453499566bd\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1ac97a56a421934ce697ae993b11b63fd966a995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1ac97a56a421934ce697ae993b11b63fd966a995"}]},"branch":"refs/heads/master"},"2931ca8687154278af913644544e16c6e172ddf7":{"kind":"REWORK","_number":2,"created":"2019-12-03 13:19:55.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/39/697039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/39/697039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/2"}}},"commit":{"parents":[{"commit":"7c8bbd2942a58ccaff0a7040fd5247400144addd","subject":"Merge \"Support QoS by meter action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7c8bbd2942a58ccaff0a7040fd5247400144addd"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-12-03 09:37:14.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-12-03 13:13:48.000000000","tz":0},"subject":"Skip DB table creation","message":"Skip DB table creation\n\nSince Neutron now creates the database ovn_hash_ring and\novn_revisio_numbers table as part of [0], networking-ovn needs to be\naware of that and skip the creation if the tables are already present in\nthe database.\n\n[0] https://review.opendev.org/#/c/695969\n\nChange-Id: Id625b0abab09fda022be23954ffca453499566bd\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2931ca8687154278af913644544e16c6e172ddf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2931ca8687154278af913644544e16c6e172ddf7"}]},"branch":"refs/heads/master"},"df099d462bf118b01c7e8e53129c3730e8464260":{"kind":"REWORK","_number":3,"created":"2019-12-03 17:04:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/697039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/39/697039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/3"}}},"commit":{"parents":[{"commit":"7c8bbd2942a58ccaff0a7040fd5247400144addd","subject":"Merge \"Support QoS by meter action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7c8bbd2942a58ccaff0a7040fd5247400144addd"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-12-03 09:37:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-03 17:03:22.000000000","tz":0},"subject":"Skip DB table creation and model definition","message":"Skip DB table creation and model definition\n\nSince Neutron now creates the database ovn_hash_ring and\novn_revisio_numbers table as part of [0], networking-ovn needs to be\naware of that and skip the creation if the tables are already present in\nthe database.\n\nDB models are imported from Neutron.\n\n[0] https://review.opendev.org/#/c/695969\n\nChange-Id: Id625b0abab09fda022be23954ffca453499566bd\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/df099d462bf118b01c7e8e53129c3730e8464260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/df099d462bf118b01c7e8e53129c3730e8464260"}]},"branch":"refs/heads/master"},"3c0225c994d7386d73a0c451c3bb49c7f6f2bdae":{"kind":"REWORK","_number":4,"created":"2019-12-03 18:06:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/697039/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/39/697039/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/39/697039/4"}}},"commit":{"parents":[{"commit":"7c8bbd2942a58ccaff0a7040fd5247400144addd","subject":"Merge \"Support QoS by meter action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7c8bbd2942a58ccaff0a7040fd5247400144addd"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-12-03 09:37:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-03 18:05:58.000000000","tz":0},"subject":"Skip DB table creation and model definition","message":"Skip DB table creation and model definition\n\nSince Neutron now creates the database ovn_hash_ring and\novn_revisio_numbers table as part of [0], networking-ovn needs to be\naware of that and skip the creation if the tables are already present in\nthe database.\n\nDB models are imported from Neutron.\n\n[0] https://review.opendev.org/#/c/695969\n\nChange-Id: Id625b0abab09fda022be23954ffca453499566bd\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3c0225c994d7386d73a0c451c3bb49c7f6f2bdae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3c0225c994d7386d73a0c451c3bb49c7f6f2bdae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
