)]}'
{"id":"openstack%2Fnetworking-ovn~697142","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~I554d62265ecf0e890f9545d871afe85c0344f747","project":"openstack/networking-ovn","branch":"stable/stein","hashtags":[],"change_id":"I554d62265ecf0e890f9545d871afe85c0344f747","subject":"Remove escape characters from container command","status":"MERGED","created":"2019-12-03 17:33:17.000000000","updated":"2019-12-13 17:29:43.000000000","submitted":"2019-12-13 17:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697142-1576258028815-349857f7","meta_rev_id":"04d73e595806a68573b37c8e9ea265f1d52c1524","_number":697142,"virtual_id_number":697142,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 17:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2019-12-10 15:12:37.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-12-13 17:07:22.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"date":"2019-12-03 17:33:17.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-12-13 17:07:22.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 17:33:17.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2019-12-10 15:12:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-12-13 17:07:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-13 17:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e0c53fbcf8366c22bc8a07024ce2f42b91a3c99","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-12-03 17:33:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4be0d873c788bb7d52f81c08eb957fb1149af9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:55:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c080458de5874a968a06eaa93bc6962d : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2daa90d51b244894bd07770411d2776b : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7bbbd2ffb84873a3de4bc19e2ee266 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0560824082814bf9bdfe3002553bf11c : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31af83f8bd1d4fe6945c6a485a058f23 : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6db9e6b4cc4b4c0eaabc2365c6fc369e : SUCCESS in 11m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7abfa4bb0f824419a1e10dbb2c49313f : SUCCESS in 12m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37e4861f73194c3bb43051a7d512b2d2 : SUCCESS in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65ca85f217b0724a1abfcc6ce6697ba482f5c6d1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-10 15:12:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c8e3d5b2ee058e65fd7b7df6338a10eaaa992e0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-13 17:07:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a73170ece06664c1ab2eb1af5b2ead52accff4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:07:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e65ef82f285685af191c8a0e98266c59e6375779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:27:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05da8a8c120d4153bde7237ab1d15641 : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f55d23d70ac4ce789d0a29b5bcc2bd2 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e12112f1f54a82b41d0e6db6b124b3 : SUCCESS in 8m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/80b0c1237cbb45a980184c5551ecccfe : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae038504984b4c63b04daec476047735 : SUCCESS in 10m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24c4130e0ea24c3fabe3dce2b7a9f410 : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e62cde341d744ccb7f892ac20811800 : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3edaf6363c604935df48ed8c57fcaffac6c09d94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:27:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"04d73e595806a68573b37c8e9ea265f1d52c1524","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:29:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca7d7d29556405cb75d5a75a39effc4 : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e472a1c092e5f0846d8753d4a96ea1b64007f9bd","revisions":{"e472a1c092e5f0846d8753d4a96ea1b64007f9bd":{"kind":"REWORK","_number":1,"created":"2019-12-03 17:33:17.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/42/697142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/42/697142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/42/697142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/42/697142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/42/697142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/42/697142/1"}}},"commit":{"parents":[{"commit":"ad8f245fb3ed308861cf1b71cc64e6654d9cadb3","subject":"Merge \"Enable tempest DNS tests\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ad8f245fb3ed308861cf1b71cc64e6654d9cadb3"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-10-10 17:32:53.000000000","tz":330},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2019-12-03 17:33:17.000000000","tz":0},"subject":"Remove escape characters from container command","message":"Remove escape characters from container command\n\nAnsible outputs JSON-encoded strings which does not work with\ncontainer clis (ex. docker). Write to a temp file instead.\n\nChange-Id: I554d62265ecf0e890f9545d871afe85c0344f747\nCloses-Bug: #1847707\n(cherry picked from commit 63243f6cb082631758a12e7f1f558784fae2abe6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e472a1c092e5f0846d8753d4a96ea1b64007f9bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e472a1c092e5f0846d8753d4a96ea1b64007f9bd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
