)]}'
{"id":"openstack%2Fnetworking-ovn~697249","triplet_id":"openstack%2Fnetworking-ovn~master~I886cd93ebf36c3fbca14e4534bf36041660e97b8","project":"openstack/networking-ovn","branch":"master","topic":"bug/1855063","hashtags":[],"change_id":"I886cd93ebf36c3fbca14e4534bf36041660e97b8","subject":"Remove alembic migrations entry points for networking-ovn","status":"ABANDONED","created":"2019-12-04 09:49:57.000000000","updated":"2019-12-04 11:54:26.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a8195f10e114fb4b98d1bfafef341ec00f035dd","_number":697249,"virtual_id_number":697249,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-04 09:59:50.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"date":"2019-12-04 11:54:13.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-04 10:14:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 09:59:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-04 10:14:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4154c6c9583f3d3d9589624b881d37354b2eda98","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 09:49:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"492a0f21a2063239b78d6d5d1386518996e354f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-04 09:59:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8543b2ea47db88c352b443432bb5b32d9a541ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 10:14:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b9b622e7c5c4ea58bbb6c7a8b860f82 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9deea25a5e1d487686cca44709df9769 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecab2d86b83948d49e2381526bb7a6c4 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eb106ae986a4649ae64289238679909 : FAILURE in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0beef77d4bd448968e489c23cb9cf6d8 : FAILURE in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33fbbeb9cc9f4c789c187c768ce2f159 : FAILURE in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/749879e926504a0a95535d5503709f4b : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fb71d1885a74ba1594ff5c61f9506609003d67c0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 11:54:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"214fa8f8f6c9540f85621b4d366310fb2d28698f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 11:54:26.000000000","message":"Abandoned\n\nNot needed as it\u0027s addressed by https://review.opendev.org/#/c/697039/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5425cabd48029d0bceff9df03ae5169bd7cd07ee","revisions":{"5425cabd48029d0bceff9df03ae5169bd7cd07ee":{"kind":"REWORK","_number":1,"created":"2019-12-04 09:49:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/697249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/697249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/697249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/697249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/697249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/697249/1"}}},"commit":{"parents":[{"commit":"7c8bbd2942a58ccaff0a7040fd5247400144addd","subject":"Merge \"Support QoS by meter action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7c8bbd2942a58ccaff0a7040fd5247400144addd"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-04 09:46:37.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-04 09:46:37.000000000","tz":60},"subject":"Remove alembic migrations entry points for networking-ovn","message":"Remove alembic migrations entry points for networking-ovn\n\nDuring merging of networking-ovn code into neutron repo we added\nin neutron db migration script to create ovn tables if they not exists\nyet. It\u0027s in [1].\nThat broke CI for e.g. networking-ovn and tripleo, where neutron and\nnetworking-ovn from master branch are used. It happens because neutron\u0027s\ndb migration scripts are run first, so it creates ovn related tables\nand than networking-ovn migration script fails with error \"Table x\nalready exists\".\n\nSo lets not register alembic migration entry points for networking-ovn\nproject anymore as it is now added in Neutron directly.\n\n[1] https://review.opendev.org/#/c/695969/\n\nChange-Id: I886cd93ebf36c3fbca14e4534bf36041660e97b8\nCloses-Bug: #1855063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5425cabd48029d0bceff9df03ae5169bd7cd07ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5425cabd48029d0bceff9df03ae5169bd7cd07ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
