)]}'
{"id":"openstack%2Fnetworking-ovn~697495","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I186f80d51ba52d386c48f40cf0d21ebaf51d8d98","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"I186f80d51ba52d386c48f40cf0d21ebaf51d8d98","subject":"Switch release jobs to use OVS 2.12 in Train","status":"MERGED","created":"2019-12-05 15:15:31.000000000","updated":"2019-12-09 22:29:19.000000000","submitted":"2019-12-09 22:24:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697495-1575930246946-5614962f","meta_rev_id":"118758a90f91d08535cffaa9dc6a4dd8ec6cf9cf","_number":697495,"virtual_id_number":697495,"owner":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 22:24:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2019-12-05 15:17:39.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-12-05 15:26:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-12-05 19:10:10.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2019-12-09 10:58:49.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 15:26:46.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-12-05 19:10:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-09 10:58:49.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-12-09 22:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f793d28e5a28a6aa6ca21ab0a50641e8f88288e9","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-12-05 15:15:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8ce45a019248066e723868a51c13b33205c3bd03","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-05 15:17:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7c8c6e10d8bc544926e19edfb8bf32bc07a21f5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-05 15:26:46.000000000","message":"Patch Set 1: Code-Review+1\n\nYeah!","accounts_in_message":[],"_revision_number":1},{"id":"5890985c5c9278d0fdce33f41a85f30e07a9755a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 18:08:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b182b38f80b4f11a77ef847e473bf97 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88cea5c2cdc445489f1daf5056acd95e : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693e6f01c3ae4d97ba972aa65e20f679 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d89cbc4de7164ceebbc3925c1606f3df : SUCCESS in 11m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96e53ccef01a40b48ec8be0f4931aee8 : SUCCESS in 12m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72a8a08e8e364642905314cbe2d712aa : SUCCESS in 10m 56s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/564274e9cefb4208af300922974cafe8 : SUCCESS in 2h 25m 20s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/da2251a2372d4d8a99796e0d5da4ca8f : SUCCESS in 2h 02m 46s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/db30a28af90b45a6b05b337885efcfd5 : SUCCESS in 2h 23m 51s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d26d1fdc03014f66b48d937a3229e5d5 : SUCCESS in 2h 08m 39s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e8a34b3ae0d142fcad90c8fb37366243 : RETRY_LIMIT in 4m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3818ca8c76334f5e9722408cee65462b : SUCCESS in 22m 34s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e62da50f55754a9c826499ec2dd79978 : SUCCESS in 2h 45m 29s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/0c05f11a125b410786ab54867b3f1047 : FAILURE in 2h 19m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd0ab29e0cf3455c8b4f3e7a034e1fda : SUCCESS in 9m 51s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2f319d325184418f9aeef433db39fde2 : FAILURE in 2h 42m 33s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/27d30ef2346b4f72925b13393932f3c7 : SUCCESS in 1h 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da0b09a8d4adffba629063778e5553ed74fd4500","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-05 19:10:10.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b089057f8a18dfbac1e0ad43bbfe97ab645e885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 22:04:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42dd447f3958485282da2c1a3890bc2e : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65d3a7654e054bf2831dc1dfa7ddd7e6 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9100f687f7aa450ca8298fc426d3b7c0 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5243df4deaa540ac8882ae599547a01f : SUCCESS in 10m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a4ebfe9f1054e3db544211ff6ea0679 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9426aef7701745b3a7861fc9594257e3 : SUCCESS in 12m 54s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/06063427147d4274a5962398fec35a28 : SUCCESS in 2h 12m 05s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5e4b3269925c4c90a2bbbc4e98b795a1 : FAILURE in 2h 02m 40s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ad71fc92caa54f71857a16f3945fb90f : SUCCESS in 2h 11m 54s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6cc4f8e734e44bcead627f21fb41402e : SUCCESS in 1h 50m 31s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/672fd6bb50a4433f9d261a0387056065 : FAILURE in 18m 34s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae0718032c604789b43dee4523fe86c0 : SUCCESS in 18m 07s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6f23c55530854b198ee688a83ecbeb33 : SUCCESS in 2h 37m 36s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9c6ae31ac4564b3695d59cdc388c92d2 : FAILURE in 2h 00m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ced4d3ce921d4660a528b44ea510fcb6 : SUCCESS in 9m 27s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bb0315440fae4e27b3833b4a836fc90f : FAILURE in 2h 37m 09s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6734bc308fa2410888f10f6ccb1df8cf : FAILURE in 1h 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9df54da836dbeed5c4899436049ba24130a1ea80","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-06 09:47:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c6f1abd5fbd8b55bf77ade89a1afdd1031aa66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 12:46:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a2e8dfa8764c819a2357839f7f9cbd : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f01f0c02f624222bfaf8f99a243be85 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e48b60b86ba42dba98d30182290d848 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff92f053dc924964aa8ebf3a1be9d5c7 : SUCCESS in 9m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ebdd48e27c14facb1e9e65c6cdbf173 : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/614e25525aa9472db43f484075d4f6e4 : SUCCESS in 10m 24s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/bba34fdc00544f44a4030f3f9c66b6c4 : FAILURE in 2h 53m 33s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8058ac1b2d0a474e858183f8007615a3 : SUCCESS in 2h 25m 55s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/80719b116a3b4c1397d899ce45607e91 : SUCCESS in 2h 10m 13s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/85b775f6ce604c3bab3b492972162461 : TIMED_OUT in 2h 13m 18s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/75e2bbc5b0bf4304b319d0c5bbf2dabc : SUCCESS in 21m 20s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8dcf465561e4433a782f4b1aeff234d : SUCCESS in 21m 40s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/04119e14b134402fa7b0128f239b86c2 : SUCCESS in 2h 39m 02s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/512eb95caed74e099f6aa83e72e965b7 : FAILURE in 2h 02m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eb2cefa513648ac91f059d5548c3797 : SUCCESS in 8m 59s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f5effdc0772d415a92f8d725384b47ff : FAILURE in 2h 49m 13s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d59dfe86737a4924911f6716a7551a3c : SUCCESS in 1h 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5629eb49ea6b82ee2326f0766588dc34e1cbba4d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-06 17:37:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4a43385969d00e115ccd5fbaa8acbfabc5b65a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 20:20:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc696990f93a4aafbed6e897ecd82b43 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5025329959e34fe5ad9238ad5400ca68 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2079916c4fd4d03b839f78f5e200002 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba44f75c7e894e8ba4feeb2f2657b9fd : SUCCESS in 12m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd08047b158c450d96309525b798616a : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/874146dec91046e9b3cd260978b546b9 : SUCCESS in 10m 49s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f528029827154f94a580615af4d12ecd : SUCCESS in 2h 28m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ba2739cce9454871a30d75b83a1af887 : FAILURE in 2h 12m 49s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/092d40e0d6ca4f9f8c39c550061c9b8d : SUCCESS in 2h 19m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac38c2e804444c478147ab20d7dae803 : SUCCESS in 2h 13m 05s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/201ea05088e34dedba8e34a99e1138c7 : SUCCESS in 20m 36s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/82472bf299ac418eb1a05afe42901b67 : SUCCESS in 19m 26s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3733dc063899441d89121df52504da6f : SUCCESS in 2h 24m 31s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/b1d430fc448044efb437d0fecbd49f17 : FAILURE in 2h 09m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61b1a4781e75470d9a3b06cf98bfe361 : SUCCESS in 9m 14s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c4b6cb3fa87b4df1ad3695b6698e8050 : FAILURE in 2h 34m 34s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/619ecc24a31047a4bb311887b52931e0 : SUCCESS in 2h 06m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d3c38e99799edcc3a0322a2089a3e49e042c671","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-09 08:06:52.000000000","message":"Patch Set 1:\n\nrecheck not related","accounts_in_message":[],"_revision_number":1},{"id":"7bde4a682709f0d5319cf95e4d8c78540418f1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 10:51:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee72c04bed44c43addfc2fdff5e2de8 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4e2e0e5d3a3442587c8337cd6804489 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11266a927d884d0ca7f3b1e3ab0b1f05 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86f69b112d5f46e887080dc6742012e4 : SUCCESS in 10m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3906b4ca17244d40a4d98dc6c3732147 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a9f524308cb47e6b5ed06b886ef75fd : SUCCESS in 10m 26s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/03489f34d2914a8ba0413a45db9498d2 : SUCCESS in 2h 37m 09s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/1e61218a39b1455ab83dbee35fd1cb7f : SUCCESS in 2h 10m 08s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8ad619f1167a45c4b26b476a86c17c96 : SUCCESS in 2h 29m 38s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c5c7d7d9bde749f483c0f08a613e2226 : SUCCESS in 1h 57m 39s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/43232bb25b934a5ea3a485af2b47d0cc : SUCCESS in 19m 36s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e53baf0d07024c3d914781a794a5b493 : SUCCESS in 22m 54s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/67ad12ff62534c17a24e2cd5d235e054 : SUCCESS in 2h 39m 25s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/bda0e3d347ec48e9a99a4b769fed06d0 : FAILURE in 1h 59m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/243d7157a8b94075995fa8770470f7a4 : SUCCESS in 8m 34s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7ce8cfe8f6f54e21a7ab6a2c974c933a : FAILURE in 2h 38m 34s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3a93e6fa24cf4a49b20356471ac0732d : SUCCESS in 2h 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc646b24e154dc49bd4345debba1438c0abf7b8d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-09 10:58:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae12eb92d96b1524a7aaa9d169534f2259a1e2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 10:59:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9797dee54e1638ccac8cf0d88347168c6d920bd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 13:55:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27446c5cfac84248baccf2dab5c97021 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6cc33e524eb4c2397341fe32eddde64 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e0d9f5000a489f94e970113d64a169 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55e9bb12cf134e5f9ca45ce2f93c89ce : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba76e2984bb144a4a7ca12840cdc2458 : SUCCESS in 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c31183625e7447d8e820bc5949239cb : SUCCESS in 13m 33s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/baa7b124fb654e9d8157a449b06d820a : SUCCESS in 2h 15m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/409ada76647944f5b1896fa0cc84d2fa : SUCCESS in 2h 43m 38s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/15a41cc5d0004b9baadb9abf33f4102a : FAILURE in 2h 03m 02s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ce9491a7f46b43fda238e48a557429c8 : SUCCESS in 21m 34s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e93132d5b04949e68c6a8c38ce96ad9d : SUCCESS in 22m 28s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/70d8605f55274c478b0c3f618cd6d8ed : SUCCESS in 2h 50m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d21d4d7f71aa42cc965c085b7e364bc4 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3e388981b8fb9675ad3babdb8d114b1776432d65","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-09 13:58:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0124c44fd4113ec3bad67dac58b3182271e06183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 16:30:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eff9d50f19c4396a9bc040abc35073f : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6e5a59b15f44d18a219be7df1144141 : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acbdf4f68a84b6584701ca706c66db8 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4139b0567eee4a629ade77d842afaf94 : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28fa83ccde7b4572907ecc89e8f00d3b : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/316a60f151e0483dbfa3ad6dd2f612e1 : SUCCESS in 10m 13s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/36529abc01ef4f62ae8881917d81065c : SUCCESS in 2h 21m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5a0c89c6cce0462680b526d8bc110c8e : SUCCESS in 2h 14m 46s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/69f9fd8cf541442e9261159ab23c2e30 : SUCCESS in 2h 25m 15s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a4a55b191c714f908983a9e8e19ec04e : SUCCESS in 1h 53m 23s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/11bbac12c3d1457792b20bfadf40f83c : SUCCESS in 20m 38s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/23c96b89d8544099aea78b92d4d8d517 : SUCCESS in 22m 42s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6f7418fc2deb4af187661f14597da20b : SUCCESS in 2h 27m 19s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/92dfc282542f4504b7406b3f03ee843e : FAILURE in 1h 46m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4d6bf19e0cd485796b2c8078270899e : SUCCESS in 9m 22s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/56de235c800a4dd1b44eecfedda5f161 : FAILURE in 2h 21m 58s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c47a17d837404f83ae14e368db4769b0 : SUCCESS in 1h 58m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b2fa5bcc3b20ea2962d446132a0dc6f0b3b516b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 16:30:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60ba12163c2e10244d744218f7ac456899709ca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 19:13:19.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6408c7c75cf230ffd049e45da3e9a4771fa2eb47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 22:24:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f9a7cde13bb47f381ba026e3fa75c1e : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f0b7ff73a134eb0a113a2f4497bd0d0 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b20c3bc09b5a4dfb8a0772499765a07a : SUCCESS in 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a28dc4135cf4f73843dae159259262e : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e60e935840b74336a4db762e6a1aae8e : SUCCESS in 6m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a5feb35555d4eefaa02b18457320e62 : SUCCESS in 2m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b49aed5821f4336877c9c6a2ad32aa9 : SUCCESS in 10m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/925af4feda484595820c00d08ab9d104 : SUCCESS in 11m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/672b66c7782244629dd824cfd7764fef : SUCCESS in 11m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/1e3f9f1e241640f7bf38855f4916e891 : SUCCESS in 2h 10m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f2ea23ae7c8c4b4692df9dc30f8ce43a : SUCCESS in 2h 28m 08s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1ddd3154af44b20b29784aa79c30081 : SUCCESS in 2h 08m 41s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/0512ec2e4c3c4cd39569d406e7c0e596 : SUCCESS in 21m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f79e0c48c56b41b19e07e3dcefc7bbc8 : SUCCESS in 22m 12s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e1c6b87563b84ff0b68d4d25666ecc19 : SUCCESS in 3h 04m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeb2f2ec9e9943baab938cead6438a7e : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":1},{"id":"87466e56ec928420412eef887c9dd7acedc05b54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 22:24:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"118758a90f91d08535cffaa9dc6a4dd8ec6cf9cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 22:29:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72928c0e5b24cdead800c8f6bac72e8 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a49513ec9949a0cfbd753da31219109785ca4b41","revisions":{"a49513ec9949a0cfbd753da31219109785ca4b41":{"kind":"REWORK","_number":1,"created":"2019-12-05 15:15:31.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/95/697495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/95/697495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/95/697495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/95/697495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/95/697495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/95/697495/1"}}},"commit":{"parents":[{"commit":"37a51de26cf07df76ceac3a24fb1c654b413214b","subject":"Merge \"Fallback to default network_driver value if octavia not present\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/37a51de26cf07df76ceac3a24fb1c654b413214b"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-26 08:19:51.000000000","tz":120},"committer":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2019-12-05 15:15:31.000000000","tz":0},"subject":"Switch release jobs to use OVS 2.12 in Train","message":"Switch release jobs to use OVS 2.12 in Train\n\nThose jobs were using 2.11. We would like to have tests\nfor InternalDNS functionality so we need to switch those to\n2.12.\n\nUnfortunately branch-2.12 is not yet configured\nin ovn.org repository. Because of this lets bump up\nversion in repository selector function\n(use_new_ovn_repository).\n\n\nChange-Id: I186f80d51ba52d386c48f40cf0d21ebaf51d8d98\n(cherry picked from commit 2543f4aaead1cde82469e67fedd435e7f23f158f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a49513ec9949a0cfbd753da31219109785ca4b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a49513ec9949a0cfbd753da31219109785ca4b41"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
