)]}'
{"id":"openstack%2Fnetworking-ovn~697993","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~Ie209dfdf522a32f89cdc07805bfaa3e04493c8fd","project":"openstack/networking-ovn","branch":"stable/stein","topic":"bug/1855672","hashtags":[],"change_id":"Ie209dfdf522a32f89cdc07805bfaa3e04493c8fd","subject":"Do not import networking_ovn.ovsdb.worker in impl_idl_ovn","status":"MERGED","created":"2019-12-09 11:30:14.000000000","updated":"2019-12-09 18:34:04.000000000","submitted":"2019-12-09 18:32:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697993-1575916343347-5614962f","meta_rev_id":"f36ed629f08aa420c18e4b541d4a24f7f9a8e1cf","_number":697993,"virtual_id_number":697993,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 18:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2019-12-09 11:40:03.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2019-12-09 15:49:12.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-12-09 15:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-12-09 15:49:12.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 11:40:03.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-12-09 15:49:12.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-12-09 15:56:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-09 18:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"433b43ad8f178080630577c2b411ebb5a0d83595","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-09 11:30:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b4ffb5cf25bcf2aa9b96880646a36f83e44b701","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-09 11:40:03.000000000","message":"Patch Set 1: Code-Review+2\n\nA wow, tricky!\n\nThanks for the investigation and the patch. It LGTM!","accounts_in_message":[],"_revision_number":1},{"id":"a8bb628e5985c3d5acba9cc2d335429a32a44ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 14:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf4b4d607ec4debb3ff6c5571e81703 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fe0763c2d314aba863966288eda5f3a : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da973c0a0c4d4f9b9f4da1dd9e3c6793 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c204c999e66745f9a220a393553c2edc : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25ac5ed3341d4d7d9448c71d7bed167a : SUCCESS in 9m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc979f41fd6e45e5873786d7132717ea : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee44030265074c83b0de18fce43cd12e : SUCCESS in 9m 41s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7ad61e92d88942129783a91b87b4789b : SUCCESS in 1h 57m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/33c151544b6b48bcba052df06f29d683 : SUCCESS in 2h 09m 15s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec3b9b1e7bca4d5fbf61c963cf93b59f : SUCCESS in 2h 02m 27s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e52c1cb024fb4d35b55ef9df56546a49 : SUCCESS in 17m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0b19cede6ec4cd292843a75e83d8873 : SUCCESS in 20m 45s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9697d9e6b5d34e709fa4000bd7119a6d : SUCCESS in 2h 45m 48s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/4c0fa1d884f6453faa5945e1728235ea : FAILURE in 2h 09m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34af48b10bfd40408bddd1e952d1564e : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b61047f35733d0c91939c42dda19f3a860eab8f4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-09 15:49:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMakes sense to me and tests are passing, so looks good to as stein-only fix indeed","accounts_in_message":[],"_revision_number":1},{"id":"93b8b71fa9d5f96bb7345df9531055b0358bb52d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 15:49:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"127050e129383ecaf03bcbb417a2d631e618a97b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-09 15:56:55.000000000","message":"Patch Set 1: Code-Review+1\n\nI missed the party, but verified this works on my devstack.","accounts_in_message":[],"_revision_number":1},{"id":"fa058d6eff014f13e637159f6ed38865c33b61d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:32:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84ef5d93318513d929f163ab01f5403558e0aff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:32:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7e385025b3495baafaf86b3bc13e78 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f1a7ffd5e914a43b7011958270709e1 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa2658b3f57459e99c1551181eec09f : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/635e087a690d47dcb79f6315ed23d558 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1e7f065222422e9d466755682af7ca : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1bd3f697a294470a848865a9f4739d3 : SUCCESS in 11m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fa2eae107d641d88da985104bde850f : SUCCESS in 11m 20s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a67e59c3f8304c4a95e1410d7972cdb8 : SUCCESS in 2h 23m 33s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/7285571120e14262a9ed9951ce49bcf7 : SUCCESS in 2h 10m 55s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a9ce6cf330b84862a8aaadbaa1898068 : SUCCESS in 1h 44m 38s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b96fcdeccc134c89be498be9e5887755 : SUCCESS in 21m 06s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b59aac0fb41f42fb92d1e39f515eae2e : SUCCESS in 22m 19s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7510c7dcddb8421aa0b52cca73bb8c0b : SUCCESS in 2h 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"db8287e24263d9461f4abcb5cf0f7845d75c5dce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:34:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9474e267855b419580c01c2fd4949bec : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"355bf2662725f6ab1c45b9a3b50f60ad0f62061f","revisions":{"355bf2662725f6ab1c45b9a3b50f60ad0f62061f":{"kind":"REWORK","_number":1,"created":"2019-12-09 11:30:14.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/93/697993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/697993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/697993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/697993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/697993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/697993/1"}}},"commit":{"parents":[{"commit":"ae63923ca7bd8b02d6634aa110a3746918d23745","subject":"Merge \"Initialize TLS prior to retrieving OVSDB schema\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ae63923ca7bd8b02d6634aa110a3746918d23745"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-12-09 11:12:57.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-12-09 11:30:13.000000000","tz":0},"subject":"Do not import networking_ovn.ovsdb.worker in impl_idl_ovn","message":"Do not import networking_ovn.ovsdb.worker in impl_idl_ovn\n\nIn impl_idl_ovn we import ovsdb.worker class just only\nto verify if get_connection() is triggered by a method\nfrom OvnWorker class or not.\n\nUnfortunately in worker.py we import neutron.common.config\n[2] that imports neutron.conf.common [3].\nIn neutron.conf.common there are base config parameters, like\n\u0027host\u0027 [4].\n\nWhile using OVN provider driver the same config parameter is\ndefined in Octavia process scope [5]. Mising two configuration\nscopes ends with an Exception:\n\nERROR octavia.api.drivers.driver_factory [-] Unable to load provider driver ovn due to: duplicate option: host: DuplicateOptError: duplicate option: host\n\nTo mitigate it I changed the way we verify if the trigger\ncomes from OVNWorker class, to not import\nneutron.conf.common and neutron.conf.common.\n\nThis issue doesn\u0027t exist in master (Ussuri) and stable/train because of [6].\nIn neutron it will not be visible because we move OVN provider driver\nto separate repository.\n\n[1] https://github.com/openstack/networking-ovn/blob/stable/stein/networking_ovn/ovsdb/impl_idl_ovn.py#L148\n[2] https://github.com/openstack/networking-ovn/blob/stable/stein/networking_ovn/ovsdb/worker.py#L15\n[3] https://github.com/openstack/neutron/blob/stable/stein/neutron/common/config.py#L32\n[4] https://github.com/openstack/neutron/blob/stable/stein/neutron/conf/common.py#L94\n[5] https://github.com/openstack/octavia/blob/stable/stein/octavia/common/config.py#L39\n[6] https://github.com/openstack/networking-ovn/commit/6ef5489cf70cbc140a8768727350a8556fc8e870\n\nChange-Id: Ie209dfdf522a32f89cdc07805bfaa3e04493c8fd\nCloses-Bug: #1855672\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/355bf2662725f6ab1c45b9a3b50f60ad0f62061f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/355bf2662725f6ab1c45b9a3b50f60ad0f62061f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
