)]}'
{"id":"openstack%2Fnetworking-ovn~698293","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I5ccfc625352505c5b53f18e051e4d18598d5e286","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1855903","hashtags":[],"change_id":"I5ccfc625352505c5b53f18e051e4d18598d5e286","subject":"tox: Pass OVN and OVS source directories","status":"MERGED","created":"2019-12-10 16:41:48.000000000","updated":"2019-12-13 17:31:18.000000000","submitted":"2019-12-13 17:28:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698293-1576258112763-71032318","meta_rev_id":"70b6d534fda307eb52060369c3065b77d48b1522","_number":698293,"virtual_id_number":698293,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 17:28:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2019-12-13 13:46:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2019-12-11 10:48:03.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-12-13 17:08:41.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2019-12-10 16:42:13.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-12-13 17:08:41.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 16:42:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-11 10:48:03.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-12-13 13:46:40.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2019-12-13 17:08:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-13 17:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f0a0b03f008eb8f252fd4a4e3fc09c37430c401","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-12-10 16:41:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e519ef7096d3497c9bcd144895557d5bd9f6869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:06:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a679741d6140c8bb032f30194e7ec6 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/869e6fd7eb7e430e87e155d2c7cd619e : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8698adb910e4a3abfec5fc49bbf0484 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbe7c01533ef48e599e30ec53f2dabcc : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a0fce144f514e228c39c320aeb4c5ad : SUCCESS in 9m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2728cd1530d9489789f7fe482f6e8bbe : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5698a40bfc04fd7b289484b2d21528e : SUCCESS in 12m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0c0f77f59cf40ae9fc7c6297ae63699 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2163db079490c9e39409815dcafd87ad9086d2bc","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-11 10:48:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3418794b3454fc41995f7f8545fc378bb9693d1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-13 13:46:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7673ea344b2c884aa4540695deecc20a04e89583","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-13 17:08:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8af01916447981f04dafdddcda7d85e5fb1647cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:08:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"946efbef454844460133f092e1aae6a1ee3e943d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:28:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1fe4a2310fc23d568257dee9a76503eb921f4af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:28:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16e019c8fd9b4a2d8673e6952e1a93a8 : SUCCESS in 7m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1c748129c3a4e08b9e2e9f9f650cd14 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4008e09d314edb9a8b3d4afe6db0fb : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4feb04161a18485f9569f6891ec29319 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17257d4975314668acf3553ef230c7ab : SUCCESS in 10m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed707c1854f546bda355ef2e75c73af4 : SUCCESS in 13m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/574eab950b5e439abff95e65d058c7c2 : SUCCESS in 12m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5440c6f61c174e758b7006ae13661995 : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e777162d7cb895d07ae726870dc5abcf86d09270","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:31:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a6ee6340e242748b235400fc02d562 : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6de6123ce2c64c21bbd5357c604485c5 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d603a7e105d2cbe8ca21a1ee752e1f3c81879e0","revisions":{"0d603a7e105d2cbe8ca21a1ee752e1f3c81879e0":{"kind":"REWORK","_number":1,"created":"2019-12-10 16:41:48.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/93/698293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/698293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/698293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/698293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/698293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/698293/1"}}},"commit":{"parents":[{"commit":"0f085fabf87e3717a4dc194eebe2d02762874dfb","subject":"Merge \"Switch release jobs to use OVS 2.12 in Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0f085fabf87e3717a4dc194eebe2d02762874dfb"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2019-12-10 16:36:49.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2019-12-10 16:40:54.000000000","tz":60},"subject":"tox: Pass OVN and OVS source directories","message":"tox: Pass OVN and OVS source directories\n\nSome of networking-ovn functional tests inherit from ovsdbapp base\nclass. ovsdbapp uses environment variables to define directories\ncontaining database schemas. This patch passes environment variable to\ntox if it\u0027s defined. If it\u0027s not defined, an empty string is used and\ncode later uses its own defaults.\n\nThis patch doesn\u0027t go to master as it\u0027s frozen now because of the code\nbeing merged to Neutron. The patch will be sent to master Neutron once\ncode migration is finished.\n\nChange-Id: I5ccfc625352505c5b53f18e051e4d18598d5e286\nCloses-bug: #1855903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0d603a7e105d2cbe8ca21a1ee752e1f3c81879e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0d603a7e105d2cbe8ca21a1ee752e1f3c81879e0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
