)]}'
{"id":"openstack%2Fnetworking-ovn~698979","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I09f36870122c2cb6ae891cefca3a6a4b8de92d4d","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"I09f36870122c2cb6ae891cefca3a6a4b8de92d4d","subject":"DNM: please ignore","status":"ABANDONED","created":"2019-12-13 17:46:04.000000000","updated":"2019-12-16 20:17:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ad50b90f76d13cfce6e634f9c14385ee0dc391c2","_number":698979,"virtual_id_number":698979,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-16 16:18:37.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-16 14:50:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 14:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ee99e4b143a284c222e3ee39ec9ece6973461f2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-13 17:46:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92dcaef30ff7b31ab9841df3251ac5b46293026e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-13 19:11:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad3d0d76b78a802ec559b192c5a83cca2e057659","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-13 20:38:08.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c8bb17f88019fa3b9658124b80b5f1843776e7e2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-13 21:02:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d18a32215281761493903a913f9c799ba686022e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-13 21:17:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9603df6e90e0a62afcd55d8da9eb719b2c8825b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 23:53:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4d9da7586e487987454c0631407550 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6d2cd1d19d0460cbbc3d7daf80cdf8d : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afba421a49504c07a0316b85c283992d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0431431f47b04fbbaf31fd1d6bff5ce4 : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4b8808647074cc49ac59dbba73b0cb2 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99b300dd76e64aa4820344015edfd354 : SUCCESS in 11m 41s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/55d72f8032a2411bad111fba4f17b971 : SUCCESS in 2h 14m 33s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b44d05322cba49548977752f3594c154 : SUCCESS in 2h 01m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ecba1e170d2d489992ff2ef8c0de060e : SUCCESS in 2h 16m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/00439afbe1f848ec944a692298ba73e9 : SUCCESS in 1h 46m 55s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f53e2aabd76349e0b1f49c36302c4687 : FAILURE in 15m 09s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9a3484db61140b79f53e72bdf8be6b7 : FAILURE in 10m 55s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/60a145f4c6894a27a75fa8e22485fbe3 : SUCCESS in 2h 13m 46s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/844bd129c76a4ba182ae5484e386f354 : FAILURE in 1h 53m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32f9e1d858eb462ebed426729e114057 : SUCCESS in 11m 17s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/053924dd5ab4488ea364db3e22753c08 : FAILURE in 2h 34m 20s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6c6705da83964cb98ef96741ca3733d6 : SUCCESS in 2h 04m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"745b53bf655d22b6dc31c0b0e00f6811ada4ecda","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-16 09:48:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"2bd00d82833bd132967afb38dec7b3d9f5345c8a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-16 10:46:59.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"dac681301cc6524eea4f6b5132e5ec34ef4cf3e1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-16 10:48:13.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"103cd476c4243846b090b1792e3dbf4f80ddf88e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-16 11:47:54.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"70a022b5621a4b7e678e5004135fc76e9fd3f6c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:50:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48502f05d79043df98d0b4471f8eafbb : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/273e62d6dac041488f41579180fcea39 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a480c249094e8690ba29166f59834d : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fd97ffd221c416296aa648416bfaca1 : SUCCESS in 10m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a548830c4e5b4220b1ee3575b64a5a5d : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6ec0e595fa0480a81b4e7c797070bfc : SUCCESS in 11m 59s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/c27cbf98768240dba98feeca505ed958 : SUCCESS in 2h 30m 07s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4eb8bddc08f74a3091fe503559032505 : SUCCESS in 2h 05m 18s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/4c38ed486c0a4ddb88a57eb4f31a29e0 : SUCCESS in 2h 03m 23s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/10da22f59a414b269f261fe58b7bb9dc : SUCCESS in 1h 44m 25s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a44fc655eba643b9b4f9f138de06c002 : FAILURE in 26m 14s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e2403f0bcea4fcba2fada88b84429e4 : FAILURE in 26m 07s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5369a0988bbc4efe8615d95c97316953 : SUCCESS in 2h 22m 13s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/00c91688735c4f728817c2aadb53cb3c : FAILURE in 2h 00m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2584b6a00e8845fda571eaf0ed40dc1b : SUCCESS in 14m 38s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5f004de665894ba193656a9406fad49d : FAILURE in 2h 54m 04s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/73310cff58f2431ba5670cbb1ee8c0b1 : SUCCESS in 1h 53m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0d685f52a36b93b852385b9987ed4725e369c07b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-16 16:18:37.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"f465770ee4bc1cf7360e386396b4a44084f84c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:08:38.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07239c66e264eed8fefbc312915aa39 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67b47c1d8b6d4394af5691e56910538a : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd3bdd7a244494eadb5ffa90ad805af : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb67d95d4624e02ae6a1e2e4a0fd5fe : SUCCESS in 12m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a33bb728e5f4b059e6213072c90ad80 : SUCCESS in 12m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e464e77c87c47e1b7eab42d7f4b83f6 : SUCCESS in 9m 44s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/e8df2641325b475ab645488bab0c25c8 : SUCCESS in 2h 26m 18s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8b503f46a8d143ab8d0d3ff8911f4b76 : SUCCESS in 2h 31m 17s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/c387d46dedf04115bf24d377c2236dff : SUCCESS in 2h 12m 10s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/280d85b2d30e424e96174c78cb418d36 : SUCCESS in 1h 39m 58s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ddd45fb53cb74ddfb2a3c1226d60f973 : FAILURE in 14m 57s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/cbfabd030eef45608765cde0e3e41bf9 : FAILURE in 10m 24s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/67734498221247b0a7044b0408e5e413 : SUCCESS in 2h 46m 56s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/b9f8a19eb7ad4e5ab61deb3760b5ace3 : FAILURE in 2h 08m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dbb159f43b94f13b383065cca3ccce9 : SUCCESS in 9m 19s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/81a5fa8454014f02b94584295cf5cd4b : FAILURE in 2h 36m 09s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f7b042129e5b419291441da37f38ea10 : SUCCESS in 2h 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ad50b90f76d13cfce6e634f9c14385ee0dc391c2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-12-16 20:17:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"6cd0a06d3691153ca6ee01c8670a73f29db01efd","revisions":{"4e1d9906c30a48c40fd1339cfc88391e9cf370bf":{"kind":"REWORK","_number":1,"created":"2019-12-13 17:46:04.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/1"}}},"commit":{"parents":[{"commit":"0f085fabf87e3717a4dc194eebe2d02762874dfb","subject":"Merge \"Switch release jobs to use OVS 2.12 in Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0f085fabf87e3717a4dc194eebe2d02762874dfb"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4e1d9906c30a48c40fd1339cfc88391e9cf370bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4e1d9906c30a48c40fd1339cfc88391e9cf370bf"}]},"branch":"refs/heads/stable/train"},"9cce914994a2fed87514ca71b0759f4188bc030d":{"kind":"REWORK","_number":2,"created":"2019-12-13 19:11:13.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/2"}}},"commit":{"parents":[{"commit":"0f085fabf87e3717a4dc194eebe2d02762874dfb","subject":"Merge \"Switch release jobs to use OVS 2.12 in Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0f085fabf87e3717a4dc194eebe2d02762874dfb"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 19:11:06.000000000","tz":-300},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9cce914994a2fed87514ca71b0759f4188bc030d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9cce914994a2fed87514ca71b0759f4188bc030d"}]},"branch":"refs/heads/stable/train"},"18a4ac885d826d18c0bc157c2b5d84433e6307d9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-13 20:38:08.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/3"}}},"commit":{"parents":[{"commit":"97edb8a1f4e4c7827505b4caf5f1e11c24138c95","subject":"pep8: fix issues found by new version of hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/97edb8a1f4e4c7827505b4caf5f1e11c24138c95"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-13 20:38:08.000000000","tz":0},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/18a4ac885d826d18c0bc157c2b5d84433e6307d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/18a4ac885d826d18c0bc157c2b5d84433e6307d9"}]},"branch":"refs/heads/stable/train"},"2f1adcff05a47eba201da334a16b49330b969a4c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-13 21:02:20.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/4"}}},"commit":{"parents":[{"commit":"ccb36d05ce15fd78f4ca67b4c68c3cf6bcf4457c","subject":"DNM: pep8: fix issues found by new version of hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ccb36d05ce15fd78f4ca67b4c68c3cf6bcf4457c"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-13 21:02:20.000000000","tz":0},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2f1adcff05a47eba201da334a16b49330b969a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/2f1adcff05a47eba201da334a16b49330b969a4c"}]},"branch":"refs/heads/stable/train"},"8a92930f36c9383945d4141757b1b7b2fceb3da7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-13 21:17:39.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/5"}}},"commit":{"parents":[{"commit":"66037f04d24c5cbab8be34954a2f55dc580555d3","subject":"DNM: pep8: fix issues found by new version of hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/66037f04d24c5cbab8be34954a2f55dc580555d3"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-13 21:17:39.000000000","tz":0},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/8a92930f36c9383945d4141757b1b7b2fceb3da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/8a92930f36c9383945d4141757b1b7b2fceb3da7"}]},"branch":"refs/heads/stable/train"},"3457994b2cd29771e2e504b33d7666854726274a":{"kind":"NO_CHANGE","_number":6,"created":"2019-12-16 09:48:24.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/6"}}},"commit":{"parents":[{"commit":"1ba7c5f3b5a64db92a603d232b233c8067216464","subject":"DNM: Fix pep8 errors found by hacking 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1ba7c5f3b5a64db92a603d232b233c8067216464"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-16 09:48:24.000000000","tz":0},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3457994b2cd29771e2e504b33d7666854726274a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3457994b2cd29771e2e504b33d7666854726274a"}]},"branch":"refs/heads/stable/train"},"b4fb10477c2c5ebe228a3b566efd2271e2ca9cdb":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-12-16 10:46:59.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/7"}}},"commit":{"parents":[{"commit":"9ce4e0dc408de78d824985db522da17d0f89ee34","subject":"Fix pep8 errors found by hacking 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9ce4e0dc408de78d824985db522da17d0f89ee34"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-16 10:46:59.000000000","tz":0},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b4fb10477c2c5ebe228a3b566efd2271e2ca9cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b4fb10477c2c5ebe228a3b566efd2271e2ca9cdb"}]},"branch":"refs/heads/stable/train"},"81bbba8501cf9964f6bdbf29642f1d5b3103a98e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-12-16 10:48:13.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/8"}}},"commit":{"parents":[{"commit":"0d603a7e105d2cbe8ca21a1ee752e1f3c81879e0","subject":"tox: Pass OVN and OVS source directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0d603a7e105d2cbe8ca21a1ee752e1f3c81879e0"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-16 10:48:13.000000000","tz":0},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/81bbba8501cf9964f6bdbf29642f1d5b3103a98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/81bbba8501cf9964f6bdbf29642f1d5b3103a98e"}]},"branch":"refs/heads/stable/train"},"6cd0a06d3691153ca6ee01c8670a73f29db01efd":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-12-16 11:47:54.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/79/698979/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/79/698979/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/79/698979/9"}}},"commit":{"parents":[{"commit":"b02aff6859ab9800d6b3d9e92e32388c48c35807","subject":"Fix pep8 errors found by hacking 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b02aff6859ab9800d6b3d9e92e32388c48c35807"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-12-13 17:45:53.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2019-12-16 11:47:54.000000000","tz":0},"subject":"DNM: please ignore","message":"DNM: please ignore\n\nChange-Id: I09f36870122c2cb6ae891cefca3a6a4b8de92d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6cd0a06d3691153ca6ee01c8670a73f29db01efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6cd0a06d3691153ca6ee01c8670a73f29db01efd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
