)]}'
{"id":"openstack%2Fnetworking-ovn~699643","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~I3c01bd7d1685c8a7e13a55e545e98baf19e9a0f9","project":"openstack/networking-ovn","branch":"stable/stein","topic":"bug/1838535-stable/stein","hashtags":[],"change_id":"I3c01bd7d1685c8a7e13a55e545e98baf19e9a0f9","subject":"Only set \"unknown\" in LSP that makes sense","status":"MERGED","created":"2019-12-18 09:51:38.000000000","updated":"2019-12-20 18:43:17.000000000","submitted":"2019-12-20 18:41:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699643-1576867289965-7a9322a4","meta_rev_id":"702390fecefd8d6a0ef74d9ce19ed007bf790095","_number":699643,"virtual_id_number":699643,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2019-12-20 13:08:37.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 18:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2019-12-20 10:24:17.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-12-19 19:18:05.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-12-20 10:24:17.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 09:51:38.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-12-19 19:18:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-20 10:24:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-12-20 18:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0afda5eec430ae0374a3677b522292b094e187f4","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-18 09:51:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ba2e4b5419d818126c39949b9c29562905cb3a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:00:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73a558d2ab54435aec1474a7b6b01d7 : FAILURE in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95e6cf98d0d843bfb83f9028be3ded0b : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918bb675e99044e1a8b8131956b43863 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9a39e8766b0c4915b32e36ea4c37c3ec : SUCCESS in 11m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2debb4c7af2c4ed7b1bf85eff4e30d5e : SUCCESS in 10m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a306c30818994069bb220c8d0d170606 : SUCCESS in 16m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0624efd576f04eb9a6a54a535421e916 : SUCCESS in 9m 58s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/da6a851a3cd94d9c88e8bf1e016fcde1 : SUCCESS in 3h 01m 26s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/82895f90fc2746a481b7c9ec59206217 : SUCCESS in 1h 52m 57s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/01e8b2a2273341eab83beebcdbe8239c : SUCCESS in 1h 38m 23s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/722d0263afff47b8a2a33e853602bbbb : SUCCESS in 20m 10s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2515963b7e5d488384dac4cce20196fe : SUCCESS in 22m 15s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8ee5086017c242eea308987d2a398a6d : SUCCESS in 2h 37m 30s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9b20e7ddf5f847db932862f2fcc47ccc : SUCCESS in 1h 50m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bf9e13cb5904374b3dc80cec7ebc387 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab8070c834a2112d8e0e499e1f3091d5a1f88a9c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-19 09:27:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8c633c443dd4ba5dd2d66c927f494175318ee50e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 11:57:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275b55ddff2e4bef90a9c55631edcbfa : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c04579235eb947b09eef91270b6483f3 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd931551450b46ed98e9dd1475edd711 : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aaaaa2bef8e3418d8e47558fa1e3b4aa : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72f1f8ce41c2427e8aae91e975cec9b6 : SUCCESS in 8m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2be350d9880c42549406c9016b8bdd99 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a2420f36eb249caa2e24d5af0099855 : SUCCESS in 11m 11s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e0efe8477fbd4f0e95858aed6f02ac60 : SUCCESS in 2h 01m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f35ab5a5999b4bdfbbbc9966e8eebc0b : SUCCESS in 2h 03m 29s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7d36022574d444e9ae8a2d35cd49e6d1 : SUCCESS in 1h 40m 40s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d81c5735eb5b4d31ada2c1994b17f210 : SUCCESS in 19m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/627aa8c984814ef79f2b3151c817945b : SUCCESS in 22m 25s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/caf6b82ab620418eacf290fe74048ebc : SUCCESS in 2h 22m 47s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/adcb693eeec24d35b4bf325520391dfd : SUCCESS in 1h 54m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f6fe19915dc42e5b3139db1d7b60d40 : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae10ee4a2a6d0479404bc8953ffbe6f32694ec83","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-19 19:18:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8cecce0907a64eb38195f6449ec7c8dd547489fe","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-20 10:24:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"73d9110c1db7154c50ccf5bf15a16e0e86fd9671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:24:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"055b3b023c81c6b8644c9eab9edf9f08f97a7854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 13:04:51.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edfd67e4c66e419da2537af3bfc18d8b : SUCCESS in 7m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52a1840b164b4decacc07ce1ff5c041b : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1c9007656d443c967b6be8043a43bd : SUCCESS in 6m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/24e3ed7236314d06b1eb19782a4017ac : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f6bea8715794448ba5b74799352b99d : SUCCESS in 13m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7af9f8c2f85f4e8a9d0c5b362cb7438e : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08ba771d93654b18926e977c2bc299a2 : SUCCESS in 16m 47s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/314327c993b94c778ea3a3c39834b2a4 : SUCCESS in 1h 54m 02s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ac48483de9054301aea5125dbcbd0218 : FAILURE in 2h 36m 07s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bc52f67a2486417e80b0d93ddfdf1b3a : SUCCESS in 1h 48m 29s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d06ca86ecf924baeb0b13a49d8c7bce6 : SUCCESS in 23m 27s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/20d1a52a668c4bd08cd4156ec15aafa8 : SUCCESS in 27m 38s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/66463f9afd08463a820ee1a86a488cb5 : SUCCESS in 2h 22m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5adbd5d17232c83da6e981ca59dfec2cc652a3fe","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-20 13:08:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"449124d1bd1dcb8297ed6602fb23083bcc21d379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 16:30:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8a8c6b4722481ab5d588da82343b38 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bdbf6bc904a4dda9f7424971321aac9 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb9f773bc294422a08c28313f1b29fe : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f09027f4c537425ea05aa2e572e70976 : SUCCESS in 10m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26969fbe2464486bb6cbb91e24e4b083 : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d8578ae4f024c84ab2aa139abd48599 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d83f56e4c3249bab4a96ec61767f48f : SUCCESS in 10m 18s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d9f35a5819bc488191feaf86e4c7f911 : SUCCESS in 2h 06m 44s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/0b9fd001c6614018bcab858f7110e8bd : SUCCESS in 2h 40m 28s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3f45e87b24954fc4b48445d525e1d728 : SUCCESS in 1h 45m 10s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c1137a666992495ea160ae7fee3153ab : SUCCESS in 29m 35s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f69a5b01ef7b4b4499bf74af10fe5fd5 : SUCCESS in 19m 51s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d9a053f1f5a9488eb344946dfe194051 : SUCCESS in 3h 19m 14s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/e312b72313d04b94923aa9105adff7c0 : SUCCESS in 1h 47m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f01439bffd14ad99562d068e9f9de33 : SUCCESS in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7ab4ac137bed0d8ef6a544b7dec7ebaf890dbba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 16:30:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8b68efe9b5f6192c78e3a49b806e1b5e79462f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:41:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d37004393fcc42e0a7f03aed74acf81a0caaf489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:41:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58abdd045a1442adbd5a5a590da1302b : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76c8c16fd3b14661b72265110e0d1182 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/397e3275cf8a4e54932a3404ece49e08 : SUCCESS in 6m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0a2b844e5b1e441faaf8310dac19d8e7 : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b626847daa0c49439d591d680f6e2add : SUCCESS in 12m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80f86f2cd2bb4cf3a843adabb63d1f1d : SUCCESS in 10m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c07d199c356547cb84c9edede9c8d977 : SUCCESS in 10m 40s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8f740f6b4c194fd49de583be7a831de9 : SUCCESS in 1h 45m 07s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8c2dd11a13ca4ef8b2d2b57ff90d05d5 : SUCCESS in 1h 44m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c08caa8b0dda40c5a6951521bc30f9e1 : SUCCESS in 1h 32m 18s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/25a6a028c51d45f69acbae56eb38cb77 : SUCCESS in 17m 44s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a10cedfdd08c478bb206700e83ec9eea : SUCCESS in 17m 20s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5f19c02746c4412f8aa9d431013c657e : SUCCESS in 2h 09m 20s","accounts_in_message":[],"_revision_number":2},{"id":"51d2be4a3795cf84cf2eeef422e029a7da154394","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:43:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4efca640d7489d88cadd2d7cab53ac : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86eee7957b081c75d4d4d0cd810e77f10937bda1","revisions":{"908db9d241a70b9a6b78a7d09c8d86134c32b279":{"kind":"REWORK","_number":1,"created":"2019-12-18 09:51:38.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/43/699643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/43/699643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/1"}}},"commit":{"parents":[{"commit":"a804d18d4b28e555881459a65e68bead390c534a","subject":"Merge \"Remove escape characters from container command\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a804d18d4b28e555881459a65e68bead390c534a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-31 12:41:34.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-12-18 09:51:38.000000000","tz":0},"subject":"Only set \"unknown\" in LSP that makes sense","message":"Only set \"unknown\" in LSP that makes sense\n\nPrior to this patch, the \"unknown\" address was being set the logical\nswitch ports solely based on whether the port security was enabled or\ndisabled. That\u0027s not how it\u0027s intended to work.\n\nWith this patch the \"unknown\" address is only set to the normal logical\nswitch ports, those which types are \"router\", \"localnet\" or \"localport\"\nwon\u0027t be affected.\n\nThe maintenance task was updated to correct this behavior for existing\nports (the maintenance was suppose to be removed in the T cycle, this\npatch bumps it to U cycle instead due to this change).\n\nCloses-Bug: #1838535\nRelated-Bug: #1815270\nChange-Id: I3c01bd7d1685c8a7e13a55e545e98baf19e9a0f9\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 6a89dbfe47ae1f384e08fede3bb6afae2ada7b9c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/908db9d241a70b9a6b78a7d09c8d86134c32b279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/908db9d241a70b9a6b78a7d09c8d86134c32b279"}]},"branch":"refs/heads/stable/stein"},"86eee7957b081c75d4d4d0cd810e77f10937bda1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-19 09:27:49.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/43/699643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/43/699643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/43/699643/2"}}},"commit":{"parents":[{"commit":"deb9efeb45e79e6413c8a59bff1f8bad9141477e","subject":"Functional tests: Fix setup error due to missing function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/deb9efeb45e79e6413c8a59bff1f8bad9141477e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-31 12:41:34.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-12-19 09:27:49.000000000","tz":0},"subject":"Only set \"unknown\" in LSP that makes sense","message":"Only set \"unknown\" in LSP that makes sense\n\nPrior to this patch, the \"unknown\" address was being set the logical\nswitch ports solely based on whether the port security was enabled or\ndisabled. That\u0027s not how it\u0027s intended to work.\n\nWith this patch the \"unknown\" address is only set to the normal logical\nswitch ports, those which types are \"router\", \"localnet\" or \"localport\"\nwon\u0027t be affected.\n\nThe maintenance task was updated to correct this behavior for existing\nports (the maintenance was suppose to be removed in the T cycle, this\npatch bumps it to U cycle instead due to this change).\n\nCloses-Bug: #1838535\nRelated-Bug: #1815270\nChange-Id: I3c01bd7d1685c8a7e13a55e545e98baf19e9a0f9\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 6a89dbfe47ae1f384e08fede3bb6afae2ada7b9c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/86eee7957b081c75d4d4d0cd810e77f10937bda1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/86eee7957b081c75d4d4d0cd810e77f10937bda1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
