)]}'
{"id":"openstack%2Fnetworking-ovn~701931","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~If28ec1ce8179afb47a4ef11ee953f3d7db98408a","project":"openstack/networking-ovn","branch":"stable/train","topic":"ovn-provider-train","hashtags":[],"change_id":"If28ec1ce8179afb47a4ef11ee953f3d7db98408a","subject":"Add OVN Octavia CI jobs","status":"MERGED","created":"2020-01-10 11:08:29.000000000","updated":"2020-01-14 13:17:04.000000000","submitted":"2020-01-14 13:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701931-1579007600834-d6071f12","meta_rev_id":"6307b78669972af6111ef8b40b01beea1ebfdd8b","_number":701931,"virtual_id_number":701931,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 13:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-01-14 09:54:55.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-01-14 10:14:57.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-01-14 10:14:57.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-01-14 10:00:53.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-10 11:08:29.000000000","_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 11:08:29.000000000","updated_by":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2020-01-14 09:54:55.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-01-14 10:00:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-14 10:14:57.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-14 13:13:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc5c2f154f29f432afa92bebc5ee3c225046c0f6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-10 11:08:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dae2f5e60d65c6b4bc45113a01f743478c00abf","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-10 11:12:02.000000000","message":"Topic set to ovn-provider-train","accounts_in_message":[],"_revision_number":1},{"id":"94caa264f63a962d26274a17b0816cc66e07ab86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:58:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ddc0ff9ba6451185778915fabbbcb9 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6eaacf38689417198634480821e729d : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e3c369d62c4e7aacaf3596006d2ef6 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/302f7763154b49e5a2075671c8137de3 : SUCCESS in 10m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5957e13b10c549fdac119b4a0ce9c39d : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0d4e7e7351e470c89eef9abe60e9b2f : SUCCESS in 11m 00s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f8bcea4945214d7ab493230b705f88a5 : SUCCESS in 2h 32m 37s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/875540cfcef3483987ad59ee618f02fd : SUCCESS in 2h 19m 34s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a7aa2acac45243c69dcdf0a283f51de7 : SUCCESS in 2h 11m 11s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/89a9f59c1dbc408e932e186caee5d73b : SUCCESS in 2h 00m 52s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/2d1595e671c743c4aef5e8e5de09d5a5 : SUCCESS in 19m 54s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/85f282c1e71849e89fa89334d9bd2339 : SUCCESS in 21m 17s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/09c08e7472de4ed0874831b77284385b : SUCCESS in 2h 41m 13s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/4a956a964e404ab5b0b9c20e78a8f2db : FAILURE in 1h 58m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a33c4235dc944868f672235710bb847 : SUCCESS in 9m 24s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8a14e0ae9e98437b9b286cfd39a29e76 : FAILURE in 2h 44m 36s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/154ee5e042ed4f899a16e4f68867ab0b : SUCCESS in 2h 09m 21s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfa6c2ee9d2a476f9e485aff28d55069 : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":1},{"id":"368c456d445f775bfcbfe39bf45d5c3f419fb658","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-14 09:54:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b09dac2711c8b58729f44c1688d198a341bea38","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-14 10:14:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nmaster review has 2 +2 and cannot be merged anymore with the neutron move in progress, clean cherry-pick","accounts_in_message":[],"_revision_number":1},{"id":"9c2dcf07cf01b09cb703ac3f2c288b0b8e8b6f3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 10:15:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"630b9faf12cf671a183a45e052a7028908d57501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:13:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7a2a5604d247458b92ff9c9a7b0050 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7d75a4b6e2448bdb23dba362098563e : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f1e9e603e941d29ce14503b1551821 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd8ec2eb684342348b81f6189d955604 : SUCCESS in 9m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f2c8240e2ab4b55b887819f7f623d71 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c9101f9aba345b8b4bc280657e933f2 : SUCCESS in 10m 54s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5334c8656e524bd7b4f1925a92007aed : SUCCESS in 2h 17m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/87021dc8f7ab44c3a3a155aef1540b2b : SUCCESS in 2h 17m 43s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3ba1bc370cf44989b14aef9cf69fd741 : SUCCESS in 2h 01m 59s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/44776ad4a9124e2982c9013d63e7779a : SUCCESS in 22m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d5c402f042e847d6a5b639dda1d9ea7f : SUCCESS in 22m 11s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/83794b2a951b45acbeef897136565063 : SUCCESS in 2h 53m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b61f59fde32b4a41bb33eb9e83bbcd94 : SUCCESS in 9m 50s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5d04de6604d4c638e0d4ddc6a73052f : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":1},{"id":"79f532097b8170627d877228e80a2f787b4db2f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:13:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6307b78669972af6111ef8b40b01beea1ebfdd8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:17:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738b854a8fa84979a801ce570c357c8f : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf1365f861ad05eaa01264bfea01302763c9fe2b","revisions":{"cf1365f861ad05eaa01264bfea01302763c9fe2b":{"kind":"REWORK","_number":1,"created":"2020-01-10 11:08:29.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/31/701931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/31/701931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/31/701931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/31/701931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/31/701931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/31/701931/1"}}},"commit":{"parents":[{"commit":"eff86ce13c0d199c2d2a8f9b5be1d84691ec783c","subject":"Merge \"migration: Use ForceNeutronDriverUpdate when migrating\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/eff86ce13c0d199c2d2a8f9b5be1d84691ec783c"}]}],"author":{"name":"Numan Siddique","email":"nusiddiq@redhat.com","date":"2018-08-08 05:44:12.000000000","tz":330},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-10 11:08:10.000000000","tz":60},"subject":"Add OVN Octavia CI jobs","message":"Add OVN Octavia CI jobs\n\nThis patch adds Octavia CI jobs for OVN.\nNotes:\n\n- OVN currently doesn\u0027t support L7. L7 tempest tests are disabled.\n- OVN currently doesn\u0027t support health check for LB members. Health monitor\n  tests are disabled.\n- Skipped ipv6 tests because mixing IPv4 and IPv6 members\n  is not allowed in OVN.\n- OVN currently doesn\u0027t support flavors. Related tests are disabled.\n\nCo-Authored-By: Reedip Banerjee \u003crbanerje@redhat.com\u003e\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nCo-Authored-By: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\n\nChange-Id: If28ec1ce8179afb47a4ef11ee953f3d7db98408a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cf1365f861ad05eaa01264bfea01302763c9fe2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cf1365f861ad05eaa01264bfea01302763c9fe2b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
