)]}'
{"id":"openstack%2Fnetworking-ovn~701949","triplet_id":"openstack%2Fnetworking-ovn~stable%2Frocky~Iddd89725e617b22c35e30ae7f20a7f87de296cdb","project":"openstack/networking-ovn","branch":"stable/rocky","topic":"bug/1800875-stable/rocky","hashtags":[],"change_id":"Iddd89725e617b22c35e30ae7f20a7f87de296cdb","subject":"Fix revision number race condition with attaching router interfaces","status":"MERGED","created":"2020-01-10 12:55:14.000000000","updated":"2020-01-17 22:05:59.000000000","submitted":"2020-01-17 22:05:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701949-1579298759486-71032318","meta_rev_id":"19a1cf1648835045a0647589f49b0bf18ebc130e","_number":701949,"virtual_id_number":701949,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-17 22:05:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-14 10:06:21.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-01-10 13:07:27.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-01-17 17:28:55.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-01-17 17:28:55.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 13:07:27.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-01-17 17:28:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-17 22:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"630e6ca709614145ef34861c57fb3df46290cb59","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-01-10 12:55:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f676448a86cd4fd3396d9451bef974205ce20a9f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-01-10 12:55:29.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 4804513907f7669669af273e57ceb821dc242578","accounts_in_message":[],"_revision_number":1},{"id":"65a55f97bd5d5cb3f750f668606ff7c846aaf77e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-10 13:07:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98b2a92cca9ebf50dbfa63bfac2333c626e236a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 16:20:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72b7688dcf0411b8a3af2f2a74573b1 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8587fa51509d442d8a50709b7fdd7c4b : SUCCESS in 8m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/987562c914264d12b80a305eb755a903 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4184a9a77b40424b86ed13c3c28e0b61 : SUCCESS in 11m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/70b8cac3726e42429568fa3e6d2fcfe3 : FAILURE in 1h 28m 44s\n- networking-ovn-tempest-dsvm-ovs-release-python3 https://zuul.opendev.org/t/openstack/build/f58a0d415af94965a1a1ddf7c1e7bdd3 : SUCCESS in 1h 39m 26s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1a8987bfa0ec499193b892aa33cd1a18 : SUCCESS in 1h 44m 18s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/98e81b0e2a8b4926831c2a97136b8155 : SUCCESS in 22m 10s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/5be88e0b814e4b24be6dc35a432d64ec : SUCCESS in 27m 15s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/92809791ece54e5c82948e4237098aff : TIMED_OUT in 3h 25m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c12c6accb8b54573a4c593e07cca5daf : SUCCESS in 10m 54s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/31cd46f05b48491995a4f70d089b1926 : SUCCESS in 1h 59m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dbd59adb8564bbcb0b82d39fa2beefb : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f855549e6522c51a97a85fd80913e0fcb4482022","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-10 21:16:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88ae68962f93585190eb016bc06b2e6b44f5102e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 00:23:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7518290edc4e47168b9bb3598a35144b : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eed6b0d604b441992254a0ffc9f2ea1 : SUCCESS in 8m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/17e353e07b2e44299a6eef5b3b1243f9 : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/97c7e44dc2654deda132d53931174069 : SUCCESS in 8m 12s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e3d30a3a5f6c424eae76918c73443d6e : TIMED_OUT in 2h 11m 49s\n- networking-ovn-tempest-dsvm-ovs-release-python3 https://zuul.opendev.org/t/openstack/build/e16d535f78e04801b0e6a003031312cb : TIMED_OUT in 2h 11m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/500b97d021324757b586dda9e2e77e76 : SUCCESS in 1h 37m 44s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/2988893cec25471da190a1f6a1c1b118 : SUCCESS in 28m 08s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d84a519965be43419fec1c3ad261856d : SUCCESS in 23m 12s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e228def4f83945d3bbea619d0a92bf81 : SUCCESS in 2h 35m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f5b266f14a742b3b4fe2273d75e539f : SUCCESS in 10m 16s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/900c8ee872384142987fb99ac56682b5 : SUCCESS in 1h 45m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a605a5fe33ee4e6c8e445511e1199023 : SUCCESS in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"232529fddab0434fc2e05199be08920067f664e3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-01-14 10:06:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9881204e85c9fa249704e59dc9e569ed986f4e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:29:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b03525db47f48fb8e90ad3ecdaa5a1a : SUCCESS in 7m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/411de13e0e874f9ca09d767a421fa74f : SUCCESS in 10m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/280fca413d2b4d32886775851a9c5be8 : SUCCESS in 4m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/38ae81300f4846f8ab0bb17d12ee4d3a : SUCCESS in 10m 14s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/93b9d9b8d7cb448788e111f409bf8855 : SUCCESS in 1h 42m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python3 https://zuul.opendev.org/t/openstack/build/d591b97fb1ed4ba5ad75e6914c9db449 : SUCCESS in 1h 38m 12s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/343186a3bf2e40cf8806d1ac4d3cfb77 : SUCCESS in 1h 45m 14s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/717f8cc9c9c948b983deb42b30fe443f : SUCCESS in 22m 34s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6a3b422746643fe97b50cea0e6a44bd : SUCCESS in 20m 32s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3a83657596e1463a81a2f322ee24b75f : SUCCESS in 2h 55m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46d58c346f2a4efd9f75edb92a3444fb : SUCCESS in 11m 17s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/960f8ad9898f44d7a93521aca5683302 : SUCCESS in 1h 32m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53d723e2f2ed48b2ab8103a46cd83069 : SUCCESS in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b42a01d882844e86a517c0852d8d12b8b57f9581","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-17 17:28:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fef1fb1cef97c6e07a3c8d056807716d88f89339","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 17:29:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5aa9a772f34c02d4bb689a9d8fc701d1ee7a9dc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 20:08:39.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d6c38de9d9cbf1cc35e0269dc4be43f97e2616f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 22:05:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e7c49c282f4359a472344c908a2b0c : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/695734401f5740a1a6499f4788b19a11 : SUCCESS in 10m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/18e5bddcdb2c4c19a11bfd0ea6822e5c : SUCCESS in 6m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a3babf29317849999a946bf894470662 : SUCCESS in 11m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/303689db7f79484cab8a97a7d521c369 : SUCCESS in 1h 47m 50s\n- networking-ovn-tempest-dsvm-ovs-release-python3 https://zuul.opendev.org/t/openstack/build/294b623de1914e6bb3549909960ad8a3 : SUCCESS in 1h 46m 17s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/afdf1792e314433bb758285a7892270d : SUCCESS in 1h 42m 42s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d5c75d47353b4f50ae7458d4cc16f0cc : SUCCESS in 24m 24s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/af53d03cde964a42a41dc7a4443a668b : SUCCESS in 24m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/632b3ca97144402f9eeed2efc01cd227 : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"655dfc39c9c66d6c1ed2057acf2e2c7a119ed791","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 22:05:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64ac6d274e8f3534924546bc1054404236f89f95","revisions":{"64ac6d274e8f3534924546bc1054404236f89f95":{"kind":"REWORK","_number":1,"created":"2020-01-10 12:55:14.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/49/701949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/701949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/701949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/701949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/701949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/701949/1"}}},"commit":{"parents":[{"commit":"38b8985f6bc719a302f5ada76fe90dcac786d61c","subject":"Merge \"Only set \"unknown\" in LSP that makes sense\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/38b8985f6bc719a302f5ada76fe90dcac786d61c"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2018-10-31 16:44:55.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2020-01-10 12:55:14.000000000","tz":0},"subject":"Fix revision number race condition with attaching router interfaces","message":"Fix revision number race condition with attaching router interfaces\n\nThere\u0027s a special case in the update_port_precommit() method to handle\nthe case where an existing port is being added to a router. The method\nwill try to create an entry in the ovn_revision_numbers table but, if\nthat entry is already present a DBDuplicateEntry exception will be\nraised and the whole update method will fail.\n\nThe update_port_precommit() is the only method at present that have this\nspecial handler, for all other resources creating a new record in the\novn_revision_numbers table happens at the create_*_precommit() methods\n(this include create_port_precommit() as well).\n\nThis patch is fixing the problem by adding a new parameter to the\nreate_initial_revision() method called \"may_exist\". When set to True the\ncode will not raise an DBDuplicateEntry if the record already exists. By\ndefault this method is False.\n\nCloses-Bug: #1800875\nChange-Id: Iddd89725e617b22c35e30ae7f20a7f87de296cdb\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit b5642d93a6c0ce0141fb5f61a3d853b026b7b67d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/64ac6d274e8f3534924546bc1054404236f89f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/64ac6d274e8f3534924546bc1054404236f89f95"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
