)]}'
{"id":"openstack%2Fnetworking-ovn~703939","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Iea532e2a02b7992305d1b90aa040e064901c340c","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1859977","hashtags":[],"change_id":"Iea532e2a02b7992305d1b90aa040e064901c340c","subject":"[OVN] Delete NAT entry first on any FIP update","status":"MERGED","created":"2020-01-23 08:45:48.000000000","updated":"2020-01-28 20:27:17.000000000","submitted":"2020-01-28 20:24:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703939-1580243086253-87017a1f","meta_rev_id":"c5f6996dba5d7c21b16f5277faff40c89d943e4b","_number":703939,"virtual_id_number":703939,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-01-28 14:34:07.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 20:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-01-24 13:07:56.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-01-24 13:39:33.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-01-24 13:39:33.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 13:07:56.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-01-24 13:39:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-28 20:24:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f67f1ac8bce62660fcd348f68fa9287877bef5c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-23 08:45:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92d79c0f1957d5f8754d3cd08a55f9293696248c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-23 11:50:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a3fd6665b682bfe2fa864b06497b8b54dc81eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:09:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6dc86b03e2b41db972ebc04821b8818 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06f3355e6e0b466380f7f30cd4695703 : SUCCESS in 13m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a308a162e0c7495f93973bad1393bfdb : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a988b5db2bb447cb244dcf3063e5098 : SUCCESS in 13m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caa2f808b1974fb480b8cd2fcd0b1c5a : SUCCESS in 10m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a22f040ec0ea431cab4c66d073d4dc07 : SUCCESS in 12m 23s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/7ce7f5b60859435b89676e57074729e6 : FAILURE in 2h 33m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/647462cef63c4fcf83f5058f1a2d08a1 : SUCCESS in 2h 24m 36s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/2bd055682017403381897c9412e08e0b : SUCCESS in 2h 08m 16s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/21f473f301e04bce93e057c1c04c5692 : SUCCESS in 2h 01m 00s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/43c0643970174f42a3a263227341b15e : SUCCESS in 24m 12s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4763d1ca87343dfb26c1cab0aac9138 : SUCCESS in 25m 10s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d955fa4cfb844c39872cb0374f86af2f : SUCCESS in 2h 25m 05s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/3c2b02082a3b499094f1db156c897719 : FAILURE in 2h 15m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb4d5232813c46f0a212ac50c8debed8 : SUCCESS in 11m 44s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ca5d0cf4e6b24d879ddf9573e7544025 : TIMED_OUT in 3h 07m 09s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e3dbc9650c8347ebb22fac77c0f6a97d : FAILURE in 2h 00m 45s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c05694f207b54911b2ba195ed0e65654 : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":2},{"id":"347b15f06170cc3ed3475371dcf03f72408295a1","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-24 13:07:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6f59ad1efcf5926fe3f9270a71bf753193862684","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-24 13:39:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22e46fa77454fc9851f021716798620d0d0499f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 13:39:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c0a679b0fd18207346890dea25c6a3448e843b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 17:34:02.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a689c90521346c3ad1e2bc6d111f0df : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a88193944b6c4fd3838b78b9a7d3e62f : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60430248c10143cb868a62e80cfafd46 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d2fd4818be8479b80fb3e93289678ed : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f56d500381694f79a58b6b072954faaa : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/189d329d8ba24e3fb9e7d3bcb3409147 : SUCCESS in 13m 41s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e45e60f8eafb4404a37248e238e04c8a : FAILURE in 44m 35s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/87d5d3f9cb8144618712b375f5d3f76f : SUCCESS in 2h 11m 52s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea6e4925d835497cb011c73c0dcb308b : FAILURE in 1h 54m 21s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b3d88ac4478c482988c56dd5d670bb02 : FAILURE in 19m 44s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/622020a8d6cd47c19a86e8aaac9a961c : SUCCESS in 18m 51s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4207c7d44d4b4094a6de58db92cc20ef : SUCCESS in 2h 50m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/856f4d38e8e9494cb35d1ddc7da3493c : SUCCESS in 12m 26s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe9a4e76fa4e4415a89d997e72b5a22b : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5ccb20b7809cc7f87b057f071e2eeb9afef49ab3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-24 17:43:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cda5adfca70da86915e66276a070dcb7cd15f7b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 21:15:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33a72f2f3b148c9a2239a3561d077f8 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d122283c9b74f4999c9c2f51db6136f : SUCCESS in 14m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef1d1e4caf24e4097ce0d725eb9c21f : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b4deb7118c742c0a41f951976b490f8 : SUCCESS in 10m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f56fd352cf542d4880afac5c47354eb : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d704bb5a66744046b2497e05225fe737 : SUCCESS in 11m 18s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/63612ae10c714aaa8fcb6cfbc0a801a6 : FAILURE in 2h 03m 31s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4767cbd8995a49309db88b39ca8fb24b : SUCCESS in 2h 17m 12s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/eef7e83542ad44bba78a44493c22354f : SUCCESS in 2h 30m 03s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fdaae1cbd7c4448a934a119da6d8d7ed : TIMED_OUT in 2h 15m 41s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/3bda91afe5e64e78b7e8d09b2e3c48b4 : SUCCESS in 22m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/89a8042fb3e743fe83daafbef37235df : FAILURE in 27m 22s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f8f4b53c4ec84305bdc5d3ed070c8d81 : SUCCESS in 3h 20m 36s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/696d4c8ea22648a387b139875c3387ba : FAILURE in 1h 56m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b79fde15e4dc471fa89c6719475909d9 : SUCCESS in 9m 41s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/696dac59c51f4044ab014e03fab426e3 : TIMED_OUT in 3h 05m 31s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c6b94325a773401793f46437bccd16df : FAILURE in 2h 20m 08s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa2e71222911442086cf039c6c96096b : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ef8dba16d765aa37fd5daa5cd2f3ccfac795d35b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-28 08:23:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bbe702beb1e6653a7629e3173ef144482d40d4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 11:24:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788ee49ce3cd4d6b80f537a97d0938da : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59ae34535b264ec6ab0e0649de0eedfc : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559302ec7293409982e93eb8871293e5 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35374da316ba41c1abc839f563b977c2 : SUCCESS in 12m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67fb55c7ca454063b161868f1b13bb84 : SUCCESS in 11m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/913e3efcb73f4926bf2a085798e129df : SUCCESS in 10m 14s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/57a2a9050d9549368e51b85304238f08 : FAILURE in 2h 09m 38s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b9f3809539d84e2e8d690c4817e55a1b : SUCCESS in 3h 00m 22s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/08747c7f142a481eb059ed88b6094be9 : SUCCESS in 2h 09m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0aea87e2406a41aaaffb10701723f165 : SUCCESS in 1h 48m 04s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/bb65020bf3334a9e91942da9f977fd59 : SUCCESS in 23m 58s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1c81c3f3e1147fca0f0ecc6f4421b2a : SUCCESS in 24m 06s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/967db51905294404bd0c8e2f60eb1b08 : SUCCESS in 2h 17m 49s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/4d1bcfec9d7f42678aa320102574a82b : SUCCESS in 1h 25m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/287ae747173941f8878ad576b7b0c3b2 : SUCCESS in 10m 48s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b1b32687a1e24c3cbdd6f9e38f37aaf7 : FAILURE in 2h 40m 11s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/05a6b7b45be54f5298910425f3d80252 : FAILURE in 2h 27m 56s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07524ef3bc8d40a9a1c9bfb2e32e8bcb : SUCCESS in 43m 19s","accounts_in_message":[],"_revision_number":2},{"id":"295017d5012d8eed6de3810aee508828a1580c35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 11:24:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60244bb181da5309d969b0178e3800c813a80669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 14:22:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/368ab28cb54a4dbf96e045343bdbf765 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a05e1f4fbbd44445844118a730a9580c : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd493c078a394416922bdc7e84310ddf : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c438213703947b2807649cb226316f4 : SUCCESS in 10m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad28e01bea754059816972cd9237fa37 : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/903906d0fdea49ed90bc9cbf9cd3900d : SUCCESS in 10m 48s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/109c9b4c1418440a9dc0fae66142c499 : SUCCESS in 2h 10m 10s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/7ca14325d04448ccb7bfbafd1faab2b8 : SUCCESS in 1h 26m 00s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/741d6059c4a14417a9d26389706c8eaa : TIMED_OUT in 2h 13m 27s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a7df3af31eef400a826303e5fe353f82 : SUCCESS in 22m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb4a2f6fc34544ccb155c44e0d82bfe4 : SUCCESS in 28m 05s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a48426a43e554915846cf1bd88a2b58f : SUCCESS in 2h 50m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ada24c0f88f54f1bb2ee8ace36d3aace : SUCCESS in 10m 38s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd8fee65d54242cca627550602eaaf6d : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b6cd707a0e01fad1e6fab36672ceee18fd147dc5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-28 14:34:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"84d4a60f9a0a156038a788d91ae2618939d99137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:45:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6946a7a580ac4aa9857df362935c2397 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a63fccdbbb054c88ae2240898b8e0e1e : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074242d29f8a4ec689c23474721a7332 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345064c3fd8c43449859869f19073e25 : SUCCESS in 9m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/131f5e469ede4e12a7735fe3e1345d61 : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ba1bb5d17a1484bb90c28939b316baa : SUCCESS in 10m 53s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f265873012154bffbab0054b2f36313c : TIMED_OUT in 3h 07m 05s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/beac5e8552684926980d0c2554d713dd : SUCCESS in 2h 36m 39s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ee6f7f4bde874029acac957e2e93a117 : SUCCESS in 1h 22m 55s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c550bb8e24404f288eeb5ecdcff4902d : SUCCESS in 1h 57m 50s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/1981d20be5384cf48630505c36394646 : SUCCESS in 25m 47s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bfd2e2735cc4e11be57a050fca5737e : SUCCESS in 21m 29s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/73a633a95cfd4204b09924bde24dda1c : SUCCESS in 2h 44m 07s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/cf6083bf3857442d82002ef60ac93269 : FAILURE in 1h 54m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b47d51bf7b34841be8dbd38d7307b62 : SUCCESS in 15m 29s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/49f2418437ca47d4b18aa0ec32662108 : FAILURE in 2h 38m 40s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d6f2ed05b01948d0b6c679cd6b503f51 : FAILURE in 2h 04m 00s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/110860a0631f4e83b43ecc9d05892963 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":2},{"id":"a5cfbf8a243495799145d4c6e42eb6777d061ec0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:45:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac8ef63f6e4020c212fc4df65da79a00a4829c66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 20:24:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a24ddd4f68b47799354f2da2d7e45e4 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee1c0e24938b4350820120ed34b73940 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236625302ffb4dcea3e2dba2f3e82d26 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/712c66d3209e4bf4ad4c29e4abcc0423 : SUCCESS in 13m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ffa79d16506462b92c420019ba39acd : SUCCESS in 10m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c48c92c0643343129cf40d132995a09f : SUCCESS in 10m 54s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/97791da2505f494ead3286c40afbf587 : SUCCESS in 2h 02m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/54f5661fc75b440789f5d5f6d91aabd5 : SUCCESS in 1h 34m 31s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e261b8a4f6be459fb20114df417af03f : SUCCESS in 1h 47m 15s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e3ad201f7b474678adb2bf53dd1e08a1 : SUCCESS in 20m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/86ff3107b73d4419a87c7bc39908e000 : SUCCESS in 18m 42s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/764514ff6efb40a1bf0a812126abb72c : SUCCESS in 2h 36m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b25d38ac4b33444bbce2a18c01c8cae1 : SUCCESS in 10m 35s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8dc0207cc7141188be5fec06d7a6f86 : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":2},{"id":"f616da52e8e812f106d1f8b989aef6617e163246","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 20:24:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"af096346d4c0c0e645d3aa37255471965d1eafe7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 20:27:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdae7052315c48c38d1730c87a57c762 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba542715de5fa22ae57d8b3263a6ea8c6ae2620f","revisions":{"45d4d9c8afcaa562b0ec7f0b8d599f9a159c03ac":{"kind":"REWORK","_number":1,"created":"2020-01-23 08:45:48.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/39/703939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/39/703939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/1"}}},"commit":{"parents":[{"commit":"743d853b981f7a5c1703cd9cbb50fd1760bbad48","subject":"Merge \"Don\u0027t send malformed status update to Octavia\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/743d853b981f7a5c1703cd9cbb50fd1760bbad48"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-23 08:36:52.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-23 08:44:47.000000000","tz":0},"subject":"[OVN] Delete NAT entry first on any FIP update","message":"[OVN] Delete NAT entry first on any FIP update\n\nFor now while updating FIP  check if port or logical_ip\nhas changed and only then we deleted the NAT entry.\n\nUnfortunately each time when FIP update occurs the\nmethod _create_or_update_floatingip() is used. It first deletes\nLSP pointed by FIP and adds it again along with new NAT entries.\nBased on author comment this actions are required.\n\nSo if we don\u0027t update FIP with logical_ip or new port_id,\nlike update a description, the NAT entries gets duplicated.\n\nSince all is wrapped withing a transaction and to not wait for\nproper fix (this code need sa refactor based on commments with NAT\nexternal_id column) I think thats safe just to delete the NAT entry\nin such situation like described above.\n\n(cherry picked from commit 45ae9dfb7d5acacc72fcf9f071a9db1beb0ca972)\n\nChange-Id: Iea532e2a02b7992305d1b90aa040e064901c340c\nRelated-Bug: #1859977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/45d4d9c8afcaa562b0ec7f0b8d599f9a159c03ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/45d4d9c8afcaa562b0ec7f0b8d599f9a159c03ac"}]},"branch":"refs/heads/stable/train"},"ba542715de5fa22ae57d8b3263a6ea8c6ae2620f":{"kind":"REWORK","_number":2,"created":"2020-01-23 11:50:17.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/39/703939/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/39/703939/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/39/703939/2"}}},"commit":{"parents":[{"commit":"743d853b981f7a5c1703cd9cbb50fd1760bbad48","subject":"Merge \"Don\u0027t send malformed status update to Octavia\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/743d853b981f7a5c1703cd9cbb50fd1760bbad48"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-23 08:36:52.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-23 11:49:27.000000000","tz":0},"subject":"[OVN] Delete NAT entry first on any FIP update","message":"[OVN] Delete NAT entry first on any FIP update\n\nFor now while updating FIP  check if port or logical_ip\nhas changed and only then we deleted the NAT entry.\n\nUnfortunately each time when FIP update occurs the\nmethod _create_or_update_floatingip() is used. It first deletes\nLSP pointed by FIP and adds it again along with new NAT entries.\nBased on author comment this actions are required.\n\nSo if we don\u0027t update FIP with logical_ip or new port_id,\nlike update a description, the NAT entries gets duplicated.\n\nSince all is wrapped withing a transaction and to not wait for\nproper fix (this code need sa refactor based on commments with NAT\nexternal_id column) I think thats safe just to delete the NAT entry\nin such situation like described above.\n\n(cherry picked from commit 45ae9dfb7d5acacc72fcf9f071a9db1beb0ca972)\n\nChange-Id: Iea532e2a02b7992305d1b90aa040e064901c340c\nRelated-Bug: #1859977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ba542715de5fa22ae57d8b3263a6ea8c6ae2620f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ba542715de5fa22ae57d8b3263a6ea8c6ae2620f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
