)]}'
{"id":"openstack%2Fnetworking-ovn~705249","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I3e9e734b757f67d3e4ab1c5a02f8e7f9f0c9aa2e","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"I3e9e734b757f67d3e4ab1c5a02f8e7f9f0c9aa2e","subject":"[OVN] Remove VLAN check when setting external_mac","status":"MERGED","created":"2020-01-31 16:09:36.000000000","updated":"2020-02-20 19:59:10.000000000","submitted":"2020-02-20 19:57:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705249-1582228644787-a5ae4dff","meta_rev_id":"da3323bc058d472cd5b093cc79cf627ca19f54f1","_number":705249,"virtual_id_number":705249,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-20 19:57:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-01-31 21:16:56.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2020-02-20 16:54:09.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 21:16:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-20 19:57:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8265608a7d8ff2e108818bbf8988b35b3bb4602","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-31 16:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfbf5e653b473084397f8e5b46f99ddb7c2ec7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:20:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a164fe29ce9460fb1fba867d2714f8a : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a1743ada8ec49499de870bd13c3d4f4 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30314cb8fa244959ed2ac9ac02b340c : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25250f086a69458c8496b9b34eec8e68 : SUCCESS in 11m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f59a8e830df0489d90c0358bb9ab52ce : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c0698092a814913b8c0800d1a9963a8 : SUCCESS in 11m 02s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/c9bfb51ec1334cb9bccb8c59d7dd00f8 : FAILURE in 2h 11m 03s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/513ca1a7706f43a0a6306beb4f6682b0 : SUCCESS in 2h 09m 50s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a3e082b264404fdd8cba3f79562c9590 : SUCCESS in 1h 27m 59s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e811cd5a443640d29a641c86272df300 : SUCCESS in 1h 57m 09s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a073e80d9706419dab5daf992ea7b445 : SUCCESS in 23m 05s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/92c1285d3e1243b8a2b6c6d469862730 : SUCCESS in 19m 58s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fe164d16c83b47e9ae0ca448a1494dbd : SUCCESS in 2h 51m 22s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/7d6d1bceac1144a08a346e916f78088b : SUCCESS in 1h 32m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/059b1a20702c436f82aa6a9db511a7ef : SUCCESS in 12m 09s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/abed887759364d3bb1e45038d02122c5 : TIMED_OUT in 3h 06m 07s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9843944ea99f42698ec6f46cf6a35274 : FAILURE in 2h 36m 58s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00bb01d86efe4dd2b75cee89761ac8f4 : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d53a199ad3cf27b257ce8bf9a20e533d8d72ebae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-31 21:16:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5d99e64868e04fc6694cbdacf7b920ead54407e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-20 16:54:09.000000000","message":"Patch Set 1: Workflow+1\n\nThanks Brian for the review!","accounts_in_message":[],"_revision_number":1},{"id":"01698088115a9475f007117000592c5804dc46e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 16:54:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbdf6b7f6be6aa7e4e25924ecabbe6550e93e069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 19:57:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebeacdb997634926b9e36cc580c5aff9 : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b86a8bc8f39d44208a568132eb8b1726 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6428a44303d64d7192f3272548fb38db : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4963fd2622524f51b0699ee68d8df190 : SUCCESS in 11m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/118ba6ab219b454d969f7e88ccbf0ea3 : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9306a27f52014d06b553378f0a477bc1 : SUCCESS in 12m 35s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c8e5066f47c147ca8395e947fc7e2e7c : SUCCESS in 2h 20m 59s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e6af8938feb84b729d931f360828f13f : SUCCESS in 1h 41m 32s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d7be33f97c341e5a73c2a5c8e73287c : SUCCESS in 1h 52m 52s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/46ad25fd416b4f6abd3c89174c4525eb : SUCCESS in 25m 37s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe88bc80f5e84288880a352d1d9be283 : SUCCESS in 24m 23s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6457800adbc947109d8eadbf8be0ca23 : SUCCESS in 2h 57m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35376216e0e8443c96420efa45a0614c : SUCCESS in 11m 11s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a92f5f3b6ca64aa19d311db2d47720a8 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":1},{"id":"16398ffb245fc672dffc535b8dc463d287ecbb2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 19:57:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da3323bc058d472cd5b093cc79cf627ca19f54f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 19:59:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26c7cd6b846481c9ee7c48d2bc88087 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9b1dffada3ad6c1328843b30128e72b92cfcdf5","revisions":{"f9b1dffada3ad6c1328843b30128e72b92cfcdf5":{"kind":"REWORK","_number":1,"created":"2020-01-31 16:09:36.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/49/705249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/705249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/705249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/705249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/705249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/705249/1"}}},"commit":{"parents":[{"commit":"844540e2d00b85bd31b8e0795b2acb8c81f1da75","subject":"Merge \"Agent liveness - allow time to propagate checks\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/844540e2d00b85bd31b8e0795b2acb8c81f1da75"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-01-31 16:08:26.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-01-31 16:09:33.000000000","tz":0},"subject":"[OVN] Remove VLAN check when setting external_mac","message":"[OVN] Remove VLAN check when setting external_mac\n\nThis patch reverts [0].\n\nThe code wasn\u0027t accounting for VLAN provider networks, as stated in the\nbug #1842988, DVR won\u0027t work if the provider network (where the FIP is\ncreated) is VLAN.\n\nThere was also an incosistency in how the external_mac was set for the\nVLAN networks. Upon creating the FIP the code was checking for the\nnetwork type and not setting the external_mac attribute in case the\nnetwork was VLAN type. But, if the port went down and up again (e.g if\nyou reboot the VM) the event handler that set/unset the external_mac [1]\nwasn\u0027t check for the type. This is how people worked around the DVR\nproblem (as stated in bug #1842988).\n\nFor more information see bug #1842988.\n\n[0]\nhttps://github.com/openstack/networking-ovn/commit/c5aef51edc9843db605303ec8bd8610b6c55e9c2\n[1]\nhttps://github.com/openstack/networking-ovn/blob/eda5d7f80d877601170631c5f5485370ea701f42/networking_ovn/ml2/mech_driver.py#L794-L800\n\nCloses-Bug: #1842988\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n\n(Cherry-picked from Neutron Ifb795626dc9c2ac4f0104f491dd38c9b4cc902c9)\n\nChange-Id: I3e9e734b757f67d3e4ab1c5a02f8e7f9f0c9aa2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f9b1dffada3ad6c1328843b30128e72b92cfcdf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f9b1dffada3ad6c1328843b30128e72b92cfcdf5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
